)]}'
{"id":"openstack%2Fkayobe~901230","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I8ad027384d9d062fdd363b10fd7bcebe22d775e0","project":"openstack/kayobe","branch":"stable/zed","topic":"selinux-zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-17 10:27:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-17 12:52:24.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-17 15:22:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ad027384d9d062fdd363b10fd7bcebe22d775e0","subject":"Disable configuration of SELinux by Kolla Ansible","status":"MERGED","created":"2023-11-17 10:14:26.000000000","updated":"2023-11-17 15:23:53.000000000","submitted":"2023-11-17 15:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901230-selinux-zed","meta_rev_id":"b703977a6cb1445017e7e609b725c0a252592431","_number":901230,"virtual_id_number":901230,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 15:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-17 13:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-17 12:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-17 13:47:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 10:15:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-17 10:15:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-17 11:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 13:47:30.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"2f471c039f86f706a192a51c65be44ab9159443e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 10:14:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9a0b3c39c1a67afad6b2345cc3e168896a59283","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-17 10:15:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6288c961f490f9a9a8a183283be706a9e97625c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 10:15:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ad683422a0eaaaac1ea11f599b9d58a50bfbbc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 10:23:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47b70655d38d29db1fd54cad7f2516a33c5a4bdb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-17 10:27:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ca85f0bade7efe40a9ecafbdd3795b2c6b0dc54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 11:46:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea5f58d9e944fa58e0dcdb596bc0d9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7781451cd7cc4b4c9d5e509c1f2d9349 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21540a69508478f81c885e86984116c : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958da85e79bf47728d79b8c3b5eb678d : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b2b2e6f74cc41f4a616129c3eb6e43a : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ebecdc096484d29b25d6e2358ec456e : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad2c9027cc14d2dbfd5016e707b5bae : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb5e5ca053f4fd490d596f4758c858a : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/110823872b55476298ce02d9f43b4cd9 : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0e5fa8112e04775be9f986582e977d3 : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb14a77d65054e48a95ebf5bdc6a3465 : SUCCESS in 14m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9da84d13124c450fa7451b621ebaff7c : SUCCESS in 55m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fbf935a4b2fc424b99502df98e6a00eb : SUCCESS in 48m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c65544f0bb604524a2ea1a3e019f7b05 : SUCCESS in 46m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/77754f2effca4ae4b7de8312653b1eac : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/767e1fb04e5d432793e9755d5abd8d59 : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f67accf04b404ffb9d7f81e7fad68ff0 : SUCCESS in 16m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e021e237d13436993f303c55f332918 : SUCCESS in 16m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c15d839abc654003ac9e5ffd57db65ea : SUCCESS in 1h 30m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ec482233580d4f9280f23a25c28feeb7 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02245c14f80d4dce80f03b349f9363c6 : SUCCESS in 14m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f1df5e24288341c38ad5dfbb965df593 : SUCCESS in 31m 42s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3cdd6c791e841d5b836f750f4313150 : SUCCESS in 22m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a13029a4916e440fa44ccfcffd008d1e : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2894604036cf4b90bf2b86fe66a987f2 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48b265184c6b4782ac115480986e323b : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3b8ecabfb6b41f7965dd7d728e03737 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d4dc0e248f2db11c938ff317689f93581c3b8e55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 12:24:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3b3a261a6f057dedfa2ac54fb752e874111da00c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-17 12:52:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d7521d853d698770c8824a0ed0ac442f44bc0df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 13:45:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ad527c8fab4915a3a688bebbb65bb9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1341f7a6689f49c4bb88c63324a00551 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0627a17de6544b0bc80c935340b02d9 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffdc6257f7a64391ac45c68647e860b8 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a482906b5ec44999d12433e37d0a041 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17401ca7f82148cab2b6f4651baf3bb6 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76591a6af1454f6d801ecb86e1ff5476 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/078fcad617f640b8983c449a1c5951ca : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4cb96f26e204b99b8e44bd9de6210c5 : SUCCESS in 3m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/653700731a044addbfc30efe6bcda744 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f458b2e5ac7d4375b4a60d66c8027a52 : SUCCESS in 13m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6b153bcc4e7b4db581b6f05857b02abe : SUCCESS in 50m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c7fc3c75d7fe42c18ba79a9931d6ac57 : SUCCESS in 48m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/286c209172ee44fea3ed40ab881a8f0d : SUCCESS in 46m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f357b60da1be485bb90c2fb759ef513c : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b7131e4a76e848ac95036501a5ef6e98 : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/80cd2871d8ee4fc38acad8c8f88900ea : SUCCESS in 19m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/783f164fe5b745d4a28b2ddffb8a4221 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be4c8a65c4649e1844fc2f2e4edb461 : SUCCESS in 1h 11m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/067a7dc113c044d98bbb752ad25988a2 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62d1d31999ac4189a2bebb4f6f309bf3 : SUCCESS in 16m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/70c88e57d3fa4d3184079aa0001ce0e3 : SUCCESS in 26m 14s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/405fa1164c7240a18eb49faee48dec58 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d4a2b1b128b4ce59369c8d0ca334f18 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd3fadaf8dbe4d8298b767f3934098ff : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84ab340364c04b3da14225692684cc9e : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff75ca58edab4bf3af629d7471ab61ad : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":2},{"id":"69e2e1468aa209ec9129753e76356cec286cfaf3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-17 13:47:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1292b2a53ef73c26f3a9de2d054bdb7da4fc2926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 13:47:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4fabe9dae9b54c6eb2e40295621678831e2bce68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:22:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09949e2c289d4349bd9d1bfffd3194a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/977e3b418141461394c24bf4d64f73e5 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e082f7a327b4a71817247d4a0142e2e : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5fb9c88e498450888bad0c88f04369c : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526d8c153aa14fcd9aa69777ceacfc25 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9233f7e1f9246aea6cd63fa6dd85ffd : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7bcb903503b476faab1c23ea4593f8a : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a3fa4e436390440dad4eff4402781d15 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9070687db54849b3976f0d443de5238a : SUCCESS in 16m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fcbd5d74a79440c98ea45b13123ac472 : SUCCESS in 55m 26s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e5dd846a11b494888e5b14c01b42088 : SUCCESS in 52m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a294255da2cd46678b1192aa0738b2d2 : SUCCESS in 59m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cabdc25ce40d43f1811d29817657df5b : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f98f1196d7f84eac98747aeae424b6ba : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c344597028ff4cabb68b6d2de272080e : SUCCESS in 1h 29m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/de444496bbe24ce9bbc9ae4d1833390c : SUCCESS in 15m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3f4587480a04da29ff14bceb2c5ae88 : SUCCESS in 15m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17483db2ec94f0787e5225bb83bc345 : SUCCESS in 27m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ad783f4eaa34e2088eb5eb97593043d : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8f558834614920aae4f55008a51b80 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79db39d88ba743d195262515ed824e29 : SUCCESS in 15m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/910bf07044bc4714b8b5a5db254ce100 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"82e77e424305de7f3749e3556e544650be82d4c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:22:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b703977a6cb1445017e7e609b725c0a252592431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:23:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/439f55e276d4422db24300613a64dd8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fecb5508944517859f308b7bb88067 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12871c0e2a244829bd918e1dbd97c788 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22307eb73e97833c25521fda5ff8fc6f173116ad","revisions":{"64babc1f87b1f5d2c2ec6f98b6d2e7ab3c927022":{"kind":"REWORK","_number":1,"created":"2023-11-17 10:14:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/901230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/901230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/901230/1"}}},"commit":{"parents":[{"commit":"6783b0a39b743aeb3601d386a9ca07bad74c67f3","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6783b0a39b743aeb3601d386a9ca07bad74c67f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 10:13:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 10:13:35.000000000","tz":60},"subject":"Disable configuration of SELinux by Kolla Ansible","message":"Disable configuration of SELinux by Kolla Ansible\n\nWhen using default configuration, bootstrapping servers with Kolla\nAnsible would revert from permissive to disabled.\n\nChange-Id: I8ad027384d9d062fdd363b10fd7bcebe22d775e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64babc1f87b1f5d2c2ec6f98b6d2e7ab3c927022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64babc1f87b1f5d2c2ec6f98b6d2e7ab3c927022"}]},"branch":"refs/heads/stable/zed"},"22307eb73e97833c25521fda5ff8fc6f173116ad":{"kind":"REWORK","_number":2,"created":"2023-11-17 12:24:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/901230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/901230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/901230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/901230/2"}}},"commit":{"parents":[{"commit":"6783b0a39b743aeb3601d386a9ca07bad74c67f3","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6783b0a39b743aeb3601d386a9ca07bad74c67f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 12:23:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 12:23:36.000000000","tz":60},"subject":"Disable configuration of SELinux by Kolla Ansible","message":"Disable configuration of SELinux by Kolla Ansible\n\nWhen using the default configuration, bootstrapping servers with Kolla\nAnsible would revert SELinux from permissive to disabled.\n\nChange-Id: I8ad027384d9d062fdd363b10fd7bcebe22d775e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22307eb73e97833c25521fda5ff8fc6f173116ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22307eb73e97833c25521fda5ff8fc6f173116ad"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
