)]}'
{"id":"openstack%2Fkayobe~901239","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I06b0bd4634ebb8b78baa23656c891333fd96b88b","project":"openstack/kayobe","branch":"stable/zed","topic":"selinux-test-zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-01 08:14:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I06b0bd4634ebb8b78baa23656c891333fd96b88b","subject":"CI: Test SELinux configuration","status":"MERGED","created":"2023-11-17 11:58:08.000000000","updated":"2023-12-01 11:19:00.000000000","submitted":"2023-12-01 11:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901239-selinux-test-zed","meta_rev_id":"60d64b624cf5e7a1b6271cf3b3b9ecde2a14273c","_number":901239,"virtual_id_number":901239,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 11:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-30 16:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-22 12:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-30 16:01:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 12:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 12:09:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-30 16:01:06.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"6ffb27afd489cbff5c2f7604e9fc721d26dc5f1b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 11:58:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48d1e6cb0a49fbf35b6d36f3b28e5705412c4f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad35670e5631422f9f8305452e708128\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e2bb16909384c219dffe52fd8d19754 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4294a0065c8d44689931f2bdb2087325 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81b4e0e8eb504fe9a14636b79ee63051 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9450fe7fdeb42b8b06ce5698d46f46a : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c93ac36a364d71bdfced7244f3c80c : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37cf7c483824d7084acc6c20290d640 : SUCCESS in 6m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b093bac98e3c452d89fd053f3665b490 : FAILURE in 16m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fc439e3ecea84cdf9ba70775c66370a5 : FAILURE in 15m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a046baf9db34096b29c5f294baf9be1 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"84dbeafe90f35a33706d06825a9c9d5f332b0aea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 12:24:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"368adfb59336fb952304f783b0b902b3b0792030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:44:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a16d30b74114cdc858624cf7da1a823\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beeedf7c73664f3dbdaeb5f4e2586c40 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832f4d3e74e841a6bdec3b7db1f307ca : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/587dfa0202d04e8a966f1bb1c81ca8e6 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f636326995724751b9a003c4e0ff3940 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c868e4283dc420bba25362ac0ab0764 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88666d8645c94c91bd3fefdb19e560ae : SUCCESS in 8m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eb99648ca1ab41d0ad3ac6521a2d8f2e : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fd7bacaedb1d4431a0e84e9e0ab72721 : SUCCESS in 17m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61bd973a4e054628ad34ac3348105742 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3afb34f7eb7116101988c2537bf11b5d6c7f535a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-18 07:30:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e9bcf8c415490018d457fa10c553c3d019d7d253","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-18 07:30:36.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"45fdf51fabacfafe6236b91a8e42229fcb376b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:16:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bf8a1a670b644038810c5923b46c903\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a019bc384c8e49c6ae126f5b2855e70b : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92517c5250904987a38c33e2b4b1df95 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65762d4c5f6348faa7ef72642f4c8624 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a20c1c487f04b2687f02282aa75e36f : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e768a37befac4fdbbca5d5db38a155d8 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84a2d39b2b7c40c890a67a65d47d463f : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60854f24159a45f7b4db47a005c7b122 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1fd1f6fe6dfe48bca5ace90ce1f0ec26 : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/638c077e5d4848e1b467ecb04e0f5a78 : SUCCESS in 13m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9aeb22d7e57f49028d9bf4d0f657cafa : SUCCESS in 53m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fd366a60fe9b47e0bb201b6c2f4f7c93 : SUCCESS in 58m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aac356ab370248448d1a9ad3ce5793c7 : SUCCESS in 54m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/14c81e17ee874b00a8eaada7a6787427 : SUCCESS in 50m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9eafba9c5a4447dcac41f838269bb2c5 : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fa85c8b307ae43d1839eea30b254aa5a : SUCCESS in 16m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e122f7c402e54061aa5d0457db44c300 : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3182e412123c40a5b24d057a129b1a43 : SUCCESS in 1h 40m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/21fa437fe54f418ea14a84cdd11708cc : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bab4b062ddd43de8160cc63e7500762 : SUCCESS in 11m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/92bb5e0f35344cea923f4230d2b0e4fe : SUCCESS in 32m 01s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85e3c518c4f7408f9274b5c9be75e4ca : SUCCESS in 22m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d98b5607958f429c872d96fa860ead71 : SUCCESS in 9m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2232c4a678b94506b9b77772ac020a49 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ad3ebd1302e4b24bab9003d98b766b9 : SUCCESS in 20m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/072bf2fcc30e4e5b9d6bd0b65d5201b8 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":3},{"id":"32d0b566df43985e293ff10779b829e2eabd7c28","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-22 12:09:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a490de47a96e88a880d7430cb2727943ce6f7180","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-30 16:01:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"86d34007f09878888bfae99fbfc9e36a7471cd72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:01:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f792fd25db0636bd7ce220c1e0b9690c017bb041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 20:51:26.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23e676d95b7c4e6ab22c83e08e5e663a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fc980883c64eba8a02a4ff85c8525f : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fae2fa2c5aba4b0197da373356cde93b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0d9a4eb36b44452a97d43a79a1fbcb9 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3585b4fddf4acbbdcf6c3d8d04dbbf : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01beabc852484cd28bfab5b9b91235a0 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a2c585c81934ca1b4b25515a6a1f806 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af4436da583e4a69a1a1fa80df6be2ec : FAILURE in 5m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/734fe2235d74473596d73e1a0d282b8d : FAILURE in 15m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0354f56e98c64cb6826aff45b863ca0b : FAILURE in 15m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bded05eaf8f6434e9fc018960ec60cf9 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eddb153ed11546899121d3c680f18a5a : SUCCESS in 14m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f7bab3e1a8247b8888120d17beb397d : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9639ccd6cd2432db0575522e6a21e79 : FAILURE in 15m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1197f126c2b54e6f9a50fd43eba5fa11 : FAILURE in 12m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a12b8cf3db645599e81cdba0e103af1 : FAILURE in 10m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e2407067698459d841ee0d2c9dfd83a : FAILURE in 12m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/698a6d71de7441ecaf00f4d8818c5eb4 : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2d2ecdd698c4a30857f8379fa4083ec : SUCCESS in 12m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a04fbe28bf1d4ccc9ac8f05839c8a358 : SUCCESS in 13m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b09bdae773e7430eb693f05254cad3a7 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5d4be75d8a57fe0ecbeeb91793ea6487276885f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 08:14:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00f925ba668a792bf75b80fef353b5d97c88a90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 09:47:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f6f8ce84db4eefa85f0203cfe9b9ce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ec80187a8d418e8a82aff3e3a81e0e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20b5aa10ef78447f91582d04bd013661 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c0524603264f6aa301d8ddfc4acc87 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bc3020009b34d739e0c236377b892e7 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ff3815385a741e69b96e01fc3fa1f8f : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db80d04ba45148f091e3245bda979656 : SUCCESS in 8m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f74f376d9ba4ed6987d8022699a7584 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42f69d16bb664a9e89eafba92999f0a2 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e844d50c6ce4c24a09912c47a75d4a0 : SUCCESS in 14m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fcbc04d32f8049e0bf24fbf383ae269b : SUCCESS in 51m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/00a874785adc4644a51b436c1394438c : SUCCESS in 54m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b635c792f3e3493aa8e09880a52115c6 : SUCCESS in 38m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/115e46d89a27464695870d1d38229c42 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/133815dd4e3c4716837e621a4f2c23a6 : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dc7b98a04b0448e3829192a230bf681a : SUCCESS in 16m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b186860c96ae4625b03249ca5777aaf7 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22fd2782b75e46daba1f57b3c302c4dc : SUCCESS in 1h 28m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7c53834dcbbb4992bfa9495f592ef09f : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/413d832e489f468f8189559b716e570f : SUCCESS in 15m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e0252e10772949a7a140a033b6d3b577 : SUCCESS in 31m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b796003b7ef14b4ab385bb19f1706e4c : SUCCESS in 26m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d0876d8c45744ce0986822e637820d46 : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/111db62a6fa7423589b8fac687d43610 : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c43d55379af74e38a52ce7aaf89662ec : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f7fc7a12c514dbc8fa74437177c1668 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":3},{"id":"26f23d4d1bae7ad65105602792c48fe69ce3f431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 09:48:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ec0545e5829fbdc08103ed25dd775d7f0c5f68c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 11:18:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d36f70375ca413f85f2793d30f853dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c3c0ea84594610a996c2914b365125 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b5bd59c55a4bc8959f2955091650e1 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd325aeeefe4481d94b5c7f1d9bfa346 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da57e58d77e4c92af9416d647aee796 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83c3843a2f0f479887722d0421a3fd7b : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8725d8643b804b4dab01ce22109660ce : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8fe9fb077584fbc8b6185ec4b46b2a1 : SUCCESS in 11m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1527fbc7fa7f44fdbc3dfe531d46ccee : SUCCESS in 54m 36s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c0cf32c33924ed89649aca66d3a6aee : SUCCESS in 51m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0fb660a6136f419c8e05c95572773216 : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0d63e3ac15c444b3ad1ae96d8ed9abf5 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e033cd983d3443c92b5fe2ceff29487 : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/512fb2865ccb44e0b2bf824d0ca18819 : SUCCESS in 1h 29m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/95343027442e4358ba4a9b2e244b98d7 : SUCCESS in 15m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc4c4fea459a4c30b487a4ddc3248154 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df9cc550a8ad4075bb945fb32a1e72c2 : SUCCESS in 20m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e33ed400eb7945c4b9b7654d8160d145 : SUCCESS in 11m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b8e5d61b11428797d740039576eba8 : SUCCESS in 8m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c859ec2fa7bb498584e41c70b57bf5f5 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0139793cbc345dd916f83f2869a17f7 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d954302608b2cf4ac00d486193b28da6ef646c31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 11:18:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"60d64b624cf5e7a1b6271cf3b3b9ecde2a14273c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 11:19:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca873cf4a39149a9b363d47b18499ba7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb286b2f592e49e0887c4fecff0e2f17 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0f5994167d8c73b3b9bcabe355af4fb84b0fae04","revisions":{"8e470a88af33c0691360dfff1fcd0b09de7dafd6":{"kind":"REWORK","_number":1,"created":"2023-11-17 11:58:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/901239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/901239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/901239/1"}}},"commit":{"parents":[{"commit":"6783b0a39b743aeb3601d386a9ca07bad74c67f3","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6783b0a39b743aeb3601d386a9ca07bad74c67f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 11:36:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 11:57:54.000000000","tz":60},"subject":"CI: Test SELinux configuration","message":"CI: Test SELinux configuration\n\nChange-Id: I06b0bd4634ebb8b78baa23656c891333fd96b88b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e470a88af33c0691360dfff1fcd0b09de7dafd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e470a88af33c0691360dfff1fcd0b09de7dafd6"}]},"branch":"refs/heads/stable/zed"},"5d1a73e9034f6cfe116bd1ab236cd0e2ff358486":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-17 12:24:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/901239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/901239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/901239/2"}}},"commit":{"parents":[{"commit":"22307eb73e97833c25521fda5ff8fc6f173116ad","subject":"Disable configuration of SELinux by Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22307eb73e97833c25521fda5ff8fc6f173116ad"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 11:36:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 12:24:34.000000000","tz":0},"subject":"CI: Test SELinux configuration","message":"CI: Test SELinux configuration\n\nChange-Id: I06b0bd4634ebb8b78baa23656c891333fd96b88b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d1a73e9034f6cfe116bd1ab236cd0e2ff358486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d1a73e9034f6cfe116bd1ab236cd0e2ff358486"}]},"branch":"refs/heads/stable/zed","description":"Rebase"},"0f5994167d8c73b3b9bcabe355af4fb84b0fae04":{"kind":"REWORK","_number":3,"created":"2023-11-18 07:30:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/901239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/901239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/901239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/901239/3"}}},"commit":{"parents":[{"commit":"22307eb73e97833c25521fda5ff8fc6f173116ad","subject":"Disable configuration of SELinux by Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22307eb73e97833c25521fda5ff8fc6f173116ad"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 12:19:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-18 07:30:16.000000000","tz":60},"subject":"CI: Test SELinux configuration","message":"CI: Test SELinux configuration\n\nChange-Id: I06b0bd4634ebb8b78baa23656c891333fd96b88b\n(cherry picked from commit 8da3481ca236ade0a0a801f18d82153d457a7948)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f5994167d8c73b3b9bcabe355af4fb84b0fae04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f5994167d8c73b3b9bcabe355af4fb84b0fae04"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
