)]}'
{"id":"openstack%2Fkayobe~901243","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I06b0bd4634ebb8b78baa23656c891333fd96b88b","project":"openstack/kayobe","branch":"stable/2023.1","topic":"selinux-test-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-01 08:14:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I06b0bd4634ebb8b78baa23656c891333fd96b88b","subject":"CI: Test SELinux configuration","status":"MERGED","created":"2023-11-17 13:11:23.000000000","updated":"2023-12-01 12:18:01.000000000","submitted":"2023-12-01 12:14:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901243-selinux-test-stable/2023.1","meta_rev_id":"d9ec011dd13eb51d5b3830303e8215b32491d48c","_number":901243,"virtual_id_number":901243,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 12:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-30 16:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-22 12:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-30 16:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 14:51:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 12:09:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-30 16:00:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"6c705fe7eb6d185fa6bbb97da5e7e396d86799b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 13:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7699b58dc8bb1af1e6f0feccc9e23cbebfbc166a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 14:51:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f679bfa9ff2049ad9ca3455d64a21d9c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6278ee43ed5649aba74bf11de04c5e22 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6beab2ee10c4c8094da608d6d4fd7e9 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b86a73453c50486287051054a8ca8d4f : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a783fddc9433434ab3f8a1e885dba5ca : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d9b13b9be45460196092d75f55d2654 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42eae4ce2c2d4afab60d28e3d35e71a1 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1722ced955184251a4a09f7c54e6e497 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/adaa8e31330c4d6a8273a0a03c80b467 : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a9677b796e5b44d595ef139623a86fef : SUCCESS in 55m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f36497752454ba1a3778f2043dae37e : SUCCESS in 55m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aa7123b92b6e4637a0c60359ac16b2a0 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8b3292d2c41a4b25b3d34e3f8cd39f3d : SUCCESS in 15m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d44f6f1375524fccbbebd77291ab74e8 : SUCCESS in 16m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/145bb747577747ff8801af060c3dd6bd : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e071bcbeccb147fba1690cebc9c4f357 : FAILURE in 1h 20m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ffc06bb3874501b17a42efafb2899f : SUCCESS in 1h 34m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4489272ab3bf44d590778fefafc3a830 : SUCCESS in 18m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43c493094c2746b882577ea3ae15d08b : SUCCESS in 15m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/23782502657945019e042d6d0dedf1a7 : SUCCESS in 34m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/af3e72b809764e63979d9e8ba87015f6 : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84698f06ff3e41379d368142473bcfe1 : SUCCESS in 23m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e486df632c44d72813d9fef637c3872 : SUCCESS in 9m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e8a2af415554c47a0d8092f606e65b0 : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/773157fb978d457b8af1055e0e32e0ab : SUCCESS in 18m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/210edcf76523415fa65421274be94272 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a8987d6a44fba2495d192019f84c1048c8a088ca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-17 14:55:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd25856760ef1081f01ed75925b3a9d87cc82e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df738275886f4d7c8217da5609dc541d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3651de227123429a8230f05dde6793a1 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067010e6a842420a93f802e443900b5b : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f83d94d67d044e70aeb160fe3ae2f38e : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8825eace573440e6b6de53dd602877d5 : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fe7ea0c778147c0aa0a58fab525f0e2 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea458febfb814bdca7e25832065b8da2 : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1b7a3477e444c7987b5aabddf80e652 : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f9a31af3fe394742b80159cfbf41bb8e : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/80aff24fc5da451db503937f3c737314 : SUCCESS in 1h 05m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30b9e94abf0b42288771dda1a24745c6 : SUCCESS in 58m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7e290efe307e486198d70915b0e65def : SUCCESS in 1h 10m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e3dc863524d4fa68b2ea75e9329210d : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/038702e45e644f9b9fbc9647ab80a2a3 : SUCCESS in 16m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a39b989df2d04c69ac26d5518c1330ab : SUCCESS in 12m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/915ecb37a18f46f6a04b03824b9623e8 : SUCCESS in 1h 39m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9918bf24ba3a48abb7dcd3b348188234 : SUCCESS in 1h 42m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6428268e505b47d3b9dc0a265d3ba350 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4c2dde01ce04ab58653c39d1a378765 : SUCCESS in 14m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/269ec36a2c8b44a0841fd82469a197bb : SUCCESS in 31m 34s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a4531bfe2ba4a4086ab1483711c0f0d : SUCCESS in 29m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e273d1efafd4f8b9f2fc183929abf6d : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92078f0b622944d6b8dfd120e21d9b99 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce27506a6974d42aa8adf97e19f4c9b : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef8d281e935c4db39fdb1f9cf7509685 : SUCCESS in 18m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c003cde4dc3144cf9990e3d332b6fa02 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"517dd5430960730b74bebb3d9b3cb4ef5188df17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-22 12:09:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10434795f6ed7b8da507cd3c2bc6294dd3d62428","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-30 16:00:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5ae26f9bf08287c87c23bb6f19590111c00e477","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:01:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c513944f8d392c093b3160e743de86cf0f40f402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 20:52:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/725edbecc064486e8db0d4ec9e649279\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3535c4451554f4da7f144132aa30da6 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8075b28c31054e64b1de21b8728e544b : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2ec6016d9045a890fd6b4e13f9d58d : SUCCESS in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49116ded74d1416daac42909a6e4c37b : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9256db444f845ea80f88440d85e34e5 : SUCCESS in 10m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe7a1b359a02483ab95b5dd5359bd44b : SUCCESS in 15m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/78610b5401be46ca97f9f94ff43b66fb : FAILURE in 20m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e167e8deeff4e4dbf1e9e0414b73483 : SUCCESS in 52m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/311d48583f0b4387acce7de014d49891 : FAILURE in 20m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/865a905983bd4454b6b7b051b681503c : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae18c4abd12f465a8037892ba11e3ba4 : SUCCESS in 13m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9619396251dc4defb7d3024ea3f471ca : FAILURE in 22m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ff939a43e624a3b98e985ecd43c7530 : SUCCESS in 1h 36m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/85e33324422a4bef910307d2e0d8fe3e : SUCCESS in 18m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bfcb4e6e9274f8a80bba595e2efa290 : SUCCESS in 14m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7656c1e7c257432bb50c8a62b1949132 : FAILURE in 25m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3d117b353d648329f83fed9c5caf881 : FAILURE in 22m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a5539e72a2094f84b5546a788bddc18c : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a233d5bc050e4ed49d9068da2d94083b : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/63c7e22d9b9f434ba5e08cf7e2e4b084 : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d2e401605c46f2a515a98dab667c0d : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c2116e9cbdebe3c3a503577c2bc618bd87da9d51","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 08:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b27ccb26fa69ed9fbb59961d9b542bae44f6a5cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 10:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff82d4dca7cd4fdc83d9cc549c5f610b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f57fa31f3bec419fa9e68a4cb1b89429 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f852d143c8c40c59e1d1f3729b1fb14 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82eb9cf867bf4a13a6a6e1216827f62d : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09c79c62017a4f259bcafc5da163a7c6 : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1647ba88aed24596bcf8a68ece360bdb : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4862b3a27ff42228796895f59312414 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a635f09b4c804370a32c100dbbd7af6b : SUCCESS in 12m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/edc478ee600041e5aa93caa7ccf24772 : SUCCESS in 57m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2562be9372a94f62a59c5404454d1ab7 : SUCCESS in 1h 02m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4025cf2a85534f5ab264d9bfd46c574b : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8f1bff209b984a579677c547ec8352e1 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2a0b8ef5e0234d5caceaed9f0f867082 : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fff1dc5c25fe403b8ca131a9668acf10 : SUCCESS in 16m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1a7f13ce4ca423bb9c8f3a39c2d1404 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/993deb710c8242d28111429e35bfa03d : SUCCESS in 1h 52m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c61c9b02ccdb45649a27430a88ac037a : SUCCESS in 1h 35m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4302e04741144484bfd94deb7db77289 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d880c83560242e28300978627b14361 : SUCCESS in 14m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/34390da94bcd44768d8e0ea7f9e60916 : SUCCESS in 33m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5c1ec5b304a849839eb26455a000e8ff : SUCCESS in 33m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecaa98d892084770a1c9a7d7af2fa053 : SUCCESS in 21m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e46bc383e6a64b8295aa500215406ea4 : SUCCESS in 9m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5eb689bab674df1a9ba621952a13158 : SUCCESS in 13m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/72b45f83f5bd4d78a92390260ea9f400 : SUCCESS in 12m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3851eddac0154f06a6d5c2d695d3c0e9 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"966f8d37ab5bd6e1ca6e7be7cb8a05e498d6da6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 10:12:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a2852a569667e6aba7c31e5f67239f3740aa25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:14:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfad8aff640455281c8d648d451633c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9363a1b3ff704aaeba1075740d54d717 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32232face2f4de48448cf0cadb53194 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d14d3b79dbd4d70bd708966e5056c19 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a29bfba45fd475982ca3e4a202ed539 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdf76ed1f4344e7584e4dcf84cb2b74d : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c46ebc47b66a4e619abfa3950017a430 : SUCCESS in 15m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8e8bc6f13dce42f1b5424dd0f6993f9a : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cccc3d8891e4e519b34da2287b39274 : SUCCESS in 53m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/067fd79d2b8a4edb917ee0fa6e9a8dec : SUCCESS in 1h 15m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8926458e89014df78cb822339bdcfed4 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed44dbeececf45458c5809f58dc87961 : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/12976558d0684ee58faf1b885d163ffc : SUCCESS in 2h 01m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f30b07ddaf64b4b8d54d45391565c58 : SUCCESS in 1h 45m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/343e8594997149929255a45ec432e08b : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cfb608222d34ac5b015f60b9f27aad2 : SUCCESS in 12m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3177e74fbf0145a2a14c473bd379b56a : SUCCESS in 22m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/684fadea52bc462584fe8c98c29a5ad1 : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2aad7708f9847ff9992c9666e861651 : SUCCESS in 11m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419c2fb926e94f10a8303347ddc9078f : SUCCESS in 9m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23cfdca8e47d4a529fd0f14b9cf72b4d : SUCCESS in 19m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b149ae89e504b40b53f05df00288fbf : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2e21ab471348f2851736a980bb29b87ff66402bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:14:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9ec011dd13eb51d5b3830303e8215b32491d48c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:18:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5903e29c26c34ecdb4f61a65e8ffd571\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7feaf578bb9d436abce5718c0410c05f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c","revisions":{"2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c":{"kind":"REWORK","_number":1,"created":"2023-11-17 13:11:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/901243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/901243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/901243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/901243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/901243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/901243/1"}}},"commit":{"parents":[{"commit":"7f6d33a6dece10228bdbeb95884695f2df7fd79b","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f6d33a6dece10228bdbeb95884695f2df7fd79b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 12:19:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-17 13:11:16.000000000","tz":60},"subject":"CI: Test SELinux configuration","message":"CI: Test SELinux configuration\n\nChange-Id: I06b0bd4634ebb8b78baa23656c891333fd96b88b\n(cherry picked from commit 8da3481ca236ade0a0a801f18d82153d457a7948)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
