)]}'
{"id":"openstack%2Fkayobe~901472","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I994fa1b73f2bfa08a386d0b788a842eeb5cc688a","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-22 16:01:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I994fa1b73f2bfa08a386d0b788a842eeb5cc688a","subject":"Run elasticsearch exporter on opensearch hosts too","status":"MERGED","created":"2023-11-20 17:22:13.000000000","updated":"2023-11-22 16:02:41.000000000","submitted":"2023-11-22 16:01:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901472","meta_rev_id":"45afcae7e3dea3c17e58f7b4ae980380f635ee0c","_number":901472,"virtual_id_number":901472,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 16:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-11-22 07:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-11-22 14:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-21 16:03:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-22 14:25:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-11-22 07:09:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 18:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 16:03:58.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-11-22 07:08:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-11-22 12:09:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"fe2e5d0dbc83f8bd78eea45407a103e335cff862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-20 17:22:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87b7a50049024f8036a33c7858734b1ad6d96d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 18:34:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7de9bec2b3e4cbf8b20ab7a6865ee3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73ce5ba5cc2745cf853e58a745b4f9fd : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06bf58b7a124dcdadd128da38e4a315 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3c214b76a4422cb84cfe57bb8df77b : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/352e6b2d772e40038b3141d00bb472e8 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b3112e1fd449b6916cce67f7301d33 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77f5b4e989f4085985a8ff66ebb16b4 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9cdce86e92c43c4a8cba3ce1c33b8a1 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bf83789387b4b35b5f7a3597e8790c1 : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2df04a34b16455da451b97db86acf42 : SUCCESS in 15m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a74125537aa94171813af6772c5e2f79 : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8dd1d1aef133413c9bc78729227a09c4 : SUCCESS in 48m 11s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f7f10defb6004ccc806c59bdf17b5bd5 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/af1a64e698354bcf9ffe88b564083c07 : FAILURE in 46m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc0e807a8aa740978f5f2246d4d1e5eb : SUCCESS in 54m 50s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39cb9a8298cc49eb88ec8760f523734b : SUCCESS in 47m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b715236b9e9345b9bfd470ae9c0f64d3 : SUCCESS in 25m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/27fb184988934062a52c6067fa3f37fc : SUCCESS in 22m 44s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/50ab36e8c45e4f0e94e7a368017667f2 : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c3a64cda2d64bc4b32158bcca79247d : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa71b6193e614094aba2d351ff6279b8 : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ec249e89e7b4afcae4407585ce32c04 : SUCCESS in 16m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ff4f079edde24fa58179d7b29844e18c : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9a536ca8b7341f88a157f2dbc90f174 : SUCCESS in 59m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f0d8db9047e144f5a156a37e7ad71c05 : SUCCESS in 22m 25s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/67bcae85e7dd4c929eb1f53073047c5d : SUCCESS in 19m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/813b8360e6fe4c5e88e617fb3f0c97d2 : SUCCESS in 21m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79d4631aaabe49b68af4c91599c52955 : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94da36eac3594e328a2de0956a34ea2b : SUCCESS in 15m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/655fc9a3ac7c4f90b30624dd7291cea7 : SUCCESS in 18m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2c0f8bc9b25c463a80f2fd49724acf6c : SUCCESS in 16m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b0e8b871c5649a6bfce135100a378a5 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d55867a78ac2413c8a6b5f82c8bb41ae : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/510fb2be0c4444148250b0f2fa005b18 : SUCCESS in 13m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/85f6cb240c9a4465a14db7e6dbaf87f5 : SUCCESS in 19m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/55e190c8b1c640f7b3bdbcb2c6d30cb0 : SUCCESS in 18m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6b69b6d9c4764499922aa8783fd14945 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5946713e59654d9bb3c62e1096a21e37 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca3b5596a74496faba8c71e039fe4ce : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bec99a58f04cac69c06f4e698e015c150f9822b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 06:22:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93e4c690fe8d095911ed84c607f4a9f01b27736f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 08:07:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79f2c7d384794f3ca818d6aee6d9c867\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1195cb12927d4244a8ce05d1b5ad3d5b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/412e415edb8a4a1c8c451a25f2a9bafe : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/472514983f194742b18da18da8134b42 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d75461b64b554e8281759b6b845117d3 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0e7130caa34a1495e4f3e05e6f677e : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0fa0d97ef8a4f4e941c9b316693e514 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a293fdfb97924a71a46657ee80115d45 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b0767089a034e8f833182e231539156 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7280e5c393de429e8decafd7e45adf76 : SUCCESS in 15m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5af0ad0a8f2a426a8da667d5f64e70f0 : SUCCESS in 57m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4d4705ca66de4d70bb5034393fe282fd : SUCCESS in 50m 53s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/51f6c32c765d413fb60e587b2bef9015 : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/40e099edddf0497e9b782e144bef2aab : SUCCESS in 48m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d7056c8eddb4f20b5185b4493e4ecde : SUCCESS in 50m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9eec53ed6154d6cb043cdc58e4141a9 : SUCCESS in 49m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/352508b488604c75bb0c2e83944812ca : FAILURE in 12m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/563d92d1314042f286f665a6b12b8d1e : SUCCESS in 16m 48s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/66efddea28704583aca4f0d6ef4b4784 : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e9bd65b74cf84f0c9c084bddbe993948 : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158bf3c4b96946f19c3b512dc03056d4 : SUCCESS in 12m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4dcc458e0c24e958bd3de850416f825 : SUCCESS in 15m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6b82cbc8656a4ad2aca255a2390bc7ef : SUCCESS in 1h 23m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6c6eaa6cf6e74318a2a314e8249fecda : SUCCESS in 1h 03m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/799651bc493d4d25aaca096585b2d252 : SUCCESS in 21m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9029e6eaeb814735ae6b05bd3dcf487d : SUCCESS in 21m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/40e5063dff8e43c0b52f6b2ca41cd51c : SUCCESS in 20m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e76ca6458464be994401b84705c7a0d : SUCCESS in 13m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a6e17fa71dc4dc99fd8f23ad75d1f1b : SUCCESS in 18m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/83d12194d0cf4abe8703b2a2dfa46058 : SUCCESS in 15m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e50c642bb2064e2497baccc76d1d5061 : SUCCESS in 16m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/103606deab5d4582a2e497251e282a21 : SUCCESS in 16m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b3415d1aee8468d90ffc88fe9068649 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d4cfb8582b24953815cad28f6a7a723 : SUCCESS in 13m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ea6a10bede34d29af42b6831c0e8753 : SUCCESS in 18m 11s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/27f16b51bf7d4302b95a042a0fc6da19 : SUCCESS in 20m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be176437b7ba453095504e24002f34c7 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b85f5f3e3f48dca14322bebfd1458b : SUCCESS in 14m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b75e752dce145f0b5a306def87e16bf : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9bc101cc6a61d196a6f36bcb434dda1c77df3db6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 09:05:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e298b36a7340a718db1fab446cebbae1d225cac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 10:50:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7850d1af5e434546882008a96546556d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52395ec04eec4dd1b85dadc90768538c : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b45d0cd72a412aba267ae7f5d4bb3d : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f138aea0498844f1b0aeee9ad31dc9e9 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc60bcdb9d794ad3af1b761ff2325615 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c6c29cd926c4f40b6fe84d716bd1a31 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7105f56730c4a8a86d83dfaa97349ae : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71049a88d54940469aa6d972f3085e55 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17689383333a45c099c95c0691755956 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/669a64d704ad4ab9ab385d225dc4640b : SUCCESS in 17m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56861666403440369b1c9568829369d5 : SUCCESS in 1h 09m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/84ff417abe8841adbcec1a19ce587753 : SUCCESS in 55m 15s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e47f0382362143a4a53c1903cc9fbb36 : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e58a76c3c4114055ad4182b02985f27a : SUCCESS in 51m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1f348b7c1b54883bbd447c12fb2e70a : SUCCESS in 48m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e4b3036b34640739855a226eb2b5c76 : SUCCESS in 50m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bad4cd76359e401894ef2520059f8052 : SUCCESS in 22m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7eaa9587d5af4a419d9e4fae04298691 : SUCCESS in 19m 47s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/df48157b25d044e0a7d55d74d4217213 : SUCCESS in 26m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/53c10aa4311e4c648f25d4be672f510b : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4522020760e04fdd9a0682224e13dd10 : SUCCESS in 15m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/162fd59602cb467c9e4bba50ca37c1f3 : SUCCESS in 13m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e738f1e7db64065a104269eb3fcef4f : SUCCESS in 1h 13m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/728e54ddc94444c5883d0eb7243d0759 : SUCCESS in 1h 03m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/696a5e87838148c0a92da03b18cf85b7 : SUCCESS in 24m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/fed886351a7f4324befba241d22c1193 : SUCCESS in 18m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ac74e8b7f5314d1aa8adff68c76df6e1 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd9ff14cc40a4d68a223337e838b98ed : SUCCESS in 17m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b422b5f9bcf54f6d88d36fe6d4c5830c : SUCCESS in 17m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9c99d3c4c4db48f193b01a358bccf21f : SUCCESS in 19m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f82753b1f3244178bb02c22a03dd7543 : SUCCESS in 17m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c24103d4bf9e40719f19f6a7372e4879 : SUCCESS in 19m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c4874bca027455b827c06b6337643fc : SUCCESS in 11m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b8540e989914cd294accb394069d9c0 : SUCCESS in 10m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2ea9c4c5f6e046c8a6b5db63e3578aef : SUCCESS in 20m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2c2e67000bea461790408db1636df553 : SUCCESS in 19m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee77653081cd4fe0921740f851516474 : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57cd5bacf189406e9f8f7dee3257798c : SUCCESS in 15m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b5c34bde4c64825b7fd8413df07eaf8 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"188d7f3cdd673d5d1e8709ed82995cbc0bf87131","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-11-21 16:03:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc18d76fb245a136c8716cebc5a96d63c34b40a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 07:08:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8fdd2d57a2731044675c005203f729f32962d9f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 07:09:01.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6e8dd7f35720dec68ee901e1495d65d2b39cf83e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-22 12:09:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"68a7e1fdb0e0e74eeeb48f5a5c96addfec5d6048","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-22 14:25:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7935b5d1a42a7c61bfad340f601b63bb28dbca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 14:26:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"782060428febea82599644fa14cb96646994586a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:01:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a314634f4b7437b9f537728771fd1b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012294e103b6491fa14eb3ef94b7665f : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5faacfcc7aaf4e8aaa26a153694b0e9b : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d347b31a644494b8ae041ca7d2d25b5 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ee19ada98743efa908ff1177aab55d : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb3c090c10e04ea8b0b0e5db8adedd53 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b192410e8bc646bc9387571e66c58175 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c20e34be9f64e198189dd39427e10bd : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33e0d7881fce4c589bcf45daa6a7c0f1 : SUCCESS in 16m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/974b1e3e098c480187a3061ad4758db2 : SUCCESS in 1h 13m 30s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/63c824215f874c4f822e768475960886 : SUCCESS in 1h 13m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fac94987bf9146dab1f8dcedb8282a7b : SUCCESS in 49m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9f0492f938142a7b498f82f5a84a809 : SUCCESS in 42m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7d18e49c1d147ed8f2722a78275a672 : SUCCESS in 1h 00m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0720290c2e4446598170e148d63d519e : SUCCESS in 23m 26s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/37768dcb940f4d4abd448de0f61a4022 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a5a9a1eb421b42238b1a570ddb790275 : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c5aa70e48c147f3bdff114a045f9d9d : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2892ad1d9f548f0bbea699d701c3ac5 : SUCCESS in 14m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c127499b6c3d4be485b4cb2b1454de94 : SUCCESS in 1h 24m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/68a1e1ebbb4f420e9e775333444a03d7 : SUCCESS in 1h 03m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/261acd541bd5488793bae81730f93829 : SUCCESS in 22m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/198c5f0d629643799d9e0ed85c91bf5c : SUCCESS in 20m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/36740d8843ae457e9a57ce790f936881 : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5524b8f35824528a06338234ac2b52e : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c08c5b8ee0d40bab67d059f670076eb : SUCCESS in 15m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8b27a40245e7499ea8194f22f2f15c65 : SUCCESS in 20m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/125dc0a3e31445cd8aa3bf7ced4dca62 : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a45fb9a5ef9f4c53b0157201323145f9 : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14679ffe62bb46b3a5c8456aeb84eed5 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76a176d55c45474aba278a93e2598561 : SUCCESS in 12m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/69ba90f4834847ba8005a4b62af539cf : SUCCESS in 21m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3a58a0eba4364100abd61f616276229a : SUCCESS in 23m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18c4adad1f1a47a1a048a24940cdfe28 : SUCCESS in 19m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39fd033cf744448799f142bcd5cef869 : SUCCESS in 14m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc6303be7dcf4056aa0ce0f5d798e0cc : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"24107c81513bbc4d51850fdeaa0c31fe5c37796e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:01:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"45afcae7e3dea3c17e58f7b4ae980380f635ee0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:02:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68c73d4fa0f429c9650117dbe5082b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4e77d06e7a41e1bb1ed0bc36c750f3 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bc226c28c824ac791baa260aea01dc8 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f41f170235da4f83da40fecb18a15a1787362e3","revisions":{"4f41f170235da4f83da40fecb18a15a1787362e3":{"kind":"REWORK","_number":1,"created":"2023-11-20 17:22:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/901472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/901472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/901472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/901472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/901472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/901472/1"}}},"commit":{"parents":[{"commit":"45884ad0c062cc42929bb9b4ed42fad2e7f19c8c","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45884ad0c062cc42929bb9b4ed42fad2e7f19c8c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-20 17:21:58.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-20 17:21:58.000000000","tz":0},"subject":"Run elasticsearch exporter on opensearch hosts too","message":"Run elasticsearch exporter on opensearch hosts too\n\nThis synchronises with the kolla-ansible multinode inventory file.\n\nCloses-Bug: #2044005\nChange-Id: I994fa1b73f2bfa08a386d0b788a842eeb5cc688a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f41f170235da4f83da40fecb18a15a1787362e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f41f170235da4f83da40fecb18a15a1787362e3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
