)]}'
{"id":"openstack%2Fkayobe~901473","triplet_id":"openstack%2Fkayobe~master~I03f83db6c2b447e4b121671b7ff2e8c130f10971","project":"openstack/kayobe","branch":"master","topic":"provision-net-cleanup","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-21 06:17:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I03f83db6c2b447e4b121671b7ff2e8c130f10971","subject":"Remove code setting Ironic networks as non-shared","status":"ABANDONED","created":"2023-11-20 17:52:03.000000000","updated":"2023-11-21 17:02:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"347cce978959146cf2e9f5ae0bc576e6381f9814","_number":901473,"virtual_id_number":901473,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-21 08:46:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 20:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8554117ab0b050e255f63be07e92ab4767b120fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-20 17:52:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f892e58f4b0acbc44fc53a00b152cc9bad24bd23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:13:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/762fde7b5e464ef898d2299736ebadc7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76a05ee7b34f44e5bab5af4d524f1b5c : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980ecd0d20af4017a44f59a76fb34a36 : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10b869b3b8d14125a7c2d0c596d3d6d9 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f14a98606794bf489b250cfc24e8820 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3cb75014be24e689793ebfa47781ea8 : SUCCESS in 10m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d557a6fbb2fc4ed69d2f26acd392e5af : SUCCESS in 56m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8b1b874af743476f97fba2a3a212d958 : SUCCESS in 59m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/515d792908bf488b81317b19df3dd1f9 : SUCCESS in 51m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a1cce17aa9b74f309e2a4566e4b2a92b : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f8d5563ef1647299657f3615876132d : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e99abc40911a4f81aa53a3a5c4067c13 : SUCCESS in 18m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9385b66e8d2482e8f4a7201822cfb24 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/230a5adeb1604590b38b0f8178f97d5f : FAILURE in 52m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2defab9faada4a0bb1cf41ce37d4378d : SUCCESS in 2h 15m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d21aa090279b4d198a2267d62b76ee30 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6474452cdf4c4b24b872cfb954fb3cd4 : SUCCESS in 15m 47s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f2170840dec146a199b754558b7ed703 : FAILURE in 18m 49s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2f30d92dd10248aaae138114df65ef5f : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24ac33c692d9476e90b5fd2646044cb7 : SUCCESS in 27m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fda9c002e1c84849b0e932b9c8c66c86 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/346fe785a86a4021a76cc06152e0722e : SUCCESS in 10m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8df853f50a8f46bda2a38d42e11965cf : SUCCESS in 17m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7ede9d4186045a19cd2ea772e3c7c3e : SUCCESS in 15m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8221568cf85d4de08004132ca8eecfaf : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2f1531c339e4108a5506d47aa333f81 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fe2441317494ef8a83582cf2bb60170 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d94de6d9115f74f2e96a3f02ba2126fa72aa34b7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 06:17:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b49f163607db8e1995d1577096334f5a2717080d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 08:46:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9de76c25ca4c698baea876c7348295\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f345da91c954445a194c4be5c4c6bce : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdb8dd68fff4587b9e6c00ef00b6612 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdf8ad211f78442380db3a1a139a792b : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fa568f5c36949ec9efca7a6fb04a558 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c92b781c7823415cbbbf9ec9e2309982 : SUCCESS in 11m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8632375ad8544946b25a728933fbcdd0 : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/591db8f2aec341488f60bb213e2581e0 : SUCCESS in 1h 01m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45b85b7e142541718d65eee27a7d2f6c : SUCCESS in 50m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1a3ef13a003b4628a8fcebce556dfb8a : SUCCESS in 1h 11m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c403b42d986f454ea00a298fd093267d : SUCCESS in 16m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a502ee33fd004f4ea30dbd3dd5a4356d : SUCCESS in 17m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b95e999606e84f25828cf1597d1b16ec : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5972b512ccf84324bcb16ecd6fe298c6 : SUCCESS in 2h 20m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0b7b5f2623744d8bad60b63ee4e4b5f : SUCCESS in 1h 53m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/97ff72e09df94a94bbd490edf0c28892 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cac1550af044c59a8195173067b0b2f : SUCCESS in 11m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1bf43e75dbf54c9f85d1f516477e1da0 : FAILURE in 33m 47s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/897e7cb4d1324ebab46d807062f8b48c : SUCCESS in 34m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2a1cd2228e84b8096deaedbb56df7ba : SUCCESS in 19m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5532f163c91d40db8a9ab73b4cc6af2c : SUCCESS in 11m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10df7a424534450eba9da47ba6661e1b : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/679e8cee3483483d8be009fb410cc405 : SUCCESS in 17m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e9b1496eeb24680933f341d1d2cb6ce : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa51b6a1570c4bd48261e3b900dde3c8 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/added04a5f29477cb8b4be8b943f97f5 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f925dab60774b449f4ac10d9fce4bb4 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"347cce978959146cf2e9f5ae0bc576e6381f9814","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-21 17:02:18.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/kayobe/+/899936","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8df05f57d2c9680a31535e971f69fefde6575b70","revisions":{"8df05f57d2c9680a31535e971f69fefde6575b70":{"kind":"REWORK","_number":1,"created":"2023-11-20 17:52:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/901473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/901473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/901473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/901473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/901473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/901473/1"}}},"commit":{"parents":[{"commit":"8da3481ca236ade0a0a801f18d82153d457a7948","subject":"CI: Test SELinux configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da3481ca236ade0a0a801f18d82153d457a7948"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-20 17:48:54.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-20 17:48:54.000000000","tz":0},"subject":"Remove code setting Ironic networks as non-shared","message":"Remove code setting Ironic networks as non-shared\n\nThis transition code was added in the Rocky release, in 2018.\n\nChange-Id: I03f83db6c2b447e4b121671b7ff2e8c130f10971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8df05f57d2c9680a31535e971f69fefde6575b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8df05f57d2c9680a31535e971f69fefde6575b70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
