)]}'
{"id":"openstack%2Fkayobe~901754","triplet_id":"openstack%2Fkayobe~master~I21016445e33f9b7cca9faf7abf2d60bcf042fb55","project":"openstack/kayobe","branch":"master","topic":"prechecks","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-11-23 15:34:16.000000000","reason":"A robot voted negatively on a label"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-07 15:24:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I21016445e33f9b7cca9faf7abf2d60bcf042fb55","subject":"WIP: Add skeleton for adding host configure prechecks","status":"NEW","created":"2023-11-23 11:49:00.000000000","updated":"2024-02-07 15:24:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63fa8ef841804a1f6e8eafe1bfa82875eb58250e","_number":901754,"virtual_id_number":901754,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-23 15:34:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-23 15:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 15:24:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b07a4e971ac62b1e108db0aed087ae406e29fd6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-23 11:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f72956aefdd95c95f90df5d83d798a3a0ed08a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-23 12:50:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d50837d44483dee61ca600d051da0f309905b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 15:34:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4683281001c841849601f21091eb9803\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b901e14b5af140cdb1136d7450f6feec : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d4f075fb6f4f42a017aa5b7ae708df : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2464f6a189d14adf844419d8fb5b0685 : FAILURE in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c27010d7221c4c6d9d29c86f8ff778d3 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/76ccb7593b4948a9b2e10768a3c6e2d4 : SUCCESS in 8m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/16cbc6c8c85047398ca66eec4d198623 : FAILURE in 9m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7d59851a2c134d78a5d002785a59f3be : FAILURE in 7m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2b65615084f45f68891ce4979e30920 : FAILURE in 6m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e407d7e689a24648b4e010728a7a58f0 : FAILURE in 6m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/572fd72d91b748aeaba73329a60199f4 : FAILURE in 8m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b3702ccd85514f168a11c39b5d919d90 : FAILURE in 8m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27ef273d128349218a6425e1a55eea52 : FAILURE in 5m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/16acf07c175a4e09a75030df4bca732d : SUCCESS in 2h 39m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7dd54ea8f7346569249d009038f6c7d : SUCCESS in 2h 18m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/050e1848e9034a818787a4ba4b36c188 : FAILURE in 7m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26dbab2572884e23a4a03e82244524e1 : FAILURE in 6m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2b142a0221ba4950858b08067d76f879 : FAILURE in 7m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/31d0407a2b4b41d9be5297d391a878b7 : FAILURE in 14m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6df690d21b59447885a2b0cd487d5a53 : SUCCESS in 26m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9976f0256af4c919392529e39472240 : FAILURE in 8m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adedb677223a457f86b7d39a0b3898d4 : FAILURE in 7m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b77d90c83f5400ab646bf40d93d7ed8 : FAILURE in 8m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3793f2a0df5647d28186a643e8453449 : FAILURE in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa221bfc5834a76b728399d0ae2359e : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ceb8005a50204914bf59a24aaf3c707b : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/122c800b55ec40598c72d62bab05ef59 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"759a72ec374d5ea491cdd08a4bbdffe8f801efca","revisions":{"a5142ff6e0dd6c2f49addad88300c92ae8b7e6f3":{"kind":"REWORK","_number":1,"created":"2023-11-23 11:49:00.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/54/901754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/901754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/901754/1"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-11-23 11:46:34.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-11-23 11:48:31.000000000","tz":0},"subject":"WIP: Add skeleton for adding host configure prechecks","message":"WIP: Add skeleton for adding host configure prechecks\n\nChange-Id: I21016445e33f9b7cca9faf7abf2d60bcf042fb55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5142ff6e0dd6c2f49addad88300c92ae8b7e6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5142ff6e0dd6c2f49addad88300c92ae8b7e6f3"}]},"branch":"refs/heads/master"},"759a72ec374d5ea491cdd08a4bbdffe8f801efca":{"kind":"REWORK","_number":2,"created":"2023-11-23 12:50:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/54/901754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/901754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/901754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/901754/2"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-11-23 11:46:34.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-11-23 12:49:49.000000000","tz":0},"subject":"WIP: Add skeleton for adding host configure prechecks","message":"WIP: Add skeleton for adding host configure prechecks\n\nChange-Id: I21016445e33f9b7cca9faf7abf2d60bcf042fb55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/759a72ec374d5ea491cdd08a4bbdffe8f801efca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/759a72ec374d5ea491cdd08a4bbdffe8f801efca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
