)]}'
{"id":"openstack%2Fkayobe~901941","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-28 11:04:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3","subject":"Improve neutron images regex","status":"MERGED","created":"2023-11-27 13:30:22.000000000","updated":"2023-11-28 11:05:35.000000000","submitted":"2023-11-28 11:04:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901941","meta_rev_id":"caabf1fcd78e72c12deed4e4d64b33e963477cda","_number":901941,"virtual_id_number":901941,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 11:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-11-28 09:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-28 09:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-28 09:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 13:30:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-27 15:01:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 09:46:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-28 09:49:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a5f368136587dec0bdf9713942b04e0b8314e1ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-27 13:30:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17cbe530eb9d9087cbc2550881df869d569a1787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 15:01:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655bab04cfcf4df99f0b28a851797cf5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c775110fd58a4322a3f4ae476a903bb9 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a83794c4174d3b9cad83522cad6ffd : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd58c2ea5824153a34f37df27203825 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea5bc421e72644eba78137e61048b751 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bd99ecb26f40f7be539221d2250057 : SUCCESS in 11m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43917bd826544275855936e5ed87a49c : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bc70497da064fb2a15aa3d29c9ae071 : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b57816dbb7d64c85a6581483d75c19e7 : SUCCESS in 17m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0c5a59383a744abfa8580b700d0c0e41 : SUCCESS in 51m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/15269e0dcba74d8fbc558b572d52c69a : SUCCESS in 50m 33s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/aefbcceaac784eedaa3aeaf39d14f556 : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/40eb4b68aeec48c181fd9ea54ac822f4 : SUCCESS in 46m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d50ba91da604159b442680ec2c4bba7 : SUCCESS in 59m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d89f330fa66e49f2b1e26d59cd0614b6 : SUCCESS in 56m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0014c80566e34680814ef217e07f0a41 : SUCCESS in 18m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7ed05428cd604c2c9b4227f9936b3531 : SUCCESS in 21m 06s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/70864d00fcfa44029a58ba23e1874d8a : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f8243e2372154a3d9a64a0da962d9f74 : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c7a765b766476595199ef3c88c1789 : SUCCESS in 15m 15s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52f6a33cec244ebdb195c1e144ec1ad4 : SUCCESS in 13m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/382ceff6b0774199afc76dc9e025ec0d : SUCCESS in 1h 23m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/65453683c94e4962b5f4ba684155088e : SUCCESS in 56m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4bcc13a24eb64ada8d4ce1351337df6a : SUCCESS in 25m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c5e26219a8474cd9b93a0c75c51629cb : SUCCESS in 21m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7708e8e3d4794039af329cf9a2dcf5a6 : SUCCESS in 21m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba3c10994af04e1e9b8355729b6b9f70 : SUCCESS in 15m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bef6137f56234caa9ee81a4df0c49a56 : SUCCESS in 14m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/201030b3b97940ae8a6478fad5a5a07e : SUCCESS in 17m 49s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9ebaf6758ae4407b8cc5cb0427107dad : SUCCESS in 17m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/95ff92c4783a4226943123c3494ea3fd : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/402edd2fe54e4ab5a1066c75dc19c6a8 : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87810ebdcbc8466a86a63230cd899528 : SUCCESS in 11m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4c3381851143405e896b525e33689ab0 : SUCCESS in 17m 23s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eae7d30d1792468db2bc1e67e8553dfe : SUCCESS in 22m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d869e820f69244f7a564f5de8b4fc14f : SUCCESS in 15m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f4838fd680e46f8be24c14fafaed0c7 : SUCCESS in 13m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31b43de63645442f9687d655d22f166d : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"67cf2d78e5b042385e9e1e75cff34273cfbff7c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-28 09:46:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f158448ab6c990cb0e1d6f6675e1fddbb204115","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-28 09:47:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a2f3c6d1b8ecf6eb9f438c0339dfc38f0952a86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-28 09:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"674cf9b3adae5900b229ae2d598793204db04b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 09:50:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c63e0c42adc4f1bdf1ba802c2d417ac0e72bda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:04:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d1efda4c0d4e239627f16417191a2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1aa95ee63b44f483dfe2405a894fa1 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/126829881379472b850843d199c99e5c : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/073e3e29e74a4d1c843c32596f8da4a0 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31a75d4fef141e08ea027bbaf1fec45 : SUCCESS in 10m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6302aca7a5bf4bf2951606380aca81c4 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c86fc77ccf3481e9c666656064b8674 : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/daac8fddea56462882680f6b08e8bb0b : SUCCESS in 21m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f225baabc2a645b3bcf9af6d154bff78 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/907a57e4012242b6baab58b851ed18cf : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd7e2229f09446ab98a34068df78f148 : SUCCESS in 53m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/353225aa1faf4702af1ff33cf2961d9c : SUCCESS in 52m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55968c1afdc46cc9e69cab910d69bf4 : SUCCESS in 57m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1016feee96b04378821f28eef6b16602 : SUCCESS in 25m 28s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b9567d7c2f5644f38766fde19cc7b3ee : SUCCESS in 20m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/83db18ab788c4621a85f219c927c8c43 : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d7f22655f04d0fa28876205930cd76 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f58455c9362438e90c7c85347e83a2d : SUCCESS in 16m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9c00ad3f6c5e4a5b849220816531d69a : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a242f6fa41ea46fb8a708322a3948c89 : SUCCESS in 58m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ccb3ae3a64b04f52babd1243e465febf : SUCCESS in 23m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2d79015c192c4bc481a570309424be03 : SUCCESS in 18m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0682c2a2629e4ad5a43629c5de3708e6 : SUCCESS in 19m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f1ca30db2a43f08dda2c362649e7dd : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0256e5542da43188bbe1567c9904a72 : SUCCESS in 16m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/025e48f4dad047c695774b607db79561 : SUCCESS in 19m 16s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/41c83d9a923e4d0da293179d2e6135c7 : SUCCESS in 17m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c96a3b9710724ee3b9a848ba4de4f63c : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd376d70a32646b587fe35d3d472a201 : SUCCESS in 10m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1304b9e1fee4200b4059b9d5dd5b182 : SUCCESS in 12m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d673be4fd434ecb84e2f97cb8fddfa8 : SUCCESS in 22m 59s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8ce67594b88f4a8ebab01f9977b65763 : SUCCESS in 17m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f7a970b878a54aae84f749c07c233b7a : SUCCESS in 23m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e14e008e61d4aaeb27ac119874fb505 : SUCCESS in 14m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00664fdb90814c2f9ad4a9574de41021 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"777135b7209d9e81481f6eb2d5057b357a8283d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:04:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caabf1fcd78e72c12deed4e4d64b33e963477cda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:05:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a70a3f510974117b955be26dceda800\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef0ee7c717b4b928df76d0f07ff964c : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"698c281a5914469329ea738e44df306841b6ff64","revisions":{"698c281a5914469329ea738e44df306841b6ff64":{"kind":"REWORK","_number":1,"created":"2023-11-27 13:30:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/41/901941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/901941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/901941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/901941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/901941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/901941/1"}}},"commit":{"parents":[{"commit":"00f19424d8dbbd54073ba82dc54dd245be5e712b","subject":"Merge \"Run elasticsearch exporter on opensearch hosts too\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00f19424d8dbbd54073ba82dc54dd245be5e712b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-02 06:43:22.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-27 13:29:44.000000000","tz":180},"subject":"Improve neutron images regex","message":"Improve neutron images regex\n\nMatch ML2/OVS agents only when OVN is not enabled.\n\nChange-Id: Ia2ba7414202de6750b18ada47e7a5bb0b636e7b3\n(cherry picked from commit d9ee59ea1bdf21f50f5798278b5f6c75bf638a04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/698c281a5914469329ea738e44df306841b6ff64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/698c281a5914469329ea738e44df306841b6ff64"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
