)]}'
{"id":"openstack%2Fkayobe~902222","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I73fad29af6a57b29afea46947d1ae90546253a2a","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-07 14:56:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-07 20:38:14.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-07 20:38:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73fad29af6a57b29afea46947d1ae90546253a2a","subject":"kolla-bifrost: Add precheck for inspection allocation pool","status":"MERGED","created":"2023-11-29 22:33:19.000000000","updated":"2023-12-07 20:39:20.000000000","submitted":"2023-12-07 20:38:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902222","cherry_pick_of_change":901748,"cherry_pick_of_patch_set":3,"meta_rev_id":"1f7aacbb5065dcbc2117232746de8e1d85590e6e","_number":902222,"virtual_id_number":902222,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 20:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-07 14:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-29 22:34:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-30 12:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-12-07 14:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 22:33:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-11-29 22:34:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-29 23:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-30 12:38:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-12-03 13:06:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"dead5ab0d58a42461f921560893df02b3e46ecbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 22:33:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1fc422385e2e5c327e571b98138e56884034433d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 22:34:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ec9e27745263faffcc99c2d5d6679c60e0439931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:53:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e168b082074943a1a785e6ba03597acf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c4425fe8016465db7c07fd670e5dc28 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8136d66bb6d8473aa1c5cffb78283f97 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064ee4fbb3034fc6bdaf5819d8c4f302 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe82a0f53e2b45ca9c326a3df81576de : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/658b05638f0840358fadee9b1b76a126 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7578338608184f76bd73ff2054a9533f : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96a33410acf74895a068275c4563d61a : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe98d3af509f49a2baccd445b75de3bf : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c4c6c4b57db4d7991be00e9ef4d0012 : SUCCESS in 15m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/84d93fb1a68a4c509dac2fc34e5418c0 : SUCCESS in 52m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7f34d73f6a6544babad6c4c7eb92e295 : SUCCESS in 54m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad105966fb714823b8a928afbf5143d2 : SUCCESS in 47m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3188276917344f718ae4e5e83b807c94 : SUCCESS in 57m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/063377a871b845a98806455d6a0d0477 : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2848aaf8016c49de82d92a3df07804ab : SUCCESS in 19m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e26ca39eddd8488cbd82e7ec5e53f795 : SUCCESS in 12m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/742103b7b8bb488e9b707a7369cd717b : SUCCESS in 1h 17m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5ae3d8418f974004bea01b2d7ec1f47a : SUCCESS in 14m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20502b1e07fc46869a5c47bc75bec0fa : SUCCESS in 13m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8322cc9d23274f11a6bd9fb39e5ba1db : SUCCESS in 27m 34s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72dbddd6f62d41e0850cd376100d869a : SUCCESS in 29m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/682ed4eaf1dd44cd8029274f46167ee4 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d73c1ef0f414c4eac8588111a550b33 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d606dad8c37f4d46ae4f48a7cedac48f : SUCCESS in 18m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a123b7c29f944c9eaebbab02bad9ef89 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7afe08fbb19e0b9a4e0a89312f228bc3e5a443f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 12:38:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9855ea8517e7fc32fe3b19f8ea727b17421631d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-07 14:56:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9273c732305c3b28b628462a42d47cdeadc48a3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 14:57:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e0ef92f02997f8a8057c8cd8ae7d932aed3c7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:04:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32c720391b8a42a9bcab7006f5cbe919\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa8e3a8708f4e0ba052418e6933bc8b : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f23bdd24494ab2b19c898951c743b2 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e3d1748452242b5b9f0c085b70d7464 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8941041c0ec245c7b6f4dc819e54a19c : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ee7ccc0e3ab42edbf1453d9a4d622f1 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/274c1c19b43e4ddeb90bb013af0b55be : SUCCESS in 13m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bdabeb5084b458aa09cf6baa44ed934 : SUCCESS in 15m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a6c2d18cec2641a595832611bc074a66 : SUCCESS in 56m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a23fb0a93fb4c498b7dd9fd26d1768d : FAILURE in 57m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/896d7c229e9c4883ada20e9c91519d56 : SUCCESS in 56m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f820bf70342849fe9cfe0c2456940d4f : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b1b81fd9f4e41bea85a525ca0b78a24 : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54dedecb7849406b98d40f82ccf7b3c6 : SUCCESS in 1h 31m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0887fccc7988409985a1aed5ea7a7790 : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bed230d1539d4fd5accb372e8cbae324 : SUCCESS in 16m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce471d787b7d49329107f37033daaebb : SUCCESS in 27m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08a74044c80441e9b7f98684f51fd6d6 : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2154b703b7bc4ee69d66cbf26c03e9cc : SUCCESS in 12m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cf4c0b4131e14751aa0066715a5d4b09 : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d936ba89b66c49cca39e03691a4b35ed : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bfdd84a03e0dca765099d9dff67638384f7e48be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-07 17:32:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50f98b71e3cfcd313441b96f5e186b5ebc0d6392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 19:09:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83461f75370f400a89f0786f7a53f926\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b326503ca45d4e82b44b746f9cbb13ce : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bd865fd7a74471ada88065912f2366 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/464fbf6a8e38452a8022bd0e89ddc85f : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7e21e1e4ae4135a13ad6cb37af23cc : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4dffc9431634cf390d1166fb612c6cc : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a323574fa31f476aaf91d17aed6061dd : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7ea49f9d1cc42afa34de8da1804f01f : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e63c4759b1b476ba9fda66e8ca2a601 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2eff228c76074828a20903b4f942350e : SUCCESS in 15m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0fcc9d854dee46f5b8a12a9b93dcafb9 : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/35d380d5bca54f95933ea15cffea32f5 : SUCCESS in 56m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c950a5e52e94740860c6e508804b688 : SUCCESS in 46m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a3062752d031489bb5e725fa2fae5392 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fc648bf0c7ed4da2a82b35b4cd20ff15 : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8589fd303163401dbd8be37f0d34f6c5 : SUCCESS in 17m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1aaa3a9ec034adc9a7cc934cdce41eb : SUCCESS in 20m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b146ec160dd4b308b0746c45779521c : SUCCESS in 1h 30m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/124fd43dfed84bc3bb45678dc98d8347 : SUCCESS in 20m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e28c0f5d7e9e447ebc4e898f88bfe618 : SUCCESS in 18m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/92fbfec6bac24b8abc59c098727a5ed4 : SUCCESS in 36m 39s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6475670421d440a92db3bf562bfbc30 : SUCCESS in 27m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3434e76f71144fd7893e8e8764b6ea78 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ee8cfd4eaef4264ad7620354be20aef : SUCCESS in 10m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b447deb6a97471a8b1dc9fe165a027f : SUCCESS in 16m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9a3c02d7900493a9a00efc9d37baa42 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0413aaf4f7715bc59c3b2e9c3fb5964a579f57a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 19:10:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea85934c5c201c2b8b0b0ad0560c6525b9b01e32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 20:38:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b032c78c9446729884db71b440dbe7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985e3e443ed74464b93adfc6675056d5 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff4577836ee426286fdecf028a07318 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/065123fc9d5a4b89bed38ac5582c3a98 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8fbd09d90645d29e7f813c0073e8a5 : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7f51848d4994d7d91f344955671752e : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec33fcb6cd6c40c4b4a61d1ca92974dc : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aab9353c5a1b4510b769886be9b02870 : SUCCESS in 16m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dcd33101f24e456bbfa9498eccd165a7 : SUCCESS in 52m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b120214279b49bf80352b258ba37e9b : SUCCESS in 52m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b61c8607a9164b6a8692d89af2d48d0c : SUCCESS in 59m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8c3e79b5e66b4405a66d30f139a33b7b : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9309da5933f4dc6bd67537f9ad8f7c2 : SUCCESS in 20m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b1c2bba83754ec7a853fbe211832b79 : SUCCESS in 1h 26m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a5a2150b71fe49ccae7812a80c38c71b : SUCCESS in 15m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e922aa313f401bb96cde89e75be4e9 : SUCCESS in 15m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d3f5431cc4e409bb09b0c8333865f6f : SUCCESS in 30m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7adfaf00aaf14ba9b58ed1113b771dc3 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40a28800c0204f6e904c052278ca7750 : SUCCESS in 11m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a5ac445b690244dcb3be26f6839c8778 : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ee49e1962a40a2bbb7eceae30bd77f : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c49b523b4b7c1d3f2d49af52e7350261a506bd61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 20:38:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f7aacbb5065dcbc2117232746de8e1d85590e6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 20:39:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcf53a0b4cd482985b738d91386353c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23579a3e350848c5a5b77344c73b1773 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68ac618dcaebefbcd73e336b677bfa37bbbfb5a5","revisions":{"68ac618dcaebefbcd73e336b677bfa37bbbfb5a5":{"kind":"REWORK","_number":1,"created":"2023-11-29 22:33:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/22/902222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/902222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/902222/1"}}},"commit":{"parents":[{"commit":"22307eb73e97833c25521fda5ff8fc6f173116ad","subject":"Disable configuration of SELinux by Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22307eb73e97833c25521fda5ff8fc6f173116ad"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-11-23 10:54:46.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-29 22:33:19.000000000","tz":0},"subject":"kolla-bifrost: Add precheck for inspection allocation pool","message":"kolla-bifrost: Add precheck for inspection allocation pool\n\nChange-Id: I73fad29af6a57b29afea46947d1ae90546253a2a\n(cherry picked from commit 1e2ac5eb2829dac3f637c7f6c2e00b1c980b22c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68ac618dcaebefbcd73e336b677bfa37bbbfb5a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68ac618dcaebefbcd73e336b677bfa37bbbfb5a5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
