)]}'
{"id":"openstack%2Fkayobe~902226","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I371e8a583c439264a88dbc4bdae14c472955a063","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-07 18:31:56.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-07 18:31:56.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-07 18:31:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I371e8a583c439264a88dbc4bdae14c472955a063","subject":"Honor custom ssh args in kayobe-ansible-user.yml","status":"MERGED","created":"2023-11-29 22:38:28.000000000","updated":"2023-12-07 18:33:08.000000000","submitted":"2023-12-07 18:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902226","cherry_pick_of_change":788276,"cherry_pick_of_patch_set":3,"meta_rev_id":"c0a0789aa12bbfe420d7d2e23936edd4abdce158","_number":902226,"virtual_id_number":902226,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 18:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-07 14:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-29 22:39:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-07 16:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-07 16:36:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 22:38:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-11-29 22:39:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-11-30 00:04:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 13:08:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-07 14:56:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-07 16:36:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c87bf29ffcddd207b4f6ed7bbbefa84ad2cfb10a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 22:38:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8cb2cb6a6cb4d50169ed87759f0a9a59b5a9754b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-29 22:39:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a9e4c60093bf6e10f2ddaeea3daad53ce4fec3bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:04:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139df1f0bc44403c81034bad62ea0f32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/525350cd8e6d4d33b9bb5a85d2640b22 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ad0d4e8b4f4656a4a873a9365c5137 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbd5215d393940e0b76456aeac6b0c5d : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c887829e506d4287ad700602f7194f88 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa734d1a002142c3a5073ee77f941df9 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfff90f0c3414c5a9cb1712958c6b137 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c43a51f095e4e0aabaa86e47bd40f88 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d485751061946bfa2a071088ff9b4f0 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/574ad4be2a1546e9aaac306f636beee6 : SUCCESS in 17m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/53ef0a27e5bb456f86393fa6fe8baa62 : SUCCESS in 57m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a43d19c181ef4a33a889b5e6d17e7ae3 : SUCCESS in 51m 09s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7cbb10c8e6eb41d2ad601aea083c1d00 : SUCCESS in 1h 16m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bd4a142b313641cc9916e7f103ff7ac3 : SUCCESS in 55m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f56c94470a2846c0b5754313269162b2 : SUCCESS in 43m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cf7cf8464684377999c7bd62ad34ba6 : SUCCESS in 53m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3e82b08ff7754c2ca2706d28316a1238 : SUCCESS in 28m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c170265301bc453393316dc43072f0f7 : SUCCESS in 18m 14s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2cea7f95bd8344849d7df6c22d5f959c : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/674786139c7f467abc893e9d4cbfa3dc : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5abf2a2c742845458a3a871d0186f4ed : SUCCESS in 14m 01s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f39e92538560436dbebdf9478313a510 : SUCCESS in 16m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ce427dd644144b2dbfeb166f6651c433 : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fec3746a775a40a09696cd125cd16e8f : SUCCESS in 1h 02m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3246de99ed284a4297fb2f243e050911 : SUCCESS in 21m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8525d7d92b8b49aa95c1fecaeafbf932 : SUCCESS in 20m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/052f08e4ac5a4df3886320cb49a54ea6 : SUCCESS in 19m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6563aa724cf4b49ad41fac49cc45303 : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c07c3dc0b4de4f5a9c99ebb39e878547 : SUCCESS in 17m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba029209ef684fadb0ce6be9b5984706 : SUCCESS in 18m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3843095bda1749dabe2e156caa113843 : SUCCESS in 17m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6edd3c0c17d14d23b005ccce399248b0 : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1211530e854b4687b2076d26bc574e45 : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/459e936f4e654362858e34e53534b3ba : SUCCESS in 12m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cde1ce4c50d244fc9b336b60fe001fa6 : SUCCESS in 21m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c64ecf00fe454cac8990548ec6f6e79f : SUCCESS in 15m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8643194fc98a482da553f6ead57c8dbf : SUCCESS in 19m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8a7ac14c3fa44e0927052a3239c101e : SUCCESS in 15m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ab2424631214af18070cfdeb2f8256d : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"15bc9db18b8ac9cc6d9b280d19fe5342a99c51ce","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-07 14:56:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b71d0660740836bdbfb4b08af1cf5b3cf1a5870","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-07 16:36:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d2dfdd808ac972ab9adc1eac47fde8cc91e22f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1bbf6606892ca177245f39bb24c4475ea3a69134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:31:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00cac5f1cca483c9cdceae96cc1bd3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e6c6da449543df9381f46dd940c3a7 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bbc7a668b5d47fa902453eef27b1f82 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/714e1b5491364eeea7ee07d24dbcfdf2 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258ab1abeb5b45a7b128fd14bb19fcb7 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2e4a6d8dbfb496e8d92a9338d890d1b : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/858508b633e54345a0fa66ee8e67be5a : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c94d5af5bd8347a5915e783c781cae1f : SUCCESS in 13m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/390b7042a36b4f20a5c8f7f4552722cd : SUCCESS in 21m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f4b822dc5cac401ba081ff0ca9814cad : SUCCESS in 1h 20m 45s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/578f6619a2814a58bfbcfc11f8817d7c : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f089d2003ee4408e8d2dd53a9db71831 : SUCCESS in 57m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ade04572b7a463887f0c2b3b8008593 : SUCCESS in 48m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/066d4e21c2ea4b208cde67d33c4725be : SUCCESS in 57m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/359e8d012ea04195af03210f03d91ca6 : SUCCESS in 24m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/322fb5e670634ff8b6c7e70fe3061555 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d8721d18b6134537aecc01da9fba6bfe : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74a91ce0739840f4b6de6e5ec5d1f302 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8e39ddd0bed48f080f51d90abcef234 : SUCCESS in 19m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec095efd02274781bf27c10bc656f821 : SUCCESS in 1h 09m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ac4d017c16d343f9bc96d11300a6c6fe : SUCCESS in 57m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cd32e656b1a848f2800b209a500535d2 : SUCCESS in 22m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8cee4dd11da6417e98a97437224d8e1b : SUCCESS in 20m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a3a811149e5c47c1bff67e0e9c637d5b : SUCCESS in 19m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d13731476584d56901475d2d2ba4d99 : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b8be960761d4e17bcb1a5dc867558a5 : SUCCESS in 20m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/098390d025af4ecba0a0fde27e8552af : SUCCESS in 18m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c5610384aa0d4563b554ebb5a282925f : SUCCESS in 17m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2461149016f04a9ea0023dcdd8a7d223 : SUCCESS in 14m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3845823cd70348a6a9582f27cd5ae26d : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/125ee613d0ab418e9c19e15d08383c19 : SUCCESS in 12m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/00c95be7f36e46959bb4dba00c3471c0 : SUCCESS in 22m 23s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/81ed1f066e3f454a9acaa72d0e9233b0 : SUCCESS in 20m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5585315c594e430a88e09519252c9b43 : SUCCESS in 19m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1da386b6647148048ec4578a1a00621b : SUCCESS in 14m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/963ace053ce847a38318b96b873f8ec3 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c5be94cf7334f8b3c68f65d186e8e722244ef986","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:31:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0a0789aa12bbfe420d7d2e23936edd4abdce158","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:33:08.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9ec24c132a749408832f7fdfc423fe3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663c88df29cb48f8977696aeb6d80928 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c8e5e5375b64663870a11df45c47bbf : FAILURE in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a14d05d537259e8bc9c8933a931a653b64660206","revisions":{"a14d05d537259e8bc9c8933a931a653b64660206":{"kind":"REWORK","_number":1,"created":"2023-11-29 22:38:28.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/26/902226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/902226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/902226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/902226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/902226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/902226/1"}}},"commit":{"parents":[{"commit":"698c281a5914469329ea738e44df306841b6ff64","subject":"Improve neutron images regex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/698c281a5914469329ea738e44df306841b6ff64"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-27 10:32:49.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-29 22:38:28.000000000","tz":0},"subject":"Honor custom ssh args in kayobe-ansible-user.yml","message":"Honor custom ssh args in kayobe-ansible-user.yml\n\nUsing the raw module will honor ssh args. The slight change behaviour is\nthat it will print an unreachable message and show an unreachable task\nin the summary.\n\nChange-Id: I371e8a583c439264a88dbc4bdae14c472955a063\n(cherry picked from commit ffbd1a7833b69b9dfdd56e5e4ff71e9bad925c8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a14d05d537259e8bc9c8933a931a653b64660206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a14d05d537259e8bc9c8933a931a653b64660206"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
