)]}'
{"id":"openstack%2Fkayobe~902234","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I4830500d3c927b0689b6f0bca32c28137916420b","project":"openstack/kayobe","branch":"stable/2023.1","topic":"bug/2031390-stable/2023.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-07 14:57:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-08 10:57:11.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-08 08:55:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-08 10:57:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4830500d3c927b0689b6f0bca32c28137916420b","subject":"Fix configuration dump with inline encrypted variables","status":"MERGED","created":"2023-11-30 16:19:33.000000000","updated":"2023-12-08 10:59:29.000000000","submitted":"2023-12-08 10:57:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902234-bug/2031390-stable/2023.1","cherry_pick_of_change":891399,"cherry_pick_of_patch_set":4,"meta_rev_id":"7d2b8362313cb05f264decd1662d8cdb17b125b6","_number":902234,"virtual_id_number":902234,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 10:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-07 14:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-30 17:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-12-08 08:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-08 08:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-30 16:19:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-11-30 17:58:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-30 18:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 13:05:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-03 13:05:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-03 13:05:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ae1376de17254a1ea0683f2bd10be24fd558dcbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 16:19:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c6f322b904c69bdfa7b3bc33ef359313223ea1c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-30 17:58:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"394849be701003a42212a9a29f01c819db050df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 18:29:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c77611127b24103b3c1bd7c2d413c4f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fed4235eb296404ba0af5f0de4b930b4 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fb5b87ac88476b92e56f34b0984a95 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7279906811c45d4937e868e171aecb0 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a550185682934e5da493bda9c7d5c5e1 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aeebf6c0e4a49fca677e708e5d83a92 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2974c7f1355041269e6bcc560ddd6ab5 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/468dfa1e65f54637b1ef6bfa1ea254eb : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/052de26a7b3e4587b576b05a2ed7cee5 : SUCCESS in 13m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8f545c21ad3d4c2b9b7d6cc9fcb71976 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5f0c2d1da3724873b26f57de5e5b1afc : SUCCESS in 1h 10m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79be01de71474829aa24a6a375d446bd : SUCCESS in 50m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18398c1897ed42b2bda14bf7071b8ebc : SUCCESS in 1h 17m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/670c9e59306449bcb0082a2af3882e0d : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/518de3eaa0614072b3179e14a0a1e24a : SUCCESS in 15m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fa865b072cd4ae980efb4abb8b9ab34 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc071aecf2c7412eae6cabb66f063373 : SUCCESS in 2h 03m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7aef751d5ac54082b76000c048bb0cec : SUCCESS in 1h 35m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d4815a0261ec4fa1bcae60e6b63b1f34 : SUCCESS in 13m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64ea131014394e2b82b77faef53767aa : SUCCESS in 17m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/650b60e7d4384591875488097ddff8ad : SUCCESS in 34m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/87296558c2e0466ea742f07b12cc959c : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ef24a9712234bfdbea9e57d227c6fbb : SUCCESS in 24m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c609dec8253c4f449aa4134f659f6203 : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/754a6779f77e4038adb2c2e023ee55b1 : SUCCESS in 9m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71a7d0b939a44c9bb2b47ae1dede8b8c : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f584298091294cf3ba6be94cb6651f5e : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c3603c5bf3457e3ac955768b3a1ce1e39b272f78","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-07 14:57:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68ead9e7cb15cb91f096e28ae78b24a29ff216de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-08 08:55:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"920552045289d9084e7404017baded495b41d16a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 08:56:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81d47fe7793eb74b319aeea6368f6cc969e72f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:57:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b362f3d45ca143078ce03d9ef3af2d69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01a8e595b0d4b3b9cdc19dc20193450 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21cf47bed96d4149ad178010ea1ca54b : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f6d7606bee4a3c8f3c32aeae709244 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4be26ea6254bd2b9bf3da07ae3a161 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1eb3f6b8cf57402c8e5b9c9665fe21aa : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f881cba80524c1ea1d3299bc0a6ff6e : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83328526a0bb40c59b92ba956b363ec9 : SUCCESS in 13m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c23eec430b14ee1b340a4b8d740a977 : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679a684135af4c09a42c395db9e21d45 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0bf0351887724500b52d7eca4346c4a3 : SUCCESS in 55m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7900e03fc9834c68a00729a81443d5b6 : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be5f785096cf455ea0356d7941771236 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f4f6b500c8d6483a8c767eb7057d9568 : SUCCESS in 1h 45m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2716622afac4df482e1982d178bcbbe : SUCCESS in 1h 38m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9fa99334c9fe4874bc9bf6ce2310776b : SUCCESS in 18m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4ca42e10bf04399ab38175cb94ee806 : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/91a7f548903c4bd59824a0eeb54552e7 : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2ba99f325c7441f82af88bed47c1ae2 : SUCCESS in 25m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48670a52faeb4843b085ac22f9ba33f9 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8d7c94d7aa74fbea37adb565f6eb20b : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a260ac573a2e4963a163747bde1fe746 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e01d4a3226b640d99f067f73479aa59d : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"52f92a44d48ad7b67219e6a63e4fafba4ce92c62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:57:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d2b8362313cb05f264decd1662d8cdb17b125b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:59:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55609ed61f044be3ac9291d472e91611\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbea1903fcc4c339e82bfefe19e9ae3 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6710f724ae744e30b08366595c82ebde : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a23cde010d3d9b57dfc20ebcf09b70784d10aeed","revisions":{"a23cde010d3d9b57dfc20ebcf09b70784d10aeed":{"kind":"REWORK","_number":1,"created":"2023-11-30 16:19:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/902234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/902234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/902234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/902234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/902234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/902234/1"}}},"commit":{"parents":[{"commit":"7f6d33a6dece10228bdbeb95884695f2df7fd79b","subject":"Merge \"Fix an issue when user forgot combine custom passwords\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f6d33a6dece10228bdbeb95884695f2df7fd79b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:19:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-30 16:19:33.000000000","tz":0},"subject":"Fix configuration dump with inline encrypted variables","message":"Fix configuration dump with inline encrypted variables\n\nIf inline Ansible vault encryption is used to define an encrypted\nvariable in kayobe-config, running \u0027kayobe configuration dump -l \u003chost\u003e\u0027\nfails with the following:\n\n  Failed to decode config dump YAML file /tmp/tmp_fg1bv_j/localhost.yml:\n  ConstructorError(None, None, \"could not determine a constructor for\n  the tag \u0027!vault\u0027\", \u003cyaml.error.Mark object at 0x7f1e5c7404c0\u003e)\n\nThis change fixes the error by using the Ansible YAML loader which\nsupports the vault tag. Any vault encrypted variables are sanitised in\nthe dump output. Note that variables in vault encrypted files are not\nsanitised.\n\nChange-Id: I4830500d3c927b0689b6f0bca32c28137916420b\nCloses-Bug: #2031390\n(cherry picked from commit 78702d0e3094e6d6a16a31eaf2517d4e0f25d1c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a23cde010d3d9b57dfc20ebcf09b70784d10aeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a23cde010d3d9b57dfc20ebcf09b70784d10aeed"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
