)]}'
{"id":"openstack%2Fkayobe~902379","triplet_id":"openstack%2Fkayobe~master~I50238dd694edc7ee02be4bfdc56ebf3a603e05d2","project":"openstack/kayobe","branch":"master","topic":"remove-kolla-enable-host-ntp","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-06 15:31:34.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-06 15:31:34.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-04 12:23:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-06 10:09:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I50238dd694edc7ee02be4bfdc56ebf3a603e05d2","subject":"Remove unused kolla bootstrap-servers variables","status":"MERGED","created":"2023-12-01 12:00:35.000000000","updated":"2023-12-06 15:32:39.000000000","submitted":"2023-12-06 15:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902379-remove-kolla-enable-host-ntp","meta_rev_id":"263b3d4b59ac8bc414e1fc04704c138eb811ba1f","_number":902379,"virtual_id_number":902379,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 15:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-06 10:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-03 13:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-12-04 12:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-06 10:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-12-06 10:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 12:03:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-12-01 14:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 13:26:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-03 13:26:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-03 13:26:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-06 10:09:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"226c7849a637477d8a898259b056844af6796190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-01 12:00:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8b844a204d3a0970daa59795eaaa39503682dd0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-01 12:01:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a38f936d66a7e59fd64410bbdbed3cfb54557182","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 12:03:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d116c88e03c8b0b78ccc8c7c75eed55ccc69af2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-01 12:07:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae82240c9cdc2f65e607809457f59f336bc2aee7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-01 14:26:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25ada07ec1a1879d03c609ebb441428d5d6749f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:35:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ade32638e14dde8dde6598443a4cde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831401b96b174731b48c5e4ff83a9948 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c830b95fe6439cbe5732bebaece2f2 : SUCCESS in 7m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65b44c51b7cf419a93f0d59bb2476982 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d7cf70f52c2492fb01ffcd543c8cd55 : FAILURE in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2710cd3fae34fbfa4aff5b04b14b2ea : SUCCESS in 11m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aa7bb9a21890455b8414c759668ad465 : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/18b8132fbe524ed294416e18479711e3 : SUCCESS in 1h 06m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/163bad2ac13b4eb9bee6a13f9fd68e2a : SUCCESS in 55m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2473d07c5d0348819f2f5a21bc0784bb : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9335cdbbdb96419790427043e5a0c0ad : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0d9520e7095e4df69e061a96d4bd47b3 : SUCCESS in 15m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f29ddfcf256c4f8bbfdba31b227678c6 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6d4aba4e8ba848809cb9cfd1b5497ddb : SUCCESS in 2h 23m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e0a816e04e47199860b9db0d83f5e1 : SUCCESS in 1h 54m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c7b345c807a9417b923ec9d5bffd659d : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0264e1ba97684d6a89aeaeebd85701b8 : SUCCESS in 13m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ca07e935e9134028b4011abac0d2839d : SUCCESS in 24m 56s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a2277e072f5349768b846bf77f112820 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b084089d0404322a0058b3a263b8724 : SUCCESS in 20m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ddc2e6415264bf7b576bb947372a973 : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f1ee89817a1404b850c18804933e026 : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03ac5d3b88754ab38a1aed1dba572b75 : SUCCESS in 15m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93833aff7c8e4881bcb00d0f13a96e0a : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7bf9db4038742a6b3966a474e0e6d7d : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af3f56d929342b59980501bd2626609 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2775c4c66a04e9094011249a1c6c718 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9fea1c7b220a8f84b053870fdd9180619ca04c25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 15:24:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e410e92dc0dbf4878b364ee78d8ad00e9f37402","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 15:24:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"88d94d5b5f5faccf6d58d3b1dc88fd877245abb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:02:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78f09cfe477549a583edc73d1aaad794\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9df1c0e586e4133be0c68b869f07147 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137d9f6507d5447dabaacf10b591621a : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/593657c43f204a8f89133b16cf553370 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bef8fc9f38945ea81bb2bd50892d806 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/614b52d68aa04de9aa058fc191a931bf : SUCCESS in 9m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a5ba598d8fcb423aa849b1651dd04fa1 : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/882c2475d1614f36a3a431a3f4b574ce : SUCCESS in 59m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27367377fca74dd5a0a0432181f479f1 : SUCCESS in 45m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9241c294252c4109acdea09a48c2be73 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7756cc0c5551492f9d1469bf6dc94592 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b9a1d418e26241e2a76ae03d3de32242 : SUCCESS in 15m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ff87e76eb964778ab33e390eff84601 : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/994e8fb91f374686abbf3c13bedceadd : SUCCESS in 2h 37m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b85d9172c2d43b396edbf2dc7f763b0 : SUCCESS in 2h 11m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ea621b3b62684b38a6aafc4dde84f0e2 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00bf0f3ec8b04d44bf9fb87fcb9e83bb : SUCCESS in 13m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f11f82204f63408eb6c7ea080bebefc2 : SUCCESS in 25m 35s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd1a98a3833845369862558e90d94a98 : FAILURE in 21m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98abe3e8849945c8b1d5b25413f209c0 : SUCCESS in 26m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e29f3c95c6724ba98dd513a29f45526d : SUCCESS in 10m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dc4f74760c6459aa4ff6f58687afc12 : SUCCESS in 11m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0edf349ad7e74e54904e9bb6b9647e53 : SUCCESS in 18m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d33d25335fc4a0498c4c6f8b94994c9 : SUCCESS in 14m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/976aa36bee8b4e35b43d125521474a90 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b194dd4b38349289da34e061163e06a : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57ce324f6a5a46c5afeded405eb8647a : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"26a93f569dfbaa09d22e880b93994d31aed24007","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 20:56:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3400e7cdcc2213367923cac024b24f257b7344c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 23:25:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d005349b3dd74373bb1043996382a154\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435e00adbbe5466096a0078098be6ec6 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f990644dec2f49b0b3875addd7d5cda4 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bbb687dd808e45a7adedabd17e64c344 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d3259991e884d7a8b3990aa516f832c : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d40d13200bd1426fb2c319682af14f3a : SUCCESS in 11m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/60ba21a9a8b1407aaca68a6969edf687 : SUCCESS in 57m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/00cd501921614fb38b99e5a2a28c9082 : SUCCESS in 59m 34s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adcfb9438b514292a185c80e88c7b6b4 : SUCCESS in 58m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/03cddef2b4ac4b76adf2f818a882ba23 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d8aa8535572e465e97a75f6d06098441 : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/60e813b5bb3340c2956b13c1699ea60b : SUCCESS in 15m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3b14c4f32b448248c8fb7ea4862b92f : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5756626635774f929d9abb5d60a8a5ae : SUCCESS in 2h 23m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c51c42fdf9c4c4cbbf46c683e5dad0a : SUCCESS in 2h 16m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f0113c6d973468b9c7ce065f3f852a5 : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c088f9cb54664f27b7175987631e1409 : SUCCESS in 15m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8b97f39a4d2a4a1f8c1e5810eb17d8bf : SUCCESS in 28m 56s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6a7c1cfcc204c6c8512b0bc78a7ef57 : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ae93de2bb8d4eb29c31ecd12b60e77b : SUCCESS in 23m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/309bdd9112354330ac5b055d818a0b68 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2639d25517b545c49c382f60afea1c55 : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c465abf074db48fa8c33e5d4677af053 : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b2542c488d846f985a0bbf1adabfb12 : SUCCESS in 13m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af9b763e5c245f2aa7c715cd66d4be2 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ab80a1f2a474831883a3d49baf80049 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/919407b47542467c9258f5f7e22284a2 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"129e5ae685e84564df09999aa502395f50d1b721","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 13:26:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"71e76184a83f6d5f5a85507cd8c0d47685849aa6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-04 12:23:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cf4ff7a7080ad1e7421e92df613cccab78510702","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-05 08:17:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c7e33081b6b9e7644564fe40dff9f222aef7cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 10:16:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95f814788f19436fa2ddda490ed6cebe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64f2bfc93a4640bd845f8c48737cb315 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc88c9d6f27940a6ab7271b6ec7fda92 : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a1810255c3648f7bf8c46cce299b2f2 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e5033af436f4fe8bc52aa310ab6cc29 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45da861573c84d8cbb3d5def04c427e5 : SUCCESS in 10m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/343c2a39862a4e4d833b329365e37e71 : FAILURE in 18m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c11f5dcbb52436997dcd1ab5c132009 : FAILURE in 19m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6735cb906b649f182c8edf13d91658f : FAILURE in 15m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/37be08d161f94cc68e350ef126d6c14a : FAILURE in 21m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/031a124988794440b7247e79015c5f6c : SUCCESS in 16m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1b435d9950d04b92b8976b14c14adbec : SUCCESS in 19m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8430e56604ab40de84799dbac04eca37 : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef519b9c57ef49cc86ab89f006bbce27 : FAILURE in 1h 51m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aedf43477ded43efbc41179cfc033520 : FAILURE in 1h 31m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f18216a7eecf444a9b5c6c06d68a3273 : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/330023ec4893452ea7ae5d3b01fc5b3b : SUCCESS in 14m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/366869c8076b46e19348a149cafd970a : SUCCESS in 32m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8971608c3924094ae96f277279431c1 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/891b7808d63241f3af43c724540fb366 : SUCCESS in 25m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a58aee7fe88f462eb8ccbfac5e665765 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/332032b3fbd042a2828d445a17e8fdbb : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6215730c4ce94c2a83c5a084c691ec98 : SUCCESS in 20m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfc0249e12eb4b8586c044c56adbf9ae : SUCCESS in 15m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc067dcee21f41828205256713e97353 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fca483d4b4ff46668fc2be21f6fa811a : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d5f6fb0001d4d199ef4068f687a6281 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2b0c1f49b445ae5b5c6c50946ec28eaeffc5279e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-06 10:06:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83c4b3c691ae8424490a9ec00ba9fcc813456aa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-06 10:09:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5a3d212c1c53fb181d75fbf345abf375381a9fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 12:33:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae214cbc14d4cd88057e20a15e36b04\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e79f0b6e644eb6be4f85df5bbffbb3 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddad07d26193417d835567e7c48878d5 : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/771a56bc73484539ad0635a627101314 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76cce770e8494ef397a33e5d2acdab9c : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5423acb71a6f4ff4a25086b67840bf29 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e484df4e46164dd2917d124681b0a359 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e30f6421dec6481eb9cc25831ad85d6c : SUCCESS in 1h 06m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b71fdb6acb341d9a5e568d9b914e4e3 : SUCCESS in 43m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ecc7d804e1334becabe5c306a30a9a21 : SUCCESS in 1h 14m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c76bede8cc944213ae025e2538f93929 : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4fb6ab8237814b28baa214631bf72c49 : SUCCESS in 22m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9a804e2a86f452e99a4e355c489a752 : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fa6a48a7d8fb4de7b17462147a623ce4 : SUCCESS in 2h 19m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ac651ded391469eab233439731a4a9f : SUCCESS in 2h 20m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/94b001c0069c4e4998def732ce8b843d : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31fa14551b0242978ba4d2fa9af8b742 : SUCCESS in 12m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/31accedf821c486da5353c9d3e51c2b8 : SUCCESS in 38m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9060e551cf114c28805d27fb8af7ccd5 : SUCCESS in 29m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f6125d2e8924f538ccefe9123ae3d84 : SUCCESS in 29m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c81452ad12f6473197bbabfcc9ed9d73 : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f0f8ddf041245dfbc24fd9dc878851a : SUCCESS in 13m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ac97a18d8051484884c33aa5c6c7293e : SUCCESS in 14m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b784a9f856247fb93269150055fc737 : SUCCESS in 16m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9482bf5cda04e819e3a440069ee2f53 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/194417603dd24a4aabab62458ba2551a : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b0f95eb7a2c42e79d765f0f6f204544 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"13d8e38663a81da560bc642dbefcb900aa5b1f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 12:33:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4bf746d9f00c925ecb922fb1efd166b3d2407e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:31:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84fcba68cf74424af5c19ef9cf4f0b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969056ba181d43ddaca2b824d28a98af : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfd8f1476e7e4ad0a3031f791a777efc : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bab8335958c34612898088c2c94af6c7 : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e15f4f51879149e5a2c1f3bdb03ef3e8 : SUCCESS in 10m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/72a3708e57734e09ae3eb658ebafbffe : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b50b4bd74e4748829734ff08068eca50 : SUCCESS in 54m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0c77c2b80de140378d2d5455ecdb852b : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/91cf0a46c63a49a1bd46d4ceb7061be9 : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c36b3a44b2264efaa17acb2040aebe1c : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8ae86842adf47d99d26a4e791b225c0 : SUCCESS in 2h 52m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a146481427b4d8ab2036559ae5af3f4 : SUCCESS in 2h 27m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bad487ef46d445c2a372382f406be4ff : SUCCESS in 18m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e209f3d067d4ef5a78d4ef90340d2d1 : SUCCESS in 14m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8697b909d204d768e7347d932270615 : SUCCESS in 33m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6763d92450749a79c81276fb7398c31 : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/96d15cdb990f4c5f89b8af773b3b649f : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efa4032b7c7e41dfba0e8fa63567ac56 : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b19cbd8200548d8913ffad0a6c76578 : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23480b21261a4fbfab6ccd4abb555c23 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0658f400579e678481a4dfb749178235870d9dd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"263b3d4b59ac8bc414e1fc04704c138eb811ba1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:32:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5511b8eeada94a1283251d5f0d974b02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f60dcefaa84dcea16c8e55aa4fa837 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"35405d0a2da0b83e57bfbe83a8800647767b0ec9","revisions":{"860cfc69eccae11276293ed8bc48733dfcd71815":{"kind":"REWORK","_number":1,"created":"2023-12-01 12:00:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/79/902379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/902379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/902379/1"}}},"commit":{"parents":[{"commit":"c418bb67ca25307fe5052accc4e567d664b0fe0a","subject":"Merge \"Install docker Python package in kolla venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c418bb67ca25307fe5052accc4e567d664b0fe0a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-12-01 11:57:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-12-01 12:00:06.000000000","tz":0},"subject":"Remove unused kolla bootstrap variables","message":"Remove unused kolla bootstrap variables\n\nWe no longer use these variables since we switched to the kolla ansible\ncollection.\n\nTrivialFix\n\nChange-Id: I50238dd694edc7ee02be4bfdc56ebf3a603e05d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/860cfc69eccae11276293ed8bc48733dfcd71815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/860cfc69eccae11276293ed8bc48733dfcd71815"}]},"branch":"refs/heads/master"},"3068f5a69f5a61efe00f3f42de4dfed043e98bc2":{"kind":"REWORK","_number":2,"created":"2023-12-01 12:07:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/79/902379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/902379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/902379/2"}}},"commit":{"parents":[{"commit":"c418bb67ca25307fe5052accc4e567d664b0fe0a","subject":"Merge \"Install docker Python package in kolla venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c418bb67ca25307fe5052accc4e567d664b0fe0a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-12-01 11:57:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-12-01 12:07:03.000000000","tz":0},"subject":"Remove unused kolla bootstrap variables","message":"Remove unused kolla bootstrap variables\n\nWe no longer use these variables since we switched to the kolla ansible\ncollection.\n\nTrivialFix\n\nChange-Id: I50238dd694edc7ee02be4bfdc56ebf3a603e05d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3068f5a69f5a61efe00f3f42de4dfed043e98bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3068f5a69f5a61efe00f3f42de4dfed043e98bc2"}]},"branch":"refs/heads/master"},"35405d0a2da0b83e57bfbe83a8800647767b0ec9":{"kind":"REWORK","_number":3,"created":"2023-12-01 15:24:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/902379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/902379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/902379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/902379/3"}}},"commit":{"parents":[{"commit":"8da5472ea84729334b4911080b00cf8929f3111e","subject":"Remove kolla_enable_host_ntp variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da5472ea84729334b4911080b00cf8929f3111e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-12-01 11:57:19.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 15:23:20.000000000","tz":60},"subject":"Remove unused kolla bootstrap-servers variables","message":"Remove unused kolla bootstrap-servers variables\n\nWe no longer use these variables since we switched to the kolla Ansible\ncollection.\n\nTrivialFix\n\nChange-Id: I50238dd694edc7ee02be4bfdc56ebf3a603e05d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35405d0a2da0b83e57bfbe83a8800647767b0ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35405d0a2da0b83e57bfbe83a8800647767b0ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
