)]}'
{"id":"openstack%2Fkayobe~902380","triplet_id":"openstack%2Fkayobe~master~I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-01 17:40:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","subject":"Remove kolla_enable_host_ntp variable","status":"MERGED","created":"2023-12-01 12:02:41.000000000","updated":"2023-12-01 17:41:25.000000000","submitted":"2023-12-01 17:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902380","meta_rev_id":"35e96982d74ee5a11b259f3bd879836eb6803ecb","_number":902380,"virtual_id_number":902380,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 17:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-01 14:27:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-12-01 14:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-01 14:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 14:27:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-01 14:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-01 14:45:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d5388ef8c1e495e9a1fd4f8d35cdf77925500334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 12:02:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"046477de128b871a18d7293adcacb7666b08b334","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-01 14:27:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f1771560863cca50fe2ab1b6890a5e797f5c99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:35:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f988deead80d45af8351a033938947ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2af9b98d83aa4f95b2a23e51438eb271 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a26d01480f4f6a824550fa007acb4f : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/08a85ede5fbc495c87deb7fa5218cf36 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/932973711ed149e2ab2c9e0334b1bce5 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/524cdf0bdac94805ba548cc8f9790c2c : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a68139b518294471b116a5e474819947 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/480188cbcd8f4b8b8cef63e00a230af9 : SUCCESS in 56m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f75675694e934596b50eef403e62c172 : SUCCESS in 51m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f38db1e6fa2a48aba635a45d1b36836a : SUCCESS in 53m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f972853e70db45d8af7535aa3ab6237c : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/458f37cb61074b5fa46e651553fa536e : SUCCESS in 16m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04efdc1e336341288102f64cd4db08a8 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/944540ab2ee34fb78e4ddd9f748f1047 : SUCCESS in 2h 12m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f9f23b083b04db4a6499bd7a4b3c254 : SUCCESS in 2h 22m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8d267edcc5b14f9aaf8ca5e5a8e1abfb : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c717482f0f454bfb98999c9bbfa880cc : SUCCESS in 15m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/55e508ad242a4c729414abf55330311a : SUCCESS in 27m 13s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a81e3bbd8684adead7289af9a63fbff : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9311ee4087047acb7cb2d2c8a7cbf1c : SUCCESS in 20m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b2c427d2d5af40a49fbe54bc3c19afe3 : SUCCESS in 15m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c6257afa644937b2f42605da3db276 : SUCCESS in 9m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51267850f3ea405ea1889c5e57a8912c : SUCCESS in 18m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39a461a4919f408b885806885af083de : SUCCESS in 15m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5b21a1e195e442689c0d971c84505a3 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8fed303c49f4ca4856c22bacccdf102 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/401ee6c555574c23824f6964659438fb : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5847dfd470d2177204fcd2a108bd70d973a294f9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-01 14:45:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"488f0e67edea00dbb681e271a48fc5ad3846edf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44ca4d39528c6216c3601e93c52395a77e92fce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 17:40:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c2944f5fd749119d1ace882b34916d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9acbd0ef3f45afb03c0bd32ec2a256 : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da66c556f64640eb9037b968fd5ec3d8 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd198e4e8ca64a79bdab5b2c6eae7f07 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22dfb9fe03f843e3bdcd45aa4f42bbad : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d1233ece8bbf4394b0fa187ae284cc45 : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9a5e6ac2cb54e11a6a6b8c1f08622be : SUCCESS in 47m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/85055256a6de4af09d4498a5ebb56919 : SUCCESS in 1h 19m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c9702d83d2a4c1581f3293363f911e6 : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd84a1f95e0c4e43b78c8c5bccf2ff81 : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/800a62e63eda466eada52ac58e37087c : SUCCESS in 2h 47m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5800ed762cad4a999cb0eac184263b2a : SUCCESS in 2h 11m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9e3c20b4d5ed4092a434964864632e98 : SUCCESS in 18m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b3a8586b9e44378a453f2c51553b0c : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b36d3256cce2442b83f3e200cff7d039 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48ec84303e9846e29e30ba9010b94e66 : SUCCESS in 22m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3717e030af1448fe9a815b4ab76c42c1 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ead814f318a64f3fa27d4d87be36e4fa : SUCCESS in 9m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e686ffbc7f5946e69390a6e9f50664a2 : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f861b8057c564f7398867e9af638933c : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c6b03c71151ee2568a666f2819a2f227b53f4d41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 17:40:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35e96982d74ee5a11b259f3bd879836eb6803ecb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 17:41:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75fea30e1d7d48d9ace26339b427ce78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f912b0f65154d8b94fbed9ce92c45d7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8da5472ea84729334b4911080b00cf8929f3111e","revisions":{"8da5472ea84729334b4911080b00cf8929f3111e":{"kind":"REWORK","_number":1,"created":"2023-12-01 12:02:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/902380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/902380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/902380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/902380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/902380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/902380/1"}}},"commit":{"parents":[{"commit":"c418bb67ca25307fe5052accc4e567d664b0fe0a","subject":"Merge \"Install docker Python package in kolla venv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c418bb67ca25307fe5052accc4e567d664b0fe0a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 12:02:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 12:02:19.000000000","tz":60},"subject":"Remove kolla_enable_host_ntp variable","message":"Remove kolla_enable_host_ntp variable\n\nThis variable was configuring enable_host_ntp in kolla-ansible, which\nwas removed in the Xena release.\n\nChange-Id: I737598d3bbe40b933b4b727eccc3b2a76ed97cdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da5472ea84729334b4911080b00cf8929f3111e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da5472ea84729334b4911080b00cf8929f3111e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
