)]}'
{"id":"openstack%2Fkayobe~902422","triplet_id":"openstack%2Fkayobe~master~I749a1ac30fa15cf7faa456fcb90c4ddbbef77cc9","project":"openstack/kayobe","branch":"master","topic":"kayobe-bobcat","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-06 05:22:45.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-12-06 10:59:46.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-06 10:59:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I749a1ac30fa15cf7faa456fcb90c4ddbbef77cc9","subject":"Use OpenStack 2023.2 release","status":"MERGED","created":"2023-12-01 14:59:45.000000000","updated":"2023-12-06 11:04:57.000000000","submitted":"2023-12-06 10:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902422-kayobe-bobcat","meta_rev_id":"4711a7d3a2b5de85144d083c6170f0c63c75303d","_number":902422,"virtual_id_number":902422,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 10:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-12-01 15:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-12-01 17:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-01 17:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-01 15:11:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-01 17:03:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-12-01 17:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95f4c3bf09e57ebafe845c1d183a9d1271587bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 14:59:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1d9c80e1d46bf60b3aac071cea8f0cf08aa0283","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-01 15:11:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8537cae93fdb1a6eeba853f878334d3b7681d7c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-12-01 17:03:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11e9810e33d27769ac412eef339a7fa5b05ee40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 17:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad5c873a9a0244958a4c3294e3082eb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9277d3ce7a9d43f2ba8e0c1d7ab450b7 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c70e016b984d40dda237e97fba55e09a : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d10448312cc4489a155974bf50f42c0 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ddb34a7d77a4f28974b14e0f2a49862 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8902fdbe9414be4917bc7cc9a22605c : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50ba61984db14d319705a0bf589b688c : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b600fb42de2f420d9e1ec632b6a69663 : FAILURE in 12m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/05cb1ddcb87543a8823c98e8ee4a23c3 : FAILURE in 14m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e81e7b4cb45a43d9aae569a48e9959f5 : FAILURE in 11m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/133e555e806549b6bcf379c37c20907d : FAILURE in 17m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8cd16713a7de4e2eaf39431f1f420a2f : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f89fe1c360c048078347d8a5f3640b62 : SUCCESS in 16m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/536d563dc4e846099f242243173471e5 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a517bf93c67e471d8328072d8a1a5bb4 : FAILURE in 1h 47m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b288ce8d9774d75b12511a148ae101c : FAILURE in 1h 34m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/362f87a53b6f4b058edab1a37ee7616b : FAILURE in 13m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05001c204f3c42bbbde1a51d61bc39f1 : FAILURE in 10m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5ec2687192a5498c9b84b9107295e72b : FAILURE in 10m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7f4517dc9eb24d6bbfed2dce681b28ec : FAILURE in 24m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38c0e9216f6e407ba156f03518ba788d : FAILURE in 20m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab103ce86a9e4cc982eb858f7ec71eef : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/924aaad107264999b890bd23f1001098 : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a82c4debb3c4ce3830ccf4c1224643d : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa1e0210106b4f4a86f023b9c2c6a827 : SUCCESS in 14m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03e14c8dd99846faaee919191d347b5e : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b497e78fde4f4dbf912b006823285d43 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04ee741aaf7c483292b924a46c87b8a3 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9a15cd12d4ac81ec2735ea7c898508d572fcef3b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-02 08:14:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dd0b7c96f1ffaf9843a38d3aef7edeb8255b76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 10:40:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20a0985f63694080b5fbf8039b6f202b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9adbcb0dd0043188456428ae7cfe6d1 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf93995847024092b421cb3a549d7c9d : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df99b376958a48ba9e041ea99ea8ad59 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b33843876da74c0e999fcedaa6b18540 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/324d32d2158946948d7df6a5b9d1ac04 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a01430e72c14d9fa26d2c1bd5db607b : SUCCESS in 9m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e8962e1efcfb42a8a96d10c58affba1e : SUCCESS in 51m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/09a4a622990a48c3b6d32918a336e051 : SUCCESS in 55m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f95471d14d842e7b2e2deb4f141c33b : SUCCESS in 52m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4412d797540d4a47a4009f0c92cc8a9c : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/37f21c9c50e04f5b97f9b584c3bf94f1 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/82d8f040505843ad896518d482430c1b : SUCCESS in 17m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/733faaaf7da4447abbdac1e849c01279 : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/92172e7213e24a389e8be2cf8a382b61 : SUCCESS in 2h 20m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ada8d60d558944429a55403f85920f37 : SUCCESS in 1h 54m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3f2e79d8c7cc4972a16e5796434eeedf : FAILURE in 11m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02b9b13a6d194f92afb6f72139ebcaf7 : SUCCESS in 14m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/11486f2f48174ded9d5b5867009ba644 : FAILURE in 13m 07s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/642ddeddd604425fbb62e4b66e3a74d8 : FAILURE in 25m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e97e687c33a412094d1c8504e14298a : SUCCESS in 23m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c493d5e9540a4073ba7e4adc4bd070e0 : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ea98fa236914b10adcbbd5be4a42677 : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5966d8142d4e4287967703f939166911 : SUCCESS in 19m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc03a02781164b6f98c506f97037d604 : SUCCESS in 14m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b28397166cde45a09f62b342d4dac701 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94c275f742994c27b4c113c777f0fd2f : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ff3f467898b462dbedb98f29f3941eb : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"deade1978d1fa559a114699bb21bf1fe9fce4fa5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-02 14:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b293c3cad5bc9fe7a7e4e59f1a692673bb32845c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-02 14:42:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d41df60a3757c1cbb454c7666d5838ec18ecc760","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-05 06:49:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"1f2dc13fe10d03b8ef131ece6f4b686f2daaaa69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-05 06:50:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91b5a468c9aea39ce24be0f89b3f35372b0a48ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 08:50:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/670a3ab09d524f3f8cf88e46a29e73f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de07c2f5a54a40f58a4c6cb750c48ff1 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c2dbb455754091a3536d2e64ccf5f4 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ceb32c122cc4c7da2e497da7971e184 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/328e9a2f74374b41bfa397e66e2e4096 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22d4c5ea15724e52b31f72b735ce4790 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09c34714567f4fa7b782107582f6137e : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ef8bc392897848b4b464943c05a07f3b : FAILURE in 20m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8f7e181bc0554beeadceed2b973dd851 : FAILURE in 16m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e994d0b3709d4d9dbcc53375d5f1ea1f : FAILURE in 13m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d949c9a7000e465cb6c43f1ce7766f3d : FAILURE in 22m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/09b0111eaab44eafb9bbb8a409dcacd9 : FAILURE in 15m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6be4992c86e64e7ab0466894e94e3b78 : SUCCESS in 15m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3145da279faf440487c52fb3a73a7d54 : SUCCESS in 12m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/924d40d1ea724ff7aa362a10721a5762 : FAILURE in 1h 54m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6c93e12c7754d2c9172be03bca7b473 : FAILURE in 1h 16m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8d074622e566441d97ccf7d5bde5d6bb : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7bbba5597d4f8f9babf3a9f8e0cb38 : SUCCESS in 12m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/296c6b0eb1f448a2b6186ab1dfc28eb8 : SUCCESS in 32m 08s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/918059ea5cdc47868b741090936d3ddb : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/856aa660b1ae4cb09104edb7c8947959 : SUCCESS in 21m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59ec9694bad54beba2d79886d3aafd79 : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc9626463560405b8a2995064bfaa019 : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/54afdfb24c3b460e8ee16439c6c18164 : SUCCESS in 18m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0101efa89e344fbb4a068a01f8af5d5 : SUCCESS in 14m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc69440e2eee4b5fbcaeadc44ce3d1f7 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4993b0e78a6e4dfeacc52a47f1b4d7d2 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/183f79d4d13e4a309a710aa0339e25a2 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6e895cb2ea04bbb295909099f523e2ca85289ad1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-06 05:22:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecb6ffabb406f474f81a5743b0658f4f7e726544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:11:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d5b32a76b64886add1640b5dc2f683\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1843bbcf032e46d393e5e4af8ae63934 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0910c8b03d8842f9824b2b0e4af26a94 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/388c001bcdd34a30848f5b2fbd176db8 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5434d4a487a94939b64cef5a7d31bfc7 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16119986ed1247cf9d61906ecf8b5aa5 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/523a61baa7424d3d8a7b7157d1908e60 : SUCCESS in 10m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/980781a0597b42a0b4d3801d9873a25c : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f592dde7865b43268ec55bffe1fdb421 : SUCCESS in 45m 54s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560e8b312ddf428c83166bcf017283c2 : SUCCESS in 41m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3380712d19ea4753a6aec167a28d45b5 : SUCCESS in 1h 15m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ca91aabf9314495cac2ec3d89d3f867e : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0823b9219939404cb430c4499e4a3053 : SUCCESS in 14m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/063b4700bc654433aa94f7cebcbb4054 : SUCCESS in 11m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9eb33f65a58544fa87543265188510e2 : SUCCESS in 2h 43m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aab67cc98f2547db8493d34c620cb16b : SUCCESS in 2h 11m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3be23467cd2048bbb576de1b769e2835 : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3acd80e5c014bdcbce2d0c8952bf5d8 : SUCCESS in 11m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1a77b4cbc0094cd5bcb16a814ffe29c9 : SUCCESS in 32m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6982069c65124e5fae52d35ca049840f : SUCCESS in 30m 19s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f25f4d7b85d46ba933f983a258a5509 : SUCCESS in 20m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e85f2faf57543c387afb6bb750d6b21 : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a759b3090d4858930af339da509801 : SUCCESS in 9m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ccb4380e2e044e69bdb8154a3cc1bf7f : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/048bf8e8652046eda00a69366c1140a2 : SUCCESS in 15m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7975b6a41e5f474cba75a8e5d096265c : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a61638be1d2f4adf9ff8e2ba9d34d0f3 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f0ac0a5748146e6b4cc7de2c7baf3d7 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"82615f21e9c9bbcd9558b78891f391617e8b1376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8091230d3a557b14c05743ae29b1402943251a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:59:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240e763eace94e0ab83ce88fea2a1920\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736a230f636b41e195ebc2ad239b6de4 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35b52ca01da348f7b445a08672af1542 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de8f45d13c9a4b07ad066dffcea7b1a7 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b7c851178754f3c97a3b8ca13b9aede : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ecc1a446230d473d991fcf1fc39da785 : SUCCESS in 11m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2e63724f42684554bda8e36bbf6d6842 : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a72da06f1d145dfb00b8c14e44dc2b7 : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd98d05d06744491855fef284e6ac57c : SUCCESS in 1h 10m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/67af6ab7bf864bc485873dcc6f0ddd6e : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea9efeaf3b6c44f5b7daf7b6ffaf3102 : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8f66bfaed6344318cd74e20fe0dce3d : SUCCESS in 2h 42m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68cedbb64e614c89994db72ec9e56e0f : SUCCESS in 2h 08m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1228724e845043299a574a08dbd8f21b : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0441b960eba04629be824e1d3d7a3517 : SUCCESS in 14m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0cd75950a02f48f885169f1d0f7e346b : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2b4afa1aba0457683eea4475bfddde9 : SUCCESS in 23m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ed822a3c8724a8a84a39208664ea361 : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31626098201b40e8a656d51176d5e1b4 : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05450da2c64f43adb6ea117d5d124862 : SUCCESS in 16m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1f16a4ed58c488faa0293e83b53ab7f : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b4fe5210ffc3b131174f74d40556ab3c2328c1b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:59:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4711a7d3a2b5de85144d083c6170f0c63c75303d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:04:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff0cade90a046f8ac8496c45e1fda29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e0199f437240eeb2fadf36fcf449cd : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4df195d4d36413ebc57ac8ca6c33144 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99f5b64956b795d4d2ac3f04c51db5a108da3560","revisions":{"99f5b64956b795d4d2ac3f04c51db5a108da3560":{"kind":"REWORK","_number":1,"created":"2023-12-01 14:59:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/902422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/902422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/902422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/902422/1"}}},"commit":{"parents":[{"commit":"bc24c9ece1ff76a6bc412da8465fd2e3de976676","subject":"Merge \"Remove libgcrypt package update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc24c9ece1ff76a6bc412da8465fd2e3de976676"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 14:58:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 14:58:47.000000000","tz":60},"subject":"Use OpenStack 2023.2 release","message":"Use OpenStack 2023.2 release\n\nSwitch all dependencies to use the OpenStack 2023.2 release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/2023.2 branch has been cut and RC1 released.\n\nChange-Id: I749a1ac30fa15cf7faa456fcb90c4ddbbef77cc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99f5b64956b795d4d2ac3f04c51db5a108da3560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99f5b64956b795d4d2ac3f04c51db5a108da3560"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
