)]}'
{"id":"openstack%2Fkayobe~902449","triplet_id":"openstack%2Fkayobe~master~I50d47ceeeae803545d78a40bbeec1aaa2f5112b1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-05-01 09:08:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-01 11:47:44.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-05-01 09:18:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-01 11:47:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50d47ceeeae803545d78a40bbeec1aaa2f5112b1","subject":"Add python3-devel to Kayobe dependencies","status":"MERGED","created":"2023-12-01 21:54:22.000000000","updated":"2024-05-01 11:48:50.000000000","submitted":"2024-05-01 11:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902449","meta_rev_id":"577d64b0082c9643c7017fe31e7b007ed54601b8","_number":902449,"virtual_id_number":902449,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-01 11:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-05-01 09:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-03 13:10:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-05-01 09:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-05-01 09:18:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-05-01 09:08:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 00:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 13:10:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-03 13:10:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-03 13:10:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-03 13:10:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9b93193788011e83661caa55c844ce83eed6154e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-01 21:54:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54fd6e1bf0baf67136e4af9ebe898096bbb84733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 00:02:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d1344e59f1b4911acd0adbb218c7288\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6943ce0602b14d459f7e5ca086a3ea6d : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74b0ef876b54b7b83c320a208c2fdd3 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb0f4fa6fb374804bbbe3d59adb1855b : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0abc9e07cf514a6abfd65b1a3b1aaf48 : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7a6d6f0381f4ef4aefe4048cf969629 : SUCCESS in 11m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/953a882e389f421eb91548ee35a7cf6f : FAILURE in 16m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f57d213b8f6f4b9fad37a0f2e7a3d668 : FAILURE in 23m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c96832a47c4fe89c06d4fc1767ca24 : SUCCESS in 39m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e37a86774a9f4b0d8b7abe89d37f7aec : SUCCESS in 1h 22m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a0e3e47df89141109b64f9ed3d2d02e9 : SUCCESS in 17m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d0beb24ac2ed48b99e427816528400fb : SUCCESS in 16m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25ab31227d344902aef2472901515dfd : SUCCESS in 11m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a96fed0d11c54cda9554dce76dfedb2b : FAILURE in 16m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c58a516e2984daba4bd97fa590a7fe8 : SUCCESS in 2h 07m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/42276076a3374560a6ef89cd512b84d1 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82fccc3ca374e28866a6189f9746213 : SUCCESS in 15m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/42a97212fe564bd8a263098b7fb4be12 : SUCCESS in 34m 04s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/67a89e5022af4f7c823c335b9560150f : FAILURE in 13m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/718c0c5b22ef4f56a9e67c2e5aa3b632 : SUCCESS in 25m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ed407900be843c782c0b491e8856dd4 : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe1d75083c8746c7ac39a4fd12d56bf4 : SUCCESS in 11m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1880d065661c474baa732ff5fa0afa15 : SUCCESS in 18m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aa978efbcdf412da13475df8eec9327 : SUCCESS in 14m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2dd225ca9549429dddf09a16a37f1d : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76c5207b2b674bf2ac58defa34e2c690 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c34261c65e424d6cb21331188f01e538 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6be2c90e2f49523dc6676aa7102e32d9b5aec1a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-02 08:14:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cdfc983b7a54193c2aa208d98cdfbba6a1f8243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 10:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e82dd108f6c4caabfa2b686da087f9c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33518218e96a49779dd76012cb1a0904 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c68a0d72c34b0eb985d1318842e432 : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/94ab31afcba14fa9bd1f6b4a05731397 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c7473dce07c344d48378418499ffe276 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d75be7cf5a745d8b6970a852bad4bbd : SUCCESS in 10m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c403a1756f684e01ad6d6adc45f5bbf2 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4bd17a99cedb47788a4eec80a099f3fd : SUCCESS in 55m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11ce65a7accb48018bc9677356b134cc : SUCCESS in 48m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8e2b3bce34244295b62474b87bf9a7ab : SUCCESS in 59m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4c97ab136d614994a9f868cb66169f60 : SUCCESS in 15m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/26377b7c45cf4a7a9afbc4f4aba9bb7d : SUCCESS in 16m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e50fa2dda13c4799b7e56bc285e011db : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3fa30a3923b644c096d06aed6251c864 : SUCCESS in 2h 28m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b62aeee65d9b4b628e26b8cc353e1561 : SUCCESS in 2h 14m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f51d8447e2d24f9793ee96664884e1f9 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9e07ef5f0fb4ca29a29c35294b6d8d2 : SUCCESS in 15m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7bc7b503320742eea7c5cfc886efbe41 : SUCCESS in 37m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/801532a260bf4cf3b4a11d363190dda3 : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/551ed17ce4cd4316881e064760904f5f : SUCCESS in 23m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afc9e935a5e8476cb127f565c052733a : SUCCESS in 15m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1adbd91fa49a42ab806b4d78817e2f0a : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2a74e3d62a7444ba7bcc100f3ca9f15 : SUCCESS in 18m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3a248f64a684790a4e0a9653488eb06 : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b05c2d2b0e42a2a1cd92d34c9bcaf9 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/850fe43fe4b94d9492ee642144cec965 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84aa9a29af1c43409c000b33988192cb : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"26f3ca81ea05957b5e211518b0540cec2389e182","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 13:10:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19fa5826a6f8f9f8663b209319b23dbb325a8991","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 09:08:05.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2721a71ca25a75e19a10c43d19e84ebbbfd4b555","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-05-01 09:18:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"920919a6c9ff3ca31d794f69130f5299b7648c11","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-05-01 09:18:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e1fcb5a055d67d2a27e957fc66c02d250b2c864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 09:18:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d90d0c9bc5bfb0be83102dbbf8b5f25c37fbd5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 11:47:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf394005c045406b934d98cf0cac75e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0e4ac3f992445a84af1ef911eaa748 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/56b9384e5c4a42bcb6670a3819e37edf : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/502928e65773437d9ded62ac9b821e62 : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df564d11687244a9b2015bcec5cdf4ec : SUCCESS in 9m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ce120866bd0d4f768bbebf7db636045a : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dda9a379d0a483b9292f27ea882050a : SUCCESS in 48m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8df6c4e8fa12491ebdc11f09179833ed : SUCCESS in 1h 12m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6f1bb5e0a7c34b4d8aa86a1fc67a83af : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8e6b54f7b641339988cfe5a07a295b : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a8aa8846ea44b8c8e9e83e6a7c73a8e : SUCCESS in 2h 24m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1971f0741bd4cc68ddd2dc019c1b01f : SUCCESS in 1h 40m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5231a53883a4035a36391796180e7bd : SUCCESS in 19m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9acdd2c380f04302af7486fbda038afc : SUCCESS in 14m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3fc6f96b26542ed9306ad8e5e1fed54 : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/524b3ce43de3443eb025f5eb7e711a1f : SUCCESS in 24m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b4f65fbd519c43228dc748e7923978cd : SUCCESS in 32m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e50c9b062cd14853b2ed72005f0fa447 : SUCCESS in 22m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/60b8a3909d1047ea8756b78cc690d1ed : SUCCESS in 13m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f346210422ad4d098cd3d9fa59df8714 : SUCCESS in 10m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2fdc5b5ec70547f8b4069143789c7e60 : SUCCESS in 18m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45119462f1f34ff7942469afc8c08a3a : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"be3d1f8e9e2b506c888ec7e9dd605861c429a035","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 11:47:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"577d64b0082c9643c7017fe31e7b007ed54601b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 11:48:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021737f70bcf418391f3d6c7eca7e918\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463c8f70310b405e981ad42b566955d1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"611bec8867b72974b6136f8427a080f2222b99fc","revisions":{"611bec8867b72974b6136f8427a080f2222b99fc":{"kind":"REWORK","_number":1,"created":"2023-12-01 21:54:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/902449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/902449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/902449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/902449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/902449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/902449/1"}}},"commit":{"parents":[{"commit":"b3d9b3773c9d0268932658722ab24ee513c73501","subject":"Merge \"Remove kolla_enable_host_ntp variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3d9b3773c9d0268932658722ab24ee513c73501"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 21:52:50.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 21:52:50.000000000","tz":60},"subject":"Add python3-devel to Kayobe dependencies","message":"Add python3-devel to Kayobe dependencies\n\nWhen python3 is missing from the host operating system, the install.sh\nand install-dev.sh scripts fail with the following error:\n\n    Creating virtual environment in /home/centos/kayobe-venv\n    /home/centos/kayobe/dev/functions: line 169: python3: command not found\n\nChange-Id: I50d47ceeeae803545d78a40bbeec1aaa2f5112b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/611bec8867b72974b6136f8427a080f2222b99fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/611bec8867b72974b6136f8427a080f2222b99fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
