)]}'
{"id":"openstack%2Fkayobe~902483","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-07 14:58:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-08 10:57:07.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-08 08:54:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-08 10:57:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","subject":"Remove kolla_enable_host_ntp variable","status":"MERGED","created":"2023-12-02 08:16:49.000000000","updated":"2023-12-08 10:58:21.000000000","submitted":"2023-12-08 10:57:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902483","meta_rev_id":"f2a0320a318e52c7f2bda2f5bfc895358b46aea5","_number":902483,"virtual_id_number":902483,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 10:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-07 14:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-03 12:47:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-12-08 08:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-08 08:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 10:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-03 12:47:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-03 13:00:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-03 13:03:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-03 13:03:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c082b59c572e4a5eb6810a02949a952bfaca43fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-02 08:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a9266624b0d10404dfaeab98e2eca4101602551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 10:13:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbeebecbf72474e897d0d70ba36b75a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2689cd703c9941dba94cf86ed2b8482a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905de18020f745a4ab5f7e3bb0096288 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef2a2ef920a48dea336c9f42cb6bbb9 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f94191d45a40598427bcff836d88b0 : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c828e7558bb146ef9ac971978ebb60c5 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b185143912794cc1b563f0aba3caf72b : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/143f508eab2940c2b65221ad4739d396 : SUCCESS in 13m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/70c94c11c204493d9fe14d5d517f0ae6 : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d295f2742cbb4ed8b423a6710000d56e : SUCCESS in 1h 07m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6db82b48195d43d393fdcbddfa483a2f : SUCCESS in 52m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8426b627ac734c1389c30a3a17bd90d5 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/85728832fbf946648425faa421379086 : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cca86e01324e4419a2a5f270d6691da2 : SUCCESS in 17m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57bb27b7a6ae4e35a0f5a08912db73ab : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d363bc99086c42aca42908c5a2520774 : SUCCESS in 1h 51m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a03d45655b9c454f956f964721870028 : SUCCESS in 1h 32m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ee1bb32be3444e53978b421b5381251a : SUCCESS in 18m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3fedc8077b42cd95507d89ed6b035a : SUCCESS in 13m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/20e42ec73f76453482f3689038bd2690 : SUCCESS in 30m 49s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cd79ea31d1ca420da13d517e1d87056c : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0548b93e6f14d93b6083b5715809376 : SUCCESS in 21m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/761e86aa8c4e4e04b71a34cba02df3f1 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/197455cebe4545e3bacb10422b006dc9 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4915f7ef79574298a9fe1e2d293da123 : SUCCESS in 20m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9eed0658e82a41209683220c9bb32229 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"36420438edd93db47fc1555c9efb7c9708fd7f76","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 12:47:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d58f3302364378ecd959c7c4b0cdf8b6ae316dc3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-07 14:58:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e185f68538436f4a8c1be6a1bb1ae23347157b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-12-08 08:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53a1d86bad0f8966e2ad6b182463e58486ec73a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 08:55:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f45de51314b200875fff8e284a87ffdbc316c79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:57:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018bd0d7187c46bab7ff858c93ac9a80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2774bee8f63a4ed2aab430d5085e5b37 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff609190730c49a599e8f4d07a90bdcb : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049a2e96489e46d5b09c35bff053b229 : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba41fddcb345488e96e8f77749ae7260 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/526f36ca4c8a4b87a7d2e02a6189f7ab : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ef7040d535d4a5d982f4192a4c298c1 : SUCCESS in 14m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a2b7865b49b448a58c0bc7ffdcc4a5fb : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79be36907e484e6b98c0d457e5b0b8b6 : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c64a6f4789d44256833660aa84ace1b9 : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/51babdbd651b4b4885b65f2b9b4f0616 : SUCCESS in 19m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a55cc8dff8dc48069dad9c818ecf77f9 : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ccc8571ce714b2e9e103c9fc730181b : SUCCESS in 1h 56m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1b3b8f7a5c145949259f910db42e2ca : SUCCESS in 1h 30m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/db193c0216f34953a00be42adae5f325 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca3598b4b4454d7b87dce92811382962 : SUCCESS in 14m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0f99b311e4544bdc9f35a1a5cea37c28 : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11f8c7e676ae48769162f62053c57c83 : SUCCESS in 28m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd27394a66fc4efe8a485ca8824873bb : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae88662ded26437f9c65945f9c4c7b92 : SUCCESS in 11m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a95196375a914df2b2bfdf85c82e166c : SUCCESS in 15m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e5453528b548a59748c5869402e451 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1168efaa52aa485cf86c3117d336fb2444f3ff7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:57:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2a0320a318e52c7f2bda2f5bfc895358b46aea5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 10:58:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e1f4ff2cb4f489ca3408163bfc54346\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d3808d9a5f4601a3a78a809bba1135 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32d48209394ac069d88012b4500b97cbce7b72d3","revisions":{"32d48209394ac069d88012b4500b97cbce7b72d3":{"kind":"REWORK","_number":1,"created":"2023-12-02 08:16:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/902483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/902483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/902483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/902483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/902483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/902483/1"}}},"commit":{"parents":[{"commit":"2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c","subject":"CI: Test SELinux configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cbe9b7a42fd3c51850f3b1fd0fceed06e89325c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 12:02:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-02 08:16:30.000000000","tz":60},"subject":"Remove kolla_enable_host_ntp variable","message":"Remove kolla_enable_host_ntp variable\n\nThis variable was configuring enable_host_ntp in kolla-ansible, which\nwas removed in the Xena release.\n\nChange-Id: I737598d3bbe40b933b4b727eccc3b2a76ed97cdb\n(cherry picked from commit 8da5472ea84729334b4911080b00cf8929f3111e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32d48209394ac069d88012b4500b97cbce7b72d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32d48209394ac069d88012b4500b97cbce7b72d3"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
