)]}'
{"id":"openstack%2Fkayobe~902494","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-07 14:52:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-07 16:30:19.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-12-07 16:30:19.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-07 16:30:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I737598d3bbe40b933b4b727eccc3b2a76ed97cdb","subject":"Remove kolla_enable_host_ntp variable","status":"MERGED","created":"2023-12-03 12:59:56.000000000","updated":"2023-12-07 16:34:24.000000000","submitted":"2023-12-07 16:30:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902494","meta_rev_id":"aaef26b9d35aaf3acf243b875432d5982a8659cd","_number":902494,"virtual_id_number":902494,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 16:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-07 14:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-03 13:02:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-04 01:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-07 14:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-12-07 14:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-03 12:59:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-12-03 13:02:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-03 13:02:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-03 13:02:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-03 14:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-04 01:43:37.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-12-07 14:09:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"cfc75dc5fedc9b811cc56907453d85b4a00a05f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 12:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29b7080a6ed48f7568483c4493a479c0bf875f70","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-03 13:02:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c6043e9bb911ed06117b3e17e2f4ac2b7fa7379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-03 14:21:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3014438c7bd54ff4a921fb3f1f5ec531\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f211a730754e82985402922058abd0 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29794df0ce9c4c1d857f0e9d6b026ddd : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3904e238ac7f45d29b92467c076c3fc5 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c036d3ecdceb4c3a9d107c30486216c4 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a783ec7d136743dba4f9a662ff37eab7 : SUCCESS in 10m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2b0c5c4d1494310bf132a3d44055dd6 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d55bf66ab6d40f1b82f20bacea7a263 : SUCCESS in 14m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b08fdd7099e145fea7c3eb463b2b85ab : SUCCESS in 16m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/92aa134148af4899af18a0004800f09e : SUCCESS in 1h 14m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d1d99457955f44e893b7f993ffbbe38f : SUCCESS in 50m 16s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4e316a20903e460e9f8bd7f64627d79e : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/86aeb117949d4266be59744c25c13561 : SUCCESS in 52m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46c3ff3fc9784e7691a498892f7265df : SUCCESS in 39m 40s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/573a693655134894a48c05b51f825735 : SUCCESS in 53m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d6f57c662bc94efdaa0b80fd3520e278 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cbdd99b4ab0948159923cd2722ea2ef2 : SUCCESS in 16m 59s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4366bda168af4d1abecce7fab693819a : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b6278280b9944e4939979421459ce74 : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1759c1af60514cc29cc49718a598a0ab : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbf8fff634f84c64bd8c017bf7b9705c : SUCCESS in 16m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/495e8ed9eb314e0bad5cbc58565f77b1 : SUCCESS in 1h 15m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0534ebdc14064f149ec347aebdf5f9e0 : SUCCESS in 58m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/554e15f6b04447c792f815ccacf55b0b : SUCCESS in 16m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/596ee0e26d5c43f3977d419a29833ca5 : SUCCESS in 21m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/249e1c5d004f48b89167b42739395d55 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/154e4520f13b46aab12e0bfa1bcfff63 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44a8bc9cf3fc492a8f6c5cf5dbba162f : SUCCESS in 18m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/34f4811d38ce4f749bb20c82d66ddeb8 : SUCCESS in 14m 50s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/05a36878709547ab8c0f0fa66d2b6833 : SUCCESS in 17m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6e5858a7e2c1418683681d8546597cbd : SUCCESS in 15m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fef68e8dc15848a6a606268518435497 : SUCCESS in 9m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64a3f29b077408d95ba1e758ea0a9fd : SUCCESS in 10m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3477488500984705ab5b4cc9d58bd58a : SUCCESS in 22m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/07403e9fe4584be0b97e089078298dde : SUCCESS in 18m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eac51a4074a14651876c02721b1543e6 : SUCCESS in 12m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de541af3020b48b9bacfa0c6116051ab : SUCCESS in 12m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/911fc63ff5eb47d0a2ed10d9fbaafb01 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3e5f9deb84dce4941a138720ebb0908c6af056b6","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-04 01:43:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a8699330e6644cdec4aa0e6e32bb389c5f559ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-07 14:09:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93893cd63b45136cd584d842725cf91c4311a4ee","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-07 14:52:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddfba2f792493e2dd1b3a1781b084e751d4e826d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 14:53:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d9fa0f1efd8e7ab46341be1135cb9e990043e88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:30:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e23bab1c0d25470195b4a5c50fbee73e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8767ca9f929146d081d84db8faf3e742 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d81f82a75ad547b2ad353cbbca3c0180 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4135c1f210e4975a068b9833738cb58 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b54492ac9343f2babbdc8a54cc2b43 : SUCCESS in 12m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00d1725ec3d949909fa07330b251625d : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/122bf3d19cf6445da1011a9d0c0524b7 : SUCCESS in 14m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66b21d038cd540698bc22874ba66cd51 : SUCCESS in 21m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0519c1dd46984734b0de7034535d6fdb : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8de36c2a9d52476da036bfe1df3acafa : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8673f2d6aa214306983b334d8d7a80ad : SUCCESS in 53m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56059b801dd84049a493cb9e8431c41c : SUCCESS in 57m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51c3f812e91443fda6287c3db3fe8ab0 : SUCCESS in 53m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2c2a73a101154555b3d1c28896f76dc9 : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/749dbd36849e48ceb07e63acb2ba904a : SUCCESS in 22m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9e3431276a4b4576b85a0a88900fc351 : SUCCESS in 21m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77eb8694620a41318357b3e6c7946011 : SUCCESS in 18m 15s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09706abe45e843e18fe069f17993c927 : SUCCESS in 18m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/70f63a087a4b477e8a3698d59c15bfb6 : SUCCESS in 1h 17m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cc0c075966a2466baa8caf1cfd551765 : SUCCESS in 57m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dd95fa5541ff405496ac4c709d412341 : SUCCESS in 22m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0e88941b77ba4d21935a78115020d725 : SUCCESS in 22m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bcdc39bbe67b4450948f142ef48c652f : SUCCESS in 21m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413b9779d1434a3f9170b4f62fd0e016 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4689423ebad7491c86bd79a413d55c1f : SUCCESS in 18m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0489223fd8634454afd0511a73d3fa13 : SUCCESS in 18m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0a2d5a3179ee4146b50917c29d663c8c : SUCCESS in 16m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9e75e22f235489daa013b24a12ce023 : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0041812a5b7a445db63bfe572ff7b648 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b20e7fde4f7492f889b7bf2315a9def : SUCCESS in 14m 36s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e69baa01bb2c4ef0815a8ef2b101a785 : SUCCESS in 23m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/01f0a6cc4aef49e1a73c561612eec68d : SUCCESS in 27m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5b030c96a3a48c59322f6ef53d2e978 : SUCCESS in 21m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7cbec3faf0b4c79b4dabd2171561504 : SUCCESS in 14m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60b2938f23aa4d039b78bafc466a77e8 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"57df9e92c3a7738f7a767744f78ce22643308ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:30:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaef26b9d35aaf3acf243b875432d5982a8659cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaeba94983d3459da2cbed13d4f223d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7bba7561a74f9bb431c8b657a80af0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e309ef0469fd317534ed7ef34a228298a05fe18","revisions":{"2e309ef0469fd317534ed7ef34a228298a05fe18":{"kind":"REWORK","_number":1,"created":"2023-12-03 12:59:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/94/902494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/902494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/902494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/902494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/902494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/902494/1"}}},"commit":{"parents":[{"commit":"698c281a5914469329ea738e44df306841b6ff64","subject":"Improve neutron images regex","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/698c281a5914469329ea738e44df306841b6ff64"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-01 12:02:19.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-12-03 12:59:47.000000000","tz":180},"subject":"Remove kolla_enable_host_ntp variable","message":"Remove kolla_enable_host_ntp variable\n\nThis variable was configuring enable_host_ntp in kolla-ansible, which\nwas removed in the Xena release.\n\nChange-Id: I737598d3bbe40b933b4b727eccc3b2a76ed97cdb\n(cherry picked from commit 8da5472ea84729334b4911080b00cf8929f3111e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e309ef0469fd317534ed7ef34a228298a05fe18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e309ef0469fd317534ed7ef34a228298a05fe18"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
