)]}'
{"id":"openstack%2Fkayobe~902614","triplet_id":"openstack%2Fkayobe~master~I5cb458132133d171037f8697072e6bd8858db01a","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-06 01:21:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cb458132133d171037f8697072e6bd8858db01a","subject":"Pin ansible-core at 2.15.6","status":"MERGED","created":"2023-12-05 08:15:45.000000000","updated":"2024-02-09 09:54:59.000000000","submitted":"2023-12-06 01:21:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902614","meta_rev_id":"9d2d27b3450408aeebc7fc028dc815b83e5d4b97","_number":902614,"virtual_id_number":902614,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 01:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-12-05 19:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-06 01:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-12-05 19:50:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 10:18:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-05 19:50:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-12-05 19:50:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-06 01:09:57.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"1a562efed8c9afb36424b2b65c9dc2b1e2f6413b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-05 08:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef7c41f37eaf74b287ea50324639ce9457836782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 10:18:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b86ca95fe45445ff9d60fa5cfd530147\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7051150ace66464fb151e84bd9312e9c : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a6411250194643a6cbb014f703816b : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4096a92e9f974fc4a1429fafd8f4d99f : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a13cb58f76a64e4db61352f51f473e3c : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f72d082f5307404395d15559d4875d9e : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7c56dc1e873d4bf58ee1b935d32514d4 : SUCCESS in 1h 16m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/31fb04b583a64a578a50aad9a5f2a404 : SUCCESS in 54m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3a5a367b6164d618430b33ee9258d88 : SUCCESS in 50m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4c65e6e9d4fd4c3ba9141ba1b53769af : SUCCESS in 57m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eab306e47c5041ceb6baa119743eb2ce : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5d96c2c0670747fe891d4f0f840ab16d : SUCCESS in 16m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/632fe81d69d243ddb934dbbafdde8077 : SUCCESS in 12m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/25861e6c935f472ab1d9c05b57b053e5 : FAILURE in 1h 46m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6b8126a0f36497fb8384c3c05926650 : FAILURE in 1h 39m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4439723ac84a4fbf8419e4f1a8c23fdc : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fac46db30c14c99aaf26eceb7dfac2d : SUCCESS in 14m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9ff0e55759b44781a6b5bfbc876f0417 : SUCCESS in 34m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0be999f4a54b400695689b29b4e8ec73 : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f80bc01f9fee4f63a6cf4758f41bc697 : SUCCESS in 26m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4feeeeb0e39f41efabce2e391ff1cbb8 : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2879c24d0bcb480ca877814a93192b47 : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bcfc3a62f24f4cb79d146916c9d34b8e : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce334924e3c94f148f17bec07f65a397 : SUCCESS in 15m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaf3cc10adb8475793ea1d5ba28c7616 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5af13f81bd264b8ea7c934bea2559f5d : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8c252d119c64815879a3e949da1616a : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"556f876938206af4b458d228aae12bb452b87e96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-05 19:50:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e33a0a7bae2b0d0632daca09f451ff1b504ed25","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-05 19:50:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dff794298dbc14085c3d48f4b677c0ff4a7dffd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 22:34:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d80125af33439fa450f88d264f4fbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694c444d920e4c54be8509896cf1887f : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34333076e8b40c18f635346eab66e70 : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8f0b3663b1d417980284c86a13544f1 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d72b10232d0462e92395a0317211a5f : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/92c7f3fb2b50425799d92f6e124a5741 : SUCCESS in 9m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fd0b0ed245114f918a7fef999f9b02a5 : SUCCESS in 57m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c8afc00e6a514e138fc7873227ec1c3b : SUCCESS in 49m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbb81ddd2c1f4663a2493859a3742f44 : SUCCESS in 43m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3eb2928264e140b3b61b26ef2ae45a6b : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cee9ef05e8184d81b8949302310eaa5b : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8cc6d699c8c04cebaabdcd40fb299801 : SUCCESS in 15m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/475260f7111948099ee9d39ea274140d : SUCCESS in 11m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b84597a8bfcf432eb74353cc43c319f8 : SUCCESS in 2h 39m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73d83a3e7f5a41f19c4a44ec30ee49bf : SUCCESS in 2h 24m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dfd9b364ec0742db8698b16baad1b658 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd63877456a644aa9e8d59d4dbed32e3 : SUCCESS in 16m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/566ac8c3da7445afa63bc3ef4c555347 : SUCCESS in 32m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0c747ef9f2e04ba0a938de1ba1f92de6 : SUCCESS in 31m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49024a11e32a420998bbb2c477423076 : SUCCESS in 25m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6fd281acd4d44053a9b786a93a46b238 : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1efe2ab89224a4d8eb5a72c6d9b30f2 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e8c3fa7586ce46aa8909627c0bfb3b60 : SUCCESS in 18m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9de663e18c64c75a628085be7e98574 : SUCCESS in 20m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a89de8880895496292bf51bdfbdb6a8e : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feda788286ad453e854afecf81e8febe : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c50a35db12364f88bba7e6f6fe96977d : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"31382bce224c30443804ef2fd2e05f0ea4bcf33c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 22:34:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b0926265ccf43aa5c4dd63f316957581609b35c","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-06 01:09:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e792d8a124a1a55011c3f43b15737da31533a659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 01:21:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414b6b5a4dc4465485dbaae50b9b6e7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1960826c7534b22a60ec26ebc1f8bce : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a28ae5efe0f47e5adda154879e5de7c : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8721ab34091d4bf698be1c065dd68a07 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a619d8c24945463c99480cd37f64ac3a : SUCCESS in 10m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fba2c26f301b4066a47ebec2db51266d : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c0a340f12a24a2bbd098fd1db59830a : SUCCESS in 47m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0deee85516cb45b1a022afb0b06b5683 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d65c51b901bb4e1691c1365cddeb0aa4 : SUCCESS in 15m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d1bfe36099b4ffdb52688d6e0b5174e : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c917c22230244f17b9490a84b3fd1c5c : SUCCESS in 2h 41m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f3fb05da119401ea799a90d4f138b56 : SUCCESS in 1h 58m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b9f2d7083faa49bdbee3a9480554dfa2 : SUCCESS in 20m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/912aecef9e5b43b891a96de06c755060 : SUCCESS in 14m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2733404e1ca646688111813774969b87 : SUCCESS in 30m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11d15cb2a45c463bbfe18a43ac461b22 : SUCCESS in 23m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5cf64dde109c42f4b169135c7c307365 : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d38a1684944d4cc4ae0831fb19d4a204 : SUCCESS in 13m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/416a3dc53c6840c680d6270c1a8d800d : SUCCESS in 18m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/234dd677d1f340a4b64b7d7c3f017a78 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"27f7111fa2afc5aef86a055bd54127fbd715bb29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 01:21:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"defae7aa5bb415b5d048885fda6356f246dc8b8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 01:22:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15eb3c759f2c45759afc1161d99a7cc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27098eba295941c8a5a151562589b04b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"9d2d27b3450408aeebc7fc028dc815b83e5d4b97","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-09 09:54:59.000000000","message":"Created a revert of this change as I2d39d004208f13275ad0d52079203044d135d965","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1edf231477a7fc970f0d013587257a349e20596f","revisions":{"1edf231477a7fc970f0d013587257a349e20596f":{"kind":"REWORK","_number":1,"created":"2023-12-05 08:15:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/14/902614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/902614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/902614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/902614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/902614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/902614/1"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-05 08:14:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-05 08:14:43.000000000","tz":60},"subject":"Pin ansible-core at 2.15.6","message":"Pin ansible-core at 2.15.6\n\nWe are seeing the following issue since 2.15.7 was released:\n\n    Pattern and root_dir should be of the same type, not \u003cclass\n    \u0027ansible.utils.unsafe_proxy.AnsibleUnsafeText\u0027\u003e and \u003cclass \u0027str\u0027\u003e\n\nChange-Id: I5cb458132133d171037f8697072e6bd8858db01a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edf231477a7fc970f0d013587257a349e20596f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edf231477a7fc970f0d013587257a349e20596f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
