)]}'
{"id":"openstack%2Fkayobe~902750","triplet_id":"openstack%2Fkayobe~master~I96f00721c173cf095463424c0ca8779c54943591","project":"openstack/kayobe","branch":"master","topic":"reno-fixes","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-12-06 09:36:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-06 10:59:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96f00721c173cf095463424c0ca8779c54943591","subject":"Clean up release notes for Bobcat","status":"MERGED","created":"2023-12-06 08:38:51.000000000","updated":"2023-12-06 11:05:54.000000000","submitted":"2023-12-06 10:59:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902750-reno-fixes","meta_rev_id":"f0157b6cf870416878cf664a6fd46c2176fb2c6f","_number":902750,"virtual_id_number":902750,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 10:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-06 09:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-12-06 08:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-12-06 09:36:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-06 08:42:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-06 08:44:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-06 08:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b8515105ad1bdb0cafd3a628dba6193693b0846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-06 08:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56ebce0f09cead16963b8bfa28aaf291f9e403df","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-06 08:42:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfd783ce6965b0810242608b1183ceaa490174cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:52:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a140a5f7a7034a50bc2f6e9d1f6f2015\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f83a591190461ead31996a1e8b3ca2 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4395be893dec41f28d16e011bd02ae2e : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bfcadf404814c9374b062bc93837d2ec15eac568","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-06 09:36:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3538742f5a526835fbd0d676bd28b5b545a310e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 09:36:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66d45b0f1ede8078525bd96705194136e28afb40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:59:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd7c94815fe40e1b7c0d4524f4a627b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d93354899404d00bc688945dc21bafc : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24848118340c4c0e961c1d58a70eae0b : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"66b6bab37b0055e59e79b53f0a249f5aedb12567","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 10:59:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f0157b6cf870416878cf664a6fd46c2176fb2c6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 11:05:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7431eeeaffc146febfea81ceccadeb13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfeb772f7b5e4bcaaa211b916741b1fa : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe908dccfef64fdd877c337db2208b2d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99d03686936dd1378e0d50289dd0ab8f0badcd8d","revisions":{"99d03686936dd1378e0d50289dd0ab8f0badcd8d":{"kind":"REWORK","_number":1,"created":"2023-12-06 08:38:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/50/902750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/902750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/902750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/902750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/902750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/902750/1"}}},"commit":{"parents":[{"commit":"1edf231477a7fc970f0d013587257a349e20596f","subject":"Pin ansible-core at 2.15.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edf231477a7fc970f0d013587257a349e20596f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-06 08:38:34.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-06 08:38:34.000000000","tz":60},"subject":"Clean up release notes for Bobcat","message":"Clean up release notes for Bobcat\n\nChange-Id: I96f00721c173cf095463424c0ca8779c54943591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d03686936dd1378e0d50289dd0ab8f0badcd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99d03686936dd1378e0d50289dd0ab8f0badcd8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
