)]}'
{"id":"openstack%2Fkayobe~902772","triplet_id":"openstack%2Fkayobe~master~I719fc8042742fe8b3b0312658aec39317a1bc358","project":"openstack/kayobe","branch":"master","topic":"redfish_rules","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-04-03 15:20:25.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-03 18:30:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-03 09:37:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-04-03 18:30:17.000000000","reason":"Change was submitted"},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2024-04-03 18:30:17.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-04-03 18:30:17.000000000","reason":"Change was submitted"},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2024-04-03 18:30:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I719fc8042742fe8b3b0312658aec39317a1bc358","subject":"Add Redfish rules to Ironic and Bifrost introspection","status":"MERGED","created":"2023-12-06 13:42:11.000000000","updated":"2024-04-03 19:12:54.000000000","submitted":"2024-04-03 18:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":175,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902772-redfish_rules","meta_rev_id":"bbba84b4fdd844b261d92cb45139ae70eed70242","_number":902772,"virtual_id_number":902772,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 18:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-04-03 15:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-03 09:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-04-03 09:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-04-03 15:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-06 16:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 23:17:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-12-08 15:15:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-12-11 12:59:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-20 15:50:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-02-02 14:56:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-02-08 15:07:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-05 08:10:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-03-07 12:44:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-03-14 11:45:10.000000000","updated_by":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"REVIEWER"}],"messages":[{"id":"448e0638848f9221dda027f1b5427fad8d9a0a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-06 13:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5726844d8d0d4ffcb8a950bf24f75d56527eac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 16:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d6a09b6ef849f48c92019ca2b8dbd6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96f36ec36bf841ada7db1f22281563a8 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248cfacf349a4aada92942a10c2b8d37 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a35beb21871748148b0c383b6f434fb1 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3bbec8a453e4c5285f45d4c6f8e1e33 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c3fd067333f24d02aca88c788c8fe621 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37a8b17c8bbc4b4c859c3e5e23f04692 : SUCCESS in 9m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e9502aea440940f3b62787aaa4058af9 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/caaf6e3b44374b3cb8a0c4dffc29bc39 : SUCCESS in 1h 07m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/028e0a61856b4f34afe1235c49d3344a : SUCCESS in 58m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/774381afb5c0402a928a5f51bc231328 : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e0cc1e60bfbc4b95ad096dbe8b7c5679 : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1db149534c0b436ab8873642a0822837 : SUCCESS in 17m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb816ef1d0fb4e4f94247a54c53c8c40 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/402e5eafefd244ffb9c6d667c79b6474 : SUCCESS in 2h 34m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b59e1bd8b9544367b086422c7eaccf2b : SUCCESS in 2h 12m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/410a734fde284ca68ca3d96428999854 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc92ddcba1f487f95ee4e1461936d8f : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a33a98c1c44c4a00aae7c3a6b84fbd4f : SUCCESS in 34m 00s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3144706021f9404988292cf6653220b4 : SUCCESS in 32m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95e28c89f5da475a8535233615521709 : SUCCESS in 24m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bfc630bf2c4a46258bf2e8830f824212 : SUCCESS in 11m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c46b8f43848041ec88a3ee1a654d67db : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/941319a8fa4143b8ac5c4d89178ea159 : SUCCESS in 20m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5aa15594043e403d96cba46e4111f97e : SUCCESS in 15m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23da28760ff34bc5ad7db30eccbb82a3 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1a8c08b7c6b45988695779e0e33f848 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd0164ae2a6a4205adebf3cc7808d3fb : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"489319c3dc4bd699498d35290f4d9ffb4ed7b39c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-07 23:17:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"19897370e2d7ae0837586ccb67c3526c59191933","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-08 15:15:51.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6826a17ada02e93a9d34eead7e5c8ffd4aaf1bee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-08 23:07:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60ce71c8b71361f5b2642bbaf56c328ce701265e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 08:48:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0de5dfd33d7b147cffb4069845beb3cb2ab28915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 12:43:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b0c22fd8340e224bbe52f0641297d89228ccbf09","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 12:44:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"56eddf6f9b4ab43fa66812b079c5f0a895807bed","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 12:44:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c4eec3158abfcaea5916ca578aac21416927579","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-11 12:59:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"42bb5ae284ccb83fc469defc49ff062cae675c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 13:00:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ff7e6658c8ef8b0e33489db816cb9a137a336db5","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 13:02:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8270ce503179c56ca04f98807590e06da1d37a64","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-11 13:38:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a7f3c214634e09d250d2c3f89e51e90e94175f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 15:38:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6196d9375892420b9d5a1182e2f8fbbd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89c71b6c0ae3455b82b759bc14445508 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6b9eb692484e16b9abfc76941c9b4c : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8438c8b5c3d9489b9a43427d1af3c999 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07d7a575697f4720ad120f0821807d45 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e58d6060a65475a93515ebf3cb92ff8 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d0fce5cf7f24b8581c483d52cf3ab78 : SUCCESS in 9m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/392bae007d79471da00863dd749c1581 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7c901d8f643c4f389a159111a2007fc2 : SUCCESS in 1h 01m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a04882c8e1d344bf8ccebd8352be11b7 : SUCCESS in 54m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0f0b7e7f4ca1423a9226846203bff319 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c04bdf8ce25c434d8ca668418af121e9 : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c08a43e8a7f449738ea1b089300da6f7 : SUCCESS in 18m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9401be6325624cb48f95de590df6243a : SUCCESS in 16m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/62950ac1ecc74326879b791b5af71687 : POST_FAILURE in 2h 28m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa42c7066d4643c88d5dace08ba54b64 : SUCCESS in 2h 30m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/33b7fc06e62b43efb69544517060f004 : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75d628b4576d47319043f82ed25cd7a3 : SUCCESS in 19m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b4b06e644dc7406eb3b0f12bb84e422d : SUCCESS in 32m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fcba5759241d462db56f6b644934bd5e : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec994c74926d4f39ba62dcde59ec1aa2 : SUCCESS in 30m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76a13dc797eb43af8da3c3cde281bbf6 : SUCCESS in 12m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8906236c826644d796bb621548b8a5ce : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dde0a07a0f23484a8132d984d4733d04 : SUCCESS in 15m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed82edb23dda4c82bc23ed1df33e1a6c : SUCCESS in 15m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857e33dd05bd4493ac0a69586a3e1afd : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f080fd049734c4aa5fcfbe158883e21 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32fb2e3a80ad49219a3927dbaf48699b : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"616389d52f8e2d16d1f1565bd1b83c4bd049b7a4","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-11 15:47:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d410b6f9f7fb6201a12630756b1384825ffa291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 18:24:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd4e199e6b9454582474a3d5c962d33\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31cd27b499ce440d8f7c2ad883cf0aa8 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c5e930ef014306b12c0796c2557e18 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77e46c82155418db77e794993410553 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a85edab8378349e395fe897e5c1e228e : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7303f0f06964fcf843ee4e99796d60d : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d81abcaece440a7aed4a15410bc7fba : SUCCESS in 11m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a336659eebbc42f8b33cb51b7bf6a61c : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bf424c2a741642a6bd6925b675b43023 : SUCCESS in 1h 09m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83dd1100b95d4bdf814d6c261274597b : SUCCESS in 58m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a112545ec9674887a750643963b99ed7 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a66e114d9dc4b04b11c49f13d7c39f5 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ef3dfa6ad07a47d483a78164983ca204 : SUCCESS in 18m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cba8b1c59cb406b898fc543a1ab3040 : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cc9740a5aab844f9bc52fe12c6d25ca6 : SUCCESS in 2h 30m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2e69fbfefa94a1a90e9b38225bc1a11 : SUCCESS in 2h 14m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b3d44db0a83346a2b8d077e38f2f0d17 : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c5c4cfd04314f35a5f56c2c2b18c9ed : SUCCESS in 15m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c1d3525b635b4663aa9cbb91c11a4238 : SUCCESS in 31m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f2ce34a5c9ad4835a0f44137a29508e9 : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d08df8f93847ed953083658487f124 : SUCCESS in 26m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7456053f4684d3db95fc6775e574519 : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e47dc9f1f604142bb7552399b1c5369 : SUCCESS in 12m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66c25414f4154cf3813a7d93994aeb0a : SUCCESS in 19m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2845fe9b964444db95b469e1566fd778 : SUCCESS in 15m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b2357ec4364793917a67ed02677b68 : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f2a236d1f534f40914b9251adc50827 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9c4517e12c94b2bb03b6866b233c569 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"83c3bc50478a5c7eaaf7d6caa9ad5bbb08c674a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-12-20 10:40:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"13a805031831e52ceea7427a27a7d70a2342f10d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-20 11:02:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8679cd91560c0030617d7557ba37731b13fc5a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 13:18:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef59a586a8a94d41a16091baf9e94570\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76365f54ec31427a9699b22c9db9cab8 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4d8fd6e97c4753b531741a1395f36f : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89ecb21f621d41ff94fc6afc1764796a : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0d0ed98b1a447498a7a114db0d1666a : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/759b3b968dd44a238fdb00c351803b2b : SUCCESS in 15m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba8d82ff2a8e430da472263acadf462a : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/63c94a7e996e4a8eb1fb5fe5255de7e5 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1854992ab5604114a048514760adbe1d : SUCCESS in 1h 11m 47s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c12371fe774446194b79e0f6ef39824 : SUCCESS in 54m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3fe80527d2ea447e92b0f71886acfa45 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6848509f46b04a78ba959a023f1907e5 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0a797e2435264c54adaf705b32d0bc40 : SUCCESS in 17m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35747d4d280c4e5f8af956cf1c5c2a8e : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44a37b582e114045bb00f614a700593b : SUCCESS in 2h 32m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b8cada8dfaa48198694151e965e1817 : SUCCESS in 2h 21m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba8a4b7452e44102af92e7c4917a7723 : SUCCESS in 17m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a3adffdffe148bdb465385f80746fb8 : SUCCESS in 17m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4dd531e47c8d4ed5b2be4aa0afc1d200 : SUCCESS in 32m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/975fc437348f427881f1d2f230049f62 : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df095a49c93a43c3a2ffdf8657ce2511 : SUCCESS in 26m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4cb196f9a2742858f9a04027b5c8bc0 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cc6a43cc55c4057a034aae62c7430a3 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7a3128111dd949b38f818b627cb3a5ea : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79a23eb3154748d7b629b776b8c96ea1 : SUCCESS in 15m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a55958b1868e4f5eaac62bebfbfcc1ad : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad0d8bdacad24e1e87199058503d2fd8 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/becc535c00494889b37a5c650bc4ba05 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":4},{"id":"c6e4b2a70c70875123127c3cf90ad6fa4d786d6f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-20 15:50:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5071997e0d60f85edca722a920e1288e55310e23","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-08 08:03:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0849a07b0a891c6436e241f7d77ecc92f9c6b9b7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-08 16:22:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ebf51d57ce24b9af47cfdd86a16e16af6ce9064","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-10 16:38:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b16c268ac19bd8d1e3b28a39e4fe843cbe33be73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 09:36:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6e9449a07a3c5e69ef8a9c12afb40f9b18424f92","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 09:59:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4fdbc85f98521124ce6ab2b53426158d5d14f2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 10:01:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96e983a38c33dbd0a648bb90f128b4abe1880c3b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 10:02:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2899fc7cf54c0e13933f5031a1da5ac20a8e6ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 10:04:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c92a1f55dde628e322247fbf32cbe0cda309e7e9","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 10:04:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2df431641f8460b7298d7bf45d4eca7332f2d022","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 10:05:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1c4d7df66ae8b59a7aeb7e844b9c09159bb2348b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 10:13:07.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"31eb3edbfef73b32776926c2c6a7d9a4f369dca5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 10:19:52.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ec33be6433c7d1d3806f396646c897ef1b78999a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 11:20:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"680ebfaac6d8f76f5aef9605dc66e8862704d57f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 11:33:30.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"96d648d025785de6d50c1a61f0581377cf075fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 14:23:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c4ef6d4da85d3ce4764496c2985386dc6f6c2277","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 14:32:36.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bf48a5fa43e2a38508832c514661f2a059189471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-01-31 14:33:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"521b9cddd0c3d01785d220db4e5b119928de25b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 16:54:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff0782c6aefd4374a7ed1293dbabd8d4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d1bd499995045819da5d1247fad1f1c : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35cec9ba41c41d59437dce5cd4ddd9b : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9e2af51ea474b06af4ef81662a346da : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22e7a46a23fd4fcfa7eac68a90652760 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c32eec160ad64fad87df819a55d594a0 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de392bb42b4b435f8ead1621e5c1cfa1 : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2a66bc8289de4e05aa9d309a7b83974c : FAILURE in 59m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6da0a38ff23f4a5e92a686310b906811 : FAILURE in 57m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87204ac1d6564fa58fda6b932720dd64 : FAILURE in 50m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/04205cac684f4bb2978ef0ac291c474f : FAILURE in 1h 10m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4e4067260b274481a62f4cc0b673c8c3 : FAILURE in 15m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/88f1a73340704ed487a8a005ef14ebd4 : FAILURE in 16m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/155ac2cfd9fc4be89165152ea20508a6 : FAILURE in 14m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8cdae25dc44940298c6f58757a857037 : FAILURE in 2h 18m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def276f2fbe64163bf6864b4982aea48 : FAILURE in 1h 23m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b7403c2f4055489aaa3203e2885f4107 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/357530d3240d40a89a61d35d0da49727 : SUCCESS in 11m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2b13c1c7fa36489bb2aaa27ceabc0764 : FAILURE in 31m 24s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b215cd4ca5114868bf14781b3ad71976 : FAILURE in 9m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6a4dc857d864fa9af81b261056bf940 : SUCCESS in 26m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8efbd98b058456b88120d5b2ac7e9f5 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee438e63105d4400a4b7e8fad6bf22f7 : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9df1347453de454da3c64dc3bef5ef27 : SUCCESS in 19m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ff4756b6b73467aa4f5b1123823a696 : SUCCESS in 14m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b47696f17634e438cef00cb825b2e50 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba1be99468ec4b718eae7b37adb40dba : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/754e84e2ca6c47379a29382a9cb08913 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":10},{"id":"90e63b7b6b4642b85b94ae92c006cd64a568942e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-31 18:13:56.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ee97a93d5b99b88de99cc7fc5f0eae37ff903fd7","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-02 09:00:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d3c30be5a20b8c9dc122af765d92ff2e1862347f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 11:30:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09fd869749aa4152b67fde77c8a51f2c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a13a9c75b549473ba1dee2a36eeef784 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d2c8fb5f5e4a2d99d6f8fd7025a625 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3286c25070174efaa0d21b7845f6621e : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8700ad78b5004b94b322fabc656d08e1 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cdc367527f64c0199bc2aca2a5ef53a : SUCCESS in 12m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b9f0458e6e04ca2ba8f1199be8cb96b : SUCCESS in 10m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6bfc3e6c79dd419889507270ee0b37e1 : SUCCESS in 1h 09m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/883b1fd844ae47b78c907db0056020f0 : SUCCESS in 1h 07m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de7137ff47a04a688265f14772e5eade : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/768878aa92f5414aa62852b854f22dac : SUCCESS in 1h 13m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/71bf861e5bbe412485db0eba3111c87d : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7761982297e54a9bb102bfbca0b8530f : SUCCESS in 16m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7200004cf7541d580996a13b2329f4d : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/56635c13409c4d439c5d3882a15848a1 : SUCCESS in 2h 29m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b80de7a6b160423d9f8c2d2972b9d596 : SUCCESS in 2h 07m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b75b910c63604845bcdee595f5e10986 : SUCCESS in 20m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62cf8323bcbf468591d04f6b30156776 : SUCCESS in 19m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e77d827425924c279b500700a1e9dfa0 : FAILURE in 33m 32s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/96dfab81ea374df5811ddfbd5d6c0357 : SUCCESS in 27m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbdfb9931c504b8a91a7edbcdd46908a : SUCCESS in 28m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8c4432b7d9e24d35ae680353bf7cce30 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea6c0ec8e3c4afdb01e22bd483b1ddb : SUCCESS in 13m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/180d279d3526441ab5f7140f709f8220 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5780095244c44f39bd05b4965ad575f4 : SUCCESS in 14m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc7ad547ed90497e893fad61792463d8 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d437c0b81fe64006a0b78725f8ab28f6 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6169cbb70b6442c81cfcc8982025497 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":10},{"id":"8d689694b97411cc30f931b49c024e5a21d881bd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-02 11:40:04.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8d388748ffc39ccf4bf09c92963b9af8cb17f357","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-02 12:54:10.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"475039ede9f1d2856423d03f5b39a5add78f48aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-02 13:07:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ca272c8b3ab476326c2137616abfacd7b9af5c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-02 14:15:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d66ac898dd1e4eff70d1d8e3ca2713a7e915d562","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-02 14:24:20.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2b89a314396a7584bba4b3a71dd57f3ab8ff2348","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-02 14:56:07.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c618099266f77d66294700bf2e89eedbe942ae67","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-02 15:51:24.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c43512f0b60d52d70c4676b828b8bde526abd2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 16:59:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942e2ad859594296b8db8920f2ff4ec6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/342f1e5405fe4e5c973d4e96f9800169 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8a26cd5f5145ac94e1c0e91f4fe95a : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06d6b0a37c7e4588be9d5d098474c06a : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d58a76cc8cc44cb99e939ec224011f3c : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca6774569435407d8d965eb9ca600f42 : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe19e03a6f024df989390a891e25b24e : SUCCESS in 11m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d6b555b2fe9447ab1372f8f114728c6 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0721753e459041e1ba5936230c8d678c : SUCCESS in 54m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54164f06fffc43b39e0226331769ebae : SUCCESS in 54m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2692116d24874afdb6210d877722744f : SUCCESS in 1h 17m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0ba1e31c25d44796b3333ba65ecf06fa : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/536ec8f426e84ea68b02bdf55fa3d17e : SUCCESS in 16m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0b0008f729f40599a628da50a8cb05f : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5239ede3181848f1ad657064d95e72f0 : SUCCESS in 2h 44m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f72260537d04165901759bba07faa1e : SUCCESS in 2h 04m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd589471a6e54e2ca62708ec1e0c27f8 : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d018bab5a9144668a29fd8c06ff7207 : SUCCESS in 17m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ccce205fff5c40659d3c9e5a4fa83b1e : FAILURE in 32m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7a3565fa6ed24b74b37bdba4696c4237 : SUCCESS in 32m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9b04209e32e45589c4752474a630497 : SUCCESS in 22m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ffd9d08cae8449e9a687fb2174ad3f1 : SUCCESS in 13m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8c18171296f4a5786d9488c00a709b3 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59f40d5179264b028eec5d3acac1e22a : SUCCESS in 21m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f31456dbcf6a4f9590c86676a65dad06 : SUCCESS in 14m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8949f6459b485286390709b3015f1c : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aa3ebce28754c95a10b417f3610d302 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/259bcec04ffd45fb8b7513f25eb90518 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":11},{"id":"1da830a55870c8603336b77ab380f29b0571eac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-05 06:42:05.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"5b43df78e1fd14f8d5d8087c87d7d4f256a78503","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-05 06:43:08.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"22761222acc0139f414c4383b9c09501173f27cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 09:25:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d387fd8f5c94c0ab47112afa3512557\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b0a75123274fe4a77f86c7cd8f204c : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7797a72da1423aa77968f26e3ff23d : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b9b390ee5684cfc9c3a03aa239a3374 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7bdc2d6c5c7442b39eca7339029f685b : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/181828bb74604deb8e9d22dfd5475153 : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ad1821c4b544f81b281e89d5657169b : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f3e4ad719336447b823282faae015731 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/089e7b7a95544fe2895d2362efbcc73f : SUCCESS in 46m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e466d1c085aa4c1489d66dd478ab9c3b : SUCCESS in 51m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c11ee22263c24c61b7f510ef306d3f9c : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5638e171cb8e402bba20f95801a6a819 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/41e48ca22d3445879527bde4a3da29bb : SUCCESS in 13m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9008525084c74096bdf3cd2e594f622f : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/275a9745201d42d8918d1c0d7abbe1d5 : SUCCESS in 2h 34m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6715e410f47845098f391c065b6e5628 : SUCCESS in 1h 32m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3a020c6f222d425eae8292be6b868617 : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bb145dadeaf49e2a7e83e508420d5f4 : SUCCESS in 12m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/81c4d3ea4f9947009d5d8f9963ba2717 : FAILURE in 31m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a4d8bf7e9bc544b19fc19576ccce862f : SUCCESS in 24m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b84dc9af6614819b0d843fe777e4ac1 : SUCCESS in 25m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4dbeb41c8256417887c8ece57ea05414 : SUCCESS in 11m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14951549603442e5a87b68dd8689de02 : SUCCESS in 8m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c90c9d6ed62421b804380dd8007e212 : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eac1fce9a0964b89980016b133d11ef0 : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1bb1b1a5fc4490bf6b3ea0e08fbd13 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b60963168610497eaa35956b7ebe72cd : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/724f4862bfa045b59853e773282b380f : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":12},{"id":"0384b256dee6577abbdb4a36b9bfe63011b64de5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-05 09:52:00.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"294a67e63f3a076be13c7eff65a93a25f70e197c","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-06 08:44:29.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"82fb01228fca6d6b76e904fbeff14f50f6e260b7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-06 10:29:16.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"2e5a3a0b130010d40e17ce65b0d6843218d12134","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-08 15:07:30.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d14e460d211687309bdfc91fe346369d439e32dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-12 10:41:37.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"824e628460ebe398b91b6d2cdf4c3930693f86d0","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-12 10:41:46.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6f2ffea6169602403f9144514eb058d70c1f6546","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-12 11:27:43.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"e3729c1b5eb37f975c3269e9b36368db4c53bcbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 13:17:15.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffc00c3a517a47c99cc3b555f8b1ed12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0b1b1d082a4822947b65ea6f70b671 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69958f579a34910bb87560493ab43bc : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842b3029f95640e59d3212a09dc2787b : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/667ef18e1915431081a1045e5aa695ce : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/865f1e2f9ae6411a91cb56c0a7908f4f : FAILURE in 13m 28s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce9c5ae168964d0189a3135c9818ecbc : SUCCESS in 9m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/251a8fa562134b179b5b633801354a3b : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ea578bfbf1234f269ea416faa8a5730b : SUCCESS in 1h 07m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/630b678d666c442584a9c62cbbe6e517 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/048950a7c22143cbb8d763d1a60c1069 : SUCCESS in 1h 16m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0f478bcfd4314ddd9de31f6e28c46ce4 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/50b270a22da04080a3496c6d0c35ef37 : SUCCESS in 20m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/688fae46dc6542f18df1ad2d0e19f40e : SUCCESS in 19m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aaffb4a1e4c1410e99e4303e91528152 : SUCCESS in 2h 31m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/819e7004caa0445e92314a9b082d214d : SUCCESS in 2h 01m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f78f5a8d4cb1426eaf0db17104454e4c : SUCCESS in 20m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b215c0c26b3d4c21be8e1f7adb31edbf : SUCCESS in 18m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/68c3e5936e36413da65d2a0d2babe66f : FAILURE in 36m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f051f0c287bf479a8748ab2513e768fa : SUCCESS in 32m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2856fdbc97442f0854c6195082db606 : SUCCESS in 26m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cab7ecc0f85b4551b44a1a66f7bfa543 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e3cc7e1ef7a4a1fb723d7241e4a2d11 : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77af0b66cb1b428fbbd25879cb84a144 : SUCCESS in 17m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1148ade6ce84df39987654f85befcca : SUCCESS in 13m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c051fb92e8143b48062fdf644fccb2a : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11c8393ad39940bda4d3908313aab0cb : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb30afc8998d4b8c9ea3273bbee09eda : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":13},{"id":"db466b63e0a6ce36c0407580558cb0644bb70133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-12 18:33:50.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"2ab31a902895836b98f2a7ce33d4f66f0945738b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 21:01:13.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50b348bb61e4515a2115a53dfdf597c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51621fd9e7af4b47b3339fda9e5a285e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2e51810756401b94f6dc17801ae70c : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/969714685978430e82257e66c24605ff : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2dbd2b27f3cd4e99a814ef8884a70af2 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4a3343263d9459b8b94f60df82571d1 : FAILURE in 11m 17s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/feff584a240444bfa255edb786b705b2 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cc29d35a34524e5380466b3a7f6920e8 : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1ff97fca071b4668bb70699def535fdb : SUCCESS in 59m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24106bc236694e76a093c9ecdc2990db : SUCCESS in 54m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ec49ebff1a324696b6ad7a77fecd12e7 : SUCCESS in 1h 13m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/56f44e24346d47f9b536528cab87303f : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/00c23b1c3219448587ed5d297934057f : SUCCESS in 15m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f574a9f82b854083841fd7237a83aeb7 : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1209513d30824e9ebb4b256cf7675952 : SUCCESS in 2h 21m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7b2e75fcb764e8db055863a50a252fb : SUCCESS in 2h 10m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/60ba4357d1684b99baa64539fd59fe5f : SUCCESS in 20m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12764e4178cb44dd933d19ad1cc307ac : SUCCESS in 13m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a7b3742e183a4956b45d957b2183f9d7 : FAILURE in 33m 50s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/736aa801f9904784bb16c2932f9540cf : SUCCESS in 29m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c3ebdaebdd4d198dcbd60d538a51b9 : SUCCESS in 21m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eee84ee01e4c452085d5048fa0a22430 : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/213ff453b7bb444085ebac69e0d64c0e : SUCCESS in 11m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48349a2667ba47148fd8c1dfcc648660 : SUCCESS in 19m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dfe55ade7574d338bdabf473abe4fef : SUCCESS in 16m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7f7ab149e1438093bbb67cd23db95d : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3934d59afd2466f813beb8533c0de56 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eefa86052f5344cbbb1d4aef3a0353ec : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":14},{"id":"9b79b288fae4747995803cf7e34c259a8422fa31","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-13 07:15:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d73c44a7fcff7bd88987af03454d55dbddcb069f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-20 12:32:54.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"42b8c9e8a4c568b17bf5c3418198b7eb38cdb94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 15:02:31.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6650989f9f495d9cf697c4196b408d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/200f726e6588479f9c18abfef62c871f : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2149336317476bb4897af8d8abdf0f : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32ff191481434dbab3233fc7c2c3755d : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b189a54b2f9b4672ae0d7d5b240371b1 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d736ac658de94ddc890207110fba7e46 : FAILURE in 10m 52s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0bd7865445441c2aa6f5f62f7175574 : SUCCESS in 10m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9ab469217d464126a867947299dc1bf5 : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5cb2b0f73db24dda89832bd5fa371b89 : SUCCESS in 1h 04m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b255bda56bf4e1e9d473028f314388d : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c4b368ffa00e41ebb8eb549cd345d90d : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/22818598bffb40d4b82ff8512b8042c1 : SUCCESS in 18m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/017d4ad9419e4c9393bee848155bea58 : SUCCESS in 18m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a140b7ec341448b7943874253e155d09 : SUCCESS in 15m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8ed9baa31ae249238e0e3939d7322c50 : SUCCESS in 2h 24m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bbb762608bf4eb1ba312b548d2c9c0f : SUCCESS in 2h 14m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/91cece33385045cf95be3ece56318ae1 : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/577f170fedcc442b8ad52a3781af2586 : SUCCESS in 14m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/43774288cf094f618400fda3916aa38e : FAILURE in 36m 49s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/42d76a12cba94983866c42d16a6e55ca : SUCCESS in 28m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cc0346abce84186bdc54650b9d66c47 : SUCCESS in 25m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/885f8129ccf544eaa087ead8ab9437bb : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99379cb8fe8f47fc92483791f290ec94 : SUCCESS in 12m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/47029f48564f43e6b688b37cfdeb5894 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c4e75e1246c4332aeb5ea6753adec7c : SUCCESS in 19m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff410724d3248aa8ed38a0274c8bb39 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ea175695a304f44b7276092adae9230 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5aa84fcaa0a1425e863184870b48269c : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":15},{"id":"95d42f645b1bf185f7be17cc442567b2d5831b51","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-22 09:20:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cd84a137a5ccb82122267987c2f37ccd1b7eaa0e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-22 09:21:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4e4140ccae72601a7a2910b01ba018871871e77a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-22 09:44:11.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"21a20825022b2755d60428c8bd148e2de5729567","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 17:51:39.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"848a0a095c18c5069ac8e4b33b86447c6c5e508d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-23 18:32:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"9f02ce63164c792ea0a8e793a623f5186e9b47d7","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-02-28 08:26:55.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6ecea86ad93d83527128ed1870d8d648fbbada2b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-05 08:10:57.000000000","message":"Patch Set 15: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c1cf8a0dbd9e87cd31cc3b728058a465d8962d18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-05 08:26:32.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5f7fd8102adda847b47a03601898fc1813b93757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-05 14:23:29.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"3d73490e7307af4d7204f044d3c6746b766961e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 16:45:14.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/047e8ce8d10f4655801e2fcb2a0e528f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d8f818b2eb34547a55d2326a3487fc1 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8185e7b65acc44f093715e6ee97a63ff : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8577b1cd3c6d44c7866fdc5bada2a7f2 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/125e5933b28843bda1525f1c9df03e93 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d830b01b1364b3695f540fbc2c3eca7 : FAILURE in 12m 50s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afbc06e556854ffe92daf51e25bc1e77 : SUCCESS in 10m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/df45734533dd48e78e1d0c78ae659ba9 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/630ab450231c4bb5b008f12a2437d461 : SUCCESS in 1h 04m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96fee1bfc0f244f8b12c31666db87b2d : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23afd75b919048aeb7429f5d1d63b77f : SUCCESS in 1h 16m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c0b1968055cd49e6aa8bdacfeaf15e23 : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e91b6e7bb14f40d2bc33f8c91c372ab3 : SUCCESS in 20m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d3e8af687224f22b425731ad787c723 : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/389b941f3b3c40a9ac21476cea5d41aa : SUCCESS in 2h 17m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ae5d89bab8644cbb87f58ac3fb3a067 : SUCCESS in 2h 06m 56s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/38468c9f60364f239358b7b57a9dba8a : SUCCESS in 2h 11m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e4d9ed841cd45eca062cef2f8d0c0bd : SUCCESS in 1h 59m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e41240326e6247b689bd80a25a9ae69b : SUCCESS in 20m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10480180a45d45ab9c5f28353f834b24 : SUCCESS in 13m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/93e2c00e95004ec5b3298fe4b21093d9 : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/058aa096e01a404198461cd75a8af635 : SUCCESS in 24m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25d216e0e7f243be8a03ec4261364c7b : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3c8ef08338ee4689ab199b68614592b1 : SUCCESS in 32m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/786dd39f9cd7484d9b29ce040de5aee6 : SUCCESS in 26m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9028b0064e034fb0b0a195d898188f77 : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c252f09b70147988e78955f28d129c3 : SUCCESS in 12m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a3f31633d51427c9c94db8046e3a306 : SUCCESS in 17m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b409b1e0187146e7985043c49fdae0d3 : SUCCESS in 14m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d59f8fc325194da2ba454e77cd1a2ce0 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c5aa6a85ea84f3585afcf50eec96b5d : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b08ced5395c244e3853979b6c533b8a1 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":16},{"id":"d7eccd025cdfc313b0dcd953eb1abc9486a21c33","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-06 09:18:44.000000000","message":"Patch Set 16:\n\n(9 comments)","accounts_in_message":[],"_revision_number":16},{"id":"a9c1f64ed5b68b952046995e3e82fd794074bdf3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-07 12:44:39.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"0fc8e3e163f705927b333f569e2da721de8ca65f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-11 21:41:50.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ee3535a0861c73aef9ce650dc8443375c9332b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-12 08:09:20.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"9a29c525222bd2e5e1a3bef29feeb80c17879783","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-12 08:10:13.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c976d5336498962634b91f369d47febb287a800d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 10:39:58.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba04421f6480431e93e79b001558a477\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a73241b240f74477a81089f18252f815 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f843932c46440328f5c160d20b18d7a : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf1dd0e0d4bf49e5b53cc78c57bd0154 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/622b3989f5dc4a4a811fadfbc0c68554 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55eb22bb7a494fc7b21ac9776f227901 : FAILURE in 11m 52s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b76a178c91aa446899086549b1869c3a : SUCCESS in 9m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bbc905593165454a9fcde99eac469c15 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/aba7a5800b4f4ceea7b3a7e013b2bfd8 : SUCCESS in 1h 02m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa4f3085853542c7b9f67faf9aaf96a9 : SUCCESS in 55m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cef42e73bdbf48208ad839cb971dbe0b : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d00c89df8c804ce3b9d7530b7804da6f : SUCCESS in 15m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b87a6a8909834485ae9c618160938d4e : SUCCESS in 15m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35aadb6480724017bc923c9b0f64d101 : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41ee0bbb56234b3895159782116a33e4 : SUCCESS in 2h 25m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a30e62ae0d9436ca9f1e072636aa922 : SUCCESS in 2h 09m 25s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f8bbb5c6932e40b3ab235a2f11c92699 : SUCCESS in 2h 06m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58183e5ab66e424395213777d7d4ad82 : SUCCESS in 2h 11m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9840c7bef52649738d89b30055f8a8ee : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65875f4f53ee4aecb2cb708bfbdacb81 : SUCCESS in 12m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2ce65de62daf47febfca9dfb7d2305a4 : SUCCESS in 36m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/15554de50b12463ba7ab4aaf100d9db3 : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d094df16cde43b7a98b3360ca46c8b0 : FAILURE in 9m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b740f1fb57d94450812eb5f8532c8caa : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21e4b17650ab4409bce986cb92128537 : FAILURE in 13m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5cc2a60098e4188b5a8c3ba68c071d2 : SUCCESS in 11m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e5b044675b742da898462f5ba574761 : SUCCESS in 12m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90f26173a3304291a18f552164bf12ad : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/371c0c91380e4edd8c6f9ca9e4da8699 : SUCCESS in 14m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd4f4c4590ab423dabe78e138f5b5427 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41762e64b1fc4b038d0cd98d71cd620e : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/081146d85a0f4a55b08aad03ea6e5b01 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":17},{"id":"09b09df79c219c71bbf977f54ebe7c9ab4b9778a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 08:29:58.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"66af07f80174dcb8ec12679796395a2e3742b1d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 08:39:00.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"47ce3958f2bf9b30614385ced0810a34cf42f7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 08:40:57.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"7573d94187a6108f2545710f4de3c9e4930dcdc1","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 08:47:21.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b5cbc743666c35590e6b9e1b9fd9b12f0efab486","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 09:09:42.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"81cc6799f930eced32dba27e09fb2c59a82444cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-13 09:16:26.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"b26d22bd6b2c64b3b19d6277bdbfdf692af2ad6b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-13 09:17:57.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b9bd57b1e5a02a6dfd28f3022d6888fca30fb49b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 09:30:22.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":21},{"id":"deede8ecb6ea7b2b514da6b6fe228cbefc69b2d2","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-13 09:30:36.000000000","message":"Patch Set 20:\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f70c906be4f8d8c7bba4acbbc2752c8ac36ea253","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 10:00:22.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"b537868547f427b3ca41aafb05d383095746fa65","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-13 11:00:49.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"25eaed015c056a3a5de8e0e157ec0e31c9130948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:34:20.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01db541162c74c42a13e98f872aea5e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42a973481019444c93ed5c22be5f77f8 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1476e5f191c44ef8b8ff0d16c5c8cc33 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08605395f58d43c8879bc4dc183f3817 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a6f60ad10554cd8b59d6286bbbed1b3 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e44312d503e48dcba62c277487ab318 : FAILURE in 11m 24s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdcf3fcd81534483970c6616ef9a6e3a : SUCCESS in 9m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c7479ea863774617995cef2847c2e902 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/73b302cfc191454597aa4ce495d0e26c : SUCCESS in 1h 01m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9f140d812d486589c24f8ce128e45b : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4edbdbdcf5b740afaa72ae85210256ce : SUCCESS in 1h 15m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/add9224af6254ebd9a3742555e944d9c : SUCCESS in 16m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e449050911cc40faa3b4e6b383ee5293 : SUCCESS in 15m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe2e4b665d5a442dae42c75e2e68e493 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/93559ecce7874eb090cdbfd564561a7a : SUCCESS in 2h 18m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c170bf968844993a5745cbab1097b6d : SUCCESS in 1h 46m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/84268e677ac24763a007c90ba8ddd0b1 : SUCCESS in 2h 29m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4412e46069034c85b93c764f9d74f9bc : SUCCESS in 1h 49m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9fe9ef2b2ca645fd85cb5d764c97900d : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d625292d30945acb71c0680f9665206 : SUCCESS in 14m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e8f899678328480db1a0d92bdf3dd92f : SUCCESS in 39m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a72f7dd058b34935845b2ac0def175f2 : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e27b5fc02424c86b1cc917f7e899043 : SUCCESS in 24m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/276b060bc8c14cfda17103eeb818fba0 : SUCCESS in 31m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f27ff0c049445b98b2b94252db4f15 : SUCCESS in 25m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eba3a453569a476dabe55ad94a0b2b5a : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5357e0aa21fa490c82b1010c2a8d90f5 : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03de4d682c2a4300b89fc4e9a01258d2 : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9933338699d24e968ffc3cfd0a2710ac : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c10de9a0dea4dec92cbf331c97a58cb : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3823ff3f0cb45bfb21d337cb33204f6 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4751d1268cf74c0a9b8e6880a3426946 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":21},{"id":"63477eae244e0b61541ea70f4438dc7504c51b09","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2024-03-14 11:45:10.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"d2b0d55b05676a020104bd70dcfec08d85c74dd4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 11:51:01.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"6ae59ddad3b3bee0c8df072e20a1b179a34a0ce3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 11:53:08.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"537f429723686f7fd5e23d70b3a561cf7a9b9d1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 12:00:32.000000000","message":"Patch Set 21: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":21},{"id":"1a42470295d68afcb22453448449eacfa91f7d6a","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-14 12:21:36.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"3c21ed23712f78c42dde9ed8fd71dfdc4d3aabf3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-14 12:26:19.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4e2c7dcd8b36a4d7ea93184d3431664be3a6ebf1","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-20 11:30:51.000000000","message":"Patch Set 21:\n\n(7 comments)","accounts_in_message":[],"_revision_number":21},{"id":"fe77a7dd8afc9f4b934a3911fbfdf633f370652e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-20 11:31:13.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"88458d676d5bfef964bb4828aa1568487dbe0d3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 14:08:36.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99ed72e8b4c54455ae3318c43bf59fe1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bec743fc85a246f494afe271fbae283d : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01abe51fda2349f98fdbe17fd7330c1c : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/514523149e7749cd99f2487a174e610b : SUCCESS in 2m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da0a1108cd614bfe83aadca16e49be9f : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58a7919645e4488da3083296712d6617 : FAILURE in 11m 34s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d8c467986ca49578269f2c9e15eab08 : SUCCESS in 11m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b9024a60e214a7c87d3d72d434d716f : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/879c538ed4694ddbbf0876803c42b8d9 : SUCCESS in 1h 03m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f84c9a9c1f1430d836af3a6bec453d2 : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c435b0947ba2445ebbd8a4c325806777 : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e86f9309111644fa9d9eb4cc83aa17e0 : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d277b36a50664974b6e6e93ce4018b20 : SUCCESS in 16m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/139d859268f44f52a2cf6e4cedb15786 : SUCCESS in 17m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7fc67ec22b054f9a9b656e4813975986 : SUCCESS in 2h 28m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02809a9089864635869bf045d3cb91ca : SUCCESS in 2h 13m 54s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5d529a567baa4d8c8e9bb8e51f7f9f63 : FAILURE in 1h 27m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8ea7636a430470b88e4f65164cc6d08 : FAILURE in 57m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af892cf658614e88bd862e1a2b2e219b : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5af97a2e20144a23b32cf3a376f54d0a : SUCCESS in 15m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/03f75e88e82d44eab643b0fc90dc3243 : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc202021d5f846a2874062f84b2314c2 : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3197c43a552947139c60378922aaa01c : SUCCESS in 23m 45s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ef8bdab6148042869f0af0a44d129fa1 : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71245b2ada3246b28db9cffbd7e6e426 : SUCCESS in 27m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1b508124ea54732a1462030c9ee7186 : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4131dfb078bf44df9215050cbb8a4a6a : SUCCESS in 11m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8bcb85ab40f44728abad770f201eb07f : SUCCESS in 16m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c419ec6bf049c08e5e477e89e1b62b : SUCCESS in 14m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4903fe28bec4d62b18e1ddd9da127a7 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71049e46d6844e8b827934e98a2e525a : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ea4fb86d81f4cd9a918a360e42ea3eb : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":22},{"id":"f0bc25abb8ef2245def0f8ad525b434afd740fe5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-20 14:44:14.000000000","message":"Patch Set 22: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"74128ce2b90e72107e61c7fa0c81d930a14289c0","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-21 14:51:12.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"62756745143b145f3e039fa4bfd21963afe44955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-21 14:58:12.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"a28066551cbd99d3a9cc3f05fe74b53392891395","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-21 14:59:12.000000000","message":"Patch Set 23:\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"6037e79d5e9c8291277b3164a88fb4e53aa5ac2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 17:15:48.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8eaf96848f7491fb8f8aead69729f6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3eea9e14e5415fb4a4b032aceb9b3a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe3acffd0c94193a850fdc74f2441d2 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e8a3714cc3149aeae135dfe6b49cb74 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b1b4e0c1a6f4790a7e360f635d1e16f : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19175f56e0f84b5fa9710e2111096eb7 : FAILURE in 11m 00s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ebeb98c99f9e449a90b19d10f7e4ded4 : SUCCESS in 9m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8524ee8bf55e451aa6292c5bac2fc6e0 : FAILURE in 21m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5bd3d2eb066b43ada768532616c794f0 : SUCCESS in 1h 09m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1898bef68684afea8014b8e1ab55661 : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/42ad333eed7d4443ad1fd3a204bfd189 : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6cdafdf8660b44929679f74df33cca36 : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/14115efbc903400ba04fb9eb49b738cb : SUCCESS in 15m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86a2bdc77b0f445b86082227ef4c52ec : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d827251d0334cd18a98bea789c5db44 : SUCCESS in 2h 15m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4eb4dfd04cac434883bab8c5e5f236a9 : SUCCESS in 1h 49m 07s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4eee1225d6e84de785f6ce6079812473 : FAILURE in 1h 28m 36s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b4f93e7a4d546aba5cde064eae2c700 : FAILURE in 1h 02m 29s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7bda2e9585fe49c2b7773616a7c4d9c7 : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed130bd5e2c44877bc74192419b9cb2b : SUCCESS in 14m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a8db5d4df89f40169d70e24df1338db0 : SUCCESS in 37m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/76901d0f92194705be8c1ce1d5e8ff57 : SUCCESS in 33m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/573f90f9c27746dfb003c9767c12fc9b : SUCCESS in 24m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/928cd17fdc3942288e9597f98f5a8556 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b30e3ad0b7d4c59bec7b2acc4d0831f : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f546d0ff9e6468fbe0f0be6c25d5b60 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/241bbf5f23a04cab8e0f92b820aa845d : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/acde5d1ffba04329a4af5348c867028b : SUCCESS in 19m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35235c289e2344c3847d874cd87f71fe : SUCCESS in 13m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee069a750c6450aa29841ab8b089e9e : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc79c393aafa444ab867273d0e559dd7 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36a8b249116a4db8b57b0bcc91575695 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":23},{"id":"8777bf459896f3f84259ad77f84d8924cdb9aaa5","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-21 17:44:25.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"56e8769c7fdc5f468eae24f0f8dd763d071337a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 20:24:15.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84bb56ddaec04c318809c4dcfcda877f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47855f0f4bbc418298e2bac24f6f2525 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2447ed90de481bbc907e5934db1ab8 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da80e1111868492f8d614b9f36398dcd : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0bd66a440e8e4f218806688b26408f35 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa3e7eb0001a417f99d20c0b121ff0c6 : FAILURE in 10m 04s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29c187b44be84ed7b70207519e213bcf : SUCCESS in 8m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/71440bb29d534742a7fc0e0de9393675 : FAILURE in 21m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/53a344de46c94b0683b8726ac01689c1 : FAILURE in 20m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/818cb66bee574fd98ab0140ebbfd0bdf : SUCCESS in 45m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0e2edb689b77492086fc64d64b501106 : SUCCESS in 1h 46m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52115fc9276646bfa9314fb23f4024cb : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/41eb196ea6af4874ad1cc1135e3d1af9 : SUCCESS in 15m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e863d2b783d4ac08b2c0e96dd0abba0 : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/818554ad28fa49efb8f2603b75b0cbf9 : SUCCESS in 2h 34m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e55cd78975de4e9d95d3bad6433a3fdb : SUCCESS in 2h 09m 13s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7e1dbcbdf1504604bff771fc40853f62 : FAILURE in 1h 15m 11s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ed53a84344b4a4f840855924e70759e : FAILURE in 55m 31s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/063adbe1792c44738a48fb5312bf7937 : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c443dce669624915923f0c6cd616faed : SUCCESS in 12m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/77dd5aa2a792405da971d00fe8e32acb : SUCCESS in 44m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/582f99e6ff37430283fb5829eb30cd37 : SUCCESS in 32m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7dc21aed0e4dff8e3b429bc1b66d58 : SUCCESS in 21m 53s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2662fb1ea45c477d99ddda1e3b8ba270 : SUCCESS in 32m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff9532e8ffe14b9cbe60ee5e0b35b236 : SUCCESS in 23m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bedd2c8f629c43a0bd4f5251646b39a2 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/136653c9a0e343328918bf5091fee0cc : SUCCESS in 10m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69dac6a30ab448e6bbf5b896f759eea2 : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4898a414f2584e2a8920845a0afc5a39 : SUCCESS in 13m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ec979c91fa64f859b31c150aec695bc : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ff988cf12b040f284dbfc4b9805d26e : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa99ec9ebbf74d32921d4620c3b4b365 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":23},{"id":"3cb5cc911a61f97bdbc8948c641e2404b16f86d6","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-22 07:57:29.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"ef0ad7d8ef93436d080ef70670f2badef2df3956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 10:28:26.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed096a565414366935205fabd700cdd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50e5b21fc37642529d14f7ce108fe009 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4bab36f0c24fdda0ba805c34cf39dd : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a26d23509d44a629063cfa6351e926f : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cab00d47188148e8804f7ea57d6efe4f : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc74f6c27f9644f19651c8dbc4c7282c : FAILURE in 11m 57s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ff5cf08e81f4831bdb20d4b088b52cb : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ccdf3f0e198642c6b9e4ad62fb36ac52 : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eae76c10fc5c4c84b32f2ab4aa710d51 : SUCCESS in 1h 03m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3808c387491d4e9b90d51d6822585c52 : SUCCESS in 52m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cdac0da9dd654166bfc8049a31c7e7dd : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/40f13f67010441a5998538e07d6a2985 : SUCCESS in 19m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/42257ed0b2934bb69ead19aa0f7425fb : SUCCESS in 15m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d38c7ca2d4604126b35f7c8f0f7184c4 : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6f9ece5d387458da06f181bcb943ccc : SUCCESS in 2h 25m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea7dc390ecc7413faa2d7e1299d951e0 : SUCCESS in 1h 44m 54s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b8a22a3095b74040b66f6e5430f4ae71 : FAILURE in 1h 14m 05s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a99a815cd6849bba650272545e1e3fc : FAILURE in 59m 03s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f56814318fc740d7ae036a7a60b1f8b6 : SUCCESS in 17m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a4cb6b0e57549ba80de5852c4048720 : SUCCESS in 12m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f1b2dcaba8b04bba957fb659c096bdf4 : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1cbab3e82cf247a2bfe3b39a367c9b9f : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ad799b7233347c0b5727d82e0e5efec : SUCCESS in 20m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bd46939e00d34590830d247a7745a60d : SUCCESS in 32m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/760441a92f694a03aaecc20181ec8850 : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd5616540d684824a79d9d6fe53018f9 : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c5bce5b29d4c2183a982885eeff17f : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc12128515844c5f96586d8a9fb49f5d : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e46a381a7a34a1098d7832cc1417b36 : SUCCESS in 14m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e92c04585aec4afd831a52e50f82cb1e : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da9edf47d489462e9b41ddbf910844eb : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9fb9f3502204fb4b84c9530ef2b1edd : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":23},{"id":"3b19f4a3120d49480cd7798710e5e4733984acd6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-22 10:59:23.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"e9ae3c9562402b22f0acdab3fe626895956dc459","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-22 17:27:05.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"8331967598eb7e0558e4a36d3c999900e119ddc7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-25 11:50:38.000000000","message":"Patch Set 23: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"3dcd86a77566b7e6fd679b529c77137a971a5a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-26 11:34:24.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"77c1b0557fa2280b7abc9d6d1d929a0ca2dc8bed","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-26 11:35:00.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"8fdae39c1e24b7a0fe2249427f8f3facc5d057c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:14:04.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be33c9c877264133a735adc67a7e997f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/475059a651ef41f1881f982e445eb558 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc983f436e274f5fb3ec784b0e5f3d73 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a27570a4565e4990814f902832145117 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31758b48e5124d29b17a0b49f11d2fa7 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3988205b25bf4240a2c67a674b63a8b6 : FAILURE in 10m 33s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31c472d0a1264d95ba72e55a2196f859 : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f5edd9e7880446c8ae3bc3824a3efbd8 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/be562f36a36c4a1da5be5d8caf3b4f26 : SUCCESS in 1h 04m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db31d0ba53ef45bc926f9ad1c5563de8 : SUCCESS in 55m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/beee974f81ee499fb7dfce96c1a4396a : SUCCESS in 1h 12m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dd7747accf894c28bc162dbd11f8224b : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cf30e9c89f6242848247c619bf6aab05 : SUCCESS in 16m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d78f0af29b44b789707cf6d8a8b327a : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0644d079183d4dd59f5122e436bb5600 : SUCCESS in 2h 22m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d41afc1adb7c4f20a365bd9a112a29f0 : SUCCESS in 2h 13m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9841e8a95ae849b0a8b9921f9a15fb29 : FAILURE in 1h 19m 42s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2393462ecf0487ebf09af8ce8f22a1f : FAILURE in 1h 18m 16s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a326c95affa24c3fb327174cc19064be : SUCCESS in 18m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0338da46b774f0aa90b3b65bafc2d25 : SUCCESS in 14m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3322f57dd2e44b8cb53cd27862711fd4 : SUCCESS in 36m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d6fa09091c349b9879ff41c37d7a5ee : SUCCESS in 30m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f35334574644119ad2fd6f0ecb2a375 : SUCCESS in 27m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/97780fa033da4cad9dd75f42527d9e38 : SUCCESS in 33m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00dc9adf45d04b0baae832766a0b872f : SUCCESS in 30m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8d5a69a40db4e6297de21186dec7cea : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aa473edd7924811a584a9e51ab5dad6 : SUCCESS in 12m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1505b672e9c841e8ae333175ac087bc0 : SUCCESS in 19m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d87126c2e654550a8e5417f3219c526 : SUCCESS in 15m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9203a281344642ac8c7d1dbb81920d5d : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/032f7210d73b4c7bbbc60d7e89b407f9 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f9c202ee0534b949e5efc71c81225cf : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":24},{"id":"f2fe10cd671d4dbc04b5d8b410ec42227111575d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-27 15:01:11.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8a994f50f65f95f92ace542f5dec2298218f1dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-28 11:32:46.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"70583e349bf4881d245758b12ae15f73b62a2a40","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-28 12:14:07.000000000","message":"Patch Set 25: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"d59026247271e40124dcacf6c660c287b5c1f8d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:12:12.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e1510daa5c43fb82eda2e70dc7fb09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d09abbe94b498f937da005d15e6079 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5d21c53f3c4e01a64d447744eaac48 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1727873779e84b8aacdc405e70b21bfe : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf62e47f06144a16ada3009730b12ec4 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/580d7df4e3634558b6c410c3c57c3899 : FAILURE in 12m 49s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6d28d8e72da42958c308075b2258fd1 : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a14533532683412fb5c8620388a1bed1 : SUCCESS in 1h 08m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e2068e286f7a49d3ba61ea2ef24b99d3 : SUCCESS in 1h 03m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a07a5da75141cd8e506ff96300688a : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ce0ffddd8ef4ec7ace7c222dc6d2696 : SUCCESS in 1h 18m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e3dca18f609b45f0b96b63cbd63be5fe : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/55219debffc14c2c9e236e58377d881f : SUCCESS in 17m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93b901ed4de24724859b479187f9a99e : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8493c6eae99f4c4bb87684e9c0579132 : SUCCESS in 2h 32m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c884567e0fc4ec8af2259662cdf7123 : SUCCESS in 1h 47m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0d879961d3f54a569b224c5ed6b49d48 : FAILURE in 1h 22m 56s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2af6691a53e4b6aa45a66c7e66fabef : FAILURE in 1h 14m 39s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/900bf33509134e1f8633b07d98f18ee5 : FAILURE in 17m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5f3b5b3b4bf412aa1e642ffc8bbd738 : FAILURE in 14m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dce9455379d04a499d828d89a21478c1 : FAILURE in 18m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f7f15c95984d4898b0cb0a9e96154a8e : FAILURE in 29m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a5cf32e6c1b4781ac03353b66c4c234 : FAILURE in 30m 08s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9e6fe2c65b90485b8663f2fe5bef3b53 : FAILURE in 30m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/354eb6c31eb448e08eb5fd747826e214 : FAILURE in 27m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f7398f9f432d4d11b2e1016d8ff10558 : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d291858a67ac4dac98192654374a8d6b : SUCCESS in 12m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52a0c9e1097849b5a9086a1685b1f496 : SUCCESS in 17m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b3ca8f10d414a6b93c96d6aa0e7c767 : SUCCESS in 15m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cce745dccf147efb3e1241149804dcf : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c4b94334d949dda792b37a10fb54ce : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97c7602c656d4eeb838604b4112b211b : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":25},{"id":"db324b47b415ba557708f90c65db073c51d294ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-28 14:30:30.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"a76b15ae44d078796389634e5d672e411fea8cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 17:00:31.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3b1c40ff9b44caa67c7aee08f0aba6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b269bf2b0ae400ea75cb86150f6e49b : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1785b71794479684c8fd5b4a1944e8 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/affb64bf6a8543f082d5ee4f6a7a646b : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d712e171d33423ebecc6a7a5ade8c0e : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/062b7d138e0b49f6a4029849b973d748 : FAILURE in 12m 07s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22cd317fddad449385d1dc99debd3bea : SUCCESS in 11m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d4913c50e4194c0aa411e66a0e9ce8b7 : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/10322217dd85436ea0049d14a7075342 : SUCCESS in 1h 04m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9941e5bce1c4492994d51f065b32e911 : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/36b6623b05eb42418219cce8c4a68454 : SUCCESS in 1h 09m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/923efef2df1c4e2c91185c94dd88f6d5 : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/45eca850d6c54d0ba96af5059037ea31 : SUCCESS in 16m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/675bf76532a7445093c31f6523f2a0e7 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/66460b5eacad47818b169ed249f73cb0 : SUCCESS in 2h 25m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec8f00deb1284dc1bc227ffb367877a5 : SUCCESS in 2h 08m 52s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bd51a30bdc854be4a00f6ad389f8581f : FAILURE in 1h 21m 33s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c16ace1a787b45dbb62a142f1ac5ad26 : FAILURE in 1h 09m 10s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4ceca6987482437795556f0ef14880bd : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d230ba258547a8b11b7824a904a85e : SUCCESS in 19m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bf630f3c433d4434b7d52a6a05131c5f : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fd8a1108f30b4e9086a6778bc4c9b1f3 : SUCCESS in 29m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d13930b187a4803afbc5d6b768b5248 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5edb7b9f58ed42e58b065e422c8803f3 : SUCCESS in 35m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bbc3dc7fca54c88becf6e9a908e6100 : SUCCESS in 27m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9589ef406e174e409d0dce0d3a65afcf : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e579b7945d114c5d86248a7298a4d3d9 : SUCCESS in 13m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ce84528216a42289eb9aa71a93b0c3b : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b7167a55494445e8833876ffc13db55 : SUCCESS in 15m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/697e6562788949ad9b67d2a26a950b00 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0818468c136c4f1ca28cb4a9b505180c : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/762a2daf62cb4fb6a668bdbc125a8bf2 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":26},{"id":"d0fe60d81e24266d5b7807b1ca4efe957c9c8049","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-03-29 07:46:36.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"278b98fefe0365795acac57393d01c37345096c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-02 10:58:38.000000000","message":"Patch Set 26: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"86e1619e7074d08d3d770725c60f2b68d905db7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2024-04-03 08:21:35.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"b7ef9a8bd0b1288364746cb80b2e1de46875b35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-03 09:28:38.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"c870bf02cb33ea73636cae328a9253d4fa9ce5bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-03 09:29:40.000000000","message":"Patch Set 28: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"b1dd2d96d76239c20d37f901600b1d2236c55c80","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-03 09:37:13.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"183b04f5d82b2013be8b9e99984e278c60df84b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 11:55:23.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17c244fa43847df8b2a782fe9209123\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d11739180af742b39de70c90df0b2b39 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba789e6e08b4641855f76ebf466e5f7 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c33db415d454250959da4259bd93c24 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19e03ab349954a5ba125ef735d751e2d : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/974d39c1223643c0b7e4dc2dd3cf758f : FAILURE in 9m 49s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef14c0d562d041cca0ec959d1f5c47dd : SUCCESS in 11m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f9816279b9e44f8881d499dd0c294b17 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/51670656af2d4cc7898aa95cfc973939 : SUCCESS in 1h 04m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38f51e5d981145cb83df3bea8ec09ff1 : SUCCESS in 48m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/69405c1ff74444df8c4ec927781b7ba8 : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64b074e59ceb48759fe00cb84200f083 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/da14ae8d7a474b45995696f698a489f4 : SUCCESS in 15m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8531ec712a345e88f99c18e6aa8b9b7 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ee89b5df24d74efd87ed0a6bed736840 : SUCCESS in 2h 05m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1519d9da15114c659d08eaa790dc4a50 : SUCCESS in 1h 34m 17s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/13203e1989ec4cab8fa182a6907b2001 : FAILURE in 1h 35m 44s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b9c4e776a3f465da68ebc256092646b : FAILURE in 55m 45s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/aab89f9fe3d2441591e471c58ca04669 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/014ad09f72a54cec9d2e60e90807e2e5 : SUCCESS in 13m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2a485dbf51f443f2bd7f739653bb3d0c : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9dd63160a3fd4402a720b2b3787fe1c3 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7305fe1f08d4507b8bfd95405539181 : SUCCESS in 24m 00s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/97e58412139e4896bcae9b742c425d8e : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a165d7b19b9b4137b0b28bd1ebaa9bd4 : SUCCESS in 25m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cad5d4f1849d44d2a5d6815edac710da : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54c933a9a8e450280ca13c0a7124c0c : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4628abaccdfa4b78bb781e6968463491 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f93770296fa4976820a0d458bee7e45 : SUCCESS in 13m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed897e12299e46328883c88d86c3c60f : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f9ed59119da492e936cc5c9557ef155 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2452d8fd7244cae839ce4baa37fb2b9 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":28},{"id":"0cfca3b51d46efca0751c1a3d4e7f11d8ed07a5e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-04-03 15:20:25.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"252c06e9b26baa83f2548d990fceedc665a879d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 15:43:47.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"b9fd2dfa88daac3b498577cd497a9fa3d9111bcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 18:30:17.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e86c42fa48343998a96a10d9b5f404b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a16cb2db224b128d1d8562ee4a55fc : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80fd18ca0167454084ac40d6add5fd5d : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e8fb5b538ce431abfa913172d3e5ff5 : SUCCESS in 4m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0dabe27767b046699f5d03abcb11fe09 : SUCCESS in 10m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7fd309a9bbb746f389417b4ac73a7060 : SUCCESS in 59m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6658493d3d0546f0aacd8207639efa57 : SUCCESS in 54m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/90337431d2e64a9cb62c583f4ecfcb7c : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/64288ced3a424ed6bed9ea32979d55a2 : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e2209566fca4a2ca506db7e7b59a780 : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/562812e9cc724de089b1dda19fc1bc11 : SUCCESS in 2h 34m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef4ee2d1b57d4f4a96d9a7e87c1e8474 : SUCCESS in 1h 58m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/312ce0c3332e46fa982b8fe9111f65b4 : SUCCESS in 17m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31b5749c61fa4e039c1593f197475d1e : SUCCESS in 14m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef4ad5f1a2964a09a98f02c7a6ca159a : SUCCESS in 28m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f81a8858cdc48209c4d24b23a70d38d : SUCCESS in 23m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a64a1e3d01474a88a02fb51f5cd93cae : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3b81af3ed244500b7279fe61f24f6dc : SUCCESS in 25m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fadb4863873f4f34a7006ec6180b1232 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fad1fa2899d64bc4b0ca7be9ccb70d3b : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c2a3bf6e715a4fb6a2468f7ccbde0a22 : SUCCESS in 17m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc88429585f944c0a1ceed046c973fc5 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":28},{"id":"645631fa5dc1c7f05b3e74140c37848b3bf954c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 18:30:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"bbba84b4fdd844b261d92cb45139ae70eed70242","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 19:12:54.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c77cfc3a6b4f0e84bd03ad9558ae34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad05eb933ce64401928c3bf3241f9d4e : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a6babad62b452789e3b7a7a1d995f3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"be2b1a4eda39892e56fb5317a074d377ca2ba157","revisions":{"40576194dd95d196e9634e17833d870d012db323":{"kind":"REWORK","_number":1,"created":"2023-12-06 13:42:11.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/1"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40576194dd95d196e9634e17833d870d012db323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40576194dd95d196e9634e17833d870d012db323"}]},"branch":"refs/heads/master"},"176b2890fc3f88479080d269b351c5c2c53a3210":{"kind":"REWORK","_number":2,"created":"2023-12-11 12:43:34.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/2"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-11 12:43:19.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/176b2890fc3f88479080d269b351c5c2c53a3210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/176b2890fc3f88479080d269b351c5c2c53a3210"}]},"branch":"refs/heads/master"},"f2537bb5b2ae2c5754dbcefe18fa63e119876813":{"kind":"REWORK","_number":3,"created":"2023-12-11 13:00:29.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/3"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-11 13:00:20.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2537bb5b2ae2c5754dbcefe18fa63e119876813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2537bb5b2ae2c5754dbcefe18fa63e119876813"}]},"branch":"refs/heads/master"},"77ea23cac1f01922e949b9dc533e3bbd2912078d":{"kind":"REWORK","_number":4,"created":"2023-12-20 10:40:43.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/4"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-20 10:40:33.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77ea23cac1f01922e949b9dc533e3bbd2912078d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77ea23cac1f01922e949b9dc533e3bbd2912078d"}]},"branch":"refs/heads/master"},"bb4dccda008615f184627824d0a69f9415918a9d":{"kind":"REWORK","_number":5,"created":"2024-01-31 09:36:22.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/5"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 09:36:12.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb4dccda008615f184627824d0a69f9415918a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb4dccda008615f184627824d0a69f9415918a9d"}]},"branch":"refs/heads/master"},"e451d65bd6ab5222e8afb57929d76bfb01411f10":{"kind":"REWORK","_number":6,"created":"2024-01-31 10:01:57.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/6"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 10:01:49.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e451d65bd6ab5222e8afb57929d76bfb01411f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e451d65bd6ab5222e8afb57929d76bfb01411f10"}]},"branch":"refs/heads/master"},"0d48eed039454445049430cc2695dfe869723780":{"kind":"REWORK","_number":7,"created":"2024-01-31 10:04:30.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/7"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 10:04:19.000000000","tz":60},"subject":"Add Redfish rules to Ironic introspection","message":"Add Redfish rules to Ironic introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d48eed039454445049430cc2695dfe869723780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d48eed039454445049430cc2695dfe869723780"}]},"branch":"refs/heads/master"},"df16e86eab784e9d4b0c4cd206df451d3c575eb6":{"kind":"REWORK","_number":8,"created":"2024-01-31 11:20:31.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/8"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 11:20:00.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df16e86eab784e9d4b0c4cd206df451d3c575eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df16e86eab784e9d4b0c4cd206df451d3c575eb6"}]},"branch":"refs/heads/master"},"d7eeed99cd541c6503c23927c7ecf572ad89ffa5":{"kind":"REWORK","_number":9,"created":"2024-01-31 14:23:59.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/9"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 14:23:48.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7eeed99cd541c6503c23927c7ecf572ad89ffa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7eeed99cd541c6503c23927c7ecf572ad89ffa5"}]},"branch":"refs/heads/master"},"96d4c3f2ce70441f1836e58f704bddb5acd85e18":{"kind":"REWORK","_number":10,"created":"2024-01-31 14:33:58.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/10"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-01-31 14:33:49.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d4c3f2ce70441f1836e58f704bddb5acd85e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d4c3f2ce70441f1836e58f704bddb5acd85e18"}]},"branch":"refs/heads/master"},"3a04debd7c61a346a572379adac2622729f10e17":{"kind":"REWORK","_number":11,"created":"2024-02-02 14:15:14.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/11"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-02-02 14:13:19.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a04debd7c61a346a572379adac2622729f10e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a04debd7c61a346a572379adac2622729f10e17"}]},"branch":"refs/heads/master"},"37bfd155855a998582ddb4a7443216e82823e04c":{"kind":"REWORK","_number":12,"created":"2024-02-05 06:42:05.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/12"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-02-05 06:41:46.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37bfd155855a998582ddb4a7443216e82823e04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37bfd155855a998582ddb4a7443216e82823e04c"}]},"branch":"refs/heads/master"},"d4c302763e8195e3e86a8d12ac6e2aeb8fb76328":{"kind":"REWORK","_number":13,"created":"2024-02-12 10:41:37.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/13"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-02-12 10:41:26.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4c302763e8195e3e86a8d12ac6e2aeb8fb76328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4c302763e8195e3e86a8d12ac6e2aeb8fb76328"}]},"branch":"refs/heads/master"},"518fa2eaed468482a8bf1e166a55f7a0bcdc87c1":{"kind":"REWORK","_number":14,"created":"2024-02-12 18:33:50.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/14"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-02-12 18:33:24.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/518fa2eaed468482a8bf1e166a55f7a0bcdc87c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/518fa2eaed468482a8bf1e166a55f7a0bcdc87c1"}]},"branch":"refs/heads/master"},"f80fada2d18e21aa9da0a56eb4a3fefaca8c3e94":{"kind":"REWORK","_number":15,"created":"2024-02-20 12:32:54.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/15"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-02-20 12:32:44.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f80fada2d18e21aa9da0a56eb4a3fefaca8c3e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f80fada2d18e21aa9da0a56eb4a3fefaca8c3e94"}]},"branch":"refs/heads/master"},"440683f9f98e9f0ebab018dc268c3d6ee5a25e78":{"kind":"REWORK","_number":16,"created":"2024-03-05 14:23:29.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/16"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-05 14:23:18.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/440683f9f98e9f0ebab018dc268c3d6ee5a25e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/440683f9f98e9f0ebab018dc268c3d6ee5a25e78"}]},"branch":"refs/heads/master"},"8f652e02342dff0cdd11d68973b6413410a46e33":{"kind":"REWORK","_number":17,"created":"2024-03-12 08:09:20.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/17"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-12 08:09:10.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f652e02342dff0cdd11d68973b6413410a46e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f652e02342dff0cdd11d68973b6413410a46e33"}]},"branch":"refs/heads/master"},"c6708c95b4b9cdc4fbb139f6b293e8dbb66130a6":{"kind":"REWORK","_number":18,"created":"2024-03-13 08:29:58.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/18"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-13 08:29:49.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6708c95b4b9cdc4fbb139f6b293e8dbb66130a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6708c95b4b9cdc4fbb139f6b293e8dbb66130a6"}]},"branch":"refs/heads/master"},"cfc6b8e02c1ddb829bd7ce9dd038c4d6df9db1ae":{"kind":"REWORK","_number":19,"created":"2024-03-13 08:39:00.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/19"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-13 08:38:49.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfc6b8e02c1ddb829bd7ce9dd038c4d6df9db1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfc6b8e02c1ddb829bd7ce9dd038c4d6df9db1ae"}]},"branch":"refs/heads/master"},"bafcde83954aa63ed572c699c6df41cdd820d43a":{"kind":"REWORK","_number":20,"created":"2024-03-13 08:40:57.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/20"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-13 08:40:49.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bafcde83954aa63ed572c699c6df41cdd820d43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bafcde83954aa63ed572c699c6df41cdd820d43a"}]},"branch":"refs/heads/master"},"d0b39080e45b6361fbd0727ada645e762d7b2449":{"kind":"REWORK","_number":21,"created":"2024-03-13 09:30:22.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/21"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-13 09:30:11.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b39080e45b6361fbd0727ada645e762d7b2449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b39080e45b6361fbd0727ada645e762d7b2449"}]},"branch":"refs/heads/master"},"f23ff547143a2bdfa65297f8a272e1b20560dcda":{"kind":"REWORK","_number":22,"created":"2024-03-20 11:31:13.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/22"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-20 11:31:00.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f23ff547143a2bdfa65297f8a272e1b20560dcda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f23ff547143a2bdfa65297f8a272e1b20560dcda"}]},"branch":"refs/heads/master"},"e599dca1e8f51b7bf217b6bc0cf5bf9633262f06":{"kind":"REWORK","_number":23,"created":"2024-03-21 14:58:12.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/23"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-21 14:58:03.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e599dca1e8f51b7bf217b6bc0cf5bf9633262f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e599dca1e8f51b7bf217b6bc0cf5bf9633262f06"}]},"branch":"refs/heads/master"},"cc656e7e42e5e078da850d3b40e61058b4481b85":{"kind":"REWORK","_number":24,"created":"2024-03-26 11:34:24.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/24"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-26 11:34:16.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc656e7e42e5e078da850d3b40e61058b4481b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc656e7e42e5e078da850d3b40e61058b4481b85"}]},"branch":"refs/heads/master"},"e1b59815a7ba122211a14baf630c1ca90e707c59":{"kind":"REWORK","_number":25,"created":"2024-03-28 11:32:46.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/25"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-28 11:32:36.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1b59815a7ba122211a14baf630c1ca90e707c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1b59815a7ba122211a14baf630c1ca90e707c59"}]},"branch":"refs/heads/master"},"c78ec0b4ef87dc8fb892782924bc42652202522c":{"kind":"REWORK","_number":26,"created":"2024-03-28 14:30:30.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/26"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-03-28 14:30:22.000000000","tz":60},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c78ec0b4ef87dc8fb892782924bc42652202522c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c78ec0b4ef87dc8fb892782924bc42652202522c"}]},"branch":"refs/heads/master"},"7cec11ddb7c3b55470440879cb9fbd19a712fa0b":{"kind":"REWORK","_number":27,"created":"2024-04-03 08:21:35.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/72/902772/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/27"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2024-04-03 08:21:26.000000000","tz":120},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cec11ddb7c3b55470440879cb9fbd19a712fa0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cec11ddb7c3b55470440879cb9fbd19a712fa0b"}]},"branch":"refs/heads/master"},"be2b1a4eda39892e56fb5317a074d377ca2ba157":{"kind":"REWORK","_number":28,"created":"2024-04-03 09:28:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/902772/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/902772/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/902772/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/902772/28"}}},"commit":{"parents":[{"commit":"46bba8eb27d9a601bf74e7367f4f20ef769e1cb7","subject":"docs: the page \u0027Editable installs\u0027 has moved","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46bba8eb27d9a601bf74e7367f4f20ef769e1cb7"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-12-06 13:40:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-04-03 09:28:35.000000000","tz":0},"subject":"Add Redfish rules to Ironic and Bifrost introspection","message":"Add Redfish rules to Ironic and Bifrost introspection\n\nChange-Id: I719fc8042742fe8b3b0312658aec39317a1bc358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be2b1a4eda39892e56fb5317a074d377ca2ba157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be2b1a4eda39892e56fb5317a074d377ca2ba157"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
