)]}'
{"id":"openstack%2Fkayobe~903142","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~Ifd9f9cd389ac09448fc91dba74f8f09595698b55","project":"openstack/kayobe","branch":"stable/2023.2","topic":"fix-bifrost-gw-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-08 18:52:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd9f9cd389ac09448fc91dba74f8f09595698b55","subject":"Fix gateway assignment when seed SNAT is disabled","status":"MERGED","created":"2023-12-08 10:51:59.000000000","updated":"2023-12-08 18:54:03.000000000","submitted":"2023-12-08 18:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903142-fix-bifrost-gw-stable/2023.2","cherry_pick_of_change":902113,"cherry_pick_of_patch_set":3,"meta_rev_id":"fe4fb8ea7afecc48b474233e9099f33bafc20ec8","_number":903142,"virtual_id_number":903142,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 18:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-08 12:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-12-08 16:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-08 10:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-08 13:35:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-08 16:08:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 10:51:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"CC"},{"updated":"2023-12-08 10:56:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-08 12:18:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-08 13:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-08 13:35:19.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-12-08 16:08:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1faa1f370536979160672a3bd74518dd6bae54a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-08 10:51:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"797c0e0ea1a843021867f03c5c6ad5b449ddac15","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-08 10:56:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d7f696c598aec29b87e208cf70062ecbdda82e2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-08 12:18:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c60f93a10d525df033df47d51e8eae72c796458e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:10:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/564be274abdd48ae80d178089c3912f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47a7d8ba91e4ad1a69fe513e928f920 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac3ef77b92a44609e7c27dcc0f6722f : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67f03766c90487d8039bfd566cf486a : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f866da19a4d48d194cef107d584bc49 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/079bb1fc4b4a431cb5f7bf07b36b78d2 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74b6253b887846b8a0c18db720134e92 : SUCCESS in 12m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/66e1e5eea5d24256a7b53db6882d7b4c : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fb77ea412dc84c648f1dcf89940632f5 : SUCCESS in 1h 01m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee9af7b98cac44b9a41afc5be51b2cb4 : SUCCESS in 51m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2ccc568c4c4c45cd8acd9f7b2512b224 : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6f73b35bd30f433087afe123a871eaab : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4d5d533b1b62498da6543b73e8930598 : SUCCESS in 16m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efd0d99a2a0f42918bccce4e62abc925 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1ff6ac5b65b4409196c8e00b490cdc0c : FAILURE in 1h 25m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd77e56eb3e5457996653c4cdb8c190e : SUCCESS in 2h 12m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/05cca67847a8449b82aadd2d911cb9db : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/160232dba0cd43d58b5d374422aedeab : SUCCESS in 15m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/44e48d71ac7f4233a968fb2e56c2b0c5 : SUCCESS in 37m 23s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6842ce386e4e4c60b9a9660e5c14b30a : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95546f3e46a54e19bf4fa36e6b135429 : SUCCESS in 22m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/abadc638c3c44c179e74590737c55ecd : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dfc2423769c48cca57d2c3425727c7c : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce9339e5665a4b9b9887ca310df82d6f : SUCCESS in 17m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1e96c6a585d4565a262a71fdb6d4f6f : SUCCESS in 14m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f94316327544f0a123a8b3152af1cb : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b81c2885704341f4a8f8bf91113071e5 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9782a13e8a7a443d87036b236042a019 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"aaa300b496c57a551105b8749ffbd64a084c011b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-08 13:30:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36f26f31c0cc2bf316ab81a5ba74ac6673c79237","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-12-08 13:35:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fa67ee3986ad50a024fde98d98566188579822b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 16:03:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/099dd8abf39c4dc4a92a02502097865e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/377c798fd2984917bfdf6243cb206485 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1777d0470e14450fba025cd402bb6214 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35accf129e4740d2b543c4585175c507 : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/553496d17de34931a04d7c6e56e27e44 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c53647aa5b1f480cadc28956896c7887 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/349031b46d1b4584aa8a31fa540e55d7 : SUCCESS in 12m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7d98a955766b4fe0a6929ed436c02346 : SUCCESS in 1h 42m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9950bc7ddc91485382a3d24fa86f52c9 : SUCCESS in 1h 04m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd7e1058397b46dfac8a96c434a2b56a : SUCCESS in 48m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7dfbbaf841704a22a3cff5e225bc7878 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/97f9ac6779924662bf59abe6fee601e2 : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d48a7ec7ee434dfa8293ea1a8864acbe : SUCCESS in 17m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2241d798dc4df08c19a5da8fab61bc : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c626038fddd4e1b8083d8f6b30208d2 : SUCCESS in 2h 27m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6685237fbe954c189088c47e6147db0e : SUCCESS in 2h 19m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/11c5c82683be4ff1bbcb12be6fd5f3ec : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/511ea0758b744bdeaa9b991cc73e35c4 : SUCCESS in 15m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/405543398db24d3097ae20a4d114c537 : SUCCESS in 34m 41s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f94748bd5a0d4ce7a769b7d983ca4c31 : SUCCESS in 31m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c38bb107c0004c078c0161c25e51b8b1 : SUCCESS in 27m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/abfabfd296564464b3ecdfb1b7b4dbc5 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/799908b8da8b481eaab33117414f8213 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a582150f735b4453ae2e9cfa4f64c87b : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a546dfe0f72f4220bfc8419c0617061f : SUCCESS in 15m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e8ed9d83d7c4315a7e108a00ac8bf0b : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b541e0f6db44efbbb9edec4b3f07646 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1fc569761834721adb4550923d04652 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2e8ce69ce0645ee6e98afb8a46bad5c186d9d79e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 16:08:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bbf4ca4ab98229b763a3929df66612be79767ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 16:08:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"487274f558fa45f12e924646a385f27a0bffa07f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:52:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ce24cf89184575afcccedc2b1a9c73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99293c2e2d7649c7954ce750627734f6 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/881f758ffcef448c873a0a8b65d9bd2a : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f413073b2bf4274a48c6c179b77f97d : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/490056a11dae4eb8b6aa160795a026c0 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e917f4f727d4b4caf715284a19185b2 : SUCCESS in 10m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/35f1ea603e36429a9a6082d085324c26 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a135c8096e842d89d7ab65d934f9100 : SUCCESS in 56m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2ec2161f532e44fd95a70e1025ce52c2 : SUCCESS in 1h 14m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a11328a021f46c3a06724bbd4abce18 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e645d54f3ec48b699a50bf2f16bbb2a : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ab894ccbd1d48bf89754981cd904ba6 : SUCCESS in 2h 40m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40d1401ffe634e74a05703b6e03a2c0d : SUCCESS in 2h 02m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/899acfd1efa94bd78e122f99c7a5ed94 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e71e6834959f445abf285a32560a9907 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b4deda3303b489fb86c4627e9d6a630 : SUCCESS in 33m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfed609602874cc3b85dabee195995a5 : SUCCESS in 24m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc0cd915467348498fef7ffe45127bc6 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2ff52d6737a40208b23a192172e8b8e : SUCCESS in 12m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a754264adad4bdd9dc4a8be9dcfe61c : SUCCESS in 16m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91244180351a4e66bb3ebb7d9b915bb4 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"95b69021fe28ea7b743614673655d997fce5ddd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:52:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe4fb8ea7afecc48b474233e9099f33bafc20ec8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:54:03.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bd0d5d6673643a4aff14c3e73d4f453\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc8f4799ac04e8b8b2873239558b5a2 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bacc4a83a0f94b2e81281a75bac679ff : FAILURE in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56e592b6489dbfc21d2071b24c1ca3767bc52e91","revisions":{"56e592b6489dbfc21d2071b24c1ca3767bc52e91":{"kind":"REWORK","_number":1,"created":"2023-12-08 10:51:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/903142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/903142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/903142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/903142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/903142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/903142/1"}}},"commit":{"parents":[{"commit":"c318ffec56dc187ad006dec953d49ea6be892d7a","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c318ffec56dc187ad006dec953d49ea6be892d7a"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2023-11-28 21:55:32.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-08 10:51:59.000000000","tz":0},"subject":"Fix gateway assignment when seed SNAT is disabled","message":"Fix gateway assignment when seed SNAT is disabled\n\nFix a logic issue when seed SNAT is not enabled. In this circumstance\nBifrost was generating ConfigDrive data with the default gateway unset\neven when one is available on the admin network. This regression was\nintroduced in Ib847d3420dee374cec067cd8af519b510be04120 [1].\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/898434\n\nCloses-Bug: #2045926\nChange-Id: Ifd9f9cd389ac09448fc91dba74f8f09595698b55\n(cherry picked from commit 49f91043c6c3b0549a8625baba1e6e218eb69160)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56e592b6489dbfc21d2071b24c1ca3767bc52e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56e592b6489dbfc21d2071b24c1ca3767bc52e91"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
