)]}'
{"id":"openstack%2Fkayobe~903145","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ifd9f9cd389ac09448fc91dba74f8f09595698b55","project":"openstack/kayobe","branch":"stable/yoga","topic":"fix-bifrost-gw-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-08 16:08:27.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-12-08 18:52:44.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-12-08 18:52:44.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-08 18:52:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd9f9cd389ac09448fc91dba74f8f09595698b55","subject":"Fix gateway assignment when seed SNAT is disabled","status":"MERGED","created":"2023-12-08 10:52:54.000000000","updated":"2023-12-08 18:53:54.000000000","submitted":"2023-12-08 18:52:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903145-fix-bifrost-gw-stable/yoga","cherry_pick_of_change":902113,"cherry_pick_of_patch_set":3,"meta_rev_id":"b3c9089cecda00b8bd6d2d23bbde258d42d985dd","_number":903145,"virtual_id_number":903145,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 18:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-12-08 12:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-12-08 16:08:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-08 10:57:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-08 13:35:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-12-08 16:08:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 10:52:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"CC"},{"updated":"2023-12-08 10:57:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-08 12:20:15.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-12-08 12:21:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-12-08 12:21:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-12-08 12:21:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-08 12:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-08 13:35:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"97a8cbc9238b1a51e4b10931c781d335988c88d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-08 10:52:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f34c0f75a4f373eec59bbe56e88456a7c7702db3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-08 10:57:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60fa63a7d8cbee1c5e063fe4de32075ddf77b8f0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-08 12:20:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3493166a9ba64d8d10763b5ae1a494e3fcb4830a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 12:24:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cec9033e2b34b45a776c6915cc3dcbb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4860573dd9f546469e1b1d581dbcc423 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3008d21ae64a48bbd600b0525f4965 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46fa2af9862f4fa3ac43a8b075e258ca : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/132d5ebf7df545fbac9a5b472b0a5c71 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82802cee3201431caa847bacc90904b9 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b150724ae18d47b59c9c4e8435633f2c : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30ab9ff338884fb0aef7234c93f3347f : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91d998ecf67345dfb99f09561c36b6dd : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55ec1f9cc8ec44a2ae831f7860d69dc9 : SUCCESS in 20m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1e00978fbc24460a82bcdd7c3cd8def2 : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/78312927a1d643a68ee5113a3e3c405f : SUCCESS in 52m 39s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/77c1a428b1834b61b9a78e8525afb944 : POST_FAILURE in 59m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/789953eeed2e49d0b1127a1ebb914a60 : SUCCESS in 49m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e22df6c8e01c4b6eb8fcb41ba9890b85 : SUCCESS in 51m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c0e251bb3de44ce9361529aa44d2aee : SUCCESS in 53m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/201467ce4cb745b5bc48b2852644ea20 : SUCCESS in 29m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/53bf22df74fb4cfeae66f0de9de22fcb : SUCCESS in 22m 40s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/984beab22ccc4c9c9a2601654b729495 : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/492dd81a0b884f64954654efdf037ccc : SUCCESS in 23m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c83461403064ed3a99d4234d243e3ac : SUCCESS in 16m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19b833a4616a40919be6a2540b4a56f5 : SUCCESS in 16m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/43a8bc976d0c491da69f39d282e26548 : SUCCESS in 1h 24m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/87b94ab7a5184e2e859c462dc24738db : SUCCESS in 1h 00m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3372a96f4f0b42869daddce339518a0a : SUCCESS in 23m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/42ad900de5544f338b5a02347d1c7c63 : SUCCESS in 28m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1973c8b8f8d04f448c469ba25485504c : SUCCESS in 18m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82a498a7692e4feeaae7d95f981a52e2 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b2373c0c9824447900bed80082ed030 : SUCCESS in 20m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c109871c424b4ff59eda58ff28c678d9 : SUCCESS in 20m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a2beea74ff384e4cbe8b20a64cdb2f97 : SUCCESS in 15m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d0130a993b1463488b005349d4f7101 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d611de52271148ee9256103734ba2126 : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb49c2c3ca3d4fd0bb0d499625217aa4 : SUCCESS in 14m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2aa8f7d1f1d42f38e5d8984def6ddfa : SUCCESS in 22m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f5722e4f871746d291b9dcbce2856f2f : SUCCESS in 20m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e345560e09a4cb3a34a006a1317dccf : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62f1f8f9904b421c9da9311e61cd7250 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68b923fd10dd4b7e81a2c2ab6c67ae25 : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0ca46f8e0986cecf144453c586083cc49ee3a255","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-08 13:30:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a6197fe0e289ec629e2a01cae2f3fe8e1eda903","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-12-08 13:35:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6787ae2622fa2144d1bd3e802aaf1160fefc7fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 14:55:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b524897be9884cfead0a2e1f5a7b0813\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6260b269d5d4229bbf97b6f00a6d8d5 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217e5ebfebce48228493eb99e287afe3 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50b5542d99ab46a1a2398956170663b8 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8054843e60694ae08afce84071ca955f : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95454e6dfa2144db8d8320ffaa62a164 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a37d49012cb24b029bb7ffb6f9cd21e6 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9af21b14d9a649e18f77f82c0314b931 : SUCCESS in 9m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e19f41c7bb10419d90d35b4a82c76c35 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30a4d322cbaa4131b9ffde797a6373d8 : SUCCESS in 21m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0c0ba235518342c082ab05a8924f0bc9 : SUCCESS in 1h 13m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3657286008024f1cbc9eb46dab658a7b : SUCCESS in 53m 28s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ae14f5108d1e49c9ac5cb01966ef7ec3 : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/619c5666368647b8af2cd2a3e09d01e9 : SUCCESS in 51m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccb1b4160ced4aa48980219e6ce9eade : SUCCESS in 52m 58s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca379f2104cb4b15bb6ca097a319644b : SUCCESS in 52m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/40501ad529204eb99245845a6dbe1e60 : SUCCESS in 26m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/aef063a531704ffdaaca3ca6d86ef224 : SUCCESS in 20m 28s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/17f70a0409f74364a4d63525bd18a799 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/91540c21cbab4ed59cf46e4ed0e03a06 : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba890fd0904b47a486666157e281b0be : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71a70fb4df244810b069ed4656360a5a : SUCCESS in 16m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6992bdc640cc4f909b677a4e5a703808 : SUCCESS in 1h 16m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/04f3128a3ea5449694fca18502b9ee2d : SUCCESS in 1h 02m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/261f0ae972f74cbdb98868bdf58ba555 : SUCCESS in 23m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/731756b571a5416fb3a9fac1842ff12c : SUCCESS in 21m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/848f353dd49041ebb8005b52719f4b8b : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f328b8b6235748d6be2193206b112b08 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbf4148ad75b479087c78dd49b2dbc4f : SUCCESS in 16m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a50f6fb02c0246eb8ded2858eb1858b2 : SUCCESS in 18m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cbaa0fb32ebf478eb737a0a4a51d35ce : SUCCESS in 17m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5485e48e756446c7b9bd322028474eba : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58d89b6854445a5b9cee9b2fdb30172 : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf4507186faa461299abe911069ff134 : SUCCESS in 14m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/072dc6f1cbd04c8bb18d7ca309d7d4db : SUCCESS in 22m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c30e8a3ab9c642de9eca1f2f8a3ce689 : SUCCESS in 20m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f2a30ba17b44ef1a900be352569d0a7 : SUCCESS in 21m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13277536cabc419e9ed2b338a1d499a3 : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4b2d83b5f1e42deb07085fa19c2c5ea : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a84f01f7b516c91171a3b12113720a67790e5582","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-08 16:08:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"483b23439a240bd3126c57ddb57c6ec40cb87468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 16:08:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07f164c1eae61cf0a5b2d72451efd3eed34759e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:52:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564e31fe042341769d8edd1f4a0f61f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8168b3b42518415186e86969c4b4beb8 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff2f57a4cf74b4c804980156977a8d0 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92cd04d45f88445592a4ea4e2fb559e0 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fa5d041f7c4e86b41294af74d99767 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fd4042c28cc44ec9e15f17bb9128510 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d34238b3e1d0437782e4b6137683e3a5 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/745bd1e45b534ae28580089f35c33f68 : SUCCESS in 18m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ce4e928870c432d9c30abc483dd29d9 : SUCCESS in 48m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3966660adbcc411a8bb10d198dd0019e : SUCCESS in 1h 16m 03s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/584d586ad7ff42518ea03992881008e1 : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8ecdee21816545bab308f0f921e56559 : SUCCESS in 48m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c85cc7e57f864379ab1ef99923b9a51a : SUCCESS in 52m 12s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e92851f8676140afbaaf962861c2993d : SUCCESS in 55m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f898c0c2c63f4b3d84d0073830084b29 : SUCCESS in 26m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f6d593aabc8c4b398999adca8eff7479 : SUCCESS in 21m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d452b4ab76e244618e3ea95067d1e210 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334712875926448d9a08091828289407 : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbaec2b6f11249cf9bf2cd5355e6c409 : SUCCESS in 17m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87a43cc25acf4b4f87078f8e6375b3d8 : SUCCESS in 1h 15m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7d1ffb0588194801bfd8818ded5475bc : SUCCESS in 55m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d4cb8594dd7f44dabb80f8f5e8217442 : SUCCESS in 25m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/defd19a1604a48a49a4da24816475a67 : SUCCESS in 22m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9b0eababa4c843918eb01b2a9d03f427 : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13b2e75e3dc1461f90c5c1ca18f35a21 : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b131ecd0c994494a8764152fdc9e0fae : SUCCESS in 18m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0bba6fbd3f9c48c6b69987968b1ddc52 : SUCCESS in 18m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f6c2ac2e5a274dc3a2fbfaa2c6b34ce1 : SUCCESS in 18m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0585516f19504c0bafad9041a898e338 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ac6fd93d31f411098dbcf9e027acc15 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87b85af0399243c592fbbe00eacbe463 : SUCCESS in 13m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ade6962b3e24b2e95fa46a8ce3ecbad : SUCCESS in 21m 18s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3afa51accecf456398483ecf0ba13cbe : SUCCESS in 22m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73c2e5d3a87a4fbe813bf2d0d6362768 : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a32641a5f1cf44da9c7e104ba6f0e8b5 : SUCCESS in 17m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea0ad91ceedf4711851b157e1e246278 : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4527419e7c8c5ee4c50a0f0f63b557d8f1bbbcbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:52:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3c9089cecda00b8bd6d2d23bbde258d42d985dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 18:53:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdfaa5fd94404d53a206f7969e3929b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9658911dbf4ebfa217936536185198 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e3e0871b505407a8fbe8bbd97671859 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c625dd590e0b0a1cdcfe0ec6b34dfc147b6b7c6","revisions":{"2c625dd590e0b0a1cdcfe0ec6b34dfc147b6b7c6":{"kind":"REWORK","_number":1,"created":"2023-12-08 10:52:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/903145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/903145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/903145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/903145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/903145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/903145/1"}}},"commit":{"parents":[{"commit":"fcdaffb70c7a4ec8cb69daefe547372c58027eec","subject":"Merge \"Clean up release notes\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcdaffb70c7a4ec8cb69daefe547372c58027eec"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2023-11-28 21:55:32.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-08 10:52:54.000000000","tz":0},"subject":"Fix gateway assignment when seed SNAT is disabled","message":"Fix gateway assignment when seed SNAT is disabled\n\nFix a logic issue when seed SNAT is not enabled. In this circumstance\nBifrost was generating ConfigDrive data with the default gateway unset\neven when one is available on the admin network. This regression was\nintroduced in Ib847d3420dee374cec067cd8af519b510be04120 [1].\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/898434\n\nCloses-Bug: #2045926\nChange-Id: Ifd9f9cd389ac09448fc91dba74f8f09595698b55\n(cherry picked from commit 49f91043c6c3b0549a8625baba1e6e218eb69160)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c625dd590e0b0a1cdcfe0ec6b34dfc147b6b7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c625dd590e0b0a1cdcfe0ec6b34dfc147b6b7c6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
