)]}'
{"id":"openstack%2Fkayobe~903249","triplet_id":"openstack%2Fkayobe~master~Ib9b9b531973dd6e32fd15b645e33384b00e64fe0","project":"openstack/kayobe","branch":"master","topic":"ci-bifrost","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-10-07 15:25:01.000000000","reason":"A robot voted negatively on a label"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-24 13:10:59.000000000","reason":"Change was marked ready for review"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-12 10:09:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib9b9b531973dd6e32fd15b645e33384b00e64fe0","subject":"CI: Test overcloud provisioning with Bifrost","status":"NEW","created":"2023-12-09 08:10:30.000000000","updated":"2026-10-07 15:25:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3f1b03fdd68ca36d2b8ff415aaf4dcc5f96b4472","_number":903249,"virtual_id_number":903249,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-07 15:25:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-09 08:56:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-21 00:11:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-16 11:40:09.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-06-11 09:24:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"50c0184e2d6a9d5a351c101bdded6baf4d2e4907","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 08:10:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73775ae30a5297506e0ffb7b028a8d97a13699fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 08:56:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3efcc94e3f4c5cb811fd42c8e6555a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94c9d782daa44fe6b6430c4489173d62 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cfdd79334849cc92e89524d1dc5df8 : SUCCESS in 7m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a588de02dab3481a93b6cc7417d7e70e : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6c9995d4ff528922894c5cec8826918a00139d95","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 09:00:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"720da9355bbc72e3d1184a9ecd6629fa4d2603a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 09:45:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/715279dfdd43437ab42b291030eb2d5a\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/50e84e1e2bfe4180b3e8000bb3650d20 : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6424847c00ed22573557b15232a580d5f1a87416","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 10:18:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"35af40425edfd3589ed9121f280d187e2afb16b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 11:08:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b099d971abd848fb9a7650101ba44ad1\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/279c7073d45f4184b452457859028141 : RETRY_LIMIT in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"df34ffb2ad83b382f6287f5cb3010258fcf1f826","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 18:46:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f7b28ab5186d840494b8a881530e8464c425c54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 19:21:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6296ba635a94aaa8316bd670e826f3a\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/74dc72607419418583877d71212a9c34 : FAILURE in 29m 59s","accounts_in_message":[],"_revision_number":4},{"id":"69806c3bcedf8349d7b0c976851d091109f7d796","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 20:43:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"82a0ec00f2e1599309d2b05f58b319f746da3ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 21:04:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37bb44162f8948d783cb0baf333dc8bd\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f1df5125abf74dfd8599cb9566a5ee8f : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":5},{"id":"afe7337ba375d8f82c7d130ae4b7dd078ea06555","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 21:06:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"808463d605b05e8c9dc54486b970459edf79ddf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 21:49:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed2dd7bef114ffa93e011a9078fbca7\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/246e4ca512b7464c9f886aba42a8850f : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":6},{"id":"b2b854e94343c9142fd6b1cf8a18e398fd9f9317","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-09 21:53:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1dadcdbdfc1f3bf2b995af98350babdac6ea646d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 22:32:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9c8f6b7034c4a948dbc533cc8ca4b5d\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c85a47d92ff44bb7823b85cc2c9d685d : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":7},{"id":"0587a6cc324fb1e1e27ecfc8816879f88e718517","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 08:09:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"9a921ce8c9c8dfa56b54a31a96d0823f3e2ed372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 08:53:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f940c370666a4de6bca92204399073e9\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c61dfd18966c45c99f48f8fe3efa57e7 : FAILURE in 38m 05s","accounts_in_message":[],"_revision_number":8},{"id":"019e478b73693bbc5b521d034a6052cc92a26797","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 15:08:41.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"70538b0f85c99481dfb9d8148e34d06a69eabdd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 15:51:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34479a6a4cd9473b844e1c7aed1babd5\n\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b940d0c9e30c466fb3b114c21503be62 : SUCCESS in 36m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f3b06a98c64fd59b3881b5e82dcb31 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":9},{"id":"07e69fc6c513032b88ef789fe82160a023536ad1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 16:05:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"65946cef8904ddc23e06592abdda8573a85be67b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 16:05:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"91a9d1ec5fea19c8b081f0974451bc5b911096fe","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 16:07:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":10},{"id":"d99a7d3720166317d1c9c818c364068acefa9844","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-10 16:07:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"31dffda6f3ebfeb2b079e91870f0588046e202a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 18:59:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04b28c445924b5d9bb791bf03f071f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfbf09d9bfec47a391d52bf3e8a14ae1 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f270e0e8564b46b59c110811cc89d5 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e94766d5cc44dd2ad4ea1959a6da707 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/810cca9bb70644c0b0da534a0bf9e2bb : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edfa389eb1fd4a0fbed957f593d5acd4 : SUCCESS in 8m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a859aa600659445ebecd8ef6051c9374 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fa49070898f24692a87af948f0e5efaf : SUCCESS in 45m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/775ed7783c3745c6b77bb58b96a582ad : SUCCESS in 39m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c8761115c764a9c995e12fb42deeca1 : SUCCESS in 1h 13m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d9c4fb2560074aada4761ef5b3118782 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9635befa21c94d96ace0396c31eaeb32 : SUCCESS in 13m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f986377f8974efe8c1f2b72b0f03890 : SUCCESS in 10m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c7eeec80c0854bae9a9f2bb27fb1716a : SUCCESS in 2h 46m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a77c930507443349ea6236a3961f3c3 : SUCCESS in 1h 44m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3a834788199242428ec9c8ba191af301 : SUCCESS in 37m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1628421892b04f5987c02612bb10632b : SUCCESS in 23m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/355effcc353343a88ff45af9d9e36475 : FAILURE in 1h 07m 30s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad878d5adc35484da482c0b2e1ebccba : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be450cebdd9649f2b880c7b402572fce : SUCCESS in 20m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb11872e875441c0ac1b127e23df92b3 : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe202a5f687743cb864ba4ee70e14a8a : SUCCESS in 8m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9da9aec3dc5341269543a2fdd3c22198 : SUCCESS in 18m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89a69f1476ed4214835d15ce7beaa072 : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/212fc259299e41d8b616db7f26d26e82 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d51dcbf509b9458fa962ac34187cf27c : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4f37610585b4bb19f21318b5f8ae09f : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":10},{"id":"fda6ace1d0a8a43bd86be2e26afc8cf2e5a330ab","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 11:40:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b4d636b28e412df9673f56c93d8b97aa3c50f9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:06:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da592f22c12f4d5a8d16ba2994d11b2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6094304c49446c3b3d286cc1ed5ce6c : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68d3cc74be443afb44ed526ec2d14d1 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d1d0e64ff494b4b8ebbb06dc881e16c : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6170477721924773b69276cfa39f56d7 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0692c95918d4a4db6707812c4d5981c : SUCCESS in 9m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d4e62f1bac6440c0b61bf4dddcbf9092 : SUCCESS in 53m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ea7ed714d2ab4725b9cc61c77331fc19 : SUCCESS in 1h 00m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82848527ed394f0a8d2d96c85fd5374a : SUCCESS in 48m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4f508a4107ba46b28972ef72a193cdbe : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9366e4e1d9e14c149648b6f70058b045 : SUCCESS in 15m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5aafca66d8704d8f9c532b6632449c3e : SUCCESS in 17m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3790e89e8904d328ce3313b7c828862 : SUCCESS in 14m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/376c844c4cc847989daec148ed9e80c0 : SUCCESS in 2h 20m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/399dc4b25b9b4c6a9fdb18882bddb0af : SUCCESS in 2h 16m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/884be84b20d44a28a0366d0a891b94cc : FAILURE in 1h 17m 51s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fac59a4d7806466e8b6bc1607cd3629f : FAILURE in 56m 56s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/93b5659241784853b11338896ea38bf2 : FAILURE in 37m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db3dd9e2fbb84c50b4da24bea9022b72 : FAILURE in 30m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3c0d79d01e544f4f9f674b9910f629fa : FAILURE in 1h 12m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14c294d68625496098faefc45cf5f19b : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c28e96b29a20454cbc839e10ee39166b : SUCCESS in 25m 22s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fb2f71d6db1c45abaf7d7556ceec6c9d : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b92c2b169ed4818bbeeb05809245553 : SUCCESS in 28m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2848bcb56114ad7b87be0e901a978a2 : SUCCESS in 14m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53dc68b32b8e4a9185fa51af27e37209 : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c309c9fb45d4117ba40f73dbfae8de4 : SUCCESS in 17m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b6f371523374193ba6b61a3ca0916b4 : SUCCESS in 14m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41968379fb1047929c4239adf1d9c308 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52a5939475e243788da3817315be12fc : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62bb4496a2bf497b890e4663dcb1fe68 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":10},{"id":"d0ebb6a66c14d2e43f56836c4500c9919318a5d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 14:55:52.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"d5f89f8cef63c5ab1b2d5ce4c02d1dc65ad4ba0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 17:19:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e67f30b397f4ffb91f65e31bcd8fb10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a12cede3bf344ed99b3acd32a19f61bb : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a4de12340b406091de055dec366594 : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12d50d07d19c4ce891ad850ef67ad3ce : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05f290773d65419da3dc47391d2d506e : SUCCESS in 8m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/939bfadac7c6447080ef392c8c376e10 : SUCCESS in 10m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3b43b42d2e6a4933a554580cbd0db273 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/796bd588faf645b39c5b5643148cf57a : SUCCESS in 49m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f22b294feaca44b0803dc358bca3e616 : SUCCESS in 41m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/51c11badd87c49e18e59a35ebccedc41 : SUCCESS in 1h 09m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/54b2983b043c4c999ba0b39a096f7941 : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1f029659eb7742059785026cc45fbb50 : SUCCESS in 16m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39dc4548b346457dbb7314766a5312ec : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b516ea7927b545a2a1f64576aa00d447 : SUCCESS in 2h 15m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8984a0d55dd4e73b903dff66ee22ecb : SUCCESS in 2h 02m 49s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/578451e911ad4137aa6cc2c31ee3972a : FAILURE in 1h 22m 11s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37d07057ef7546f8b827241a63abf511 : FAILURE in 1h 05m 42s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a3de87eb0bb44a0096e52e419850fb07 : FAILURE in 31m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1d43d6682c448c5b574d6af41749a4f : FAILURE in 29m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/11339c56eeab45fa9573aef8ade833e4 : FAILURE in 1h 14m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8038b6c69ffa405ba1c446da5e088fcb : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67db03c098894ffebe2e5209e6bae205 : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/50ca09ec61124c5692885b9d7ae34e24 : SUCCESS in 32m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d914312d22454bb398bbdf05f76a2a : SUCCESS in 20m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/591d163a36a2429481cc3d99a3c8647c : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd411475f78c4f7dacc93dd9824a7085 : SUCCESS in 9m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02f888fb4f7345dba017593441ff56ea : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d69a707dd78448a4bad09699e94322a2 : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e4713ff6e974595b16465f7a776b6b4 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd84e64453c1414ca8d4bc229b975dde : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a64b8fac3551424bab7a15e905ab2e1b : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":11},{"id":"1d77bdd981abc2f582ca8133f58d523264a299d1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-19 10:21:31.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8f35de879581fc0cb0778ca1154a26f6dc713248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 11:46:21.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/458fef1f4104467781570442c9dfcd68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db38887385b842fa8a46bc8843953f05 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c7311f6b584104ac3be7b4796848ec : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35a9ada7d42c4a9ea4212936820cb1bf : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06271f430e8743f6b90c1d72165a9f9e : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0dd4dda1c14481d826bad77d009908d : SUCCESS in 9m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cc6bb0273b134de78dd6416db82cb090 : FAILURE in 19m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ab97c493943e454c96046938b28d069d : FAILURE in 15m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddfffeb5b8a34ee691fb46ee6cb42909 : SUCCESS in 54m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58759d72f4ff4469acaeeab37be571a6 : FAILURE in 23m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/896fce41e245434e928d0590e9c7f7b1 : SUCCESS in 22m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/78c5e20abde64e59804ebbd1acd8093a : SUCCESS in 16m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b651532b0b34365ab76de89570a1eee : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7dd9a771605e49fa8d3c3b5c338c74f1 : FAILURE in 1h 19m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e1ff2987a0e4b6584f49bde49836b3f : FAILURE in 1h 05m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5ba87e33b6f64df1ae070bdfb225464b : FAILURE in 15m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abc37400d33d4a098014b54478998657 : FAILURE in 13m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/03600dc6e4e447c8807d9aaf0aa151f5 : FAILURE in 14m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7167b94048b7483d8f35e339b33c43e8 : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/982925f24a114d5f963aa5f8cc1b8b59 : SUCCESS in 21m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/53a64119fe7645d3b76ccd003949fe82 : SUCCESS in 15m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00fbc4a7ca2a4b4286941de245785f12 : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/21869f61297e4eebbc5d6c675605379c : SUCCESS in 23m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6630641540b47e38ffc4e14ef799c73 : SUCCESS in 12m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a6ee65241864342869c3308631b8095 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c99f8fa117214cd9a4bee865c8a45623 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e13f063e409478483f7988e46e1a72b : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":11},{"id":"c2298a2ea714a8936c496ff8c839fdf9eff717cf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-13 16:40:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3a05f430d29255a78de93238d59e0a287284e1b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 18:11:58.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5218daadf6654d7295d6186637e6ce68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d290fa4a8f9345a88e628b26b359b2af : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37d6b0b64d3481795b9837a2dd9c26a : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c986e807bd52487eba122212ef408314 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0be94e2813e646febb3a73cd4592630c : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/459d23083392484db3152037d4bfeeb9 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72ef09103f74b87aca18a877bd3e947 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b18aa3a5755149fb91de2295284556f9 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfa54db17a014d7e97b0fd79da4a9931 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ed120c233aa446aadade65df7891faa : SUCCESS in 9m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/21f76d342f3043269f75a0b99f3fa6d1 : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c62b33155ebf4a29b27791f5c0f4e23d : SUCCESS in 52m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e9d22c65d414467ad3758912d081165 : SUCCESS in 51m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/78d62c41c22a499c858060ee754384a4 : SUCCESS in 59m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/adb2640f2d47411caf469c262884a1d2 : SUCCESS in 20m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/97d6ef66ea224cb38f2696d94776da74 : SUCCESS in 17m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49e78cb1d1f648daabc0e08b1b739e6a : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/211c9f5532d943ac94d8625830e01d0b : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d4a293029f948538ff2d8351e1c1350 : SUCCESS in 1h 26m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5d3226821c494f9681e3fd6178180754 : FAILURE in 18m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db5d2c8d70d447ff98bc0178182b483c : FAILURE in 13m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a2ad157bded849a38ec59847c680956b : FAILURE in 1h 14m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/201ff526032f4d779e44d33aa28428ca : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d912cce557234dc5a92c868e10585d57 : SUCCESS in 23m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea1761ec761d48aa824dddee8a8fade6 : SUCCESS in 10m 41s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/295889666d9a466ebe160cef1160324d : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b86ad5d23b7472095a993a14bb0a0a0 : SUCCESS in 10m 54s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/371ad70952454c87b97d01ddc09745a4 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c15365c97b2443bae46268a6a493326 : SUCCESS in 19m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf6ee86455af4a618f08284594c4e626 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":11},{"id":"51e45d18ea32e0ca892d49a4b670a2e2b8cbe80e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-16 09:03:04.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"4c6b6f0667d8e8097a43c8a39a0f36e21b154b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 10:39:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d7d9808165f42b7a2336d73e2855180\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02f6627696424be8bc0fdc49f10e8cc6 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41462bcbb6e94d8397b2687487d53943 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e024d27495a4ecfbe5150e022f08ed6 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ebc9b8878ee4c5a8c0d0827f92d47bf : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d633315e7d4464f8adbad3304aaf5f8 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e5259a16f141d58a935d0094e9393b : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87c5ba723669416791c3b2e52cf6a35f : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/913eaa9543d34de5aa911c1e3105508f : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fa1644e16c545c2a1822e843307c8cb : SUCCESS in 8m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/39f6536b3bd94216a0c963825c57cf49 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6a2f4ccb814845acb7cc935f3d02e6f2 : SUCCESS in 51m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a912e97e5f54c2194b3bf3e46505733 : SUCCESS in 45m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75e1206fb1de4b38abc8ba36e0cecb73 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e9e7b7e24e454466ab3d4f4dbb50a441 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/09f61b575e684939ba0c19600b1e18d6 : SUCCESS in 17m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2658c10f92384cf29cb6c1cd8819d16a : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3c4d1c7ae3d640cbb50d0e3ffa149188 : SUCCESS in 1h 34m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6739702476246a199bfbde6e3144b92 : SUCCESS in 1h 30m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ded39a4a6f0f457fb6d29d6966e7b845 : FAILURE in 15m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d71757e81f74314a791e82467a505ec : FAILURE in 11m 54s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/46a85a028bd64be6b28070d93af1c7dc : FAILURE in 1h 22m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5daa157ed37a43e5bc6f894c34bd88ea : SUCCESS in 31m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b847e2162e5046e2b48e119e5b87799b : FAILURE in 24m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/327e7148ffd6474e87908414937a96dd : SUCCESS in 15m 50s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cc513774c96c44019b92475a488b137d : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d513253ad0d84be4bb0e7090ed0dc746 : SUCCESS in 7m 04s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/8dc2f458e726460d9b2047baf31bef84 : SUCCESS in 9m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4acc9beed2f04d73b821a29ff99c7822 : SUCCESS in 27m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b081bebdb790462cba20cbbfeee7a139 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":12},{"id":"e2f9e708abce730ffa4d084d36f9425630088ecb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-16 11:15:58.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0c6adb47597830f09846c20be33a912043893a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 13:11:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9fe714174294ccc856394722f1a160c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccbc0455fe541c59628d24e92a18da8 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be7fc8f57cd4b759d0bb35d1921837b : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8afa1d02c4824c9e8d9f02359a2d3c26 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b458d9c4b1a643279d672ee22cd5e502 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/874c9d73ea4949159f8299947c5b4568 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6641c0b6f2a406b9249458efb3f99ab : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d416f1805aa54e609aa899470c70ccf1 : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf8b1da5912b4294936c5b9946f1bf84 : SUCCESS in 7m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e73ff96bb7594f8292b398361dbe3d18 : SUCCESS in 6m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4213fbc954c14ab9a1c86b7a0786ee24 : SUCCESS in 38m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/57798a0282dd44e1aa98eff00b911e6d : SUCCESS in 36m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/553534851fc743f9a18a8178bfd26d0b : SUCCESS in 49m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2900164bd4a047ce91b5aa68fb7df9ce : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8552af873dbc43efa3248dde6827f6bb : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/87a4385857bd493dba52a539ce749cb1 : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b93531bae0a4a29b6c26bd1ba46b259 : SUCCESS in 11m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd874c2981f04b60a788e44b90069bbe : SUCCESS in 1h 53m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a0a4ae9de3a42149bfc3335ada0f787 : SUCCESS in 1h 32m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e408cd4b502b454d91e9ea0820817e22 : FAILURE in 16m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5d7153708964cf895f9dd2e411a56ed : FAILURE in 8m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/524fb9adcd5740b28a160785b8a16848 : FAILURE in 1h 19m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1ee39035f10a4598aa54cba473a48522 : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/380cfbb5d0e540c2983d3a94450b1eba : SUCCESS in 29m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c6880a4ca974e6aab038eb06828d0ee : SUCCESS in 14m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f862244ed0f44a9f83c826835f3ad7c5 : SUCCESS in 18m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8cff9bef24d4769b6f037c0f7d741c0 : SUCCESS in 9m 53s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/95a6bf51b9a7489489380be0ede69ff3 : SUCCESS in 12m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c161e845c71d4156a53a13f374f537c4 : SUCCESS in 20m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d57bcac8afb146d695f2ab4175b65a64 : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":13},{"id":"541d926415c6b8523cbe224a4b44865f0a4e6f38","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 08:31:02.000000000","message":"Uploaded patch set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"3f2cea74a5fe3853b869cd084f98a775f8f96473","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 08:31:23.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"53bc2686087af46b42a1bcdc819dbc0964b08029","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 10:03:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f46acf17c507530f7574ca3bc873aa19f5567f60","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 12:43:03.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"b08652c59d9272628ee457a15dc03325763e55ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-17 13:54:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"17c836dc612c4dab7be77fd1bda70c8b04c1dfc5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-17 14:58:22.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ba8ec28043ac601d4210dc728c26e4a952c70aae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-17 15:01:49.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"4ccf906c44225e286fdb11c1d4507436632859b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-17 16:17:10.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"dd22791b514b92ff4d4cd1e334c100463e12aea8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 17:50:34.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ff73840f82c83bdec70a0b425356e5783c331de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 18:08:09.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86f2945422214c79bd07eba4d4a045ac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/305b33c1e5d44d25912a12a7b8c95cd9 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c1ce3506c345a8af1628aa0af3f953 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cb59afb626c4ed09811f176f5132940 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9961eab88e44a7e90a1e759808241c2 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/211eb60960d644b5b250b4370cb5e14a : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581440e032f14130b6b4eda3aa823ac1 : SUCCESS in 8m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2030ab32613140858b3ed57cbf214f4d : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3c808885e424825982ea3ab3d110e1b : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/953c837afbce4cb3917af56376094a17 : SUCCESS in 11m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e890b4418d5142deb1bf50c488c98372 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f3f854a7ec5b4539b10f41ab210ca2d9 : SUCCESS in 58m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078360c04f624a5b852082a2d0aa9a13 : SUCCESS in 50m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/70571e9793184bc19acd4bc75ade33f2 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/622e150753094cb6afb82f8680c83158 : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5610d2dc2ff54248955b1d75606753cc : SUCCESS in 16m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8e99407590d4dee8ef6c350268e88ff : SUCCESS in 12m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d1b38cee4fd340e7aa25d5f2cbe45e9f : SUCCESS in 1h 45m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf517081d644f17a1fd9fae01f91936 : SUCCESS in 1h 35m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0ef4a01717f8420e80a40eaff1fbd56e : SUCCESS in 46m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a39c3260e94b02a83745c44c95337f : SUCCESS in 36m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f066c36b0a946448d3fde915b4e9f98 : FAILURE in 1h 17m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3dc14a531be94fd7b2d784111ed38862 : SUCCESS in 33m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21e851f4f7db49d1ae94ec7a204de988 : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/94b382fa53e34ea19d346543541720b8 : SUCCESS in 19m 15s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e34cf647a5e949e4b2ce0dc65b520d20 : FAILURE in 14m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/717d594cc7ce4f12bd33e7e847c8df43 : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/ea3f95f6120443738fcd89ecd419fc1a : SUCCESS in 13m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2e4a0debe4046b7b25fc3e9310add10 : SUCCESS in 19m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/866f33fefa114d86a2fb4553ce3fb6d2 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":20},{"id":"a7ec1a6d67d08c291813c1a9a9cfdb59b2a7ec7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-18 07:24:43.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"134b953f297b031b4d4ff639d4e61bd7bf2da6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 09:26:17.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/798470d75fd5412cb938beda11349f3e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f58a0647fee4c9eb6f6391de8193e0a : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0db8316b64419d980317b1775adbd8 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34096653716c4e9cb5cf5e10d789a38b : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a9eebedf0824730800e29326335b040 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97e3a5d822b74dbab83621f178897fd1 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d768530bff447339856317332bbcb55 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cdbba9dc668940dab48473f687570404 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a9008104abc41df97af90aafa8f1672 : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/744ab89413fc4b33b26b6bbbbcad5622 : SUCCESS in 10m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/54ddbd505f2c4fbfa4b89ada77ff9248 : SUCCESS in 55m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2653241e3b7f46c08867c37efdc2fdd5 : SUCCESS in 59m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd404814d7914fc0a6dc8eed7ecd6f1c : SUCCESS in 55m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/37253beea4df43eaa7f790a9ce55bc93 : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1758d1309608416bbb2df214e31bc52c : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ad35aa2e3c7144638ae2c1d47a2d2c55 : SUCCESS in 18m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca9764a5daff4db7903aebb5fd08fd0e : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e5953d8ab7e4f739654d2b415de5a35 : SUCCESS in 1h 55m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/172704db46cf4439b65378442012ddfa : SUCCESS in 1h 02m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e0ab5715647e402aaad4fac21d5b2b1c : SUCCESS in 30m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2027c97dcf6845f3a055f8e8cd84e47c : SUCCESS in 39m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/67b8577f095f4c47b775375976328928 : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/40e4f6c0a6fd4d9383afb052c5b64eee : SUCCESS in 32m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f473a6a06b114550afc149aa251d2797 : SUCCESS in 16m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cdd9c9f8fa7a455aba3fdb36d57cf965 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/85ade70216824474a8dbaf7ff79fd5b9 : SUCCESS in 14m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6de304e3b7c4e6180e24f78c2e21e53 : SUCCESS in 7m 40s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/03dd71aa0e384efcb84cc36428529d79 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9932400f155641daa4cdf80ebc7a9e6d : SUCCESS in 20m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29af67404933435e905e608c176994fc : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":21},{"id":"43d1ce3247b374ea8706455abb0fe2d95aef55e3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-09-18 10:35:56.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"b0bfd4064362118585e9835712945dbc3635d346","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-24 13:10:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":21},{"id":"e61243192aefa5e4209a627bcec5441edd735024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-24 13:11:36.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"108a3cc384669a54da086946c5e3037efdf28b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:55:36.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6937adab924b401aaa89d313a61f1656\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23b5266e99bd485290ed38eea0c907bd : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb45ae736be8496fb5abfaefffef77dd : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e5e925f4c0f43b3a08b098ef4c068c3 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f609972d66594e5faa7e0e2b83b35827 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77034543fe9e4616bb38124f118ee80e : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ca70902e81e44dc8a1c04923e6f7b2b : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb668fad8e824c738fdfbb75b236449b : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef3a5f28379e49de99d2666604b0cce3 : SUCCESS in 8m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/272952b455164a4891776a56c6b33573 : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1ab4d799133e47aea5f214a19d7d00d7 : SUCCESS in 1h 06m 51s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335b3fd253c748e282ba782982c4d731 : SUCCESS in 57m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6d5a5e61520c4e82b3de3cefc1eec1e5 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a24a30eb54741eb841a00b8cedd0fc0 : SUCCESS in 23m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b5d83e930c71451cb0d8dda43709603c : SUCCESS in 16m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b6c8bb89a544666a6dd92d457b0fc6f : SUCCESS in 8m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a3f17a63d6c54a50a4059f415e116e6a : SUCCESS in 1h 42m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa68249421a4a6a85e92700259bb98b : SUCCESS in 1h 09m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c02d1d4efe104d5cb836cf48b743cf10 : SUCCESS in 43m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8eb52ae65a1146409c9d65aa683b2e02 : FAILURE in 25m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c97120bf271f4dd19cd086893be0103b : SUCCESS in 36m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83052ebcef02469185326601491f95ea : SUCCESS in 36m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcc39ebc807e4e2798b5a33733bd9284 : SUCCESS in 16m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13235d8924884a9aad2bbe28b2442879 : SUCCESS in 15m 23s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/66acf86cd01e40fb830bba81f3029a4d : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8a718573ce743bfa31d505dfa7f7e98 : SUCCESS in 11m 44s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e7badbbad9014146ab97215ce36034b3 : SUCCESS in 7m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09e37ab153f648dd8415af15487a505b : SUCCESS in 21m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8509c5cd2dc46619af291a05a45ffd3 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":22},{"id":"9ee54673af8721d83fab55c231ac9b30cd469a8a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-25 14:40:32.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"d75ef64127e23480e132b2c7ad730a7c4acba5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:29:19.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea6ad87f12964f14999d486918dacaa6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c263d023c2f248628934d05e34428b91 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23df19c4d06041ab8536759ec531eac7 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87029c9845444bbcba8625c70cec6b01 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64437c59465f4c6aa9de419201563b41 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0884a045df1942acae36251e54b1391e : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ecb7a1c900e41b38fd3e6eac7a35be9 : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c4178c9f3134eaba9f3eb6b1497b645 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b15784bfbed0455181f011a01b38a342 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d400e8dac44e4bdb94272c29b2a74644 : SUCCESS in 59m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0cb4f18c1fd942d0aeb81d7c4269051c : SUCCESS in 1h 02m 09s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e9ea505b5394dc4acc1ba8598f1b234 : SUCCESS in 54m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c089fae5ed5a4d558df73e5f8e4aab9e : SUCCESS in 1h 10m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a4b30e45389a492c9f5481bfbcd0ae66 : SUCCESS in 19m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/580b2c7e61c640e3ac88276f4c49036f : SUCCESS in 17m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/909c40c583d447d4b7e0e9e40b0a1b07 : SUCCESS in 14m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4003e68f36404d09a24d262e45e55d88 : FAILURE in 8m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5befddb88b144527abbae9ee7fa366ea : SUCCESS in 1h 42m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d35f4db96d044aabc3ccb71108c55b5 : SUCCESS in 40m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02cd70afbe4c4eed80c2102ad3a84df0 : SUCCESS in 41m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d83f32a43fb6443e803b12e79908ec92 : SUCCESS in 42m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/15e6a75871834e41bf7a2f1e779917b3 : SUCCESS in 31m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42e7d57e036c45e7a536b5daeadeacaa : SUCCESS in 27m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4765abbc1b1142a2b2903194702e70e4 : SUCCESS in 15m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dfc205b194054294a13e5a2180bf84bc : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2f02cf1bed349089799c142985fc2f3 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a82ed2aea9644ee9a72a854fbab83ba8 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/474b4801ef5648c7b7bc5ce9f7137a75 : SUCCESS in 22m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bc414108c69495caca0cbafcabf1f28 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":22},{"id":"033f22cfea4c49cfc2785d5da3db0e60d373c383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 21:58:24.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"43af55d6d9e38577be70ff9a3cc0bd3bbdeda451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:59:50.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b590ff00cac94060b250569d9182e5a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81bb3497df72489980397a7f01ba90d6 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2623df8543454eb215fcfe416eb826 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b643a3bdf24247d7b48d3399bd6c1237 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88c9b2da6a442db98995762df6651be : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/229724a6e0f1465ea702ac7bdd99e8de : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/674943ecb0b5443fa7d64bd993787f13 : FAILURE in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5f8b71354f548619815c37161ca0957 : FAILURE in 6m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/04acac9f1f324f65a73eb105eb1131e2 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b6acadc47577494aa52649c8e7ddaa13 : SUCCESS in 1h 02m 10s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/035e2ab50e8640df895856b85781c048 : SUCCESS in 55m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8663695999564239a9a9f3fbbdc68cd4 : SUCCESS in 1h 15m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a4a58ea4e79d492d9e781424a3a0a6f3 : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e06d66b559a2437bb8f0b8a2c7cc8938 : SUCCESS in 19m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b37191ad4794e06b97e13cab46f1f5b : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3a4f1a6cb3934967a5489e1413c580bb : SUCCESS in 1h 52m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b08414b492ac44f8bbf9694441c4acea : SUCCESS in 1h 57m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1f5ef13bd51c4c5d9e226347511de9d0 : SUCCESS in 45m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004bbd79cf47404c82ed894e0851a6bc : SUCCESS in 41m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a3550214a19041e8a30b3e8a798be9b7 : SUCCESS in 41m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5493c3d34ee45cbad6c363e73ecb9c0 : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64d68796488d42d4bda077400c7ff7c2 : SUCCESS in 26m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3079a763eef84609834c6a9ef7757a26 : SUCCESS in 16m 52s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c30f5020e285438baae82ace2bf08465 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad8799e2d9a14ae59a603b5d339cfbc1 : SUCCESS in 6m 58s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b4c211c3032445d3b57a6a0140a010a1 : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/953fb22b72db451f946a3da1ece041ff : SUCCESS in 21m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4edf8139b1624c3382c0d4893dbd64e8 : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":23},{"id":"6bcd71c6a168424479524856a4a3877b694fc7ad","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 14:19:37.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"b0112da9c1de8e3d43460fa8e5d865a412f1b3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:06:09.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed2e1edbe050495ea25cd4579dc6d4ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d28587ac907c4aaa8dcd6f2b5208de39 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa3d094f69041eb9d0e42b9f77d29ec : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6654a2e4e87450fb9b016cae42c72c6 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb187e22e344120b5ff0d459bda84d5 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8f8499bc4d8445dbb42d5f4d8b892c5 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed0ff13e2ff84fbcaa57d859818c5e69 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/453a4d3056f647f99f46352beb5214a5 : SUCCESS in 11m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bf8d5f4e3894484793a39016edca9752 : SUCCESS in 55m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bdfdc4e16e41424f95e189682e3802f7 : SUCCESS in 56m 11s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e28c4aad41fc457c8821e2aa780b61bf : SUCCESS in 55m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/af83bc881f3543238b3b6db70bf1be88 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/49f9e93290b14bb79d6a36a369b46652 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/27304d7e25e7481083e739110d5018f4 : SUCCESS in 17m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77c4a89434f2489590563012041d458b : SUCCESS in 11m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cfb9336329e74a84bca25d62782dd619 : SUCCESS in 1h 45m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3763a8a20aa54b0e96ba2bebc924969b : SUCCESS in 1h 33m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af3ca78939bd47f4aa991caf76f5d21e : SUCCESS in 40m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9a738e55e9847f4b7cfc00e585281c1 : FAILURE in 20m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/39f5b6bd04004654bcf8bc6b0efdff5e : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff091a87dce642e2acededf843793efa : SUCCESS in 32m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/096218f1c376416f841a2f4e14dda8bb : SUCCESS in 23m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/395f304732624e039f32e9449fdcedec : SUCCESS in 15m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/77068a18513346babef2a1d89ca02ac8 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b621f03a85fe4e3fb1b60d9e04959a66 : SUCCESS in 10m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/266a7c4b9859404e9c1903bb83e5a702 : SUCCESS in 6m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a93d274a8e5c43d78e40b2debc1168a3 : SUCCESS in 19m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2342cf823f9b4466af117bdadd750542 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":23},{"id":"ddd80aff8a97a14b0956389a46289934ca66d5a1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 16:27:52.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9d568abe90ab6b4de47a26a327aee6e8516a8efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 18:33:54.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f14b767e0f2d4305a102ba3578c43a4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a102170bcfaf48618523de782ac4b7c9 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a41b9906894eae8a683a65f5158013 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbf6d3c27e01427885d5972bbe713c78 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0cd43fdd5c42439951358b771a5730 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e70783f6b364b119a2947620de0c3b5 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c66ee717291642dea6d15a92e75a001d : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8314264064d84d09a180b3d60b7730d3 : SUCCESS in 10m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/065beaf42abe41939d5227bd1f7eb7ae : SUCCESS in 59m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/52cf204ad10c40a18c0b64feb4368063 : SUCCESS in 1h 00m 22s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69e0a1abb8444a55a31bd7ff4b04ea38 : SUCCESS in 59m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1b5752279a15420095a66f063b4e61db : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/692a4571cbd1457d9bdaaf93d7d0bca8 : SUCCESS in 19m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1699fb59596e464f8c38d0970d81a3ef : SUCCESS in 18m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e1b8aeb1d7c4cb2b52d86be546bea78 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/561f030a7de14b7986fe938e42be4f3b : SUCCESS in 1h 45m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7f5d6dcd56c482dbf71b525b59640f8 : SUCCESS in 1h 49m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/affbb9ac97c94aa5aa9c895d0c50a1d6 : SUCCESS in 43m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1911feb3c81949e085bc33b9514f7341 : FAILURE in 25m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/01fcfe1a7d2a442792dda3bf3f1eeec9 : FAILURE in 26m 21s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2934d105572040c78b30358adc8274ce : SUCCESS in 34m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b82fb1ffef24db1b5be058498ee31e4 : SUCCESS in 25m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/06ab0a7da86845728b7c77895765cf73 : SUCCESS in 14m 44s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4e3a37c055534f9fab732143658a0e30 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a456a10a70424b0a8cd52eac1efe5183 : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d6453709c7ab4ddfbb354edaf7c4052b : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4a908234581445a1b2475d396526d0d5 : SUCCESS in 21m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/000d9901db2543e8937dabafaa727e47 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":23},{"id":"e20db6428574fc7708ae21c4d60480148812f879","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-06 14:44:13.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"57822cec3d7fcfacb2a5e9293c7ba71fae7600bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 17:43:30.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78cf2f022e534a068f57d35352fa878c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/545afbc0f9a34069b9f79f79b4dc9f7c : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb25a25522454619b8b64f5606da3ce2 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c909903a18d74f30bd0e0a55501fc923 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44daa652a519418aae7e37f428b507c3 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/933c6a5d09a641af8d9400e886875ea5 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81ea42feaa643ee8b687c8a6c7776cb : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a15838ff30464a25892af1856c569b46 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff2cd5a23d4a41bc968a75c0cf3d1b89 : SUCCESS in 17m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71dafaf208634eceae17fc7f0936709a : SUCCESS in 6m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bbf3c5a9ffa04ad8b5676f6759dc70fd : SUCCESS in 54m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d3911b353b584a49b9f2e2cf51bad5d0 : SUCCESS in 59m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12f168d193f147149ca48ff566e5120f : SUCCESS in 54m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e8b2b561b804e849ca2023c0be72dc3 : SUCCESS in 53m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/905bf8c7effb41eaba804a7ad76a7162 : SUCCESS in 1h 04m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ffbe021b1112476d8a21bb44f349140c : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcdfa6057687479d87101725be16e842 : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0f5269c9278a47adb4f6ba0240915b4e : SUCCESS in 2h 10m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21e3da86ce8049fa9aa72c52ab289825 : SUCCESS in 2h 04m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0a8d15a4ad354d8fadb31637bf6aed32 : SUCCESS in 2h 13m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b94a494f8720432c8a65d38a1f70dd15 : SUCCESS in 2h 32m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7cad964fb7fd4b50b0e1ae55a4b25dd7 : FAILURE in 29m 52s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/65dd577630d1413d94f9e89eecee730c : FAILURE in 16m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc1f4ffab9194e0a896d394918ce6f99 : FAILURE in 14m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/44b3d1f4f56c437aa8139c8d83f3a9e0 : FAILURE in 27m 19s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fdb0c759cbcc42d78b31a493871928a7 : SUCCESS in 37m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad0cf90211894617bc5272b578352df0 : SUCCESS in 35m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718b024b03414f80a55d3c6036574e2a : SUCCESS in 27m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/08f823b8e6e048e693c4bc41e5f22237 : SUCCESS in 34m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/63e485d303894b39a4d89557ad7460b1 : SUCCESS in 12m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b68be0b54df446cdaa91e80c19b5b0b4 : SUCCESS in 11m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee77dfcc88eb494e909dcfffa7a3e619 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7f7d804878874ef8ac95522388347ab2 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3bee5eb31e7b44afb1b4c13e02501468 : SUCCESS in 12m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5baaa63820b4949846fa2ff7b0aadee : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":23},{"id":"fe28b179493344e6b2ae3975837c653b5cf364a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 07:09:25.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"85eaeb4962b0ec5eeee5c538aeaee8fc8006917c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 07:09:33.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"e11bb54ff24a0c135ddc3e3bbc5c0a63a5c15f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 10:09:49.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9470f735b5a749e9a97fb7d20095c1aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6573def3b2ac4df486764c280a8f2771 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6a640fbc5848acab2fc5441d61c62f : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3b31e309746420b802c2a5dea9fa413 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cff0ae9a2e5e46e682422ddc67b31b08 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3224478478940cb8a1bc128aa594f9a : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f561c76215b444a9617ff5b476426d4 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ab4bcbdb61d474d92b1e72caf89113c : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d16a4148b35540fe875fafad27bc9179 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9e7d1a2dc514a5f87c9e0f3a4207ae6 : SUCCESS in 6m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff3e3b24f03a4d24b04b4faedac510b7 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b3c765d74bc40c08a9b4c8d2435b861 : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc4a966e3b624d0a945ab112b389df2d : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b707ca19036e422185b5f2f429d110f5 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/518400e49f8f49ecb48e8fb233edff6a : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f97d9d94f26494685aad5be0d56017b : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/199532c1cefd423ebfaea5403312378a : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/338bea70ad0a4bafb3a9ac94f46c6127 : SUCCESS in 1h 47m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9034d213e78546a3913d7d33033685b0 : SUCCESS in 2h 03m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b616749a04534bb39c78dc4cf3b129e0 : SUCCESS in 1h 57m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b9b767694e44671813e99769854f6a3 : SUCCESS in 2h 54m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c78e176b40464fd4a2fb9c405398ee1f : FAILURE in 23m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/da429682830445d4bc708b4514c3e2d8 : FAILURE in 22m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8a19db3cee549f2a4b507f3fb8ef637 : FAILURE in 29m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c239c78b920e41e7bb4a3638ec5f6157 : FAILURE in 27m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bddcb65f1ace4949b6e4170ef8290070 : FAILURE in 1h 08m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/193bab789aef4561afe265639fe91cbd : SUCCESS in 33m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b25584cf4bbe4f3992aa1a0ae79ce526 : SUCCESS in 31m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/969170298ada4d6c8e2bc4bd83acf71f : SUCCESS in 32m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8269e9b2e13e4ab49c327949cc2cedd1 : SUCCESS in 12m 15s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b0150846531a4a7cacc4121ac8961db3 : SUCCESS in 12m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19c0131d5d2f4bb589b32ba4c5738fc8 : SUCCESS in 9m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d4cf7157f327416dbaceaa0cab29cf5e : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b589e15502fa45908e1cbec8f6cc7439 : SUCCESS in 11m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fad1321c7e5b41048212602eb773d23d : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":24},{"id":"7eb38a8139a5e40df5113420fedf5b1ef0a47e00","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-30 09:46:28.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"347607bc5d2901b3bee10a721b25032ca673032f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 12:40:04.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30ca87e8ec5f4953aac9f3c5e208c954\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680e36ab71a647e1a9d670407020f6a1 : SUCCESS in 8m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07ead79624a64f48b2fe15001cf8b97f : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aff3e94beab486d95aa6ef84a4fe3b8 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3260e565794407a5302e2095481480 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80bd0a574ea841f2be5a9b6ea7b031cf : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f6966dcf13044db8497dae6da2c5eea : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/695a0c60d8b44f85bc49bed273eef02c : SUCCESS in 8m 30s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acd33af39ff44731853ce6b382725530 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c38bf899076406fa376f47621d09684 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea7d56862eb44a2b9cd79aa854b12152 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8bb2767f321411ba58c90d534cfc062 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a220a907d4b446a8936ccf55d3fe3ae : SUCCESS in 52m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ef4fe256bfb4057b3b32a2174774dd4 : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66e202ca398c489d8533ba007770f9e4 : SUCCESS in 58m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4399665564f3427aa3863f1b14e189c2 : SUCCESS in 58m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e362bf7d92524586b334f97de47a09e9 : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a7f0f6a7c804275b4f10d3f5f954c27 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0f13b9507eee4011a0f185e8cc7ae905 : SUCCESS in 2h 30m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bd3bbd0d75642488b59a0de7d384f4a : SUCCESS in 1h 42m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2bc5e744126b4335b1488f071e363717 : FAILURE in 29m 34s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/baf389f1202544bfb08db7a052a156b9 : FAILURE in 26m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6b6a6707c094ef5859f9b7161e0ff8b : FAILURE in 26m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0cb6572351b84a9ea3bbab5c10173f0c : FAILURE in 20m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2e5c0e71210746608dab19b4efa7e1e9 : FAILURE in 1h 05m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/07fac88dc2724e408dc38f00fedc28d4 : SUCCESS in 34m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52e75a6508924ea581deedb1128fab3f : SUCCESS in 26m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cba5b7f4b26549e5a1c519de66bfaada : SUCCESS in 13m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dbfa453561cd4ec6b138aae677957914 : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6594231c853a48079a19598a93de8e3b : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b8aeb294d196446280fee6f8623ffe62 : SUCCESS in 10m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78ab142d71e6413dbbc0a1f0a1a6caba : SUCCESS in 12m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/881c49de9b5f482582215bb5e42966a1 : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":24},{"id":"e7d5a947fdee95ca09fb9d758cf7a2a32a83698b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-13 08:48:02.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"227973c249d608252b0e8474ed822e439daafe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 10:10:34.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a391b83c02b74f03bb1febfc45678f44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14033edf1a954b13af63a67f69459a70 : SUCCESS in 7m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e7c203bd3650431db263a8603e519b37 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ddeffe860504e41a5f71e569b5361d1 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c685006a154c529525991e067b3dfe : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56392254677940d9a1eea74821a120ce : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2caa5c369f194dd1855ab905c2f82337 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e20fed4a26c04512915f76fff64c915e : SUCCESS in 8m 10s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/833ac6b33f6f4be096fd3a889c7809d0 : SUCCESS in 2m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a69396b106140449a2798b96c46501e : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8f661e1c4c2490d979a2ae26d5e7222 : SUCCESS in 5m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f8cd69e89494dbf8b532c51afead4a1 : SUCCESS in 34m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17fe8245524c467ba7bf71c902080cdb : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c5b69ee296e41c199014a0004ee379f : SUCCESS in 9m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d501973b0fe4f50830f45e9caa7e9c6 : SUCCESS in 12m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2f5a890f8c4d15af2a81f829f0e30c : SUCCESS in 1h 07m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/204b5db6451b48f9a7adaab28f1084fd : FAILURE in 32m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/be9ae39ce3f04ec4be75a8c699eb70e6 : FAILURE in 30m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d407f45e87d4ccf962f19f57f0d3eac : FAILURE in 32m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/136748912d1b4b9cbbf56f4ca0c83728 : FAILURE in 20m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bc7a43f9aaf14c959fe04ccea6c8acb1 : FAILURE in 1h 09m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1bad46a92c5342c487ef4d61955fa0fc : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cfcc80ac930460992b74663c50d0824 : SUCCESS in 27m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/865c83cba8a84229a2d7145679945a3d : SUCCESS in 10m 35s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b6b34a021bd647b99eb18be31c39b9d9 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da9719bfff5546178fa3d83f9ca41580 : SUCCESS in 10m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/36ec3516f2674c2c88bef2ad59fa8c62 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ebf9af3e62043e0afeca5aabf4c0c20 : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4dbe03cc2fa4df680b33cc36943dc1c : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":24},{"id":"b60740747ea3e7ede2cdf22f4f63b02acd742400","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-05 12:29:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b7281767fa2c258bb428b4bb0d74603e32bc41a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:25:03.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f8591d3b6b14fa98d2ff4ab37fe0b52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffab9c8c42942c8a8ab7a7a321ec6fe : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aec9bdd73a3f4ab7b54ad2dfbb9fc29c : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b78941d4494a4716b37ef5dc381ec01e : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7e0f99501a4c9ebad002f4a5e42f79 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ffd7fcdaad648d2ae1f9df4de9f75e5 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8582f08cdef54b218c93b3b5a9e2343b : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b28e0a916f24b65b0f4a254b151f516 : SUCCESS in 7m 19s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/012ab5eeb0fa47ddb3114cbfb6a3371c : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f833ded698d2400b93a422f86f2ed879 : SUCCESS in 6m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c299b839cba641e8b2f510010b013831 : SUCCESS in 11m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c221eacda773447198b31172fc4518f1 : SUCCESS in 58m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/abb9a81074dd45b0b62d08ea4b1f7e40 : SUCCESS in 53m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5e669d9ce8e94630b7b5af2870687efc : SUCCESS in 10m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e11d88a74d52464fafb703e32342c84b : SUCCESS in 9m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b099670bbab448b78250fd023c9ffe8b : SUCCESS in 1h 45m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a10cf96b157844a49059e07c641ddfa9 : SUCCESS in 31m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d76bdef3e1643249a022a6f32f73a91 : FAILURE in 16m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c4eae9aad394169ac7c9d50f5d25cb6 : SUCCESS in 28m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/96573b4d6be04ba2bdf87df156d90c80 : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81a6694f9a6944ca9e28afb9df44ec12 : SUCCESS in 7m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a521e9562da94062b8e6cdc92b59f450 : SUCCESS in 11m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bf5c9712c24d95a7b444080e85ae0a : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":24},{"id":"0b1d7284645885dedb900885786f240a2286cee3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-15 11:47:19.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"2310eadb86a7a033028dd48a208a44a544e01d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:38:53.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/228e52df3c6145a5ab0563508942c91e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0eaeef82b724357a8c5c15c726f2d32 : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b4e3ea9b0ed7488babc85b8f6d0b27bd : SUCCESS in 24m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd61689b892d483a8229cd02491b351f : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1d9a91c6c34ab7b152f32ddb532aef : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88911a663cf44568a26e2e59ea9c996e : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/304c0ac4a72042abb88cca3da48288aa : SUCCESS in 4m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/918bcba4f94d4803b6ac4a166557bea0 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d01fb950f6e0408b81f2023dde52d22c : SUCCESS in 25m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0847dba92f004deca8f01936224b6f25 : TIMED_OUT in 31m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67cb3f1beda64e65a079d0efa1bf0822 : SUCCESS in 17m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8ebf07bbfeb46c69737aef24c09968d : SUCCESS in 42m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/38b073c3b46f4c2aa3f25316061be389 : SUCCESS in 47m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f810807f4c84358b6ccdb000b410f38 : SUCCESS in 51m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea0f3f8439fc4908934f1275699885b6 : SUCCESS in 49m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ed58272c2c64abb8cde3e820a3ded83 : SUCCESS in 49m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fd5cf38d8d6444baca73d2cf3e6d1e2 : SUCCESS in 49m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c63db8df58a54e7fad2da4486b47ffdd : SUCCESS in 40m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbc882005d14af8be473699d2b1ad18 : SUCCESS in 50m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c4ea314a4f11413ab18a6d54e2726fc7 : SUCCESS in 1h 31m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/77e85da580114f85a3940ba9f5fcdeaf : SUCCESS in 1h 35m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f2d17ac7cf1642ba8c63abe196154035 : SUCCESS in 1h 40m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15030299e5a14efebcecbde6d02a99c1 : SUCCESS in 2h 07m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ca3654978514442ad57e5476bcc21ac : SUCCESS in 2h 11m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/774e1579f7114dc7926ab78feb593922 : SUCCESS in 2h 50m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/62bf62dd6dd648e29128dc2a9a7d0be7 : FAILURE in 56m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3fdee892fe8b47d483a1cddc1df31f61 : FAILURE in 53m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36ed5a62eefb48d29257736d489cfcd5 : FAILURE in 53m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1eaed261233f460ba67e1c1cd87bcaf5 : FAILURE in 53m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/35024b80a6564e9b8c5b3279ee0ed07c : SUCCESS in 1h 04m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed1598ae2969458d89be4e2efff84337 : SUCCESS in 1h 17m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd5f7aa5e56e4e93b8f506a9e0315ae2 : SUCCESS in 52m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d519aa17fb4605b46e5dc513db1ce0 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":24},{"id":"5207632ab7168720c68eb0cf515d3a5bcb86eece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-10-07 09:17:24.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"3f1b03fdd68ca36d2b8ff415aaf4dcc5f96b4472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-07 15:25:01.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2925955471cb4d929c4e470d7c2e9317\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f789169a524174b5912307b2ec6890 : SUCCESS in 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc76bce8b5034e14becbbb1b39260e5d : SUCCESS in 9m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd00a0338df94b38936293bbbacb55fb : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283e30159a034576be013ab0311c45d6 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f7d109f1192045d298575934b69b2f20 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ec79e00be84d098487e9d93c3879f7 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ee16d400e6c40ff87796a53441f7872 : SUCCESS in 9m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8aa6198b4524f41afeadce40617a209 : SUCCESS in 20m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/077563733b514397afd877f040f439e5 : SUCCESS in 13m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dd0f739d685d4255add9abb8ae55b2a8 : SUCCESS in 17m 41s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/17f0c49f632d4fe5b5f97e65147157a4 : SUCCESS in 41m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/368f48b2524749408ad419be07fb03cd : SUCCESS in 24m 07s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db95294847e043fcb687acc89231f7fc : SUCCESS in 26m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fbe85baa4b03484f8222f1ae78dd9123 : SUCCESS in 33m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/560feac545b6458ca77e966d0a43a0f9 : SUCCESS in 27m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fb0b0193ab964f29863e1fd5ef02d3f7 : SUCCESS in 51m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ebef01027ce4177bcffe6aa7ada9060 : SUCCESS in 19m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5ac7809778a248c7b11de1485f374e66 : SUCCESS in 1h 26m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b8e76ef6e56b40239378e1deee495a04 : SUCCESS in 1h 22m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/19f77cf5a48f49ed8c17791dabfae774 : SUCCESS in 1h 19m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/055519249a6e43f289852c0e451e5887 : SUCCESS in 1h 28m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dde94f0d30464f30a3dc11f67b3fcbb6 : SUCCESS in 1h 27m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/29fe82a81a9140fa81e0a4fdfc7ecb34 : SUCCESS in 2h 04m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c03be2b2e3014b66a0c72745fe81e4fa : SUCCESS in 2h 01m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d0209506045d4b0e87ac86efc768af0d : FAILURE in 36m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cbd2898e9a834522bbe52a020cc55470 : FAILURE in 35m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11f4e7e58e2c4fa280a036e8137a9fcc : FAILURE in 29m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d922110d53f240b1abe2478cbd9e9d1b : FAILURE in 28m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2fc496412edf4125a36456a3454286db : SUCCESS in 38m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e9e5cfa128747138ae884358b8ada95 : SUCCESS in 39m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/85ad423bc419490fbac96f79988f8f50 : SUCCESS in 41m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd702c8705574d3d84dff89d37fdfd06 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"664574890bebaebc3afabe61f645d50cea20964c","revisions":{"5c892b180799f9d73cd668b221fec046e56f464a":{"kind":"REWORK","_number":1,"created":"2023-12-09 08:10:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/1"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 08:10:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 08:10:10.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c892b180799f9d73cd668b221fec046e56f464a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c892b180799f9d73cd668b221fec046e56f464a"}]},"branch":"refs/heads/master"},"bd1c87f1558c6696ceb6cca9e5810363276db598":{"kind":"REWORK","_number":2,"created":"2023-12-09 09:00:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/2"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 09:00:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 09:00:35.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd1c87f1558c6696ceb6cca9e5810363276db598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd1c87f1558c6696ceb6cca9e5810363276db598"}]},"branch":"refs/heads/master"},"c45bfb8e6ca6a54abccadbce73525e1bfbf810a4":{"kind":"REWORK","_number":3,"created":"2023-12-09 10:18:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/3"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 10:18:40.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 10:18:40.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c45bfb8e6ca6a54abccadbce73525e1bfbf810a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c45bfb8e6ca6a54abccadbce73525e1bfbf810a4"}]},"branch":"refs/heads/master"},"5b292d78ef9bce3ea5904f7431a908600d92cf37":{"kind":"REWORK","_number":4,"created":"2023-12-09 18:46:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/4"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 18:46:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 18:46:00.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b292d78ef9bce3ea5904f7431a908600d92cf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b292d78ef9bce3ea5904f7431a908600d92cf37"}]},"branch":"refs/heads/master"},"218eb24cb5d983367714f50223d249a96d1359d3":{"kind":"REWORK","_number":5,"created":"2023-12-09 20:43:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/5"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 20:43:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 20:43:43.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/218eb24cb5d983367714f50223d249a96d1359d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/218eb24cb5d983367714f50223d249a96d1359d3"}]},"branch":"refs/heads/master"},"5b9cefa66d57309c08931823c08669e90926f755":{"kind":"REWORK","_number":6,"created":"2023-12-09 21:06:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/6"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 21:06:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 21:06:36.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b9cefa66d57309c08931823c08669e90926f755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b9cefa66d57309c08931823c08669e90926f755"}]},"branch":"refs/heads/master"},"026feb4343a637a88183d81b6369f5270893f562":{"kind":"REWORK","_number":7,"created":"2023-12-09 21:53:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/7"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 21:53:49.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-09 21:53:49.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/026feb4343a637a88183d81b6369f5270893f562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/026feb4343a637a88183d81b6369f5270893f562"}]},"branch":"refs/heads/master"},"8e56ccabc34131ed72807acdcc9242495966ece6":{"kind":"REWORK","_number":8,"created":"2023-12-10 08:09:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/8"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 08:09:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 08:09:04.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e56ccabc34131ed72807acdcc9242495966ece6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e56ccabc34131ed72807acdcc9242495966ece6"}]},"branch":"refs/heads/master"},"a1592568dd1a21873ea7b3643b22614df1489090":{"kind":"REWORK","_number":9,"created":"2023-12-10 15:08:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/9"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 15:08:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 15:08:21.000000000","tz":60},"subject":"[WIP] Test overcloud provisioning with Bifrost","message":"[WIP] Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1592568dd1a21873ea7b3643b22614df1489090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1592568dd1a21873ea7b3643b22614df1489090"}]},"branch":"refs/heads/master"},"bf04dde6938e835282a25acd7ddcc250630e04fa":{"kind":"REWORK","_number":10,"created":"2023-12-10 16:05:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/10"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf04dde6938e835282a25acd7ddcc250630e04fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf04dde6938e835282a25acd7ddcc250630e04fa"}]},"branch":"refs/heads/master"},"f5bbc23de1c2a46794482aa8c257290d7b1dbc54":{"kind":"REWORK","_number":11,"created":"2024-05-16 14:55:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/11"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-16 14:55:40.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5bbc23de1c2a46794482aa8c257290d7b1dbc54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5bbc23de1c2a46794482aa8c257290d7b1dbc54"}]},"branch":"refs/heads/master"},"a8b646449d4bcff0a09fefaedf60110db2da296e":{"kind":"REWORK","_number":12,"created":"2024-09-16 09:03:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/12"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-16 09:03:03.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8b646449d4bcff0a09fefaedf60110db2da296e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8b646449d4bcff0a09fefaedf60110db2da296e"}]},"branch":"refs/heads/master"},"00daae80ce5ff021d1a428e6758a95f264d0521b":{"kind":"REWORK","_number":13,"created":"2024-09-16 11:15:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/13"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-16 11:15:58.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00daae80ce5ff021d1a428e6758a95f264d0521b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00daae80ce5ff021d1a428e6758a95f264d0521b"}]},"branch":"refs/heads/master"},"488516d8c0057838f28ddbd3833108a3226ee4f7":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2024-09-17 08:31:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/14"}}},"commit":{"parents":[{"commit":"49f91043c6c3b0549a8625baba1e6e218eb69160","subject":"Fix gateway assignment when seed SNAT is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f91043c6c3b0549a8625baba1e6e218eb69160"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-10 16:04:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:30:52.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/929603\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/488516d8c0057838f28ddbd3833108a3226ee4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/488516d8c0057838f28ddbd3833108a3226ee4f7"}]},"branch":"refs/heads/master"},"0b347661808bef5bf51fc4870a0865ed82b38d2e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-09-17 08:31:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/15"}}},"commit":{"parents":[{"commit":"6b37698527ca614aea1d262c0f56707274ea60c8","subject":"Merge \"Support skipping SSH keyscan for older switch devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b37698527ca614aea1d262c0f56707274ea60c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:31:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 08:31:17.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/929603\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b347661808bef5bf51fc4870a0865ed82b38d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b347661808bef5bf51fc4870a0865ed82b38d2e"}]},"branch":"refs/heads/master"},"9f0937c56aec38b783becbec59cb55d66f7642cb":{"kind":"REWORK","_number":16,"created":"2024-09-17 10:03:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/16"}}},"commit":{"parents":[{"commit":"6b37698527ca614aea1d262c0f56707274ea60c8","subject":"Merge \"Support skipping SSH keyscan for older switch devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b37698527ca614aea1d262c0f56707274ea60c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 10:03:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 10:03:27.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/929603\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f0937c56aec38b783becbec59cb55d66f7642cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f0937c56aec38b783becbec59cb55d66f7642cb"}]},"branch":"refs/heads/master"},"9f740f7941e037e079a6911329c5993b3f245573":{"kind":"REWORK","_number":17,"created":"2024-09-17 12:43:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/17"}}},"commit":{"parents":[{"commit":"6b37698527ca614aea1d262c0f56707274ea60c8","subject":"Merge \"Support skipping SSH keyscan for older switch devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b37698527ca614aea1d262c0f56707274ea60c8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 12:42:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 12:42:45.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f740f7941e037e079a6911329c5993b3f245573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f740f7941e037e079a6911329c5993b3f245573"}]},"branch":"refs/heads/master"},"f78d0f403acdb21a428b71851eee0d0f91495781":{"kind":"REWORK","_number":18,"created":"2024-09-17 13:54:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/18"}}},"commit":{"parents":[{"commit":"903126c01836f2404879291bd130060bdd75f336","subject":"Merge \"Avoid shared IPA image cache on Ansible control host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/903126c01836f2404879291bd130060bdd75f336"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 13:54:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 13:54:20.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f78d0f403acdb21a428b71851eee0d0f91495781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f78d0f403acdb21a428b71851eee0d0f91495781"}]},"branch":"refs/heads/master"},"92f86303b1fcfbac35abf859ded8db236a6e9f68":{"kind":"REWORK","_number":19,"created":"2024-09-17 15:01:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/19"}}},"commit":{"parents":[{"commit":"903126c01836f2404879291bd130060bdd75f336","subject":"Merge \"Avoid shared IPA image cache on Ansible control host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/903126c01836f2404879291bd130060bdd75f336"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 13:54:20.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-17 15:01:48.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92f86303b1fcfbac35abf859ded8db236a6e9f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92f86303b1fcfbac35abf859ded8db236a6e9f68"}]},"branch":"refs/heads/master"},"75772f91da3c33a125044bcf24c03de783717465":{"kind":"REWORK","_number":20,"created":"2024-09-17 16:17:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/20"}}},"commit":{"parents":[{"commit":"903126c01836f2404879291bd130060bdd75f336","subject":"Merge \"Avoid shared IPA image cache on Ansible control host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/903126c01836f2404879291bd130060bdd75f336"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-17 13:54:20.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-09-17 16:17:10.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75772f91da3c33a125044bcf24c03de783717465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75772f91da3c33a125044bcf24c03de783717465"}]},"branch":"refs/heads/master"},"de113e833afce3b7e1609d1859d58bb7443750f5":{"kind":"REWORK","_number":21,"created":"2024-09-18 07:24:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/21"}}},"commit":{"parents":[{"commit":"903126c01836f2404879291bd130060bdd75f336","subject":"Merge \"Avoid shared IPA image cache on Ansible control host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/903126c01836f2404879291bd130060bdd75f336"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-18 07:24:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-18 07:24:10.000000000","tz":120},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de113e833afce3b7e1609d1859d58bb7443750f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de113e833afce3b7e1609d1859d58bb7443750f5"}]},"branch":"refs/heads/master"},"1cb0801fe628c25c0774f6cf2d6268c3c756480e":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-10-24 13:11:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/22"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-18 07:24:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-10-24 13:11:36.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1cb0801fe628c25c0774f6cf2d6268c3c756480e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1cb0801fe628c25c0774f6cf2d6268c3c756480e"}]},"branch":"refs/heads/master","description":"Rebase"},"03e21e30afcb05b9b4b65c3dba37736e85dfe749":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2024-12-10 21:58:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/23"}}},"commit":{"parents":[{"commit":"3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-18 07:24:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-10 21:58:24.000000000","tz":0},"subject":"Test overcloud provisioning with Bifrost","message":"Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03e21e30afcb05b9b4b65c3dba37736e85dfe749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03e21e30afcb05b9b4b65c3dba37736e85dfe749"}]},"branch":"refs/heads/master","description":"Rebase"},"601aaef331807049150876b9aa7e0d6257ac63d0":{"kind":"REWORK","_number":24,"created":"2025-05-12 07:09:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/903249/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/24"}}},"commit":{"parents":[{"commit":"66640c0ae5b3010ce592e5ddd44bee99ee960341","subject":"Merge \"CI: Fix Bifrost log collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66640c0ae5b3010ce592e5ddd44bee99ee960341"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 07:09:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 07:09:08.000000000","tz":120},"subject":"CI: Test overcloud provisioning with Bifrost","message":"CI: Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/601aaef331807049150876b9aa7e0d6257ac63d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/601aaef331807049150876b9aa7e0d6257ac63d0"}]},"branch":"refs/heads/master"},"664574890bebaebc3afabe61f645d50cea20964c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":25,"created":"2026-10-07 09:17:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/903249/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/903249/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/903249/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/903249/25"}}},"commit":{"parents":[{"commit":"408c3d5358f700030b54dc37eed8e4161a81ce99","subject":"Fix cpu_mode and memory for dev overcloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/408c3d5358f700030b54dc37eed8e4161a81ce99"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 07:09:08.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-10-07 09:17:12.000000000","tz":60},"subject":"CI: Test overcloud provisioning with Bifrost","message":"CI: Test overcloud provisioning with Bifrost\n\nChange-Id: Ib9b9b531973dd6e32fd15b645e33384b00e64fe0\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/664574890bebaebc3afabe61f645d50cea20964c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/664574890bebaebc3afabe61f645d50cea20964c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
