)]}'
{"id":"openstack%2Fkayobe~904038","triplet_id":"openstack%2Fkayobe~master~I9965642b83be146155ed9d3a485de489aa38dcb5","project":"openstack/kayobe","branch":"master","topic":"ansible-facts-interfaces","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-12-19 22:03:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9965642b83be146155ed9d3a485de489aa38dcb5","subject":"[DNM] Testing ansible facts","status":"NEW","created":"2023-12-19 21:01:12.000000000","updated":"2023-12-19 22:03:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f80a5e253e19070e2ff4f53384f675637072804","_number":904038,"virtual_id_number":904038,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-19 22:03:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-19 22:03:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d168feb622cab2fa8d8f4c9168ec652606716f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-19 21:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f80a5e253e19070e2ff4f53384f675637072804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 22:03:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c41dcab0d80d499789ca68f57e761fe8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7acc95c10e324f8ca7d2fd9585c7a29c : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeed845aee5e4c5f961b9f97833f31fe : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51c61f42a6fb44cd9c19f4c701f523b8 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e9b9807191b473598c7665562312a01 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1da7c10a41ba45ec9288a73a465b3ef6 : SUCCESS in 8m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b47d37fb2f0a418cae2d04ddeb162a94 : FAILURE in 18m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/377ad7d0acb04bbe84520c3d9c102fa5 : FAILURE in 13m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adc5af20e7a1473c8ebdaeddc4309ea9 : FAILURE in 16m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dd7899ef97a945b09df14d062da54186 : FAILURE in 21m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eb45e4d2d39549a2a38c63d93c701ee7 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/86b7460460d14e9a91a7ff22f9e178ed : SUCCESS in 17m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9bf4442544d455b91d6ad785aab027b : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29bd253e90af4e98ba5f65be43d876f7 : FAILURE in 1h 01m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/101a6fa2368a4cf0b5d65de002f14e17 : FAILURE in 47m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f061d7da1b7d43209eee6bb2f013a7aa : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5916f7ea4f7246d0b975554f45db6055 : SUCCESS in 15m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/010fd13faee94cd09bbb28a0bb313a19 : SUCCESS in 33m 43s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b3469782c2464a94937e3039b20325e4 : SUCCESS in 30m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2c33af6bc7b4393b26628266c0b196b : SUCCESS in 28m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18acd84d6cd0455eb39bf48e5c1c228f : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79e657b828f645d783ed333040ed2afa : SUCCESS in 12m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/855520285d8d460a9e6dea329151d6de : SUCCESS in 18m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60a3bb60a0b3409f8353208c2631d970 : SUCCESS in 16m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d62685ee11e47e5acd75695318da0aa : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e033416f1b946e1a518805e23b1244b : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd3786670fee490488c1b6d345c76105 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c488da593c2276d63a7e66b6c34eb81b220a57cd","revisions":{"c488da593c2276d63a7e66b6c34eb81b220a57cd":{"kind":"REWORK","_number":1,"created":"2023-12-19 21:01:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/904038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/904038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/904038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/904038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/904038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/904038/1"}}},"commit":{"parents":[{"commit":"bb2af07567152832700a2c2971cd72ff4e3b77d3","subject":"Merge \"Revert \"Use OpenStack 2023.2 release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb2af07567152832700a2c2971cd72ff4e3b77d3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-19 21:00:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-19 21:00:45.000000000","tz":60},"subject":"[DNM] Testing ansible facts","message":"[DNM] Testing ansible facts\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/904037\nChange-Id: I9965642b83be146155ed9d3a485de489aa38dcb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c488da593c2276d63a7e66b6c34eb81b220a57cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c488da593c2276d63a7e66b6c34eb81b220a57cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
