)]}'
{"id":"openstack%2Fkayobe~904265","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36","project":"openstack/kayobe","branch":"stable/2023.2","topic":"kolla-git-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-01-05 14:30:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36","subject":"Ensure git is present before checkout tasks","status":"MERGED","created":"2023-12-22 13:14:44.000000000","updated":"2024-01-05 14:31:32.000000000","submitted":"2024-01-05 14:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904265-kolla-git-stable/2023.2","cherry_pick_of_change":903138,"cherry_pick_of_patch_set":2,"meta_rev_id":"ebe1cf8cf263a17c63afd5658b919d0829562d48","_number":904265,"virtual_id_number":904265,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 14:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2024-01-04 09:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-01-05 11:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2024-01-05 11:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-22 15:25:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 09:30:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-01-05 11:57:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6f36477d9f99f492b3dcd35bc234184d4a6261a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-22 13:14:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7cfc4d4783e26d50f22562d58bc3771f210c16cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 15:25:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdfea400cbfd40c4a00897ddb1acdcd7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22dfb1653ae7453abe94931f437cccc0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baea5bb86ba94c5b8c1e03f02dcdf1ee : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8ae64b3b7e542bd9e86cfd9a11b6c58 : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e696f6e22f144158db69dfeea482ac3 : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1d7d6b9cf7141f391d79e9fc7074f63 : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bfe40eae36a9486dbf943d8ae6a6065b : SUCCESS in 1h 00m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b10596b937d14670952ff3ebff07e16c : SUCCESS in 50m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b36c45f3925f48dea4fcd225ea01ce06 : SUCCESS in 50m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7c67c242349d4287a58fcd12d74f7f86 : SUCCESS in 1h 26m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8aa093f4fe1a4b7895886639644948fd : SUCCESS in 16m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/83c53028d4474858af6307c3b2dc2f9b : SUCCESS in 15m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc10cf8d9a184abd9a73203d179cc7c0 : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c7ab9bcea47426eb60d26a2f870e47c : FAILURE in 1h 26m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/247b4ce5d8b34f0592de47be4ead0b00 : SUCCESS in 2h 08m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/938b391abf4f4042a71614304353a63d : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53e6af5bb0d845cb9d48d87a70125e59 : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd6aead1252e4beab277f4eb2f49b105 : SUCCESS in 35m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5ac8a680c6e14c0484929cca3e003a55 : SUCCESS in 30m 37s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5926347d0504450980a0397bd180311 : SUCCESS in 25m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e35ab042d4f745e4bbad031147af2859 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b294f7cadf0e40148855ddba059c9daf : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea313ca4395a438999ebcb8b3932f0a3 : SUCCESS in 14m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51bf3d20e3204525893f65f9021e3664 : SUCCESS in 18m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e85f5bf0eb1044358b575a40371792ef : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e00cab6f0424b9a8e8bba3ab9d157ca : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/882f63c6d5e64bcb94561ded92d2fa31 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"be23bf4bd8ce75897a1df4f8d716542d64959fbc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-22 16:10:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f5ddfaf77d53649fcce61882052ac1c5789ae75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 18:22:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4caaadb733e4bd2a603d3b61fdf15fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8121a310abb43dfbd4e5322ee881568 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6d5bd6f341406b9a678a41b529e35e : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8962116a2b94239b7ae61bd5ba644d6 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c3a35eea20946e6931e6f991a34fc74 : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d1c37322e5044c696f193a7da93e54b : SUCCESS in 9m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/230d01dff9a8408b92dac685c39a5902 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5878d4f549a945829a8f81554d362d69 : SUCCESS in 1h 08m 54s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e050d15dddee4a808120eecb7ecec8e8 : SUCCESS in 56m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f9456a9de0d04d468cbf56dc17cf7f91 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/26464ee8659a4ded83f440bd706b5a80 : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/124acb595bf842af87131bf163ba58e6 : SUCCESS in 17m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7adbdf7059b44289b0ee3dd1a102edd7 : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d92aeb708ddc40009073a6181a70b2c4 : SUCCESS in 2h 10m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfa4c2d4dc5d4f8193c0585766c97d01 : SUCCESS in 2h 10m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dabfc02ac30c47fd99e4ea1fd7f37501 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79228606cb76494995b399d6560f7d5b : SUCCESS in 12m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0e4792880b0848e49144b28dc1487180 : SUCCESS in 34m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f7d6a488d1f84a01b39313265ec66821 : SUCCESS in 32m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e372c09f7dd46d9bc3fe3b4e27a75bc : SUCCESS in 20m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/427efe315e504b12817d376fa4155491 : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a8fc56426e1435b9b5c65f812f073ea : SUCCESS in 9m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b01ca4d0fc18422fa866860480d27a01 : SUCCESS in 17m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e138aa57fe24d55929a2e891a16c4dc : SUCCESS in 16m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2709d39cbea048778ac5a38ccb8bfaa7 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51887fa6ac294ae0b066d71f8e213670 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19f6c6ce98394e229004b305c763dd00 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c1c1804888d9b82c048f689ca2dc336d168730a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-01-04 09:30:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12804d8110fc12742505022dc015d2fde9c52736","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-05 11:57:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d98a37bb3a9fd5bc3f68d2cca329b2958baab87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 11:57:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b14782623bd9871139022039be6820ea0fd6dadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:30:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb419dfbefc44fd2984eb824b5e9e1aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998cf6760aad4f5494d8011d5655e7cf : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be24caf41d5a4a8c839308ba9106e0b3 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eaec6fcd4bc443949bd2b9349014e6c3 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/196812b10726440c902ac99fa19026c1 : SUCCESS in 8m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3dbef09199b4493f99fe49df710e51a6 : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/462af912205448798e33149be0626955 : SUCCESS in 43m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5362e834d7d34ca78e82bbe7120b07df : SUCCESS in 1h 12m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/78c01512d2b84ab294dcadfe548d65df : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e41330ae2b1943ddba3b75e9ad45b4b0 : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/817de6e758424da29bbd21214514dfac : SUCCESS in 2h 28m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91c9b4f637054710a566df6b406ed1ba : SUCCESS in 2h 15m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b04862cdbe044a1f90a2372fe54a380f : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8a8b0c97f07475ba7935d15d2b6f902 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d369559edf864bd0868b690aade7b548 : SUCCESS in 35m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/022fadcef32044b7b05b348bc45b5ce5 : SUCCESS in 29m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b39989823de4c79a2461f0e8c7e5cec : SUCCESS in 15m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c609d035dd2e4ac7984b955da42a3523 : SUCCESS in 13m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c706e885743e40698a083252896d6f45 : SUCCESS in 20m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e31764f53a854ce886fc0109fa8b22be : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0b835284d89bdfe47977426784dd5ae494399601","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:30:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebe1cf8cf263a17c63afd5658b919d0829562d48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:31:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3309c7148794023b64e9e041153b232\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10d64dbde2a4fe89184a9d6a4b1af0a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c759868cdea6de997a01a92d3d181e3fbd4c496","revisions":{"3c759868cdea6de997a01a92d3d181e3fbd4c496":{"kind":"REWORK","_number":1,"created":"2023-12-22 13:14:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/904265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/904265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/904265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/904265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/904265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/904265/1"}}},"commit":{"parents":[{"commit":"45c0b1868701aca7b5f080dc72306f009a6c1019","subject":"Stop NetworkManager from overriding resolv.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45c0b1868701aca7b5f080dc72306f009a6c1019"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-08 09:35:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-22 13:14:44.000000000","tz":0},"subject":"Ensure git is present before checkout tasks","message":"Ensure git is present before checkout tasks\n\nThe latest Rocky-9-GenericCloud image does not include the git package.\n\nChange-Id: I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36\n(cherry picked from commit 0df1a461dd8e6439f2e2d4877d94e789785f111a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c759868cdea6de997a01a92d3d181e3fbd4c496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c759868cdea6de997a01a92d3d181e3fbd4c496"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
