)]}'
{"id":"openstack%2Fkayobe~904268","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36","project":"openstack/kayobe","branch":"stable/yoga","topic":"kolla-git-stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-01-05 14:30:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36","subject":"Ensure git is present before checkout tasks","status":"MERGED","created":"2023-12-22 13:15:39.000000000","updated":"2024-01-05 14:31:49.000000000","submitted":"2024-01-05 14:30:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904268-kolla-git-stable/yoga","cherry_pick_of_change":903138,"cherry_pick_of_patch_set":2,"meta_rev_id":"dd858344347893bd475938bbd2a0adbf992f067e","_number":904268,"virtual_id_number":904268,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 14:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2024-01-04 09:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-01-05 11:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2024-01-05 11:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-22 14:54:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 09:31:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-01-05 11:57:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"676bdc593f4a1c3fcd153798d20348b6813f07ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-12-22 13:15:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"353e7966b819259c5d9b0bcc6176d424d8e3e934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 14:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0738d7f0d5e4b75842de9e74f2f9849\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22e278346fc247389d170599c9b271ab : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c2498cf03d4fddbc89c370c6f212ce : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740bc58255254321a897736d2ff66e6e : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01c7280511d14945a68e1200243571f5 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28c85e18a634750b5b84c44453b7116 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/196279fb03294fc4a58d0b9919c8b43d : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f53d9306c33f481787a020af55dbbb02 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e7337b116804affbd1c803e9d44d2bd : SUCCESS in 15m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/76f137a4d46f45f3813cad27b1801b9a : SUCCESS in 59m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/986fe3b5c2eb48a5ad7d98a8564bda59 : SUCCESS in 55m 49s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f06b71d9f6314370add2e03b2c2d46b0 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d15638195032473c8996df1b72dc393f : SUCCESS in 55m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/878df30bc97e4960b324ffc85eb23a4e : SUCCESS in 44m 54s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84a9c6439b2c447b8488b8ac848714e0 : SUCCESS in 49m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/419c0891364f4782bdb8e03abc2ba25a : SUCCESS in 27m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7e8c7b6e5cb94e6593aea00f90729970 : SUCCESS in 23m 32s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a329dd0fb2ca4859a661cb2b2fc2a495 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/54c75ffd850f4287b2de87a2bd737c27 : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0aefcb57f2e413490fdfb838de6fa59 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d76f4b9da32a4af688a9d0c12ce50287 : SUCCESS in 15m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4a246f990dc847e6a7cbdd4bb1c2f0a5 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/88c8d48e68604b5d8ff5683fc6442442 : SUCCESS in 57m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/935631a770844165918d650a1c4d84ca : SUCCESS in 18m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/885317f48deb46fb8bd4d38e2060bc7e : SUCCESS in 20m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1bbb7ac6eda6416cba2c532f1f448bca : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25b8e9bda5004dd39a0718afc8bf1394 : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cda410e3237e49cbaec2b3020598f16b : SUCCESS in 17m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c71d340b5f7f4d12ba08a9f2e0e45cde : SUCCESS in 19m 13s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c9b7780624be442e8d87f9d4c3f253d1 : SUCCESS in 16m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2cb340912caa44ce8501cc6024234ba1 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae1340ee34f4eec99b0cee0476e8c10 : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b34835e426b742baa9301546a2ef4797 : SUCCESS in 11m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3b0e96d403724d56bcfc1759737bded1 : SUCCESS in 22m 57s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/035f0b80dc5c4450a488cbaf40767d1f : SUCCESS in 20m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f698ef651be4395960eb57ded86003d : SUCCESS in 18m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76f22a0b3f0748a0b1d1d3b2e79f14a1 : SUCCESS in 13m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960916c01d644e02863550df6e9e3006 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1ce4d6812879160ed1e819cef54e1837344949e9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-01-04 09:31:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b19b6a6e7785f326c04c767bca64170c0afe8b0c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-05 11:57:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a6c27ee82c6003deccec44ba532c08d45df6637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 11:58:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24341e460248917cef72754fd5ec8832385ca13b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:30:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/270cb58d7e7d43f2bcc390cfae994363\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa51c3af0a44038a428f832e17d71d4 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca45fc6d5be4f958d59ae2f54c41c15 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e04103db81d64800853a848638f0c61b : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7fe8ca6f1f4685ba73b0c955874c52 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bcc2ab101d7455cbe29e1df511a9739 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa439b5598a843d5b94eb0b5272e899c : SUCCESS in 15m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11fd76afc74e4c6c8d982c34e515a44c : SUCCESS in 16m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/461ea307594e4b06aa93e235ceb52246 : SUCCESS in 57m 17s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/df4831ae5a294bc4a9e111ec0c41cbbe : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/31305e86d2f44bdc9ed060a0d797c8e2 : SUCCESS in 45m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fad706c1a524a9aa7243dc7ad088afe : SUCCESS in 49m 47s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0edd8b189b6b478bbcad16706e6c0cf4 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0397d1d35d6142a1876481e3d43b13ab : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3ad5feebdc294f6abd8984929e7225b9 : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/276c2a203b0e49d4a44e4e5008f45a4b : SUCCESS in 22m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abc5919d621a44af9fad3aed56fc52cb : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d944a88c0b84cf381351e737e3732db : SUCCESS in 16m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bc536f30b06d434fbe4d1c86ffab03cf : SUCCESS in 1h 20m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cd74b59df2464c2eb9f0d506323e26c6 : SUCCESS in 56m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6ace23cb0f1b4e3da358dbf0ac881b2f : SUCCESS in 23m 44s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e76ee1c3561c4790934a671d0a9fac86 : SUCCESS in 23m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/765254c115d4474882ae9ee305717de9 : SUCCESS in 25m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90fb54ed45834a5baa129468a335c0e3 : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b82180e5a4ac42d58a09be7d046badd7 : SUCCESS in 19m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5459fff8e882429d9fa5122b5d14f2dd : SUCCESS in 17m 37s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6b317d700c514fe8af50b7a7af6e27a8 : SUCCESS in 17m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f49719cb3004b278e03a170070a9575 : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b008eac20c56428ca513a5683d858405 : SUCCESS in 11m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3a62f30df84ed59921d0562f1ff2f1 : SUCCESS in 12m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/92cff0aa7ee142238236c37490890508 : SUCCESS in 20m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e94a349fb2314bfcae0d84fffdb535cc : SUCCESS in 22m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08d4d2cbfcc84383ba6f993fd80aa7f9 : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac347bdd9fc74a108e7e3bae9a64ceee : SUCCESS in 13m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83e2d234f6814e719aad9cff63045100 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"26661a5143f796699fd6fe68dd04b9a605ef34ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:30:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd858344347893bd475938bbd2a0adbf992f067e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:31:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d56e7d2bc84be99a54c121e51696ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4823df759ba47e499c5d20f0607a7af : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69ba8785b36e584368180ad4fb461f65aa5e59cc","revisions":{"69ba8785b36e584368180ad4fb461f65aa5e59cc":{"kind":"REWORK","_number":1,"created":"2023-12-22 13:15:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/904268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/904268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/904268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/904268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/904268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/904268/1"}}},"commit":{"parents":[{"commit":"b9cea681d18bba7cfe4a421468e8ba85fd916a36","subject":"Stop NetworkManager from overriding resolv.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9cea681d18bba7cfe4a421468e8ba85fd916a36"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-08 09:35:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-12-22 13:15:39.000000000","tz":0},"subject":"Ensure git is present before checkout tasks","message":"Ensure git is present before checkout tasks\n\nThe latest Rocky-9-GenericCloud image does not include the git package.\n\nChange-Id: I47a54b822d45ab5e1e64fa5e0f8111c0edadcb36\n(cherry picked from commit 0df1a461dd8e6439f2e2d4877d94e789785f111a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ba8785b36e584368180ad4fb461f65aa5e59cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ba8785b36e584368180ad4fb461f65aa5e59cc"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
