)]}'
{"id":"openstack%2Fkayobe~907293","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I6f62081d71232091e26269d4d7877203236c459d","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-02-01 14:29:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-01 13:27:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-01 17:30:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f62081d71232091e26269d4d7877203236c459d","subject":"Deprecate devicemapper due to removal in Docker","status":"MERGED","created":"2024-01-31 07:42:44.000000000","updated":"2024-02-01 17:32:05.000000000","submitted":"2024-02-01 17:30:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907293","meta_rev_id":"d874a4d69c0e75e11a05f97719ebbdd1ec3f3976","_number":907293,"virtual_id_number":907293,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 17:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-02-01 14:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-02-01 13:27:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-02-01 15:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-02-01 15:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 10:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-31 10:15:58.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-01-31 14:05:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-01 15:02:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"0a38f77a3dea03aee1da63b29f81aec9aa528c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-31 07:42:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3888f671df5a6f3b1bfdf352448ac2316e829d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-31 08:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3aa180a992badf16bfc281878fbce7a5b1ce29a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 10:02:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a253132abe184b248f1bf065b5e009cf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97e8eab32734034a24858d3fa2c18e6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d4ea2d1e7b424ead8993f193ccdf8a : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8cd21977867446ab0a5d178505fbcbe : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071cf9824ceb4ea1bbd271050d024c11 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebc7366929114ad29a2e93eb1f2b7674 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b8732e140847678b4858e137221a8d : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9aea2a458e2438fa3839caae2e2fc51 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/359ca8620e014b9aa37d67cb28b4ab37 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3acfcdc5628744639c80b2559d3b3323 : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5905084d32114145a9ab0ea027b09402 : SUCCESS in 14m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6fe9341e88f54fdcafa1ba74b0f39eb0 : SUCCESS in 55m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a95adaaa2b814d249c23437a89977473 : SUCCESS in 57m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b84b03c44ff245ea84a7c7c5ad05a5ea : SUCCESS in 52m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/88fcd91f97264d8383504b7403726835 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fc099709200642228af88fad900c6bbd : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d07ccb404bda4fc3982ef0130cecef3e : SUCCESS in 21m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4bbbf509af4441e976fb5cd5b4d5ca4 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4c1486c4f6b4709beb13593601d7270 : SUCCESS in 1h 38m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b02451b5cdff49fc96a8c87e77d43b5d : SUCCESS in 17m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/771f3555bab54e478fb7737ec7a0d0ab : SUCCESS in 16m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/450ad41619ad4a0d931fe12a2f1533be : SUCCESS in 35m 37s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9e528e0c10448aa9235f7480893378 : SUCCESS in 27m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/438b81a29e244c8587ae539d2fd3e07c : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f2ed96272af4a91b9a6f98735397e38 : SUCCESS in 12m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a26c4b7dddd7448a827c1f0898453925 : SUCCESS in 14m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59a076c0c0164ae18f60ea638cf7fceb : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"32d0cafe53bd093af11d4307673c0710f48164dc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-01-31 10:15:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21920144b5d4ece1108c0cb3e36b96d8e11b3276","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-31 14:00:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"85b47e9a2d90576dd4594edf20f5e2006a6ba433","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-01 09:11:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2fd1511488af97e077002c5af67910b9b99b69e8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-01 09:12:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"a9c35b2a8d9999e9b863d717104f4114ccd8c21b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-01 09:17:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"795aeb2db518ddb526f1f2cdeb710fd0b05b700f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-01 09:52:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"027d7a332bd7d43cdcfdfdcde3aaa1ea7e8c44f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:59:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7279e916c65b41e1a17a19ca97258d34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c4c0b12df804dcd8f24cfbbd5f2e0bb : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892281b540c549de9f0992a1253a4cbf : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a13443e8bd1c4bd08b1da5c7874525e7 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b078ce5022494e36bda6ef9e65eee084 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e44646f7bae4569995aa84ca7d6c4fc : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc9d27d7484452eb36fc3464a4d0c3f : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7405d720855941328510e76dddbab4a9 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f5edebe55a6f4e23a96edc00a4c238f3 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f643fd4f6b446cd84d5f943e0e31b41 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e024c78bddd24000b2d29e86bd1890c2 : SUCCESS in 13m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0142006df8e94cedb68d581570d41ee7 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/df87921ab86d4f3b99814fae7dcc0817 : SUCCESS in 40m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c246e7c7012c44d089210907104f2e97 : SUCCESS in 53m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8ca66f5314c74deda7a0b029f3096d8f : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1f74a09aec3a462a905ae9ffef59ebed : SUCCESS in 24m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2cd0afb4f8634a29b77ea9307a4eab0a : SUCCESS in 21m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db86eb6b03e436a92daf3525eb9bc63 : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e02478a6c764194bd4aa8c0b31bb164 : SUCCESS in 1h 42m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/10ef4b7a45b8450da8d7f87353c857c3 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f39039ea78dc4e88b24d437fe5299579 : SUCCESS in 16m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b36a53e660be4ceabcdf30881c721d82 : SUCCESS in 37m 17s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8e235c4c474503be39f357a5e0005a : SUCCESS in 27m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65f5eda0d6d34c2085d750e4860ba9a1 : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5ce7cbdd37247f687778420a427af82 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73762124dda3481ab89e64792f644ec4 : SUCCESS in 17m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d81e82475dd4587a167b36bc2df113d : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":3},{"id":"437c084ce9707fe712c6ac7d8e3886b91eead70d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-01 12:29:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bb8b77725f96bd4117540f912c942a10fb543fc6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-01 13:27:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b3bd1c5e191d8f565ed0d84185b04319654249e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 14:03:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a835b7489a204b7bbc1749eb1d39fc87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4295582df107416999af7f8be08ef34f : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c2b110b55549db80482f29195d90fc : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b8a57b9cfd441de8433b2ceab000ef5 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c64303a1f5ff452cb8d1f8605ed15371 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38a7dcfd078a4757a496e13915ba571d : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7af86cbd7df4c2b8c06cbf343172351 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727bbce2fb654740b191b37617c55b61 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/470aae71d65240c9b67e8e8627b8b06d : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/627303d5a66d40d3a5a2984ef0e56dea : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa29d8596f8f4162a44211716cb94bc5 : SUCCESS in 16m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd7bad14567543a3b4321075c46135cb : SUCCESS in 51m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9dedb9c2f1044ce6992fff9cc228851b : SUCCESS in 53m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d556a22510e4af1bcdfdb84a7bf0a54 : SUCCESS in 54m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7c9075c44e56446ea9f15413a6267462 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d732809cad0c43d88f6ea0ddf361f480 : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/91d2867e52f54424a12348ede3586bf7 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c138ab0c32524efeb51e82aa0c888a45 : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e41298c98311481e994093371df3f944 : SUCCESS in 1h 25m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8276097668b44d0090b65a66ff8a495f : SUCCESS in 15m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01c3c7f19f4c41da84b1e77c12bfe60c : SUCCESS in 14m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/70fdec9563d64297ad92644bde57f197 : SUCCESS in 32m 09s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b675cb55c116416b8012c1e58f6f5fda : SUCCESS in 29m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d8dc0ba728c43d3baf2ec6634ecab59 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9912699dcda6440b9d553c3024c72097 : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc196313b9634758ab60934e7c6c3604 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49c76451833d4b8a9abb89ff386b91f7 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":4},{"id":"b27f3549e7926dad5aff52d4dd15aec479e752e5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-01 14:29:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf0895deb7c6692784d7ebef9494ee10b6e4bf3c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-01 15:02:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7f46cf5303e4e445186fa935c9a1e654ed1be765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 15:02:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3092b87617249eaf58f1f794d57081788c30045f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:30:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4887f3f95351429bba2e6a9008db39d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3389430ecd1b446aaa9588d7f22fa7c4 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/904e43f6d50f42168c297812f38316ed : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b443377227e34f4a96463559170536f1 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04f9b8bff8c4ffea4e413f84d81598b : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfdcdd60978442ae8e48457ac6df5a14 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cea4ea5db2c4446e94d32eadd9d79b04 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd798ac100cb4f4cb333b73cade44f80 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0099731105f64ab2828446cb5c5fa295 : SUCCESS in 15m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b2f00eb4b7da46988e08ee99d2f0106e : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f85ca9c39ff4849ae0e92784525becd : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4c6000afc34f4466ad90a13a29e56d22 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/631e9620c72b4da8b60283b33cb98372 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/436c58c786834f0c97bf33385ab96c3f : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ef5035b3d17432183cea06751102aff : SUCCESS in 1h 37m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/29f59346f24a489982743847dbd7a2f9 : SUCCESS in 20m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/169cce9f88d6478a8ce76e4aa33e8836 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4235dd0300842cfb9d1ff54c1cda461 : SUCCESS in 35m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5007470305064079ab29ca4077f1363a : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/621a34612f4343e186f214076c47cb1a : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b8feb0e18f1847bf807a5e2ebacd2108 : SUCCESS in 20m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e3a973aec4b43868224d491621f2dd6 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":4},{"id":"b002e9fa9a654a1d6a95259ed8a8597848ea2654","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:30:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d874a4d69c0e75e11a05f97719ebbdd1ec3f3976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 17:32:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f4e37a3bf045778f299b1fb45a46de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d123a335e23d4e2a9ed0ec7e4b6e43e2 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0150cf839d5d4c62b245e8224a0129fa : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781","revisions":{"e9c1a037df6bc6053c71af739e56f3db3d4ccfa1":{"kind":"REWORK","_number":1,"created":"2024-01-31 07:42:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/907293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/907293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/907293/1"}}},"commit":{"parents":[{"commit":"f6cabd0189ec59dac0aa4b58ca2262167af4c713","subject":"Ensure git is present before checkout tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cabd0189ec59dac0aa4b58ca2262167af4c713"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 07:40:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 07:42:32.000000000","tz":60},"subject":"Handle removal of Docker devicemapper","message":"Handle removal of Docker devicemapper\n\nDocker Engine 25.0 drops support for the devicemapper storage driver\nafter a long deprecation period.\n\nSince devicemapper is used in some CI jobs, pin docker-ce package to the\nlatest available 24.x release. The master branch will remove\ndevicemapper support in another patch [1].\n\nAdd a deprecation notice indicating to the operator that they must\ninstall a compatible version of Docker Engine to keep using\ndevicemapper.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/906386\n\nChange-Id: I6f62081d71232091e26269d4d7877203236c459d\n(cherry picked from commit 4326762506ecfe5dfea3fffa4093166b1d74f19e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9c1a037df6bc6053c71af739e56f3db3d4ccfa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9c1a037df6bc6053c71af739e56f3db3d4ccfa1"}]},"branch":"refs/heads/stable/zed"},"16d4710bcb776a0497bf4cb8b17c1149aaa09a28":{"kind":"REWORK","_number":2,"created":"2024-01-31 08:22:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/907293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/907293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/907293/2"}}},"commit":{"parents":[{"commit":"f6cabd0189ec59dac0aa4b58ca2262167af4c713","subject":"Ensure git is present before checkout tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cabd0189ec59dac0aa4b58ca2262167af4c713"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 08:22:20.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 08:22:20.000000000","tz":60},"subject":"Handle removal of Docker devicemapper","message":"Handle removal of Docker devicemapper\n\nDocker Engine 25.0 drops support for the devicemapper storage driver\nafter a long deprecation period.\n\nSince devicemapper is used in some CI jobs, pin docker-ce package to the\nlatest available 24.x release. The master branch will remove\ndevicemapper support in another patch [1].\n\nAdd a deprecation notice indicating to the operator that they must\ninstall a compatible version of Docker Engine to keep using\ndevicemapper.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/906386\n\nChange-Id: I6f62081d71232091e26269d4d7877203236c459d\n(cherry picked from commit 4326762506ecfe5dfea3fffa4093166b1d74f19e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16d4710bcb776a0497bf4cb8b17c1149aaa09a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16d4710bcb776a0497bf4cb8b17c1149aaa09a28"}]},"branch":"refs/heads/stable/zed"},"913af162365fc065339e782a0cf073ddd898eeb5":{"kind":"REWORK","_number":3,"created":"2024-02-01 09:11:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/907293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/907293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/907293/3"}}},"commit":{"parents":[{"commit":"f6cabd0189ec59dac0aa4b58ca2262167af4c713","subject":"Ensure git is present before checkout tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cabd0189ec59dac0aa4b58ca2262167af4c713"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 13:57:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-01 09:11:12.000000000","tz":60},"subject":"Deprecate devicemapper due to removal in Docker","message":"Deprecate devicemapper due to removal in Docker\n\nDocker Engine 25.0 drops support for the devicemapper storage driver\nafter a long deprecation period.\n\nSince devicemapper is used in some CI jobs, pin docker-ce package to the\nlatest available 24.x release. The master branch will remove\ndevicemapper support in another patch [1].\n\nAdd a deprecation notice indicating to the operator that they must\ninstall a compatible version of Docker Engine to keep using\ndevicemapper.\n\nNote: this backport is amended to set packages via kolla/globals.yml\ninstead of zzz-overrides.yml. This is required for Zed and earlier\nreleases, which still use kolla bootstrap-servers to install Docker.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/906386\n\nChange-Id: I6f62081d71232091e26269d4d7877203236c459d\n(cherry picked from commit f018e0a8ba97a998fad977d16cb1ea3fbb4e8335)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/913af162365fc065339e782a0cf073ddd898eeb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/913af162365fc065339e782a0cf073ddd898eeb5"}]},"branch":"refs/heads/stable/zed"},"fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781":{"kind":"REWORK","_number":4,"created":"2024-02-01 12:29:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/907293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/907293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/907293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/907293/4"}}},"commit":{"parents":[{"commit":"f6cabd0189ec59dac0aa4b58ca2262167af4c713","subject":"Ensure git is present before checkout tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6cabd0189ec59dac0aa4b58ca2262167af4c713"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-01-31 13:57:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-01 12:29:24.000000000","tz":60},"subject":"Deprecate devicemapper due to removal in Docker","message":"Deprecate devicemapper due to removal in Docker\n\nDocker Engine 25.0 drops support for the devicemapper storage driver\nafter a long deprecation period.\n\nSince devicemapper is used in some CI jobs, pin docker-ce package to the\nlatest available 24.x release. The master branch will remove\ndevicemapper support in another patch [1].\n\nAdd a deprecation notice indicating to the operator that they must\ninstall a compatible version of Docker Engine to keep using\ndevicemapper.\n\nNote: this backport is amended to set packages via kolla/globals.yml\ninstead of zzz-overrides.yml. This is required for Zed and earlier\nreleases, which still use kolla bootstrap-servers to install Docker.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/906386\n\nChange-Id: I6f62081d71232091e26269d4d7877203236c459d\n(cherry picked from commit bd16d2931d79d6df86f9e400e5387d924d99df30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
