)]}'
{"id":"openstack%2Fkayobe~907588","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I7efe883fce4117a5167b6db4a711004d77a44f81","project":"openstack/kayobe","branch":"stable/2023.1","topic":"enable-both-ovn-ovs-image-build","attention_set":{},"removed_from_attention_set":{"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2024-02-05 11:20:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7efe883fce4117a5167b6db4a711004d77a44f81","subject":"Make matching OVS agent independent to OVN if OVS is explicitly enabled","status":"ABANDONED","created":"2024-02-02 14:40:32.000000000","updated":"2024-02-05 11:20:57.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7ce33cb333e511193303b5d07fc41652db7b5e97","_number":907588,"virtual_id_number":907588,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-02 16:48:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":1,"date":"2024-02-02 16:25:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-02-02 16:40:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 16:25:19.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2024-02-02 16:40:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-02 16:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b638fa7b88c045db25d504cc8551cfb120c77af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-02-02 14:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f9c6211baf4011c7a9ca979ab11796cace4f964","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-02-02 16:25:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d77ceae781edd0c658ed8d0b28f9dea956c362f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-02 16:40:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb4dd825517968fc12fed9383dd2e63abcb86e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 16:48:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c2a94572cf74c3599fd19a26a49c84f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6d2aef823b4d6e97970f186014a1ad : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf39ab90c024b2d8373c38277a08dbf : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b128031a2d4c45d4a901ab57f418a3f3 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8888fbc5ec54be5bb63df7d86179866 : SUCCESS in 8m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c3486a8a8f3f40cd95d0b1fd8d46d685 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46909f55ad2e4969ac8b84b8c9aa60d3 : SUCCESS in 13m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f95785e62bf44b6d809cb88e34ca9f3b : SUCCESS in 12m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/74142ea240e14be5abe2dd774c994be1 : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/184673158e1c4a78a39324d7e5a9d2af : SUCCESS in 1h 04m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ad76b30455e450f9d48afb58bce67f7 : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3a80697680554822aea7df206716d3d6 : SUCCESS in 1h 06m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/718e80bc47724d4e901c763b8c399f5a : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4bcb9652e64b4b099c17669f8b9535f5 : SUCCESS in 16m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24ed927ee375466aa926697bf305c707 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a945c08f4c344d6c9b1203a862163f3a : SUCCESS in 2h 07m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee790cd0c5784560bae145fad0c06fc7 : SUCCESS in 1h 24m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5971c56826b41cc844d5d0da9a1815e : SUCCESS in 20m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e2a711342914f968b530d1503266a58 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ce0d9cc6d7f247c4b43761fe20ab19e2 : SUCCESS in 32m 45s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/941475eaf3ea42eabaccd05efe8efaeb : SUCCESS in 29m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55866b3f808f4f4cbbe053d7f9b531a4 : SUCCESS in 23m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e04c2bfcdd544202bb88c9afd0e4f808 : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/893678721a1c4cbb9c3e894c8608a5f3 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2630b18a0c354b59b1d3f594482013d8 : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a7539027be549c1997e02b9b64ae5e5 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7ce33cb333e511193303b5d07fc41652db7b5e97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-02-05 11:20:57.000000000","message":"Abandoned\n\nCommitted to wrong branch. Please see  https://review.opendev.org/c/openstack/kayobe/+/907721","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15e8377b2f2ae4199cc9cfdb216eb1bf8b287362","revisions":{"15e8377b2f2ae4199cc9cfdb216eb1bf8b287362":{"kind":"REWORK","_number":1,"created":"2024-02-02 14:40:32.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/88/907588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/907588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/907588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/907588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/907588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/907588/1"}}},"commit":{"parents":[{"commit":"d397ceca74667ae0121b7bfebab77fd6888767a0","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d397ceca74667ae0121b7bfebab77fd6888767a0"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-02-02 14:39:50.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-02-02 14:39:50.000000000","tz":0},"subject":"Make matching OVS agent independent to OVN if OVS is explicitly enabled","message":"Make matching OVS agent independent to OVN if OVS is explicitly enabled\n\nCurrently OVN container images and OVS container images are mutually exclusive.\nThis becomes problem when all container images are needed to be built at the same time.\n``kolla_enable_neutron_ovs`` is added to let user to enable OVS even though OVN is enabled when it\u0027s explictly set to true.\nDefault value ensures regex mapping behaves normally when ``kolla_enable_neutron_ovs`` is not given.\n\nChange-Id: I7efe883fce4117a5167b6db4a711004d77a44f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e8377b2f2ae4199cc9cfdb216eb1bf8b287362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e8377b2f2ae4199cc9cfdb216eb1bf8b287362"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
