)]}'
{"id":"openstack%2Fkayobe~907928","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I5644ab2595a30cfff82f03d5ca4aa501a8b36ef8","project":"openstack/kayobe","branch":"stable/zed","topic":"bug/2044537-stable/zed","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-08 11:48:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-15 09:42:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5644ab2595a30cfff82f03d5ca4aa501a8b36ef8","subject":"Reload NetworkManager on DNS config change","status":"MERGED","created":"2024-02-06 15:57:59.000000000","updated":"2024-02-15 09:43:27.000000000","submitted":"2024-02-15 09:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907928-bug/2044537-stable/zed","cherry_pick_of_change":907740,"cherry_pick_of_patch_set":1,"meta_rev_id":"6c2659a80a14dd717bdf83d8c5ff0579fc3d72c3","_number":907928,"virtual_id_number":907928,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 09:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-02-15 07:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-02-15 06:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-15 06:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-02-15 07:17:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 15:57:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-02-06 17:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 15:16:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-15 07:17:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"db799071c10f20909b001189673bc6b62194c893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-06 15:57:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd57a4089d3186b155e86a71ad787d3f89ae0da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20a24403c220484d8c684af2796f7083\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e96c5c5958e4b259bcfbe9a95ed8704 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727b09b54f5949db9f76688cead1d85f : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e9d1f2297c472996bc489634ae6109 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/971332708b354cb8ab63371a32faa6ed : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a7e7fd46ef47d49138b7e27d0475b9 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50cb4dfa36d499fb1b423df89ad0789 : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb10ef1bdf1b45ac953f22b467c9fdfe : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49d31cdfcaed45bf93f9e81e2ea6ad61 : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/593dfcb715294fcbbf7364b9b5962d9c : SUCCESS in 15m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29cf559dc8a14c3d89d1d7ba91ac3cb5 : SUCCESS in 54m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/81ed852519904fb0bf437f78e7ec70f5 : SUCCESS in 59m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9234d4ec32df4ea5bcf3ed5f341157de : SUCCESS in 54m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1dd74b92c78447a6a11244e517511e04 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2e51c77e870048bdabd37d4ea6215404 : SUCCESS in 21m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a91d715d711d472ab4b032d314c9a8b3 : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8ad24b956cd4fa2bbc6a454cfbfb925 : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d874b13cb384b8391a028c7d79b5eca : RETRY_LIMIT in 4m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f130007ae09c43a98eac1ae3c9aa3b63 : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba584eaf8c664bbf8a655e3044d02eb1 : SUCCESS in 16m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d1e25f8a92384682b9c3e842c877ea12 : SUCCESS in 32m 51s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e517f474c304ff296e6660e38159139 : RETRY_LIMIT in 5m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/abb67b2828164034b4f2ebe7222c0d67 : SUCCESS in 13m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faf6418c090548829a309d708600f5dd : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e592a9191f0342509ea61425694c5d08 : SUCCESS in 16m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3350ed40e05a43f980157fec172f5cbb : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3c73a42b82b4d75c811a139cd8d9ec6408738fc0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-08 11:48:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"90608b11a8fdc6b12d608f796fea57a978b0dbb6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-09 22:27:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50de0ab6815ef653eb7940eb98e13778222b7c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 23:34:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c37bd27da66b401a8da8f7ca1db2db91\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa85cf32edd40d3803e8334684ccf22 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979afe5ef34c4a05afb4657d23dcc6d8 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1bb8b9c7e9498f9178acd205cda867 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f564ebf0867d4e88820fa14353d8eaf4 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8c94769d5fb48c2acc39719d7149fa4 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b917d90a71834fe2ba3366577eeaa116 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30bc47ace4d04be0b058f735bf6d8f38 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9e5c8ad703642f1a734ba5130a52737 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95f7bdbf6c564eefae895a62481acbc5 : SUCCESS in 16m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f3a119b32b146f9b7ef9fcf534ebcc8 : SUCCESS in 59m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4ff1cd625d954986b9cf2404bc1ff03f : SUCCESS in 53m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d065bf9dd16e45c3bd6cada91b5392e0 : SUCCESS in 54m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1b73692ee87a4362aade6644bc45ca9f : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2661b0a5c8f844129870d06234535cb2 : SUCCESS in 25m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5a803c6847674ebf8dd45c99e5614d6b : SUCCESS in 15m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27e5a8fdf52b4499a4ce5a810293c261 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/739dac003d424dc28c3ed003b0d9fb52 : RETRY_LIMIT in 4m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b6e07b7ff9114bfc94a7af8f0bb0ee28 : SUCCESS in 32m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/729877b7d62d499886473be4fdcfcf05 : SUCCESS in 17m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d7fde67afa714331aa0215dd8ff36d58 : SUCCESS in 42m 46s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25fc3a94fe3f4b728c99ea1464e58671 : RETRY_LIMIT in 4m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8d65a550e96c4eab9792d55504c1ba91 : SUCCESS in 14m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8b7d41186b84a9890bf63ba259c1f11 : SUCCESS in 13m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35232fe879cc46d8a4a4aa12b855d67e : SUCCESS in 27m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ebb832b8c64adb9adfd864b2e3800f : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d271e62764ffbb3c5de1d56a14e5cbb27a4d9f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-15 06:52:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe721a2a97ddb7156e27cff0a9b324da327bfd51","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-15 06:52:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ac5652c4e9dd900eaa47e6330e8409f23857dd06","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-15 07:17:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4e9aad08a40ad812a6c8ff3b6d8af8f24c54a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 08:24:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d6662238364d55837599572bab039e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a511ab93aa954ea4baaea8d2f046edcd : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59f72876d864d3ab01423d0c5818dc4 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04a3778b5e2d4f15bf0ca785689d4a4d : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a25819b61344f085dd8b5eed037ea6 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e40341858b9b444b8ef606ca4f9735df : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81bb7e8f82974c6998cc70eb82ee48e6 : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f7fb21a1db0417b8e1ae723bab233e6 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/819ab8304df54651820fd9817451b5a2 : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d25bfa86a8f46ccad92fa0881cf5d30 : SUCCESS in 11m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/817fe09c7c994a8ea4d3e9fd80f0d90f : SUCCESS in 59m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f9ac3210d8b14d7daa655708b4cb3053 : SUCCESS in 54m 00s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6371abe0a01477090b65186ae6e5e8c : SUCCESS in 40m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d87e3002aee7479e9f60dc4c96a8e8de : SUCCESS in 53m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f8326f2ce9f84e63a6e492c2bfd71028 : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2450b6b50acd41fab9095974ec0bf570 : SUCCESS in 15m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bcdb55991d34e85b4098ddf8af40053 : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e3a6c1d2184e568061c3808942fe09 : SUCCESS in 1h 27m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/64afadc992d340998fa3daf32d2106e4 : SUCCESS in 18m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76284fcfb3694b068d01761e8a50bd64 : SUCCESS in 15m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e89992b4bc764504aee9f1e88f589a04 : SUCCESS in 37m 27s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72acdc9da1df42cfad5f22d7ca80f25b : SUCCESS in 30m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7c01961ec8ee4cf79933e4e771e6acde : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7bdf36b1223491d8bfacacb7b4339a4 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0254db7c01164399a4072d688fe616e7 : SUCCESS in 16m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a4d6f2fafca4e33a6eb42c4471dcb5e : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ca5d7c723d7c82a49b92cafffef9b11bb069cafa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 08:24:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47344d9938693fd14372844ba450388e6f9197ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 09:42:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34b0032edb9a4c45b21e0120ea443810\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390226a935934aeea5d8d4d05835037d : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d662c62b12414a69ad7c4d97fe2bdb6d : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e74a89f14e641c39e3b8ecf1d18364c : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6cad0e48ce475c89b9e9723c9118be : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63c095fbced54d50817f554df9f71153 : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46a404cb67074ea288ce3176d3ba33a2 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9366d63cdbda4ff8830e74f89d91ef1f : SUCCESS in 13m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f34cc3241052483faee454aba67e0f14 : SUCCESS in 52m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba814a905aee4b709302e3b8fcce062a : SUCCESS in 50m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/faffa928d67e4bb4860252f17fe84d2a : SUCCESS in 57m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/79d9aece43bc4943b57c8e53e37ca93a : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b7694cb039f45ae897abf56d48e69c5 : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1357a10819640b9be4782f14fca81e5 : SUCCESS in 1h 16m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff253e96cc2a4aa495fcd0aeb1d8d81d : SUCCESS in 15m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12d9d9f842654b1da700eda144806d0c : SUCCESS in 13m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29d3a5b459e42c49149947b9f84acf4 : SUCCESS in 29m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dcaa5a231f30426c8b0120b10090150d : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b8c47f2010743b6ae1f2056c0055d88 : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fda05e500cf14d6594646b901db8fa4a : SUCCESS in 19m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b480e49a3c194e04839129c7f4d99f27 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a29686d772576e1406642949c2aa13b8b727c04c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 09:42:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6c2659a80a14dd717bdf83d8c5ff0579fc3d72c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 09:43:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7d83f8afff4f5691687712becba28d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5888de9bd6dd48f5811fee7e6290286e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4427530c015cf4b1ee94456a427d8a0539492f62","revisions":{"512a3b304b97fd0046eeba59394fa4f360e6828a":{"kind":"REWORK","_number":1,"created":"2024-02-06 15:57:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/907928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/907928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/907928/1"}}},"commit":{"parents":[{"commit":"fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbefef43d7d17bc77ba9a0d59dd2f2d04b1c5781"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-05 13:53:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-06 15:57:59.000000000","tz":0},"subject":"Reload NetworkManager on DNS config change","message":"Reload NetworkManager on DNS config change\n\nInitial fix [1] was not reloading NetworkManager config, which caused\nMichaelRigart.interfaces role to clean up resolv.conf after interface\nbounce.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/902561\nRelated-Bug: #2044537\n\nChange-Id: I5644ab2595a30cfff82f03d5ca4aa501a8b36ef8\n(cherry picked from commit 2a73c328d39c24772887031c6162b608a980e388)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/512a3b304b97fd0046eeba59394fa4f360e6828a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/512a3b304b97fd0046eeba59394fa4f360e6828a"}]},"branch":"refs/heads/stable/zed"},"4427530c015cf4b1ee94456a427d8a0539492f62":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-15 06:52:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/907928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/907928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/907928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/907928/2"}}},"commit":{"parents":[{"commit":"291ffd87f58446eebc4396adcac732dcfc490a4d","subject":"CI: support .gitreview of unmaintained branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/291ffd87f58446eebc4396adcac732dcfc490a4d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-05 13:53:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-15 06:52:16.000000000","tz":0},"subject":"Reload NetworkManager on DNS config change","message":"Reload NetworkManager on DNS config change\n\nInitial fix [1] was not reloading NetworkManager config, which caused\nMichaelRigart.interfaces role to clean up resolv.conf after interface\nbounce.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/902561\nRelated-Bug: #2044537\n\nChange-Id: I5644ab2595a30cfff82f03d5ca4aa501a8b36ef8\n(cherry picked from commit 2a73c328d39c24772887031c6162b608a980e388)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4427530c015cf4b1ee94456a427d8a0539492f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4427530c015cf4b1ee94456a427d8a0539492f62"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
