)]}'
{"id":"openstack%2Fkayobe~908142","triplet_id":"openstack%2Fkayobe~master~I6c9686bdf736d58738fbe88169e62833cc85d2f4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-03-01 23:29:28.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-01 23:29:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c9686bdf736d58738fbe88169e62833cc85d2f4","subject":"Support credentials for custom DNF repositories","status":"MERGED","created":"2024-02-06 15:07:04.000000000","updated":"2024-03-01 23:30:27.000000000","submitted":"2024-03-01 23:29:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908142","meta_rev_id":"7cc8d50734fcafb6c25690fa9beda43e2da271fd","_number":908142,"virtual_id_number":908142,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 23:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-01 10:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-02-07 15:08:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-07 08:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-03-01 10:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 17:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 08:55:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-07 15:08:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-01 10:21:08.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"797e233905ee8358cb59a55c5aabb89cfd8a2155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-06 15:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc3c6d2d077f2c005bd27f201d467e07398ada5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:36:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8417dfb93f1d48ad9a0244b0af2413cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/652a74c20a05476fa726e724c812cce0 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b25b916964482fa2714e2fa4765fe0 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1936bf1be3314fcda27429401f6f1f6b : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a3fb7d0d5314df59a3078c7c358a770 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c1efad0e9254a8c8f3ea8c48f5b28ba : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7db41b23f9645cbae8dc1a28535d40a : SUCCESS in 8m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/51e762b8838942fcaf19849d8126ccf8 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7d09b2de561f425fb65ee7a14d4c6e87 : SUCCESS in 1h 03m 18s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bb6c6f9cd414738b5e33551711b09ad : SUCCESS in 55m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/be6eabe30029453096c25eca6e43b0f2 : SUCCESS in 1h 20m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01b8b86b10b7499db2836cc553195e58 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ce38e73cfc9d485084621c7b5a061a0b : SUCCESS in 17m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b5235778b7245ab996237666c35b923 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9a9d0fc2158047f79269b0f3e2c048f6 : SUCCESS in 2h 22m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63371eda9d394ba9b57899bbd2a39f6d : SUCCESS in 2h 00m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b74fd7851e8e4f3babe41f6f5ffb15cb : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be29d0277fee41b2b4525532bd2cfd43 : SUCCESS in 15m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/726a99a84eca43d98782d713edb2df28 : FAILURE in 34m 39s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e4c27d0b40f946ee9cfe9dbd0f02a296 : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/974d0eb8bbf142adad17667a806f8804 : SUCCESS in 25m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a405f260bb2944ecb4b0ea8cdcbc4881 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7207dd20282147ab9b725a2c5107e919 : SUCCESS in 9m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e81186fb1f844ea7a461b11a1d59e075 : SUCCESS in 17m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b997c35d15144bca4a3f7553dae6703 : SUCCESS in 15m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05adf668f7fd42c7ba6de5ccb48ca951 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b39ef67d82df4d0c91e5fac36e6a0630 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55f3ad574b394da28a0603dd81966a67 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f091f5ac02f726a268f35482c33341337e83cbca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 08:55:02.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b167ae83e9625e208847ae6e79048ef0adc4c555","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 08:55:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9cc30021b39a1945c48f7d22ad7c15135fe0446e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 11:14:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f59a9261fef4f7381b15cc2942897d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211dae368e37450eadad7a6f805b3846 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25704a8aa3e348fb8ee6fc7e1e5f8dfa : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a773e8be8ef045878cdc77dc7de8587a : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e10604fa4b649e3aa4a1f91061be924 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60aed6ee35064d0fb3f85dd4547e67cb : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d38f2f36ab71499aa0545a3a410f785e : SUCCESS in 11m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1d02a1c8f755401881f7362e5c749b57 : FAILURE in 20m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/73097bdd2a59409aa56cc6205bde81d4 : SUCCESS in 1h 12m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2520f93b8b7b4ef3bcaba831ecf49b45 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d631a6d4f15b494190796b7ab356cd8d : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dc040d1359c74fb8bb8e0f93a94fdd77 : SUCCESS in 16m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f91ac4ae88344a509b2da06c42f87061 : SUCCESS in 17m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e10eda39864e48149a21fbfe615febcc : SUCCESS in 13m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca29ec2a4bcf4383aeb6c907e9cf9a18 : SUCCESS in 2h 04m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d46c5f7fc1744e3b8df71563b0494f90 : SUCCESS in 2h 04m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1525a0ae44cf418a896e000d8cea8429 : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/180e7f48c063453c947a2386d78a4b28 : SUCCESS in 14m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/770f1cac56a94c32862d23505ad94906 : FAILURE in 35m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/277d9010a0fb4c02925fc7e3ddad7137 : SUCCESS in 23m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/982654f9ec72468c843fddb68ede2e6b : SUCCESS in 25m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/821907ce82b544a7832341eb59ce0166 : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e3670884bdd499dac0c8444cc8da893 : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec8a3ae1ac2049f4973af8fd3fbe28f1 : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95dd2bf4ae6c4832bcb1865d56b06702 : SUCCESS in 14m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b70b71205cb048f8bbe45dc90ce7b6c1 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/264fcccade5842f8950fb2fa39caf19b : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d56c2772c9ec4cdaa1bf02bde014dc64 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"232b3a48ddfee45ef85da20ad988a20068edd4d4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 11:20:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e73df009a8f59ab2daa605d9552ae63c92d3665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 14:10:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b95fb0fbee824a1b874f61d684a80ee4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/917c46f52c70413ca7a35aa042fc11ed : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac947ae73a64c649f1a55db66904a82 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/620c0153879b44788babaa442b3e52a4 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d92de6fa60a148eb8586010c25c78b45 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84ffe83dcc18491193d8f44ec8e013ee : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c769ac3007c642dd8b38cc561204527b : SUCCESS in 9m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/234633ccdf254a5690fc514f75a41584 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cd208048654e48858c2fa4bf8a76ee54 : SUCCESS in 1h 03m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/484be3afaaa04135b5f1ccb85439e656 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c693a01799cd46ccb0128d7da448b52d : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/12618bebe6d743149b892b1aeae8a2c8 : SUCCESS in 20m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d773b0dcc2a94332b4ae38dde713e5c6 : SUCCESS in 17m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a356454e41041c2830bca18dea8c879 : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/953a8107a4e042e4aabbd3c87720bc7d : SUCCESS in 2h 40m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55e8940c28de44ccacc40fab5ae52fb9 : SUCCESS in 2h 04m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/606d59108a144e088ca417f8691eede3 : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7896f721c86748dca5c713e9645c790d : SUCCESS in 15m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b8e71c59425e4d1c8cd99592299b329b : FAILURE in 35m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1d278fa32e04713806fdc6fc24c0a6e : FAILURE in 15m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d82ffd9f7f47a987d55e68161df137 : SUCCESS in 23m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/333fdc097ed04bc99166354cfa8c56d6 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/406f66ad13964a88b76747851647683a : SUCCESS in 11m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d17dc5f4727646a0a27683be4ff14670 : SUCCESS in 20m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63327f79dda545c7b2ad34228cb874c4 : SUCCESS in 14m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec9c29eb15204ab199d844ce3f2184b8 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ceede5192ff4de5857b572217bb38ba : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c4c266815ad4ac0a10585971a90134e : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"963ad393f5c8c021fdd96be0f5a74750ed652ad8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-07 14:27:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9dd8660771a5028b04cd27d1f23128fd98aaa49","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-07 15:08:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"72dca1d1dbd3ffd21901fe17e4b17272bfb472ec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-19 12:21:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ced3ff24cac8c39d942fd5a369154a6b02a22bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:52:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8857806dd145437bb3bf390cdb588a65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88d732c361d473cae115e639b474643 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240d9e4af69145259fa95ef4cf7a13a5 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f73e5e3c53c9459ab374e7f8146a6931 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e4da137aea64b03910f4fd85be3f9d0 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/842c6b3c4dfc4daf92c506991fbd6ab5 : FAILURE in 9m 31s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b3a15f49d784071bd668db00138064c : SUCCESS in 9m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b3285efe511a4363ba881a10e9c0f873 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5eeca47f429f49559ff9397e7618876c : SUCCESS in 55m 04s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62f16542ca99428080cdf5c2af879490 : SUCCESS in 50m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3795c517b2b941a5a0254bca8df0db99 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/750268b205db433fb6bbefd4be01b894 : SUCCESS in 15m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8ebc450674e14811863ecf78d92e6429 : SUCCESS in 17m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b66aca5e03c445c8f29c0cf4a92b9d3 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9bf67b5a87de4b8ea979f66fa9c48969 : SUCCESS in 2h 25m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b064398abb34aacb23fb6af52d4e12b : SUCCESS in 1h 48m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/54cadcc8360841639bcbc800f29e6ece : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dbdeb985a844149b6d62e12da4d6b5c : SUCCESS in 13m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/181320dc3707496fa4730d7c60ed9a05 : FAILURE in 28m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0141f7a99ad84796830770387157fb50 : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bb2e88893a54eb4abfb77a3f6a165eb : SUCCESS in 21m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd0ef95bea654a9694dc95611461623d : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/736ae0dd92b14512a401cfd8c5be6abe : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59771a24c3d64ffd8e2c2cee6f327407 : SUCCESS in 17m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe1d264f9db644b1bfc176519482bcee : SUCCESS in 13m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31ac05e3427d45e4834c610c982bdf50 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38e2b039fa3641fcb6de3c06be331835 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea23895cff10431ebc60fcd8d4da59ef : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5cd688ed5c6db3561e6a453edd1a132101a9d233","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-21 11:53:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a58cbc8f60dab80c0ccc2f69b54de48d6fa73788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 14:12:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/175234169b8a4b6282463d6b8080982b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9865090a6ac4465da20387a2a79d9826 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e73e7e80044ef9b77bbdf8703f8c1c : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d5cadb0a4e14335a8f962a7cba3a659 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2747f98d476740129bd66d3791f4b600 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ab1060fe89d4b5da9baddccc2ec5883 : FAILURE in 11m 29s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba2a4ae8c92d47f3b2451acb57bc3d4a : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8c99bfddadb048a1b5e9f4000c575575 : SUCCESS in 59m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0068b14247dc4ff08a946ec58edaebc9 : SUCCESS in 1h 07m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96720bd569b14419987130e4f20da674 : SUCCESS in 52m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7816f9f79ea44acfbdab1cc1ecedff0f : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cdc324103223465a9dd856c31b40fa45 : SUCCESS in 20m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/41b085d086bd43d9a93118689eff6dbe : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cab57586e99d46e4bdef3ad60016e4f7 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bf5034574a9b41c1b06686cbffd9474b : SUCCESS in 2h 16m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf8f28c1ff0b4e728e36f2c944368067 : SUCCESS in 1h 46m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/58aee82b3ac546efb4108a85458154ad : FAILURE in 15m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b139140ba80748dc986dac06f216bd1f : FAILURE in 13m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fa425d2e554f4349afad8706ab379d4f : FAILURE in 18m 56s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dae953f83e834fe7b6b6ceec4ac58f90 : FAILURE in 26m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46ba46071fee48769b5ac9e1e4cc2f5b : FAILURE in 23m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ca97b9afbf3d47c7ba4e20978c6ab861 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964cf3e9e9da45e08189aec5e3a6926b : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/368f9196a15a4b8db9d883efc8d0bdf3 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cc1a3f0567647d0bfc7ba87a193ff76 : SUCCESS in 15m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6235a18291c34d4b9b3738c1e4053b5b : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebce2994938a418a9ca15771be87d0c8 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/639b5fa70a444bec8501da52e10940b5 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ec5a713cc4a93a9323bd1f161615ee2cc7c74f3d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 10:21:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d87ffbaa8b1dc6a74d5b66474bc3054da8193fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 12:49:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a293bc9b21b4d2f97460f1ca47c2b79\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a8cccf8554f47cfb63a1019b5148f00 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb091ba03a6408cbc95ff71bbdd11dc : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b093cf477aa407c9dfaf6efc5011758 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da777b24eab94040bf3e19c9a41d3950 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1aaeb6aae7034ae38be8089541a52ec6 : FAILURE in 10m 48s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb3412810f084697b66f3f0c0c287f78 : SUCCESS in 8m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1fd2ae4f94b34bc08225a3ebba79b3a2 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/99f8dff1cb0b42d4a134831489bf9046 : SUCCESS in 1h 01m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/127892d2ead44db8beadce833590e77a : SUCCESS in 53m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1206cc525dd84b6b944f966ae5165b26 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bec681f5f17b4df0b694e067e37f8d8a : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c616578b6e9a4b19a060f20b67c23427 : SUCCESS in 20m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c370ec96b86464fbdc8420619556b47 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/add03de1f8fc4cd19d6603f01006bd81 : SUCCESS in 2h 23m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a662a44475bc434e9b0576c7010bf777 : SUCCESS in 1h 50m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/772c245a5a3547d389e7f0063d633e71 : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d7d189633494374bd019eaf59dbe41c : SUCCESS in 16m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9e5e9d9e27654224b75da752987376ca : FAILURE in 31m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41f5a938acba4da485e39daf77b9f5c8 : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa8f3af1fdf48f493fe88acec540267 : SUCCESS in 28m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14c371f4d77b48bcb497f82dc4332819 : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8c7983fe0924091a84a578ed5f41ddc : SUCCESS in 14m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad6d031b667b4792891ad349a7693e90 : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99421003d4154856a33fdbae53b7826d : SUCCESS in 17m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4995a553a4c54d3796e16c04cb483bf3 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8330097568b34b2e8fcee1287dd2cf3f : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baff0fd413694aa4bd3339d8ffae6833 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c835bfdb29367ba58258d19d4950ea2d222906a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 12:49:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd25aec91a65f622e1621068fe62e00864185e5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 15:34:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28ae0fe37b494e7fa5c723caafd2514e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f071f8a80bff4f62b51076980563f7c6 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d44984d2f6ba4b199cfa4d0da129bb42 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfabd26ed2894828b0e690f001608b7f : SUCCESS in 4m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3789f4ddd5754df8a2d478d7d64affa6 : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/249bf415e7c445d6b857af0a1ec77c3e : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c306d6b0aa94a0dad572c1486b8db80 : SUCCESS in 49m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/354f03f173a54c8681f71cb51ec205a8 : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/87cd466d5c7a4c6884d988daed4c10a5 : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b2ad2e656054de5bac00d0814bac582 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6e3941fd43c04eb195666c56d577443a : SUCCESS in 2h 34m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1d9e175c9904073bb757d1ea7f41ce0 : SUCCESS in 1h 50m 53s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d150deda0ec24faea9f11d6cdbdc35ab : SUCCESS in 2h 40m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cd3560652714b528ec2a909e76284b4 : FAILURE in 18m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b015b7a760844c7c8d1e93d3869d4c07 : SUCCESS in 17m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c61701c1d02f4b268f528bbc15903926 : SUCCESS in 14m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d96f3e2da37f4c5087ff532af0f2e3bd : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0174e3b02c34e4ebde05d44c0341bf5 : SUCCESS in 24m 42s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e8161e954e50478fb769fa2dff2ea83c : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5421050be1d84daea832574ccfea5f60 : SUCCESS in 30m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0fe788ac858a4909ac27db12b410468d : SUCCESS in 13m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99928f35a49045838b5fffef42f89c32 : SUCCESS in 12m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/19ce7a8a1a3f429ea6138bc526d3c7a2 : SUCCESS in 18m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f9da66e133c409a935cf1c7c9d025f6 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1f61d0a3fa4ab6d220b6aec6546e8d9a6968a60a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 18:04:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c6865d035de4773766f83f23617d01c35d6b90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:36:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5266f5152eff4dbfa6da01458ce77b58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23ee0ea4de4445d6958b4ff3b296b7b9 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10c6f73585140699c6b41e7d1f1af1f : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7ed71616eef4baca1107e9f7cd809ae : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea01571a45804fb985c88ae4ca7f52e1 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bf66a7238d74b978078082a8d8035e1 : FAILURE in 10m 52s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb82aaf3be6e4cdc94c5349151a80cfc : SUCCESS in 9m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2ebae29ecab84278b7fd22d921b958ae : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f9e59119f6fc4a81b38529e462f02b92 : SUCCESS in 1h 03m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28d09893908a4a88acde738c68ee799b : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0f02cbaa95fc474a83e10e827e9b6a46 : SUCCESS in 1h 12m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/914b3a32f08248b4ace89bbdd4aba400 : SUCCESS in 21m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/456767079c9e419c9f0c6bde34ac306c : SUCCESS in 20m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b9e138279b44d1e901a14cac294365f : SUCCESS in 17m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f74c27986e2047599915fb4b77e75451 : SUCCESS in 2h 30m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c69046e22634e0cb2696eecf73e6f49 : SUCCESS in 2h 03m 07s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a501d18eba1c4d7786bd1344419ce7ee : SUCCESS in 2h 14m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4002d10e9114aca8568c92e80a7ef95 : SUCCESS in 2h 20m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fa22464dfa5f4a6586743fabe84b8e79 : SUCCESS in 19m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86d059dba16e4476896a6ffa9635ace8 : SUCCESS in 16m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5ae7bfc084b74e0884e1124571acb09c : FAILURE in 39m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/035d3cb6ef3b489e8c657635fa87147c : SUCCESS in 34m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b32fa07df30548cf9f8ee971c1b6a90f : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cd5daff533104bd28fbedf143101c04d : SUCCESS in 27m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da639434e39444c798b95da75c315327 : SUCCESS in 24m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91dc0a846e0040189310706375c900e4 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3b376bbd5bb4799a41e3dc0275b2b48 : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15d5091161ea457f84c0854b9b660570 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/815cbcc0329a4e819e625d3c2e9ff555 : SUCCESS in 13m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed804af763a84ee68fc6fabb13fc727a : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/923bf3836f3c459ea5b1b76645382c85 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b9a4b96851a45a5ae36a29908de83d0 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b013326909eea8175aab8817478a7289965f4464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:37:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87462bc967f647b7353eaaa36a10abe932942c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 23:29:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47ce7d5a90c844aa9ef93c7b8168bdd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1e8d8e7f3a45d4a299e73e0967e120 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ae531d3b6484709b133ffb2e64f8f17 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/baec4d25334b4c9ea46b4c90ed45eb16 : SUCCESS in 4m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5731fa0ef41a4f3daff4de037e2d938e : SUCCESS in 8m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dbeba562e10d48ffb64d8a8410dfc301 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b35faac08194b33bc86c5e5b37939a1 : SUCCESS in 46m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dd76be5ca4c74917b84000f8885fe41e : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db75d828e57648289ed73f8f01f6f074 : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3452665d919744bdb52626b772c68bd9 : SUCCESS in 12m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/db0865a4c524412cb2c0bbe289743ad7 : SUCCESS in 2h 43m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36946aa1d5854ba5a46e38b06f0be225 : SUCCESS in 1h 38m 58s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ebf6c2ebe5fe455ea3ef2a8ad7ae2c9f : SUCCESS in 2h 39m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/617dc6d12a3041b992a578caefcd5603 : SUCCESS in 1h 40m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1de2fd3c35bb449ea3243c7936304896 : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a30aeba1bbf34cf68e3744e2db1bd48e : SUCCESS in 12m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1767eeaed4f24ea3857274d5e796db81 : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20d146f27c764a91821d43278b35a1a4 : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e408d95249d5483980f4d5700c56cd9d : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5b9a2da44e5442893013ce69100a94e : SUCCESS in 25m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a78a4c1db284d718f7a3c69cf172fd1 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aec780b88f4d4837805d13ff95f0b81e : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/141a6ed40abd435cb46ee4d0f6c9ccd5 : SUCCESS in 17m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d42192ffce4e08bdd4ed696173f511 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9e82f57b750b7313a1fbdf6592c8025bb47949ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 23:29:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7cc8d50734fcafb6c25690fa9beda43e2da271fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 23:30:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82664ae79efc4aaabc70519dd7174121\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610195b3f9d8401b901d036d9a677e23 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f7d34e60704bef9fd522c488f08f41 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd6673477f0327a5eb07f0f6e2a647827f742a63","revisions":{"b342e37075e1d5aa88f043714e8ffafad6649e08":{"kind":"REWORK","_number":1,"created":"2024-02-06 15:07:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/908142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/908142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/908142/1"}}},"commit":{"parents":[{"commit":"e469e99eba971df38ebea814c6dfb442606c6192","subject":"Merge \"Ask for confirmation during overcloud deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e469e99eba971df38ebea814c6dfb442606c6192"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-06 15:06:54.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-06 15:06:54.000000000","tz":0},"subject":"Support credentials for custom DNF repositories","message":"Support credentials for custom DNF repositories\n\nThis allows use of repositories protected with HTTP basic authentication.\n\nChange-Id: I6c9686bdf736d58738fbe88169e62833cc85d2f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b342e37075e1d5aa88f043714e8ffafad6649e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b342e37075e1d5aa88f043714e8ffafad6649e08"}]},"branch":"refs/heads/master"},"bd6673477f0327a5eb07f0f6e2a647827f742a63":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-07 08:55:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/908142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/908142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/908142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/908142/2"}}},"commit":{"parents":[{"commit":"e469e99eba971df38ebea814c6dfb442606c6192","subject":"Merge \"Ask for confirmation during overcloud deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e469e99eba971df38ebea814c6dfb442606c6192"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-06 15:06:54.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-07 08:53:49.000000000","tz":60},"subject":"Support credentials for custom DNF repositories","message":"Support credentials for custom DNF repositories\n\nThis allows use of repositories protected with HTTP basic\nauthentication.\n\nChange-Id: I6c9686bdf736d58738fbe88169e62833cc85d2f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd6673477f0327a5eb07f0f6e2a647827f742a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd6673477f0327a5eb07f0f6e2a647827f742a63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
