)]}'
{"id":"openstack%2Fkayobe~908198","triplet_id":"openstack%2Fkayobe~master~Ib5adfa6345666cf81a78181c69b04f7378497576","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ib5adfa6345666cf81a78181c69b04f7378497576","subject":"DNM: ironic secure rbac test","status":"ABANDONED","created":"2024-02-06 20:16:47.000000000","updated":"2024-02-21 12:32:12.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"26b7e7fbb109b16155c78ec270087a3ba6aee1c7","_number":908198,"virtual_id_number":908198,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-15 17:41:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-02-06 22:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099e2e21b871403ea50fc6de6055b7fd60bf15a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-06 20:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8396007ef79ad17662d84d5ef700908abf61dc96","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-06 20:17:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5c7ca5a97fc294769f710b860e74f86dc82bb248","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-06 20:18:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e856cb99cd7f1d689b5fcb4e8e03960945870375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 22:57:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eaca072b8754f1bb9e51665ec29491d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00988219895849c48611b932ea2de59c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417da3cc0ba349949a654de094d2fd5d : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9be95023036b4fd4a15e2d81c5912bf2 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/180af5f72dee42c993490a3f9117c7a7 : SUCCESS in 14m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fe49a9e72b447178904cea8b45f3353 : SUCCESS in 11m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/80999e1b6cd14d7eb514fca4ee45e41e : SUCCESS in 56m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/08d98f51bf6a4c5fb95c6a8d705f8a44 : SUCCESS in 55m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e11fdc50501147d199e8a3fa1650c72a : SUCCESS in 51m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/387be855b5b04bfe8997dc72e4085426 : SUCCESS in 1h 23m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/51abbc1996ad461daaf29bad3c1eb4d2 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/32a57c135e9b4fff9c2d320723fdcb11 : SUCCESS in 16m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a07ce5bd85a8415f9e8b8aa280a34351 : SUCCESS in 16m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c677a7a25294c9e9576789d7afc2c0e : SUCCESS in 2h 07m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1eddd59750540909802bcb8db90f02d : SUCCESS in 2h 24m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4aab6a7fa7d5403695e17c7309c691f5 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/052a0e3c3f014f1183cfa3d16a3ddb3b : SUCCESS in 14m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/025d363b93cc467e8a02d92d1207b624 : FAILURE in 33m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b2a52363f9fa4ed786aaa4a6656c587b : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f073a644827b4f44b2ef70a97a2cbdf1 : SUCCESS in 23m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68a149e8879049fa8ece5138446e8da1 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f9826f90d6d44e699847fab6700a39c : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d97222587dec4a3a93d9dfe8d4ba8d53 : SUCCESS in 17m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81937596c5d643ca861989c45f88d7bb : SUCCESS in 14m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f997ec52470c4240adfd6304b8818513 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b96a1335b8704cd4a9486c1a85ea6437 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f288e8dd7af74ab68c6ab664fafd2dad : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1d4fd2d28ba7e040ee453935c5adae5b00e11834","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-15 15:06:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8fce76b66a3f25910e82eb2db6ac43492b729587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:08:13.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/907274 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"e5bf9b8667cd10b786d327b2ecf949d580072a1a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-15 15:09:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"717aadf9cf4faf95bf21f11e4f94134aa1139913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:15:17.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/907274 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"1169fba4ef5b0ac59139fa9e74132f20c5720bec","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-15 15:16:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51cb709eddbc742bb1300a6a9990eda70d8f49be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 17:41:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da45115af7e4e6682f15726838b612d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f559f7cef4fe41da8ddc0c017714727f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7bc69dfb1e44bb89b1ad6377bdf4af : SUCCESS in 8m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a66a64fe5ef4abeb95ac260a0c1797e : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3497ebb4974a43abafd6cee06b1f7f54 : FAILURE in 13m 08s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1345c0e6809a420f9efba0791445c76f : SUCCESS in 11m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e29ba6fbc5da433cbf7b85df1cf688cf : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4d3dcb931d7540e4834a38e621757841 : SUCCESS in 1h 03m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f56be18f19e447caf26ba2e4c5159f4 : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4f06cf39b71048f7b253af3f9c7ef5be : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f871a88bbb844b1879039902e5a3971 : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/372dab5392f4421bb68b2ddafffdaee1 : SUCCESS in 16m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad043bcfae7d436f9b76194fd0170c0b : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae6ab71189984e85b8052a3fb63da9de : SUCCESS in 2h 24m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/461ce28f46a74e62ad3d17760a4e587b : SUCCESS in 1h 50m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f5153f043b7944f58f2a13b0872aa57b : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6539001179304177a4a3ad8b1aaff4d5 : SUCCESS in 15m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b21ba312dfbe4fc9a57cada06d8ec4c2 : FAILURE in 33m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/31daeec3410b4660bd4435119ecd4bc7 : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efbaaf8994494e4f86633ef96d59b014 : SUCCESS in 26m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61e5670d7e5548b987626b27ddc451b6 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60362216a690489497301d9141d72305 : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2a04656ab3d471b8b112f20436473f7 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b024715f7d6b4485a3122cbe46510903 : SUCCESS in 13m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee549cde1e17458e9ce63a4d19b0550c : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c1e8ab8a5004d1fb6244cdc6f0dbf8c : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c65261c5d8764802873e7f487502a183 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"26b7e7fbb109b16155c78ec270087a3ba6aee1c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-21 12:32:12.000000000","message":"Abandoned\n\ntest finished","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63a49158dbdb9f5a4f67e3fee96834515243ecbc","revisions":{"91a40eda2598dabffb142015582c15f2a9cc604f":{"kind":"REWORK","_number":1,"created":"2024-02-06 20:16:47.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/98/908198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/908198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/908198/1"}}},"commit":{"parents":[{"commit":"e469e99eba971df38ebea814c6dfb442606c6192","subject":"Merge \"Ask for confirmation during overcloud deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e469e99eba971df38ebea814c6dfb442606c6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-06 20:15:48.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-06 20:16:46.000000000","tz":60},"subject":"DNM: ironic secure rbac test","message":"DNM: ironic secure rbac test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/907274\nChange-Id: Ib5adfa6345666cf81a78181c69b04f7378497576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91a40eda2598dabffb142015582c15f2a9cc604f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91a40eda2598dabffb142015582c15f2a9cc604f"}]},"branch":"refs/heads/master"},"f9c997dcd2eb84cf5f01c364f225d112afc2a4f1":{"kind":"REWORK","_number":2,"created":"2024-02-06 20:18:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/98/908198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/908198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/908198/2"}}},"commit":{"parents":[{"commit":"e469e99eba971df38ebea814c6dfb442606c6192","subject":"Merge \"Ask for confirmation during overcloud deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e469e99eba971df38ebea814c6dfb442606c6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-06 20:15:48.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-06 20:18:37.000000000","tz":60},"subject":"DNM: ironic secure rbac test","message":"DNM: ironic secure rbac test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/907274\nChange-Id: Ib5adfa6345666cf81a78181c69b04f7378497576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9c997dcd2eb84cf5f01c364f225d112afc2a4f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9c997dcd2eb84cf5f01c364f225d112afc2a4f1"}]},"branch":"refs/heads/master"},"63a49158dbdb9f5a4f67e3fee96834515243ecbc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-15 15:06:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/98/908198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/908198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/908198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/908198/3"}}},"commit":{"parents":[{"commit":"bb51db15febc195384a48406cb37571776988ee3","subject":"CI: support .gitreview of unmaintained branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb51db15febc195384a48406cb37571776988ee3"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-06 20:15:48.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-02-15 15:06:37.000000000","tz":0},"subject":"DNM: ironic secure rbac test","message":"DNM: ironic secure rbac test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/907274\nChange-Id: Ib5adfa6345666cf81a78181c69b04f7378497576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63a49158dbdb9f5a4f67e3fee96834515243ecbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63a49158dbdb9f5a4f67e3fee96834515243ecbc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
