)]}'
{"id":"openstack%2Fkayobe~908262","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I0465eafa9e4ff37c96064ea8395f0bd461035b40","project":"openstack/kayobe","branch":"stable/2023.2","topic":"bug/2051714-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-14 14:34:22.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-03-14 14:34:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0465eafa9e4ff37c96064ea8395f0bd461035b40","subject":"Fix: configure etc-hosts for overcloud group","status":"MERGED","created":"2024-02-07 09:46:02.000000000","updated":"2024-03-14 14:35:32.000000000","submitted":"2024-03-14 14:34:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908262-bug/2051714-stable/2023.2","cherry_pick_of_change":907306,"cherry_pick_of_patch_set":6,"meta_rev_id":"f6cf58ab28ee955f9106e1b685599a47636ce18b","_number":908262,"virtual_id_number":908262,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 14:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-13 14:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-03-14 11:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-14 11:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-14 11:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 09:55:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-03-13 14:31:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-03-13 17:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 11:14:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-14 11:39:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d11fc5a696ec82c60134535759e126aed477b292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-07 09:46:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4d7fe7e8b356cbc9c4e81a6acb3541d089720dca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 09:55:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fe154e73fe5b22a1a6ff66e7452026cc0bb77d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-07 10:09:03.000000000","message":"Abandoned\n\ncherry-picked too early - wait for master to merge","accounts_in_message":[],"_revision_number":1},{"id":"b36bc28739db96ee6376f758c528571ffc07c3a5","tag":"autogenerated:gerrit:restore","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:29:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"69b640cfc91c4c89757f9b59140cf0f9e731fb7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:30:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6672ba15bc5f1da13b4139ef05b5fedcf2ebddd9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 14:31:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d72c54c96b2840eabb02f7d575d7f223e93fb533","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:39:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f24444c8845890b722581dfb9889d418648ac3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 17:08:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef2e45d86334bf592745f0c272304ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54833e8ab1ce48788ff0186f5de573be : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c14604cd1c4dfa805bbb920908b1d2 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898c812cdbf644eb830fec13bdef2cc9 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/423739990fb541ccac3554dbd4134da4 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78e8c065dfb3468d8151a0997b85bc3a : FAILURE in 12m 18s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20a66b1ab724483a9db27d0e7012ba2e : SUCCESS in 9m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/64fd90134dbb4f438f5a74a42ad9e280 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7d278fd2812e4ab88d27dba602fe7180 : SUCCESS in 1h 03m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09addd5de6594de2bad3a6a36992137d : SUCCESS in 57m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c37b0e0e4f2141238a59b9405afbd9a3 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1175d4f1dcee4931ae8ed4308862601a : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7378152694b8437597b2e149f60c0845 : SUCCESS in 17m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdfd1c258b1843cfa89ae47ea1d3a5e3 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/061bf7504c234aa8800f841dad2080d0 : SUCCESS in 2h 32m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7dcee756e65457da76fa06747b76a4f : SUCCESS in 1h 54m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1585f593724246f59c3a76c2bbc3c5fc : SUCCESS in 15m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fbe9e11d5e241eaa8f892dc45e06608 : SUCCESS in 17m 16s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/64f7dda767ae4f33a79aab3cba85ba82 : SUCCESS in 33m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e3edaf5775248cf932b73de1e9bc5c0 : SUCCESS in 32m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24a15119f7464af6852c0c1125ed9303 : SUCCESS in 27m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/937ee0cfaa564cdfa3c5a6b6127e3c3e : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d987d8f76f6c402092433ad5a917e510 : SUCCESS in 13m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/36bf8e24f10b4df9993825f0cbee6a2b : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64a935d65d6f4257a7e58f2921c64ed1 : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72da8e24eefe4a329b16ed37a29d2cdd : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cabd18ddecc9480d89584d16d3d092f9 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1cb96e5ec0c84fbfb52bb604c3112b17 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0de354427cc24e2844939654a71d85e1655e860d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-14 11:14:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d7b27e26b2841026a8fd8acd9791dfff20aa84df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 11:39:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9bd707ae3b48c57d6c1ecf0faebafbacc12d7a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:40:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64c14a507c1655e0b52263180e5831c8394aebd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3c0e249b4c463e9b7f6cdfac108a0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a633c7c14141aba2794c0ac7fc48fc : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18a3f852b9e74fd28061e90e2667c497 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c34b3312e56649c6b8c6c49bc45ea11e : SUCCESS in 4m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22fc4d17c1fd46f38204a6f4d2f624b8 : SUCCESS in 8m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/33c7f15383ee442b93b3321cbd02b6aa : SUCCESS in 56m 13s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2434fb0014c6431589117fcd313e9ef4 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a354d0d2a1164c8888c88fe1ab93a4ca : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4b6e1b5355bd4fdab6c8943126ab16bd : SUCCESS in 16m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90e04acb3dc2460aa83ad564db5ead54 : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44fa410b81f443ee9c9f03cc7f37a2ec : SUCCESS in 2h 44m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de0e423de7474e328dc6fb9355f64959 : SUCCESS in 1h 36m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e390d81ddb964d5da1bdfe2714b23dbd : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdfd147d6e72440aa4efcc0e7ef692c8 : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8275195e97b74387bfdd8716c0bf4ccd : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33b34ac76f87484c848d99f4bd9e4ad8 : SUCCESS in 24m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df4bf5585e004c80bbc8ed33c3daff1a : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51be1b4e9373468ab7923d28cd9d7f12 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f72c486b25b3487997980812fd7ee93f : SUCCESS in 20m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e0efe331f13475899a9d3809218acea : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9af2b1656732575cf70e1cd6e1c499dbd2db0e9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f6cf58ab28ee955f9106e1b685599a47636ce18b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:35:32.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd306c9f15ed4547a373657b32c6b0f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696d8c0c7d0d4d2d8d2eaa10835fffe3 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66fe6dea61414eee9e8c838c92c98ec2 : FAILURE in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1bcf3e19e202e4409f94eff257e5e190bbb26d19","revisions":{"717fbc5977bfd9385f76c55474dd9a55e79982cb":{"kind":"REWORK","_number":1,"created":"2024-02-07 09:46:02.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/62/908262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/908262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/908262/1"}}},"commit":{"parents":[{"commit":"bd16d2931d79d6df86f9e400e5387d924d99df30","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd16d2931d79d6df86f9e400e5387d924d99df30"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-01-31 10:36:19.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-02-07 09:46:02.000000000","tz":0},"subject":"Fix: configure etc-hosts for overcloud group","message":"Fix: configure etc-hosts for overcloud group\n\nChange the etc-hosts role to run on the overcloud group, as using\n``ansible_play_hosts_all`` would exclude hosts when running with a\nlimit.\n\nCloses-Bug: #2051714\n\nChange-Id: I0465eafa9e4ff37c96064ea8395f0bd461035b40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717fbc5977bfd9385f76c55474dd9a55e79982cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717fbc5977bfd9385f76c55474dd9a55e79982cb"}]},"branch":"refs/heads/stable/2023.2"},"1bcf3e19e202e4409f94eff257e5e190bbb26d19":{"kind":"REWORK","_number":2,"created":"2024-03-13 14:30:23.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/62/908262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/908262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/908262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/908262/2"}}},"commit":{"parents":[{"commit":"18644a564e01a32166c183731bdea0408181feef","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18644a564e01a32166c183731bdea0408181feef"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-01-31 10:36:19.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-13 14:30:23.000000000","tz":0},"subject":"Fix: configure etc-hosts for overcloud group","message":"Fix: configure etc-hosts for overcloud group\n\nChange the etc-hosts role to run on the overcloud group, as using\n``ansible_play_hosts_all`` would exclude hosts when running with a\nlimit.\n\nAlso change the gather-facts-delegated role to run on the overcloud\ngroup. The delegated task needs to be included separately as the hostvar\n``ansible_host`` cannot be set dynamically based on a loop item in one\ntask. We loop over the batch indices here so that the tasks are included\nin parallel, rather than in series.\n\nCloses-Bug: #2051714\n\nChange-Id: I0465eafa9e4ff37c96064ea8395f0bd461035b40\n(cherry picked from commit 498de81efe88707e5fd91dba24cdee2a529dedbc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bcf3e19e202e4409f94eff257e5e190bbb26d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bcf3e19e202e4409f94eff257e5e190bbb26d19"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
