)]}'
{"id":"openstack%2Fkayobe~908263","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I0465eafa9e4ff37c96064ea8395f0bd461035b40","project":"openstack/kayobe","branch":"stable/2023.1","topic":"bug/2051714-stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-14 14:34:24.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-03-14 14:34:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0465eafa9e4ff37c96064ea8395f0bd461035b40","subject":"Fix: configure etc-hosts for overcloud group","status":"MERGED","created":"2024-02-07 09:46:10.000000000","updated":"2024-03-14 14:35:34.000000000","submitted":"2024-03-14 14:34:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908263-bug/2051714-stable/2023.1","cherry_pick_of_change":907306,"cherry_pick_of_patch_set":6,"meta_rev_id":"62861a35567a817ef9ba28fb8c4fa95707281a7d","_number":908263,"virtual_id_number":908263,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 14:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-13 14:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-03-14 11:14:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-14 11:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-14 11:39:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 09:55:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-03-13 14:31:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-03-13 16:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 11:14:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-14 11:39:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"09be0bb7436b2698769850d81ab63d5d50247523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-07 09:46:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3911abd9d21aeef6ee7026d16a8ebc8150b8fa64","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-07 09:55:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b2cd45e0566c0f141b9a22e1f6b6c78f4c94af8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-02-07 10:09:15.000000000","message":"Abandoned\n\ncherry-picked too early - wait for master to merge","accounts_in_message":[],"_revision_number":1},{"id":"235a768f5e30d7b62bb1a00ae3ea634b7cb6a921","tag":"autogenerated:gerrit:restore","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:31:05.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"62387c8623e65a8a83a590b7b9a2e9c56eda9c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:31:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cec848cd7e81b744aaf6920b253d1031402f5bb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 14:31:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57fb6fc3da8055ee1d4fd40f0d7f1f11c1437521","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-13 14:39:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"624a3f6539125380293d05ee27e45d0a265c8c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:36:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47a9b07f8d9481fa4f00b14cf3e1589\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b49aba161ce48fca36e3d02da40210c : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fec34aeeb044d58a228740ef49ff82 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3060b83c0a6048059006e4e1ba56aabd : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8e58969eaf4a0d86ed96f34bcee046 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fd43bd991344b978a5389da3f638d4d : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9986e7c31e8947f5953e389562895390 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/482b38b217c44d52977ef079213639c6 : FAILURE in 10m 53s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e486b37cf0d847d2858f4cd7148cafa0 : SUCCESS in 11m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fbe7904d34f8425da841efad5d1573a1 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5a681b0d46304353b8c02498ac271561 : SUCCESS in 1h 04m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a75e6a203704a2c906bfc81ba080a81 : SUCCESS in 54m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5c9ffd7c19204242ab54b19ed2367c1d : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8fc2251173c24a8fb634049bea95e1c7 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/363ae9da602044a2aceb6c2b6e747fa5 : SUCCESS in 17m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869ed7b7e893482b8bbb20dcbf397b0b : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53e1d1496ef54e3cab38b772c47ea845 : SUCCESS in 1h 57m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0bdc70e46904806b931c5ae80ca51de : SUCCESS in 1h 36m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/73b1ee9bbb6f48af970c1ac351b51cb4 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82edfdff308f471f9a3a05fbc96a05d4 : SUCCESS in 15m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fd5756abc21848a69a831a179f85dcc3 : SUCCESS in 32m 37s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d636cd73eea04f6ba1f27a05a26636bc : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2acb8889b0dd405e960029d8e396e072 : SUCCESS in 24m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/18b770e1b0274a2b9aec4f17af5b3c7a : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae1220919dda43f18744afe796a9be84 : SUCCESS in 11m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23a391837e004ce3872fa8a65fe6363e : SUCCESS in 18m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/681da7b8388848cc95e7cfcf8238db15 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4f768253de074f6b4755884491bceb8e149e4d99","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-14 11:14:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3cfa254dfb60c967ba986d4c11e137f8b97f2f81","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 11:39:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c964a25d9a42cc008626654741b1bc33b5f28bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 11:40:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"383d94a70f6deb486df0170a622a45f5825624b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c3595ebd18463ab7cf5c7ea30d3ba5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c266a198fb8947328df4c095f2141bf4 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd192da52d94dbdbf48dd3aed80a844 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e303f6376493481cb1144e1ceda77ea9 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc39ad9b2ba34b13bf165ff4a235edf6 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7da6625143104d46b539bf6c923042ab : SUCCESS in 5m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f82273df0cd45498591b2b3c8ee270b : SUCCESS in 13m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e9f973076a943bca865e4fb8a6b5b84 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/744dd3de2a424a9d8dfbba9513ffda26 : SUCCESS in 59m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3ddab188befa4082948182ad6fda4f98 : SUCCESS in 56m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7522af8dc4840349aec113223b0ff2e : SUCCESS in 20m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fc70acec46f49a7a86f122d568132bb : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d155ba7192645a5a7ca17adddc6a47a : SUCCESS in 1h 59m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbac2c8389294800881db0239c59a0fb : SUCCESS in 1h 39m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/85cb051e7a8849b588d1a2693853c496 : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c26a0962909744aaa7012944ef42c09c : SUCCESS in 15m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46d0674c2a8e4f23bea52a3f0c624acf : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/244c7f2da3634faeb97655f91ae5fb0a : SUCCESS in 25m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f6f683bc97024250ba7c27fb5dfdd3b1 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0803e93233644b7a910ef29bfa5dc24c : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d188f54ce7046aeaa2df085ff76fc42 : SUCCESS in 23m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b069a225cf440feb1c680e611eef225 : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2147b887dc803f6619fcd6e7c7f792ab46b47be6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"62861a35567a817ef9ba28fb8c4fa95707281a7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:35:34.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e54f587aed344f6a85270b29015287a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9c2be8aa0d4378b1627e609ffa59ac : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/581f6e00c11c48b1867abf2cfe97950d : FAILURE in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7d779533e62ebee40843976c2cd7d8f169bb039","revisions":{"342870aec071516d0b53849cf64b169f22de5999":{"kind":"REWORK","_number":1,"created":"2024-02-07 09:46:10.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/63/908263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/908263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/908263/1"}}},"commit":{"parents":[{"commit":"d397ceca74667ae0121b7bfebab77fd6888767a0","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d397ceca74667ae0121b7bfebab77fd6888767a0"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-01-31 10:36:19.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-02-07 09:46:10.000000000","tz":0},"subject":"Fix: configure etc-hosts for overcloud group","message":"Fix: configure etc-hosts for overcloud group\n\nChange the etc-hosts role to run on the overcloud group, as using\n``ansible_play_hosts_all`` would exclude hosts when running with a\nlimit.\n\nCloses-Bug: #2051714\n\nChange-Id: I0465eafa9e4ff37c96064ea8395f0bd461035b40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/342870aec071516d0b53849cf64b169f22de5999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/342870aec071516d0b53849cf64b169f22de5999"}]},"branch":"refs/heads/stable/2023.1"},"d7d779533e62ebee40843976c2cd7d8f169bb039":{"kind":"REWORK","_number":2,"created":"2024-03-13 14:31:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/63/908263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/908263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/908263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/908263/2"}}},"commit":{"parents":[{"commit":"2f4414f8a4fe5184b6d1e09595993e8bfaeebae7","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4414f8a4fe5184b6d1e09595993e8bfaeebae7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-01-31 10:36:19.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-13 14:31:20.000000000","tz":0},"subject":"Fix: configure etc-hosts for overcloud group","message":"Fix: configure etc-hosts for overcloud group\n\nChange the etc-hosts role to run on the overcloud group, as using\n``ansible_play_hosts_all`` would exclude hosts when running with a\nlimit.\n\nAlso change the gather-facts-delegated role to run on the overcloud\ngroup. The delegated task needs to be included separately as the hostvar\n``ansible_host`` cannot be set dynamically based on a loop item in one\ntask. We loop over the batch indices here so that the tasks are included\nin parallel, rather than in series.\n\nCloses-Bug: #2051714\n\nChange-Id: I0465eafa9e4ff37c96064ea8395f0bd461035b40\n(cherry picked from commit 498de81efe88707e5fd91dba24cdee2a529dedbc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7d779533e62ebee40843976c2cd7d8f169bb039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7d779533e62ebee40843976c2cd7d8f169bb039"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
