)]}'
{"id":"openstack%2Fkayobe~908413","triplet_id":"openstack%2Fkayobe~master~I0ec4ec3e9d60bb1431b44dd6718415214ad80025","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-02-15 18:15:38.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-15 18:15:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ec4ec3e9d60bb1431b44dd6718415214ad80025","subject":"Support auth configuration for Apt repositories","status":"MERGED","created":"2024-02-08 12:11:36.000000000","updated":"2024-02-15 18:16:44.000000000","submitted":"2024-02-15 18:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908413","meta_rev_id":"7af5ee20ff0bc654b5ff8e2df68240da0a9bf9d1","_number":908413,"virtual_id_number":908413,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 18:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-02-15 15:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-02-14 10:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-11 13:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-02-15 15:32:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-08 14:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-11 13:57:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-14 10:50:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-15 15:32:28.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"514f1f1015b822940e31f5bd53b384912b7beffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-08 12:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"014baea59d0205157915a9ec9c368a8384c3565e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 14:49:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f11d705bad6458faadd3c40c9fef921\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16b0f05ef3bb42b1883d9fb882d506fc : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ce489cf4fa4096ab73588627f99f3a : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31a404c6a21c4cd4b7c0db30fb22910d : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c0b9434840d4d96878caadc63bb5c78 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20a16684696144c2848dda9a7cb0e3a7 : FAILURE in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fc818123a6704b7aaf8be0ed9fd618e0 : SUCCESS in 9m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b2d17505e40c4514b6d1e48c09993662 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cdc24def47c5446782bf960631c54620 : SUCCESS in 1h 09m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a86900ad38c34e2dbd870c162fcb91ba : SUCCESS in 56m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d5365cee51b7492db1a1e40d4cea815e : SUCCESS in 1h 13m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f22f703ff0d14f689a5837edc8284a37 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/30a069854305427ca2eb35af8423f4c0 : SUCCESS in 22m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91c8e4a0d946493b93fb9fba3cddb935 : FAILURE in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9358cf30b74843e5b4829a7920200a2a : SUCCESS in 2h 32m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/539a89b8e805471897dd9a1ecb34c52f : SUCCESS in 2h 10m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c2327b94ed444b1ab828ec6852cc09ab : SUCCESS in 19m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c7240144eb645da9638a5580f1a52e2 : SUCCESS in 16m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f3a17a33779f4c939979936f9ac879b3 : FAILURE in 34m 39s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/04b0deb8df2c47058c88b1117894648b : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f64496a4ecc460eaf029c9143356460 : SUCCESS in 29m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea463839d5e444a89986b672305b5a68 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29bf3d797f5c4659a420ba9e04969295 : SUCCESS in 13m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bda2cac8b98e4c6296a6c20d00f0e816 : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10e4b9707c624d53a2246987be53b1ce : SUCCESS in 15m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/506678049649443b93b02513017719d8 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba1c2ba4f68414da0875a44c47c46a4 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/250659c8b37e41688c20fd3ba97fc83b : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"38feee7988b639e60abec5279447efdc916d2e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-09 22:27:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b2b896925379dad6e148c926157833aedb55bcad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 01:59:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aada95e196e4e73b0c584f51dc7ecc5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e90cd3c20d429ea0213bcdb2122565 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e43d91bf194b9ba09af58e76194efb : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9461581799e4453f81591cd96183c2c1 : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1931b6dc0dba4f35bfeba1142d4dad20 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/051dbc84b748468982521387ce67d7ef : FAILURE in 11m 50s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c15771b64ccc4336a95ddf89be358dbe : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5ff62bb1360e4beeb2111161d22f990b : SUCCESS in 1h 08m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/339701ce6b774ccc8258bfccb4729725 : SUCCESS in 47m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b111228462348878f6463942780a3c5 : SUCCESS in 49m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d11be07cfc81434683d410b09b3d0224 : SUCCESS in 1h 16m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cb1e3742a59a4d72b5f88e6c51042f10 : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c749dc374d2c48a29d9290bde161bbb1 : SUCCESS in 17m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bce3aae09a14d53b819dcb53377ffc0 : FAILURE in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/461a25ea56f445cd8ba3f785a85355b9 : SUCCESS in 2h 52m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d2f5cd5465240aa8e378515ab8b34de : SUCCESS in 2h 24m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9014ebe013fc4cf7ab64a1c2dc8cef6e : SUCCESS in 18m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc7dea59ad64e9fa6dea25a25d48f90 : SUCCESS in 15m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fe98db019d9b4a63876e454648523694 : FAILURE in 40m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7243cde78bd248b1931bf95d01922fc5 : SUCCESS in 34m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80e5ef8b49dd4797abe4c0f7b398b295 : SUCCESS in 23m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab4707b0f2be47108937d63bc3ce4523 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bf47ed949b54c98af2c23a3888f9ebd : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b6c6aabafe54b41ab1a5243cf364c37 : SUCCESS in 18m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1e40fd4c48e4195981ac8fec9cad2a4 : SUCCESS in 16m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ebbd68364084191afa2f85004f2f26b : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2821048df6d941b4aa29228ea18327df : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fba873cb6b934f1bb57cce85ea6b3520 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"142423929b45a063cb22706b7159142c13e0036f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-11 10:06:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"db3f1cdf9158859734a01b755c6118f31993a445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-11 11:49:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"13a0a1be33b1fb5c4d782ae75714ba3641aa3ab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-11 11:49:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c458af2bd0340d83e326742f3090ca8445fcb35f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-11 13:57:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cbcc87331c480f452eb40189906a7d01b9bd1631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-11 14:19:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c059571817843dfa30beafd5260f7c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb953910ecbe48a4b019d7bf6f0c8cd6 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110e617f38164db9a5793866559423d6 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aa51831db9c4a2a95abe055f0b1e452 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/884ff245c87e4d7e9346edac48ee81d4 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7aae0c0da2fd4089a19c07f243f95913 : FAILURE in 8m 08s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd7201561827444d83e2f1709a4f36f2 : SUCCESS in 9m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f83a1c120e2145fba8343b35c7e92f5a : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/00b84730b18a4bc1a4c077b711bac1a7 : SUCCESS in 49m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2763f8c2d33a4cc3b141416f463aff90 : SUCCESS in 54m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aa148633c75c4efa9c729639a9cf6c7d : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/78561eb2ab634cb292037078f618a42c : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/76aae6008dfc4c5ca65af8260e6c968f : SUCCESS in 14m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64579f1f9017416a90c5f3b9ee576c4b : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0368c5015e1e427e855df15eab912a14 : SUCCESS in 2h 17m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b63b27eef4c74ed5b1e61d2e09d2be91 : SUCCESS in 2h 01m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5cc3b0d79274df1a28c202285b5bdde : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11b48c1bc8b64f45a3e98f146f1f8f0d : SUCCESS in 11m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/550d142cbc1d4d91a956ff7db59a85f6 : FAILURE in 31m 44s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d37258e78c604c6aad42e5403a1e37ba : SUCCESS in 23m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e24de7bb86884621b536e9005090e728 : SUCCESS in 18m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b164b11a2e5f464698d37af5522c4721 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6db4f5c18c104a9da80732eabc506fa7 : SUCCESS in 9m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83b8f965fc8348e39f71984ff97b9612 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ab7e19eaf54776869f66fe223b7dd9 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef75f4092b524ab4addbfc3511058179 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2e02c6b029a4e9d9b7c19db4821d550 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e6512353356474fa23261364cae1ade : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":5},{"id":"e692885f3b772e99f64f7111e4a54d94b8484fd9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 10:50:38.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1bbbd0d9682a377fff9d6173c26a7e6bd371a7fb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-15 15:32:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9dd0244f81dd44ae8f976a07795453002eee894d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:32:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b9caa525cdbb63ba2ae7312193795d1225882df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:15:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/804d159568424a0a9e7c82e55b92e702\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40063330cd44f849abceef5a80edba5 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb56600a6dbf4853b72077a8515ee7c0 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/588cdfaf4b734f34b68c6445dace4a6d : SUCCESS in 5m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/faa1c84a6600434195d1898bb2ff415c : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7005d658cec443c6992cbfb3d088d7c3 : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d68c31142714992a3e5cae0392b2514 : SUCCESS in 54m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c786029fd6fc4ebbb6f8ab4ebf028cfd : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dec81e348c3c4b2d80930c74f580ef50 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f9cc3152de9476b839a59006fe8d763 : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fef44a1deb1d46458e3e389db2657990 : SUCCESS in 2h 38m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8af615c30cce43f09855f4bc9c66fa96 : SUCCESS in 1h 49m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/28d0341a008745068838a4b754482895 : SUCCESS in 17m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b273fd74cb634816993c00a304e233d2 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d08ac30a5d3d4392adc7a70d26f6193e : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09b66507334c4242aadf8edc571b5eeb : SUCCESS in 22m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8fc006c1191445798f00d3170a7d5a57 : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b6cd4aabb8747b39944e12b155bb39c : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd2d5e8f842842bbb5d88dff737b8362 : SUCCESS in 17m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78b77a11854745e0b82153513108634e : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":5},{"id":"ff415e2004c1919ab6025172f9d577f607ea33aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:15:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"7af5ee20ff0bc654b5ff8e2df68240da0a9bf9d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 18:16:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7529c0e11a4746aa965ae28ec7ccdc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89adcba594ab42e9859cc537025e9115 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49866a24e64c469689f60fde9b150f75 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"db3f22d42d455bf669e63b1f89523eb036af840e","revisions":{"f74d95760ed5889260034eda93898eeab6aa52c5":{"kind":"REWORK","_number":1,"created":"2024-02-08 12:11:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/13/908413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/908413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/908413/1"}}},"commit":{"parents":[{"commit":"e469e99eba971df38ebea814c6dfb442606c6192","subject":"Merge \"Ask for confirmation during overcloud deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e469e99eba971df38ebea814c6dfb442606c6192"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-07 14:29:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-08 12:10:59.000000000","tz":0},"subject":"Support auth configuration for Apt repositories","message":"Support auth configuration for Apt repositories\n\nThis allows use of repositories and proxies protected with HTTP basic\nauthentication.\n\nChange-Id: I0ec4ec3e9d60bb1431b44dd6718415214ad80025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f74d95760ed5889260034eda93898eeab6aa52c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f74d95760ed5889260034eda93898eeab6aa52c5"}]},"branch":"refs/heads/master"},"e7ad23c5de527b655489619ed2fc70c20cf4a63e":{"kind":"REWORK","_number":2,"created":"2024-02-09 22:27:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/908413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/908413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/908413/2"}}},"commit":{"parents":[{"commit":"7d8a43764f62e2629075e7e94b2e82dfa88a9cf2","subject":"Mark kayobe-tox-ansible job as non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d8a43764f62e2629075e7e94b2e82dfa88a9cf2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-07 14:29:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-09 22:26:57.000000000","tz":60},"subject":"Support auth configuration for Apt repositories","message":"Support auth configuration for Apt repositories\n\nThis allows use of repositories and proxies protected with HTTP basic\nauthentication.\n\nChange-Id: I0ec4ec3e9d60bb1431b44dd6718415214ad80025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7ad23c5de527b655489619ed2fc70c20cf4a63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7ad23c5de527b655489619ed2fc70c20cf4a63e"}]},"branch":"refs/heads/master"},"9d5910f1489e0136b7b5f063d1c96bb3d347ff4f":{"kind":"REWORK","_number":3,"created":"2024-02-11 10:06:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/908413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/908413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/908413/3"}}},"commit":{"parents":[{"commit":"7d8a43764f62e2629075e7e94b2e82dfa88a9cf2","subject":"Mark kayobe-tox-ansible job as non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d8a43764f62e2629075e7e94b2e82dfa88a9cf2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-07 14:29:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-11 10:05:59.000000000","tz":60},"subject":"Support auth configuration for Apt repositories","message":"Support auth configuration for Apt repositories\n\nThis allows use of repositories and proxies protected with HTTP basic\nauthentication.\n\nChange-Id: I0ec4ec3e9d60bb1431b44dd6718415214ad80025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d5910f1489e0136b7b5f063d1c96bb3d347ff4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d5910f1489e0136b7b5f063d1c96bb3d347ff4f"}]},"branch":"refs/heads/master"},"238b0ff35862c8b4acead503164506a9849ac360":{"kind":"REWORK","_number":4,"created":"2024-02-11 11:49:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/908413/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/908413/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/908413/4"}}},"commit":{"parents":[{"commit":"7d8a43764f62e2629075e7e94b2e82dfa88a9cf2","subject":"Mark kayobe-tox-ansible job as non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d8a43764f62e2629075e7e94b2e82dfa88a9cf2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-07 14:29:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-11 11:48:57.000000000","tz":60},"subject":"Support auth configuration for Apt repositories","message":"Support auth configuration for Apt repositories\n\nThis allows use of repositories and proxies protected with HTTP basic\nauthentication.\n\nChange-Id: I0ec4ec3e9d60bb1431b44dd6718415214ad80025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/238b0ff35862c8b4acead503164506a9849ac360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/238b0ff35862c8b4acead503164506a9849ac360"}]},"branch":"refs/heads/master"},"db3f22d42d455bf669e63b1f89523eb036af840e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-11 11:49:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/908413/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/908413/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/908413/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/908413/5"}}},"commit":{"parents":[{"commit":"2d0f797b3e98b7f68a6273b13555c13b7de74809","subject":"Merge \"Build letsencrypt images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d0f797b3e98b7f68a6273b13555c13b7de74809"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-07 14:29:13.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-11 11:49:16.000000000","tz":60},"subject":"Support auth configuration for Apt repositories","message":"Support auth configuration for Apt repositories\n\nThis allows use of repositories and proxies protected with HTTP basic\nauthentication.\n\nChange-Id: I0ec4ec3e9d60bb1431b44dd6718415214ad80025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db3f22d42d455bf669e63b1f89523eb036af840e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db3f22d42d455bf669e63b1f89523eb036af840e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
