)]}'
{"id":"openstack%2Fkayobe~908684","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I4f79b1b0887df72b11b768593ba45d89e75ce0d6","project":"openstack/kayobe","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-15 13:23:20.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-02-15 10:55:28.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I4f79b1b0887df72b11b768593ba45d89e75ce0d6","subject":"Build letsencrypt images","status":"MERGED","created":"2024-02-11 22:28:48.000000000","updated":"2024-02-15 13:23:20.000000000","submitted":"2024-02-15 13:23:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908684","cherry_pick_of_change":907723,"cherry_pick_of_patch_set":3,"meta_rev_id":"c7eae335079d5c914853d1b7b6e48314d59f80b4","_number":908684,"virtual_id_number":908684,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 13:23:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-02-14 14:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-02-15 10:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-14 11:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-02-15 10:55:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-11 22:28:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-02-12 00:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 10:38:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-02-14 11:56:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-02-14 14:37:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"bf24d722ae12d049a13756db99014efa29122c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-11 22:28:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"183d5c538c959cee054964c089982ace2e74746e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-11 22:29:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1ba22eadac0c98032bfb1fff393c34edad5532d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 00:59:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56a6c81c473a453a90294c77a91da0d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08166837403c44989c584b53ddac1469 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fdbffc26fa5422ca705c26da46e5e95 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e186121bd465457390e1414253293845 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cd56e09c6d94f24a8f7adda432ed32c : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b644fc422724bac8ff347de9c075bba : FAILURE in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd3ad8375ff445378d5d3321d56752bf : SUCCESS in 9m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/973fe8a4a00640d7a423d5e2604178a4 : SUCCESS in 55m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4e173781fe5a4209b54165cf4136911e : SUCCESS in 52m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc43e4d766644312ba58f1f46d4e6517 : SUCCESS in 44m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4985cbd8dc0f426e82460dced5ff8209 : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0be44c0c9eb6440bab2d8bbfb8de30f4 : SUCCESS in 15m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3e342f8237b843459fc05afe6e33a321 : SUCCESS in 17m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65ade03ee3484c66a8864700e7acf4e8 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/162aac40662a49d588388663992bbfcf : SUCCESS in 2h 25m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c87a3c796564904a51d084fdfc823d9 : SUCCESS in 1h 33m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b41ebcc5d814421eb986901a8ae13b6b : SUCCESS in 16m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7f1392ff66748eda6cb9a78f37b8983 : SUCCESS in 15m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/abf1aa337f434e97aa946cf825573636 : SUCCESS in 38m 19s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d76b1ee524444ce1a50034fdbe675a0b : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9781546a833842ad8c422bef0eea7c4a : SUCCESS in 21m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/40194461e836475d92f1f8f4b50fc5dc : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31b29164a32a46c5ac26370ce3cb034b : SUCCESS in 9m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09237ab6f9db4280bfd0c797335a864d : SUCCESS in 16m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce151e410824368961e9c49f389c72a : SUCCESS in 14m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c35af748dc54155bb06878c3763dc7b : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f35a03ba69d64c5aa5c9c5ed29aeeb61 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faeb63477cc549f4a46cc34d7d23e06b : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7e5f43fd94f4118bc55d22072b89519079e01950","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-13 11:29:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c43c7cc3832ec209ccd9c6ab1c28603049c167ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 13:56:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/150b2af25cfc4e189cefd7dbe8be04c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d1ff70e25924cfb93e0f5c9885d898d : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69855c0374784e55bd6b303bd31cc806 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/621253cbf972448280b59661ed3153e7 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73f9e2d89b2d43f3ac1dd2c740b05f7f : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb07efb97a59426587a91b93c3c63eb1 : FAILURE in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf8ea5c3dd734887953e6608837c7726 : SUCCESS in 9m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/891642299720440fbc9a40ba8660558b : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dd7585a57ed84c27b422abeeb130f0ad : SUCCESS in 56m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5953ce9d506041d8987c5bb1487f4e2b : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/97cafe27a9e0434c8417fc55960facb1 : SUCCESS in 54m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e5a6028a53da4fb7a98fd0915bc600b6 : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/046d74572fcc4f25b8f2621fc819ab8e : SUCCESS in 17m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e129b93fb4747808b773a41f038043c : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9a47e39b0b5d40c08f7f2f6a56d273c4 : SUCCESS in 2h 20m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efe6e260f6594e2aa7d30a6088033255 : SUCCESS in 2h 09m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/482455ff8a2e48089a2edf26dc0cbfc5 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8801f63f114342c48f56f35efdf85ac1 : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e72f08d4beb344f99dbcf5394f3d2f7a : SUCCESS in 34m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/578d00789d2d44ed9e59e92627f9e97a : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/137a59ce77194085803ca71a411f4b0b : SUCCESS in 28m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f4d9b3ce61b4d60a0717b690c4d6e33 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44a0d1b28c804644b5f563eced3e02bd : SUCCESS in 13m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7a4a6e5680a041f89e0e6159a53dcf33 : SUCCESS in 21m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5741f9b33146411388741fab8dbcf05e : SUCCESS in 14m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fefd5c90be864eb69d2663ab69a35d75 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af9f050a49604b2daf82f4d73252ee89 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b004195401344488b33011472fc239e : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1983c72044b389303fa8d1c090953ccb6bf6aeef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 10:38:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c4765ac44b578b28fb0dda07283d000237c5557","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 10:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c992f0b1e5f336d1b2e5bc55ad132d623276b0db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 10:45:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9430433b8c0c5fa2940f5f6bb04c5964a15c6865","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 10:46:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7c311794aae85fa080065e457b7b4543b9825f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 11:56:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20edd3e5f77d45bc6d2ee1115dfe780b06d6aa83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 13:19:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ab9576125c44a7dadedd73a4204145e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ede55c2b77404ef38782887a2c64f4ff : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210adc5da5ab48edad4693e5d4926dbf : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14791bdc5ecd4e2b85f43a5e9e23701f : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82825212643a496997cc6b61ceee950e : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be7b015d435149afbb0f7604eb6a2b65 : FAILURE in 15m 07s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02cdc51090ff4d6882df7545b9666945 : SUCCESS in 10m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c2ef563b266c42418df387e988631461 : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cabbcb732d24482b8f678449749853d4 : SUCCESS in 1h 19m 16s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21bacfbeb2b242fcbf5348d1ec676ee9 : SUCCESS in 57m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2cdd1202b5f84f37a1e392ea59c6f31f : SUCCESS in 1h 11m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/24e3636e729b4e8b974a6e638cb99ef9 : SUCCESS in 15m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/42717a2b2f584bd38e0c0a9902b21b16 : SUCCESS in 18m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb4107ae3ec641c0b2cc3a753d0e8072 : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ee1e032bd994479f9f648c35593a9a30 : SUCCESS in 2h 29m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/473c14f4a98541c7a48fe38d76732bb6 : SUCCESS in 2h 20m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f2da08b2343c4dbe8a3554617ffc769d : SUCCESS in 19m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/465e7eb197bc44bab18280d3f88f2dd0 : SUCCESS in 16m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0ea40eff917947cf920c2c85ee4a0220 : SUCCESS in 43m 18s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8bae2193813a4854a9e6a8b29b21134b : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6ed6be4734b4fe8b05ca45851c55c47 : SUCCESS in 23m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b9fb5839f2f4edb8d9de4f153befbd0 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/516cb03d82c246afabe519037e32698d : SUCCESS in 12m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb538958214945d8b163c16e8ba57880 : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5ae047ebd844cfd90742a679579df73 : SUCCESS in 14m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70e0d36826554a5085264894b4143b13 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f1178ad138460581b8be2277626c29 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa831d222bf744249f4473e65e9d926a : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3ed132d6ac83f1aa7b37b4e7102f401ef5f95dd0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 14:37:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"552daa1068f7734c158789fb9a41b4a228184486","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-15 10:55:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67bff327bfa09aa104e678fa5c0f8fd48c05e0ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:56:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8139f87f01712147d49c1760d4e84db3d314500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:23:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a89619a75014f83aaf61bf6bf2ec11d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed2bc64d0d84279b1d0ff5f5a7bc79b : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a8602605eb240048d60da93f2f9f7a0 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/361066959c77416280cfd55b52109f57 : SUCCESS in 4m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d667788f321d49109cbab522c2a82dd7 : SUCCESS in 8m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/83a8a38f36e642a394bdc8202591b264 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16117f6faec24976831c9cea3f309a68 : SUCCESS in 49m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c4734543042d43949dddfe4e120e28f7 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/749e6b5f84974760b1cee61d348345c0 : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c305f602b8714d04be045010259b79bc : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9e4835d0b3e4c5faa4f7d6b1dd4ac4a : SUCCESS in 2h 17m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f1370d368894a9c9718a78b962ca177 : SUCCESS in 2h 10m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ce61c66c6a2b4388b9c22f294d290b3d : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7eee91f80fb4000a6ca1584242237cd : SUCCESS in 13m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d9fa49776f546a697002a2676ce6d42 : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5c005ec24094b6794b7c03cd53f508d : SUCCESS in 28m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c40ef72f54c48ce813e79b1bc274d05 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04dca655283d4b69b9b916e3198a809b : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4535c18f8e3d4069a96cbd8fe3c19f34 : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e32a5bef77714dcdb24f647bd1d21b45 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c7eae335079d5c914853d1b7b6e48314d59f80b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:23:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df8cbcb6c70558f7b1e3162eec5310d07b36412a","revisions":{"7fb3b5ddf66f89e87b5b08f95f9231c02c57b460":{"kind":"REWORK","_number":1,"created":"2024-02-11 22:28:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/908684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/908684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/908684/1"}}},"commit":{"parents":[{"commit":"bd16d2931d79d6df86f9e400e5387d924d99df30","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd16d2931d79d6df86f9e400e5387d924d99df30"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-05 11:31:44.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-11 22:28:48.000000000","tz":0},"subject":"Build letsencrypt images","message":"Build letsencrypt images\n\nThese are currently not included in the regex we generate so aren\u0027t\nbuilt as part of the complete set.\n\nChange-Id: I4f79b1b0887df72b11b768593ba45d89e75ce0d6\nCloses-Bug: #2052428\n(cherry picked from commit 8006c65a54a8c8fc2d9f42e478d4e146a51429ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fb3b5ddf66f89e87b5b08f95f9231c02c57b460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fb3b5ddf66f89e87b5b08f95f9231c02c57b460"}]},"branch":"refs/heads/stable/2023.2"},"df8cbcb6c70558f7b1e3162eec5310d07b36412a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-14 10:45:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/908684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/908684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/908684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/908684/2"}}},"commit":{"parents":[{"commit":"be95804757648f28055054229ea33e0f5e2fd485","subject":"Mark kayobe-tox-ansible job as non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be95804757648f28055054229ea33e0f5e2fd485"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-05 11:31:44.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 10:45:20.000000000","tz":0},"subject":"Build letsencrypt images","message":"Build letsencrypt images\n\nThese are currently not included in the regex we generate so aren\u0027t\nbuilt as part of the complete set.\n\nChange-Id: I4f79b1b0887df72b11b768593ba45d89e75ce0d6\nCloses-Bug: #2052428\n(cherry picked from commit 8006c65a54a8c8fc2d9f42e478d4e146a51429ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df8cbcb6c70558f7b1e3162eec5310d07b36412a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df8cbcb6c70558f7b1e3162eec5310d07b36412a"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
