)]}'
{"id":"openstack%2Fkayobe~908930","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~Ica6c0b204094ed61aef8242c934eb3b7715ce944","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-15 13:23:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica6c0b204094ed61aef8242c934eb3b7715ce944","subject":"CI: support .gitreview of unmaintained branch","status":"MERGED","created":"2024-02-14 09:52:25.000000000","updated":"2024-02-15 13:24:33.000000000","submitted":"2024-02-15 13:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908930","cherry_pick_of_change":908950,"cherry_pick_of_patch_set":1,"meta_rev_id":"1cddf474e57f29cc169ca6bbe5d17949d7f7c96f","_number":908930,"virtual_id_number":908930,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 13:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-02-14 14:13:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-02-15 10:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-14 16:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-02-15 10:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-15 06:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 10:46:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-14 12:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-15 10:55:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9fc8765305408ef2a5574c818a8d5ee9ca786b28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 09:52:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ebf6828672824caab30661636ac835f0d2834a96","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 10:46:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f04efd4392967c9c0fe5338c2d814b40e2db2cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 12:00:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f720a18a5845e58f9d0337ad4b38cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41f86b85dc204531b67a56b9a926b92b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c027fdb9ce48c087cf7a3786aa4a3e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b678a7f21194a8d9d2650ff8c7b90f5 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f906c24f3d4014a43f677982ad7107 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/960d6d69e959464d9af1ad623cf07c01 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39863c8960f8452187594fd9dfa62ac6 : FAILURE in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e563326ce4dc438cbaed2f26a85882f8 : SUCCESS in 12m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69267b5eadad47c5bdaf34ad593106b7 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0126ad42527348ffa178eac0c777477a : SUCCESS in 1h 05m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/130d7fa942a74212af19b3160edc7340 : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0d7cf3447d3147e69baa8e668b412915 : SUCCESS in 1h 14m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a31357476e1042559650fd66880d56da : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/15aa2880256648bcbd83f77b9f852e89 : SUCCESS in 19m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c94e98099744100af49390ac7b3c15d : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5bf3cf8c1fc942d0a8e6bea29d3e439e : SUCCESS in 2h 00m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbbf553a02a64b46bb730b2226ca35fb : SUCCESS in 1h 29m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/363ee575d3f646fcb90d0909ec6e0430 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/336f61ff56b94689a43f7f1e58662978 : SUCCESS in 15m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/74b629575b8349b8a0c32d540b07a5a9 : SUCCESS in 33m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f118a73c5c354fe698ad0f489cbf59a7 : SUCCESS in 33m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c1e75e0d6194afaadaead8c8f78358b : SUCCESS in 29m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a5dbca74a614b3cb61328b86ea236b6 : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f2ffe491ab64b0a9522ab2ab6c9e9fc : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9db4cf9e2dbf4739822382fb13cd2c1e : SUCCESS in 18m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe637fecc1ed4dfb98cbdc171dc5d2de : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5a54e5ffa51cf8d40f2b8bf2263c2ed656474525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 14:13:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5864e4cf074d7cdd78c263b46b8e9720c151d280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 16:17:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc9d9ca09de6434094ffbb3ca31f1d5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/168b1c90f3094bc9bcd61db80533020c : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e89b964d17410eb2495b40481c2a2b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa293edc80c4584a48f9e6e37669d1a : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e534d453bee94570879be84afe601d81 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47b8d29718314ceaa9e4ed8cc895415f : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ab3e6098af746bab99d17a8700cf4c8 : FAILURE in 9m 52s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/362ebec71e3d48069ae7547308686faf : SUCCESS in 15m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6b7ad88051944573832714dae47c824b : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/36946161a5014a2d96d9e38d305e5b93 : SUCCESS in 59m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73803d30972347aa9e1714dbf76acfa1 : SUCCESS in 57m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7df276b70e644a3b959079df6b62f29d : SUCCESS in 1h 11m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9114c7745efc4ac2a166bb06459c9056 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2e80c8c0972b4a2d8a941f32afd071eb : SUCCESS in 19m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfb0457d16674abf883d7aa08af284ed : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7095b012774a4816b81ebfe264142738 : SUCCESS in 1h 59m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/382ac2735e0b4671adb8b0f4c4a65a08 : SUCCESS in 1h 42m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a665e72cf4d4d1489dece99498f4ff7 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/affa228c595c4ea5b67b3422788be176 : SUCCESS in 17m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f08352e6f4af40afa89aec0a0092898a : SUCCESS in 34m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b31d0360e46642efa3c55f8a8744439e : SUCCESS in 30m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45d1b4206c6649c1becafe03e8cea17b : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14fd4ddfa843464198f4c3e8b6e2589a : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/835db59d575441759b4864f47f2ba75a : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90c91ceb884d48008bdc428dc3f595a8 : SUCCESS in 21m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a02c23a470f14a9e8aac7d5959b0c2dc : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5178d3c552dc3c773e0eabf28025cf386a02709e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-14 16:20:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57a92b6e2fadb55ee246aa1a44fb4b280898bf7c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-15 06:51:41.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"2c2e14af4b68b18603720857850c6151542d73aa","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-15 10:55:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6cd0a37884f7d11796c0318179b0173649f21a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:55:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a93cc621a49c46269eee98005048abf732e1a382","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:23:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a7e808d63e4b958bef9fed13a1ff6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1ab039234943499a1b5932909d2063 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b0ce150100e42a489e5ce66d94abf6c : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f8b7cce97f4dc2b00f5039e871c6c3 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce30e1cfd5ab46f18f3152870508472c : SUCCESS in 4m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/343cccd5612545f08492e4c25027c3d0 : SUCCESS in 31m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6f5c0f1fc2f34917bad8eed818006a87 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dab6cf1259441c5aaa4a4daec43ec34 : SUCCESS in 57m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/48ade595d81d455b9c162aa5c7692e83 : SUCCESS in 1h 13m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/33925ba4e12e43a9b4636c2c2c13ae04 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9bf9f77c4224fb7864f59592da17ada : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/86fc1c2234e449369c9a249d66c31fc2 : SUCCESS in 1h 51m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3d04cfc2f7341e29961db200bf35cdc : SUCCESS in 1h 22m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/39d2920f26bb4dec9be75d98f326a1bd : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7681fe07dc1548608822033be73b0fd8 : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ccd6fcdb1e064662b8161d0fe8c6bcc1 : SUCCESS in 30m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/671a25566b50417f9fac29a9d6e8a8de : SUCCESS in 25m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e4827877cce24928bd13b00a02fda9d0 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82e3394970fe4e34a9e3fc8e04fa9e08 : SUCCESS in 15m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6aad5c00600d4020b4e6e406ba114130 : SUCCESS in 20m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a390eddacc514cdc9dcbbbeb64c41679 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":2},{"id":"67ace50d02e7c73ceef0702bd16992d432fb0925","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:23:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1cddf474e57f29cc169ca6bbe5d17949d7f7c96f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:24:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dad402884a54e769bc4d79302ca22d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0d413edd4f44578b75e18696ea78b5 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f1828faedf5d77a03b931db279badaa2b60af82","revisions":{"b0c655785174c0c1529d400c3935a5b5a3fe7a0a":{"kind":"REWORK","_number":1,"created":"2024-02-14 09:52:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/908930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/908930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/908930/1"}}},"commit":{"parents":[{"commit":"d397ceca74667ae0121b7bfebab77fd6888767a0","subject":"Deprecate devicemapper due to removal in Docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d397ceca74667ae0121b7bfebab77fd6888767a0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-13 22:07:19.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-14 09:52:25.000000000","tz":0},"subject":"CI: support .gitreview of unmaintained branch","message":"CI: support .gitreview of unmaintained branch\n\nChange-Id: Ica6c0b204094ed61aef8242c934eb3b7715ce944\n(cherry picked from commit bb51db15febc195384a48406cb37571776988ee3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0c655785174c0c1529d400c3935a5b5a3fe7a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0c655785174c0c1529d400c3935a5b5a3fe7a0a"}]},"branch":"refs/heads/stable/2023.1"},"3f1828faedf5d77a03b931db279badaa2b60af82":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-14 14:13:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/908930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/908930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/908930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/908930/2"}}},"commit":{"parents":[{"commit":"56ab41c00d450b8f77b74b3d070c0a42adcc0a13","subject":"Mark kayobe-tox-ansible job as non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56ab41c00d450b8f77b74b3d070c0a42adcc0a13"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-13 22:07:19.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-14 14:13:50.000000000","tz":0},"subject":"CI: support .gitreview of unmaintained branch","message":"CI: support .gitreview of unmaintained branch\n\nChange-Id: Ica6c0b204094ed61aef8242c934eb3b7715ce944\n(cherry picked from commit bb51db15febc195384a48406cb37571776988ee3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f1828faedf5d77a03b931db279badaa2b60af82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f1828faedf5d77a03b931db279badaa2b60af82"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
