)]}'
{"id":"openstack%2Fkayobe~908950","triplet_id":"openstack%2Fkayobe~master~Ica6c0b204094ed61aef8242c934eb3b7715ce944","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-02-14 09:47:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica6c0b204094ed61aef8242c934eb3b7715ce944","subject":"CI: support .gitreview of unmaintained branch","status":"MERGED","created":"2024-02-13 22:08:04.000000000","updated":"2024-02-14 10:47:10.000000000","submitted":"2024-02-14 09:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908950","meta_rev_id":"80506a8a3936492b4c664ee26a1db2b251224aea","_number":908950,"virtual_id_number":908950,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-14 09:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-02-14 07:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-02-14 10:47:10.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-02-14 07:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 00:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 07:00:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-02-14 10:47:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"eb0115c3d48f11694ad5b648d0231e72b95b33c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-02-13 22:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d733ba64afb860b1b4be4234373fec236fe3d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 00:33:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5adaaf9c17643ff9a28790b2fdb4ca2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb794135e6c444ab36ee06bd02a008f : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4384aedbe2fc4be9b65d57d3f0f460fc : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a187a4a298247b382a9920736ba33cd : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c57373d6997141cb806a2d9138a457af : FAILURE in 11m 39s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8893bbc14fd4f71bc27f0d4c7ad9d83 : SUCCESS in 8m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9b20d8e8472749a3906ab8f3526b4a7b : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8e6fcb7fa3f74fd5a4ba3b2ac73f87e3 : SUCCESS in 49m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daa398d191194156a7bc6225271c47c9 : SUCCESS in 54m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f74d51e4463e4e3e843216f77c8e2f87 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4ddc1ff157d04d8b94834b15c72ede0f : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/75e8064c18e64c1285d6f287a779ccd8 : SUCCESS in 19m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65575073631f4fc8bac91a3aeb4f17fd : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5b4899c1d41946029012c7a2a5d1f1a9 : SUCCESS in 2h 20m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba48d387741e447ebd1fab9fb2e98f6e : SUCCESS in 1h 49m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a03e1fc3afdc42669a9bb7ea52f1fc3d : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cf1c0273cff4654bdedd0db23cd2353 : SUCCESS in 16m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/64f192b4063d44509456f9357212a544 : FAILURE in 53m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ee6560be3c084048b314414126af3ef5 : SUCCESS in 26m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7b8298c2bf84151a6dbfd197f1bf9b9 : SUCCESS in 22m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84f036b79eb2444ebf4e821e801289fd : SUCCESS in 10m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e0c58b523d84315b0af5c0dc98564f1 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c504cab2c7b433db878710d3f422ed4 : SUCCESS in 13m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2f3feb26d6c42d5b6739ce77cd1ea22 : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc467a016ef948b28d73e19274f43e6a : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/480ab7d63e064c918aaaf23df5fa0437 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/054c3b2b261e47c0ba359d229a1b2ba3 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2cfe62a9494f5ad3279493a3a4726c2a76c5099d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-14 07:00:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abd2a7e2b42dbe110158a587a32fbb879e2c69b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef0a078bb3479110c6eac743fdb090b9c883bf85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 09:47:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95aff454f2af4717bc3c53bdbc8485df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b9ae0baf895436b88b1faa7850fcc2a : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8086c4b5ec754730a927b63bf32448e8 : SUCCESS in 4m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a32a9b96660432ab4e0b9ca66504cc6 : SUCCESS in 9m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2d44431267d148c4991d0e7a247a01f7 : SUCCESS in 1h 13m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0feed5a4ed1d47ce88c46b5d4610ebc5 : SUCCESS in 48m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bf25cebe55884513a2cf0e490d119674 : SUCCESS in 1h 36m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/077fc958ce5d4e80a7f95e8f2194051f : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/596173806a09450ea5d8c895bdf9d5cb : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b802dd1aa48d49d7b90630ffec56113a : SUCCESS in 2h 36m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5050de5fc9f94265a4981e6baeb8e5db : SUCCESS in 1h 59m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b649e61e26644555a1cad1b9b24ca28e : SUCCESS in 15m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b19387a72a46411b9caac6ae99df5165 : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef0645c48f8f463fa6d1275204b4b460 : SUCCESS in 30m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a180b11dff1e4755a57ec1ec98db1a47 : SUCCESS in 22m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29ada1de0c53416cbdf3042df88f5eef : SUCCESS in 12m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99319c21e22540f28f5de40087bc0635 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12835773680346548d95774b70bd4d01 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eae336fa5379427cb586394bf733865e : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2b5e1467033764ba015c83783894cbf60201e97e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 09:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"667ddc8161f4fb3ec974f2167609d737d2d4f469","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 09:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e9ca16b75644029427dc9db3d3c7c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200e73dfa15446a8aa492cbcf42e7c45 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"80506a8a3936492b4c664ee26a1db2b251224aea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-14 10:47:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb51db15febc195384a48406cb37571776988ee3","revisions":{"bb51db15febc195384a48406cb37571776988ee3":{"kind":"REWORK","_number":1,"created":"2024-02-13 22:08:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/50/908950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/908950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/908950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/908950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/908950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/908950/1"}}},"commit":{"parents":[{"commit":"2d0f797b3e98b7f68a6273b13555c13b7de74809","subject":"Merge \"Build letsencrypt images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d0f797b3e98b7f68a6273b13555c13b7de74809"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-13 22:07:19.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-02-13 22:07:19.000000000","tz":0},"subject":"CI: support .gitreview of unmaintained branch","message":"CI: support .gitreview of unmaintained branch\n\nChange-Id: Ica6c0b204094ed61aef8242c934eb3b7715ce944\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb51db15febc195384a48406cb37571776988ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb51db15febc195384a48406cb37571776988ee3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
