)]}'
{"id":"openstack%2Fkayobe~909113","triplet_id":"openstack%2Fkayobe~master~Iff0b28b770753c81ac526afd1ce6f61d77fad25c","project":"openstack/kayobe","branch":"master","topic":"fix-rabbitmq-custom-conf","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-03-26 14:04:10.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-26 14:04:10.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-26 14:04:10.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-03-19 13:57:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Iff0b28b770753c81ac526afd1ce6f61d77fad25c","subject":"Fix the glob for the custom RabbitMQ configuration","status":"MERGED","created":"2024-02-15 13:31:00.000000000","updated":"2024-03-26 14:05:17.000000000","submitted":"2024-03-26 14:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909113-fix-rabbitmq-custom-conf","meta_rev_id":"aecde214820752a2d02c6a9c8711419f804a14fc","_number":909113,"virtual_id_number":909113,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 14:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-22 10:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-26 11:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-26 11:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-15 13:42:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-02-15 15:28:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-02-15 15:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 14:27:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-02-16 16:46:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"847629119212c8a422726b0ba191c55be316c79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-15 13:31:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c7a941b1e067a8592000a2326409a541fbdcf4f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-15 15:28:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a4a551349392b7f65e98b012d16c5b4301caf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 15:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/133912487e0b438d8aecef8338492280\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070c83b30fcf47ab84960318254bd626 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135de8ce6be24fa99bf6d7b201f107ca : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f8253848c184bbc80251d98b8f5f383 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbe56b3f88224230bceab023cf45e506 : FAILURE in 11m 35s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2fb78b733a841729832a0c2b20b6259 : SUCCESS in 8m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5d71186f4b7346fbacbadc1a5192585e : SUCCESS in 59m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d17cedf5b2b84d98a88e132143eca46e : SUCCESS in 49m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f9e1892ed1b492db0dcd9d057a3cbd2 : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8cfb7f22df0e4b1fa180a9999882fa30 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b9e44a10066048a293326733fa0213b7 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/df7ca772bdd3418cbe97e636a51df62d : SUCCESS in 17m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0f374cba3ed4c2da4b7dbc2c6f57820 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3dce8d920cff46e29592c02184be5f69 : SUCCESS in 2h 17m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa609c9af4d94c5884ed40899aeb2712 : SUCCESS in 2h 06m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d994dda4d6b4ffdb8f8a2463bbe2f5c : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb498994a26b4441bfc70ce76ce175b8 : SUCCESS in 16m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c4de3680372f4a27b79bc0c0c8953c57 : FAILURE in 34m 19s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e997a6f51d946d691372e28d17f1541 : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d81dbc204353433c8bb9fc4c26df6a21 : SUCCESS in 21m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8611f28e4fe45f59c1afad85e0cc3f4 : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2680ed29158488fb80f84ca2b62ac7f : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b30864bf75fa4ac4b9a26c990b83913d : SUCCESS in 20m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b15fbd0bca464dc2b5893faf37d1f194 : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120e743e964c4db18c615f7a5bb04634 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33415d4bf82b418b9932f5f1e780677e : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4afefd23041b4c4a978c2498657dc03b : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"80e0c42f97b3a4036b2281f7df62d5164f78a99d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-15 16:02:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66a58fd341ac5ab3c3c1528263c09b0a25c29d68","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-16 09:53:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cff25e3f905beafef501b5cbbf53bb50d933ccd2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 11:59:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b29fc0f81a9c75b16a918f423d6f2cd12682f2f7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 11:59:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"bfcd5212c9dcef538cd7b98783129ca7b494b5a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 12:25:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1c53077157308956ff4752feb07c85b06b8959b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 13:20:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"08b5f56a4d63c740458856b30eb6d36bdc5be865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 13:40:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ec4c2159cc91b4d3e4744cfcb365e9984c513767","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-16 14:27:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4e8d90caf34a99fe5781ec38bf471f609ff9d3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:37:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ceb975d361f44259ca3a42296501d27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f29f472a7b7f4256857112733b667588 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39348c2610ec489da02a3ec9601b55f0 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/207e16e47ed343ddbadf8d18369eaf80 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a047bc96fb444dfa85752969da5b9943 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d55dd566c89b4720b7ee9b3f8b180fbd : FAILURE in 9m 43s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fb25710b5c54ec7a84505a833fc6106 : SUCCESS in 9m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd8a4182c2994f2c85415d39ece740b5 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/23b7a6c319454e5ba1aaa345f3a71f41 : SUCCESS in 1h 05m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0ae59e701f94a009694dd5d3b2c41a0 : SUCCESS in 53m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7a0d63b26b214eaf8b38ad1c4a6947c1 : SUCCESS in 1h 22m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a26860b3ba094b22baefb91a6613ac7c : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a0eaf9f3a5424df880a7799c2517526e : SUCCESS in 18m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9e4164e8476497082e9793bfed3575e : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8385a089d2d641d2a8ef15f288757edd : SUCCESS in 2h 51m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ba607d195f5435dada5df57f63fdca1 : SUCCESS in 2h 14m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/80a80300c2174ca583ccdfc48b8d5a75 : SUCCESS in 19m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/850422ebedc74a63b861163f895f5eef : SUCCESS in 16m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/082a3985f160485d921f7b4dec2275a8 : FAILURE in 36m 46s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a3b3e1289f964c8cb4f94c3e6e0bd45f : SUCCESS in 33m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2e7d2a8159f42f7bf1fce8f429562b2 : SUCCESS in 28m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1790b3b3c3a94728a747acbec39d511c : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c5039ce87a54e7c9e0b5813f969195d : SUCCESS in 12m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87485077ee3a4dca9547dd49394a57f2 : SUCCESS in 18m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbdba6915e6045b599a876a628ef4f66 : SUCCESS in 14m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c0870758d3f40078726662c0070cbc8 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e91d0cdd6574fb4b7c03ca3dca4b7ee : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c80276d2ab444b4aafb014434ed510a5 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"15db69fa483ae61e873093e8994a737de4614bad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-19 13:57:56.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92728aa039df40c73769d35ba4526a770497b0e4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-21 12:16:20.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e0c393e4b82fca835dc847245b75db694efb08f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 12:16:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1278e111d4e51dfc87c0884c847f3a3dab41a9b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 14:51:09.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11d000a71d4d4a58b9e131ab37489c18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b93a37d0064e23a55276c91430cd7d : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fe36220a2514788b77bee2a63c096e8 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c6b5aad4e44478b82afe08c0bbfc2dc : SUCCESS in 4m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06d36f6ffa654f0f9fcc7795cb4f1aa1 : RETRY_LIMIT in 3m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e4efcc43be2048589fa152e16c05c5fe : FAILURE in 21m 14s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5208a720b7c948e182f40474fd05b034 : SUCCESS in 59m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10f7c65c9a3f4a068a242ce4be031a35 : SUCCESS in 1h 13m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a1b045f231c64c6c94e5ee531e7de6bb : SUCCESS in 14m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bc39c6b9ec9414293db3bc49003ab38 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4073c34ce6d04e7ab0eda5c1f02825cd : SUCCESS in 2h 17m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63b865127e3a4e4eabfda20a4e181d43 : SUCCESS in 1h 37m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3a92788788234304b6c46a044adfc5f1 : SUCCESS in 18m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d1cc883fd274370b9c84a99ffb58758 : SUCCESS in 13m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8cff8a75066e4365948b902c7f21a843 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba0e85f6f263452b864ce6e833e9b97e : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0dcbe19a7ba042939d19ff77cf0e023e : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46089840a8f34f3e956954196ec583bc : SUCCESS in 23m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1c1f5d5bdcb4f54bed787cdab72ef1c : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7530ce2f5e0f4af5a0fc58c14b80881b : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3b5383a39888459599892b63e4aef8c9 : SUCCESS in 17m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc72599222e34fc5a8f022a5197f2692 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b0fe388b19382a1b1aa314007eccf6362d992f5f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-21 17:30:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c050d32d3d1367e95be78ac653643a19c31e9e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 19:45:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a57509a1a2e1487989c62825445f5b7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/468d41c369824291bcc3806e08623ecd : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a009f122ca2b497987de96328d6f72a8 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e40c8f4237242e6ad9237c2b7bfb2fc : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/924e9443cf5b49279b503958af1d446b : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be599086bd3d4676a731113e11aea597 : FAILURE in 11m 55s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d2431b9c6c4d4b12bcf58115af2bee40 : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3221ca4e37794fba9f9fea2ef9c1843c : FAILURE in 20m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ab7cb209bf4c415f8ca4d063d7dc3393 : SUCCESS in 1h 01m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6445aff8bbc54a89b181afe90c57ae56 : SUCCESS in 52m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b6e998e7d98b43359ce650bc4e594f1a : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/199d20e231a74eb492224e01254246a1 : SUCCESS in 17m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/31335c2e22624bb0a6a3294d0cc65ac6 : SUCCESS in 17m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23d93a3b4610432b8d87db59eb143e39 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5d165b081a334c8c8de31532f68fe531 : SUCCESS in 2h 14m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b07988ae664fd8beeb40d348ed46b9 : SUCCESS in 2h 03m 59s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ed84976831124e628862e96004e0cb2f : FAILURE in 1h 16m 32s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d5d46560835467c80c1a82a9f2a836d : FAILURE in 1h 11m 23s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/725cafdb3e194e689ef852172e6b5bd0 : SUCCESS in 16m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a6d9070d854494d85b187a30ece54ec : SUCCESS in 17m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4e43686e0d784131861f346e26e290aa : SUCCESS in 39m 18s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fe586bf52fa64e4a9449faa6ca8bef01 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36cc7ad11f3b4e3fbf7c915d6ea486f4 : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/817e983724ec44a8bad01595c260bea7 : SUCCESS in 35m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afd18e1bf2274e1b9591f9532c9e0561 : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e214f03af6224c61b5a07375ed3131d8 : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78252e1098c74e5e9eb3488bad422113 : SUCCESS in 12m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4344280c2fd2481e8b03fd4de79e91c0 : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47355bfbec754d2a83d917b4b246362c : SUCCESS in 13m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4427229cf1cf4f68b9798c27c30cf627 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b451330eadb4065be3280a261c6b383 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f340349e57a4462a751eb6035db7a8c : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"51dcedc7c07642fda26471a2dcf9fad6ed746c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-22 10:53:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"919d89e8053033288d0e5e0a854ebc5d5cde6288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 13:29:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ecbe4e7b8464c668799211550462052\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cc5f028fdce4e5f9c4b8e3308718126 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d08da677d544355ae67f613a984db34 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e56fdf2b8c7461b9e48058f9e7b7ea1 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06574849c07241728ecbec331ffc028a : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c74c69dfcd1479f9a4ae0183b6d8bad : FAILURE in 11m 15s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f6043e6261044fe9ff1d3bea8d87e34 : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fb664020c8c4464bad137adfa2164a45 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/36e84628b05e44cb99f4303a8dab05ba : SUCCESS in 1h 07m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1d2d3c2b9d74a7b9b0a216c4bbfa3f7 : SUCCESS in 55m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c882f7cc3b94e5fa2841d876e0a07c9 : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d9163e9492af413e9a60bf77420396e2 : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/564fdbd7c6b84d9ca7665be87b6912bf : SUCCESS in 20m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f19f55e23f29468a834d5111d643abcb : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e188b7e51344e4aa846ee54a262a1f1 : SUCCESS in 2h 28m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5317a279ec2e4471bcdf1e35a09969c6 : SUCCESS in 1h 52m 03s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a1612d9de8254aaba752cd4d5b4cfe02 : FAILURE in 1h 27m 37s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c52d86f185764e28a44e570691d29449 : FAILURE in 59m 35s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cbe1a595895746f4816c6ad37d59fe15 : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c736e5845384f428f814c6f032835eb : SUCCESS in 13m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f51f6267269d45aa863fa0deeff9773a : SUCCESS in 39m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/75e97838608d4fde890ccf1327e781e0 : SUCCESS in 28m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1ba67ab9d24a9e8d66a7f6c4985ff1 : SUCCESS in 23m 09s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/28d7c1b3f503451cb5b3ad43df26a172 : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5e1b60a69b64bc0a282a4cfdab8bf19 : SUCCESS in 22m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e558640af67f41548b66323a670c8bcf : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b957c286b9f402cbe1a13ddf61d6e65 : SUCCESS in 12m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/90e8086b59a54e948a8dbdc2895b625a : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09b630075cf549a1af80c3a0bf2a3001 : SUCCESS in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3efd70c6d28d43588b371ec1f3c64191 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/162eaaa1b3994217b0c026336c58b9a0 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee37e74c79634dbea037897d4a5b9fb1 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"95ffa92db02fecab00a51d2d5f314a1df77d2b23","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-26 11:48:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"843227002c32a0df0a705ed0dadc1f1fd75d064b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:49:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"72a69b54ff70b2864fba8674a6140a2f12692f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:04:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2307c3850c547f08105677ba4d40096\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c1d1ba053b4952b6815d14a480dcfa : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f052148e34074a7b8cb1ccf1736d9c13 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6f1b07134234ac9a0cc36952102b9ba : SUCCESS in 4m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd76b7efb8ef4c429f39cdbbf202e154 : SUCCESS in 10m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fda4f97f506b4be380979664078331d0 : SUCCESS in 1h 12m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b35a0fce7b94261bf07556b543807d3 : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5488611c29d241b884354e2fc96284f7 : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3582dc7d603e4a01857a701abd3ca0b6 : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7804edc194747ae8f6957872d0dfb9c : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c4a2be99da6b4051af44c48b811a9760 : SUCCESS in 2h 13m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e36c4355a85545d2a6ebbdffe4ebc9cf : SUCCESS in 1h 45m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bbef9ba85b5c4b309d6b0a52f2cb755d : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0494a38bbe5f4a5fbc519552de9509f8 : SUCCESS in 14m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8467c22c40c43ccb598d6c78ea2adc4 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70770926574f469cbd229f2d24c3d8e9 : SUCCESS in 22m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0a60bbdf8d7e430fb5ece18d75fdd92b : SUCCESS in 30m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22bdc1c88d9a487a9f86a75000da0441 : SUCCESS in 26m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64532fcc307a4eff99c70c38518ede2e : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ca955da3b5f46afa69bc135846ab511 : SUCCESS in 12m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e315808c74449568ff830d8bd21d81b : SUCCESS in 19m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/892253a8bbc44c3aabcaa49d6d08271f : SUCCESS in 16m 20s","accounts_in_message":[],"_revision_number":4},{"id":"ae86b238a3506f1d99d8d1edf913f3a5d43188c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:04:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"aecde214820752a2d02c6a9c8711419f804a14fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:05:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a0912476044e9f967ecf78d5b06fa6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b26b0586c6d4984a417b947518b5450 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f57790876ec141ae9494bc336c3f69a1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6f59b49ab84b0bc2c6550b351069141ae97247c8","revisions":{"ca1a7f188f2a647ef4af21e45cde6180189f25d7":{"kind":"REWORK","_number":1,"created":"2024-02-15 13:31:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/13/909113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/909113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/909113/1"}}},"commit":{"parents":[{"commit":"bb51db15febc195384a48406cb37571776988ee3","subject":"CI: support .gitreview of unmaintained branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb51db15febc195384a48406cb37571776988ee3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 17:52:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 18:02:52.000000000","tz":180},"subject":"Fix the glob for the custom RabbitMQ configuration","message":"Fix the glob for the custom RabbitMQ configuration\n\nFollowup on Id4588f4529a4522e68e22ce58711cb927fa68a9d to fix the\nplace of the RabbitMQ configuration and there are not only the\nrabbitmq.conf and rabbitmq-env.conf but also advanced.config which\nalso often customized and more other files.\n\nChange-Id: Iff0b28b770753c81ac526afd1ce6f61d77fad25c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca1a7f188f2a647ef4af21e45cde6180189f25d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca1a7f188f2a647ef4af21e45cde6180189f25d7"}]},"branch":"refs/heads/master"},"2186ec4bd0c64643dd18b6271c0019f28e0ce647":{"kind":"REWORK","_number":2,"created":"2024-02-16 12:25:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/13/909113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/909113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/909113/2"}}},"commit":{"parents":[{"commit":"bb51db15febc195384a48406cb37571776988ee3","subject":"CI: support .gitreview of unmaintained branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb51db15febc195384a48406cb37571776988ee3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 17:52:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 12:25:21.000000000","tz":180},"subject":"Fix the glob for the custom RabbitMQ configuration","message":"Fix the glob for the custom RabbitMQ configuration\n\nFollowup on Id4588f4529a4522e68e22ce58711cb927fa68a9d to fix the\nplace of the RabbitMQ configuration and there are not only the\nrabbitmq.conf and rabbitmq-env.conf but also advanced.config which\nalso often customized and more other files.\n\nChange-Id: Iff0b28b770753c81ac526afd1ce6f61d77fad25c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2186ec4bd0c64643dd18b6271c0019f28e0ce647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2186ec4bd0c64643dd18b6271c0019f28e0ce647"}]},"branch":"refs/heads/master"},"48b52be053cb3bfa5a2476166b1f0c8a658482a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-16 13:40:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/13/909113/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/909113/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/909113/3"}}},"commit":{"parents":[{"commit":"24d65075f6130e98dc7c0379f726f03ac9729925","subject":"Merge \"Support auth configuration for Apt repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24d65075f6130e98dc7c0379f726f03ac9729925"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 17:52:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 13:40:23.000000000","tz":0},"subject":"Fix the glob for the custom RabbitMQ configuration","message":"Fix the glob for the custom RabbitMQ configuration\n\nFollowup on Id4588f4529a4522e68e22ce58711cb927fa68a9d to fix the\nplace of the RabbitMQ configuration and there are not only the\nrabbitmq.conf and rabbitmq-env.conf but also advanced.config which\nalso often customized and more other files.\n\nChange-Id: Iff0b28b770753c81ac526afd1ce6f61d77fad25c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48b52be053cb3bfa5a2476166b1f0c8a658482a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48b52be053cb3bfa5a2476166b1f0c8a658482a0"}]},"branch":"refs/heads/master","description":"Rebase"},"6f59b49ab84b0bc2c6550b351069141ae97247c8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-22 10:53:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/13/909113/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/909113/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/909113/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/909113/4"}}},"commit":{"parents":[{"commit":"997a1b2c63ef2c7cf838b9873b13593b90aee70a","subject":"CI: Make SLURP jobs non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/997a1b2c63ef2c7cf838b9873b13593b90aee70a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-14 17:52:46.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-22 10:53:02.000000000","tz":0},"subject":"Fix the glob for the custom RabbitMQ configuration","message":"Fix the glob for the custom RabbitMQ configuration\n\nFollowup on Id4588f4529a4522e68e22ce58711cb927fa68a9d to fix the\nplace of the RabbitMQ configuration and there are not only the\nrabbitmq.conf and rabbitmq-env.conf but also advanced.config which\nalso often customized and more other files.\n\nChange-Id: Iff0b28b770753c81ac526afd1ce6f61d77fad25c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f59b49ab84b0bc2c6550b351069141ae97247c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f59b49ab84b0bc2c6550b351069141ae97247c8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
