)]}'
{"id":"openstack%2Fkayobe~909686","triplet_id":"openstack%2Fkayobe~master~I2c6befbdda7e684228065103feea7250a0ea3826","project":"openstack/kayobe","branch":"master","topic":"podman","attention_set":{"37702":{"account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"last_update":"2025-02-03 03:17:04.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}}},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-02-02 16:34:30.000000000","reason":"Change was submitted"},"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-02-02 16:34:30.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-02-02 13:13:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2025-02-02 16:34:30.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-29 17:04:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-01-30 08:56:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I2c6befbdda7e684228065103feea7250a0ea3826","subject":"Add podman support","status":"MERGED","created":"2024-02-21 15:36:16.000000000","updated":"2025-02-03 03:18:49.000000000","submitted":"2025-02-02 16:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":88,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909686-podman","meta_rev_id":"02389adabe88f5269fa0b6e2124501f4f759b892","_number":909686,"virtual_id_number":909686,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-02 16:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-02-03 03:17:04.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":1,"date":"2025-02-02 13:13:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-29 17:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-01-30 08:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-30 13:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-01-30 08:56:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-02-03 03:17:04.000000000","post_submit":true,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 15:36:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:45:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-04 13:30:42.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2024-04-30 11:17:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-05-28 09:38:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-12-26 18:51:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-01-16 12:51:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-03 03:17:04.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"REVIEWER"},{"updated":"2025-02-03 03:18:49.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"CC"}],"messages":[{"id":"1d3d3f7f64759548c51b6ecba3c55493dfb5230f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 15:36:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3038e4a14080c126345ab1f9d5a9b3e5fc249bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 15:36:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: kayobe-seed-rocky9-podman\n    parent: kayobe-seed-base\n    nodeset: kayobe-rocky9\n  ...","accounts_in_message":[],"_revision_number":1},{"id":"fb75f174ce25b6aacbbac2803991a44b18e768e0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 15:44:08.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d24dd3ce9d820c761271f7a0a3d53f9f1a5cd2ca","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 15:45:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"529b00e98ea3b3682fafbd106efeb5f90a007945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 15:48:53.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"477f03e18d359b8a27dade68d3d2a88f7f4f440a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 15:49:28.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: kayobe-seed-rocky9-podman\n    parent: kayobe-seed-base\n    nodeset: kayobe-rocky9\n  ...","accounts_in_message":[],"_revision_number":2},{"id":"dc9df2a975fcaef2671f620da24328f2eeb4da57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 15:55:16.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9be1b4d7401003de810307907da99659977cc4d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:15:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d43fc2d747436cac081168a476e318\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/722f4029f1cf49c68b2300d4643e20d2 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b3e57cd2874449b60da78068fed7df : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6094a14f06ac4c01b349922a2ea9394b : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8763237db2974cdba83aca349915bf08 : FAILURE in 11m 43s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6267c8504c914d748fb5d79412052fac : FAILURE in 4m 45s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5fd4fa29d07647518fbb5be5210e2519 : FAILURE in 13m 17s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b1e54ea0d1e4536a369dcffa53a8702 : FAILURE in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2546b2a9217b4d489409b54db4374067 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43f559d787eb40aeb1557a3578cf1871 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c838827989448bfa076ef07cbda2f6d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"843ef1375ebafd70ec865076b60181fe113bf5fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 16:26:21.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"25382dcbb06b7925cc50f4a24331df743cdfa0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:04:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e96dabb1e3940aaac347ef5b5f3b64c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d535a75c2bf340b78e23a3cc6757ec52 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa3eec62e424c2280b398590d9aa6f8 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8de9d9d1d2814872bc5d474dd22e7645 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16ee96f57b54489e989609869d63b933 : FAILURE in 10m 28s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e5eca7fe150431a81e84c9920870a1c : FAILURE in 5m 00s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a39fc6122ba74005880494c39065e697 : FAILURE in 19m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8d966f9cb55848888110bf23b7692a26 : FAILURE in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9388fae6aabc4fba8e35bc0d04b1b544 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b3f8eca0452422a9df87b34032c30ce : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c9c0fcb5e924449933f28adee17d8b1 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"af6cd96503a6f76215bf3c591fed2434e72d6a03","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 17:06:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93e2d596594bdcde3d82a44bf6b383ba54cf9b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-21 18:28:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"29dd61066f9f23ea2f3d453a8f91226708be8faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:53:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4f9a761d71144d5819983f197003700\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82bc0a44e89f4552a2cec56179bc924b : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfc6ae995d34a73ab2130a850fb2ac7 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f0f21bb92d948c3bd5fb599771e4933 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01edfb663d72431bbc8d9d0e4266850b : FAILURE in 12m 04s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ceedcf5e57c441eada4605c099a22e3 : FAILURE in 5m 32s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/eec4968335654480aa493b5a96cb5dc2 : FAILURE in 17m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/383c1c5fbf7c4778afd6fae1856468d0 : FAILURE in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b1560f2f39453ca6eee3106afa490f : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/492c188fd89247648420c2768787e827 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c9c9ddfcf284990b88603e0469782be : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"10d8f9078b0f3c40a238cf7c3400a292659a89f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 10:49:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"237ac3158ab32acc70ace157fb6d77c3cee4e397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 10:51:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6c7ceef0140eb522e7f03117e957ff4d6baaa934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:16:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0651bc5e30d047f58c16564095c0527e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60cb3b4b028046e38b3438fcc13dbb01 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d966cd91ef140fbbeee6b6cad989c01 : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcc40dc805be4914bec5eadefc01ba1e : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe8056c23518441695b62562d135f8ae : FAILURE in 11m 54s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5dadcdfdae214a88a7197bbd0150971b : FAILURE in 4m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d908c99c62d5413590bd0ddd48003d09 : FAILURE in 18m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0f9ddcb5cc6437d84780074a96faa4c : FAILURE in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7abb514228694126aff4180bc5d781aa : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f3b2b0c6d6145f8a05a7b8354ff38b9 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/671ff3eb047e427e83a6bf38d45d3482 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"baaeb8876e2c8fb09233baa702c4e4fd1cd64810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 12:41:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a48c1df4707db2ff391dcbe902a660b8d0cb1979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 12:54:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c9510eace1c54e1afa72b6e1b40976d1287c04db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:46:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57ce6df47bbd4e9d9ef34604721a7d18\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d12e0216f4043469be90f8deda016bd : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0e8d20fc73404e9eac73fdf7d4abea : SUCCESS in 9m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb64ca33c895491f8069bfdd434671ec : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8457306d58a4a8493792e483fbb7d84 : FAILURE in 14m 39s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c8a27fc9a084a929f7a00c989180eae : SUCCESS in 10m 56s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/eadbe5a93c234476a6a8463ff70a1e67 : FAILURE in 36m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2dfb549ac3e8478da0bc5abeac0cd57d : FAILURE in 45m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbff8bd242748278da4a4f0b6879e0c : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a0a36c3ab28499c9f5416b5d2162ad9 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0edc0034eb1449c893748ed93537cf76 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":9},{"id":"73fb12a1ef4445d0154700a5949eb28027d77257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 14:50:45.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"71f7a7a68ff77478ced308329a6300fb39c55566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 15:42:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71802fa583f7440fb9b9c1d745e31649\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/722b127234944bf5828fa8353f356a0f : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86dba113eeee4ff8ac04c6be55bc2f17 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84b855be2ca84c0985a07f904752fe17 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62bc4f98b6ff4abbbad30f14482cac67 : FAILURE in 10m 54s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/513585b52ef34403a2d246905293d3c1 : SUCCESS in 9m 14s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e16a46561aa9429ea897ae8241779a6c : FAILURE in 30m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/28ed8102b8304c37b428ff56808a8084 : SUCCESS in 51m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b950cb5da514300bc1af3f9d072c053 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c35e2d185b443c1b39bca55b2c28bfc : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b393e77777fa4a5293cc21a57822fb60 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":10},{"id":"bcc9589ac3b0908ef93e0ed2b1dc76adf5c4be6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 17:25:44.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ea8f7880840ab471467ed174f6dd8d40000a2ca1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 17:39:20.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"986b576c2dc3a8f4d60c4813dcc01dd1faba53f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 18:09:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"50d6a0e1d94d58474e36b5eaa1ab7b0aa31d819d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:58:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ef0e482227246da80fb7342a8d4f9fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c3d2b9f4ee74174a93b869b83a1e919 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8342a80cd2e345d08fb0274aee79af66 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca069f0cb9df492dbe940f77bb986186 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62a9a979c8d54f8386fcfb9c7a1eda73 : FAILURE in 9m 54s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5be3d40d64f44669f5a95cc0e7a083c : SUCCESS in 7m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/305fead3784f45058c0054781b7320a8 : FAILURE in 32m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d7acc1e5705841e0b76f51f2219d7eed : FAILURE in 42m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2e3fbe1058b43c4a5eb548675823eff : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82e4c71164c44d038db42c77b9681990 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25c50f4cfcaf4583a62973e3f4b81a75 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":12},{"id":"f8544b6537f887d3fd001309162e14ae6807f30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-23 19:07:42.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"341ef78b8d60ae8f82d2e86dc89b10a2316f90b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 20:09:04.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e5ecf56f38b47658048d10d1d543add\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2de1943c0604301aa19d789e9fb8f2e : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14f125cc28a438f97b365fbda9ca6c3 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ce84a67a73e4308bf81063e6bcd3121 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55599de0db82493ba343c5b6cc55c436 : FAILURE in 7m 59s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eda0ad01125745bdab79373ed86d4858 : SUCCESS in 9m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6db4f4d5b23c49e589aa6001b8d5abf2 : FAILURE in 50m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0988ad506811473690becd71b2f64b80 : SUCCESS in 53m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e105dbca15a485e8ac4d1a8f37e5d73 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c2516091f9a4fc98d27e08ac01208d1 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aee9950167c24354b5b6cd58533a78d1 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":13},{"id":"fadc1d13b1b077196cbab89f2c2dfd9340f4806b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-26 10:47:45.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"320bd26adf808418ef8d45c9066397e9e9bbc4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:40:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f58927d2dd4e619a4aa1323a8a102a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/643284675b904b3abf65d0d91924c5d7 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c238b2d59dc488f8c597bc3de87049c : SUCCESS in 10m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2ed848122e54fcea0e19f0e130d0786 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0ba1d40809b4da68495986f93f2c573 : FAILURE in 13m 10s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e35ebc247cd4f98b5e2e626c7c48fc6 : SUCCESS in 10m 06s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7ebd5e1472db42029a346c4679155941 : FAILURE in 33m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7f3301f9086e4524ae1fafb32b2d76d9 : FAILURE in 47m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a32d1c3f7ac482cbb2c5da9d91e91a8 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffd104fc026548c48f9512561867b70c : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f1aac7205e54f7dbfc9d10928b66877 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":14},{"id":"81be8bfe9bfb22d1aeca6b7f9270ef276a605861","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-26 16:27:31.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d3a2cad805ede91e1388c0bd2ef5e79eaa71ddce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 17:29:14.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e416b919a81441e84fa148401f6603d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c765976b29674f64a40f3c31185f5bb8 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73e8ee19585488a859604ddde6e66ad : SUCCESS in 8m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/157b5bff863f4006880071c78f387e32 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36be106c54884f92910d4da9666de3c2 : FAILURE in 11m 34s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/225277f4925b48958bce7e3c52e16d7c : SUCCESS in 9m 43s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/523c56d0d7ac41e182e7b40258c744a9 : FAILURE in 37m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c90acabb13764969bb3feded44e36b2b : FAILURE in 53m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00950038c93e4df68f3dbbeb3997cfaf : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c94d469e85194d3f856cbe025c266b5b : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f66e5aff4a243378f745548a65122d0 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":14},{"id":"19dea918e05235d187fdd76ddba97494337f9eff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-27 10:29:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8bee3b80353faa6081c62544896ddc7a09d0f05e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-27 10:34:49.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3a9286b90dbc62e475b197294e54a385454d284f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:31:28.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d949de3e1d4baf870b05db8b4a1fef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/911af31c78184a8fa9484cc52fe99470 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86672127916246bdb80c53834be0b708 : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4d2faaa325f46dbb8243988fc0edb9f : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4256e3c8d9b446f191019fbac7a58156 : FAILURE in 8m 41s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f318bcd8cddb47ef963d453ecda12457 : SUCCESS in 10m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9c25e21856c44635b59cb29053fea77b : FAILURE in 31m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/701dde12f94b4947ab18456615b2354e : FAILURE in 52m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/712b1d2052584f9f83f08af2c7c5b0fb : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a08de96c545046b49d80cc84a9123a65 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e8fd38c7aef402fb4d9f3000a921e4b : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":14},{"id":"1fc9a8e530db43623bda6e0927b2759f1f3a483f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-27 17:12:52.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"83e491c90772f572e13477c830c2ad9f58900e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:16:41.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bad7954a505145dbb9cce94c71eaba82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcd8ac020ba3491c966089c6b999b006 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc4492f3ead47589e676c58441a0ceb : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c497c35fa4447bea9de2648bc7d0b75 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14f9e243ee0e4777bcbd5054c192dab7 : FAILURE in 11m 22s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e06d173dcc040069421976c6e8b2df2 : SUCCESS in 12m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a629383799fd43fc90fd628ed4d48507 : FAILURE in 30m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8e959c1eb1154b668bbfe012937512c2 : FAILURE in 57m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c19dfa9b66c42e684d3b956fe5b52c5 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d945f9867cf41f5b40a557cfe9a0fae : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/690ee90313aa42cabcca9911e8e53646 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":14},{"id":"4b1896dbdfa949f41e04452194f0503430229667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-28 13:23:08.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"04ba122f540d6449576d98f6b749d7b4dd7d4a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:08:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c0d807f7137440b9e8a89ab96ccb6d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9969d3276e4b48df9364195dadd65d21 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7680f6e79145d282ee8b97aa79ef91 : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b44151d3401449fad739f9a0a2187b3 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bdf1712491844768da33ffd9919b3fd : FAILURE in 9m 47s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2efa21a865741f0ab79294161a36af1 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4b1ae8c0ffea49619741381481e3b8bf : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e4b889b1ffc4706b9c39f6a2d7a4efc : SUCCESS in 1h 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/394fa330230b49f3a67f6763d93ae968 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a338d234b904430bb0276410ce9c4505 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83e05cc586ca4e6d8208f4ef51dd5000 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":15},{"id":"7a44f191577531b19228ad5f6bb9622c695ad47f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 19:09:58.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"74bf4f86ab936ece1c0c81b41688dad9d3af8333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 19:11:17.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: kayobe-seed-images-rocky9-podman\n    parent: kayobe-seed-images-base\n    nodeset: kayobe-rocky9\n  ...","accounts_in_message":[],"_revision_number":16},{"id":"b4fa8f1261e60b757f9f43d4ce639a12c689ed07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 19:15:15.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"ed80b05743e757fe1c22d328a1fce612b08471e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 19:15:59.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job:\n    name: kayobe-seed-images-rocky9-podman\n    parent: kayobe-seed-images-base\n    nodeset: kayobe-rocky9\n  ...","accounts_in_message":[],"_revision_number":17},{"id":"a1a512945667501c2dc5f59834d6773774c651aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 19:22:09.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"93a40175f287bd185a61f9447135a573a557aed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 19:29:33.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"1ee4edb045e542e83ffe5179f20d2eafab213c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:31:59.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/793f535f8c6349e281b2c53e9650b4da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/741043411f684ad48918d4aa5fa4f57a : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcef371dca34005a4e50a0af008201b : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882e38ff28f84839ab1d53a4ca87f391 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae9378b8b49d432ab91bc19076986fcc : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b643c0ce2f3342bba5c8085aaca902cd : FAILURE in 8m 15s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9969fdd217d4a74b497e431c7efeee0 : SUCCESS in 10m 21s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a58e813172f34ba3bbf54db35ccde496 : SUCCESS in 57m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/72519dc189d94028a3c4b518d2da6d6a : SUCCESS in 58m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75bbd1dade2b402f9647c977bade2938 : SUCCESS in 54m 57s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/1c59ebd5a3724f60bfdbb0f55f619ba3 : SUCCESS in 51m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5863cacfc1e640e488e7873abaa4403f : SUCCESS in 18m 03s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e153a80dffc043e197ec7f374d7ac147 : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7742d738d0a14b80aee20d2d2bacbe4c : SUCCESS in 15m 16s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/e584b4643ab54ddba6a6eb21ed71f487 : FAILURE in 10m 22s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a945cb31e55495a882f02766bb1193c : FAILURE in 28m 50s\n- kayobe-seed-images-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/90ebfa728c8d4b23b9ec22f20ebf8021 : FAILURE in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b81c5975af94764bdffe722827a3d50 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d99fb585d06540a191de39094df701cc : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d047c02e26940a0a48b3291a34bd5ac : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":19},{"id":"e0c12b83c60ec54ed385e04010598756544d5dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-04 09:35:40.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"276bffc54eb37c51cb8aa5d5f3579660ec9af103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 10:50:20.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/762e2101c231446ebfca3103606085b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6919564faf9e4df78561122bc9ab68bd : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ea2e2a4f7b4659a7833c4b8510d651 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38cc787349984538ba7d9f20e4b051f4 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ee6ee376208491391a76c446475b8e4 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2753e11119b47a3a186c307945c6084 : FAILURE in 12m 16s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9cbb74057fd47eaa32a6b6c8eeaf22f : SUCCESS in 11m 34s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a2e55b1319246bea6cc992629d33ac5 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d23df164114a48a496ac16491f07a828 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a2da2dc06fd495a9e49ac7b6b74e7fc : SUCCESS in 1h 01m 29s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/08d1fcd1f8274cc59ad63d65f34f7ad5 : SUCCESS in 54m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/863cd47f3cad42e6a03bdfe34054b931 : SUCCESS in 20m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3fa72a8f1ea049bea03930cc13140d35 : SUCCESS in 16m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/595659f8d93949c9b8b4b18391306c33 : SUCCESS in 12m 29s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/f6c36340679345ab9d8597d7d7aa0c24 : FAILURE in 9m 09s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/52314ae7580b43d6a5e47fb3d4de19f0 : FAILURE in 31m 12s\n- kayobe-seed-images-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/3a3f0ce69e8b428c95ce017748ee1760 : FAILURE in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d851da302d644c90827d335c02815936 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af9a1df4153d4edf860b299996274d18 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9af4d00b67534c639a1d8a452babc913 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":20},{"id":"330452580e899d8cf2d0214894590f9aed1b3ce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-04 13:23:23.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"af231a9421acfd74fe99b16ba2fe602ab0473c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 14:35:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/093d8487f93e4229a74acab18bcc09e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e850673697e947abb4026ce5e22e6d58 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3ed5cb56904523a8857d6d44f4ae9a : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5882ef057e1d48ca9935af87ecfd8ff2 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba4f41d0ba964b58ba56f5259369681d : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c9b5f25950f4fb293d8bdca16e275d9 : FAILURE in 9m 33s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f20762c8f464aceb3300fb9d5d5dd26 : SUCCESS in 8m 45s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/45e7b035df0045959a36ae6d5bb99987 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9fbb03a8da9149acaa78f0e6d6200c70 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c10e62eebb04e16b5737c1e7cf1b252 : SUCCESS in 47m 22s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/bdc7695e6c084b3dbb2ca676b3dad038 : SUCCESS in 49m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/78096767bda940b2b69e953579bcb9c3 : SUCCESS in 18m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a55fff50eef948409305955eebe46008 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/939155d59d0c4177bac993c844c907d3 : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/0c03210059d44a0caa71b998e008d2de : FAILURE in 11m 42s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe8325333452410291b212ec4b2965bd : SUCCESS in 35m 53s\n- kayobe-seed-images-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/962602260ee741c59301bc965b6d9b09 : FAILURE in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a864c315d5054a36a16594c079b17c6f : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/512bb3749e8d44168400fb3d1982f490 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a114177c98b94bcb80099b089557d584 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":21},{"id":"ff265ce82afce036bf68baf27709d79b4b1fa1b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-04 16:54:04.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"2739e2b0cfdab7b6ba57396d32e46efbcd24c586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 17:11:23.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15f8ec48cf4f4dd9afbc3fc0c3f99ebe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa8681d1d52342e38d0d88e0a9f3796c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b8bf10009b4455aa92595044814f5f : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cae0d347e5e4e83b457349a960a6773 : SUCCESS in 4m 37s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/9d2aae377e3d46e39e0a49beebc91da1 : FAILURE in 10m 58s\n- kayobe-seed-images-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/e0d948b65a294fa78806cc3abdfa9895 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":22},{"id":"81db65034b8cf9024e5cb21710ee3a5ea71d7213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-04 17:14:10.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"fd7bebad3889afd3edcad3471e6724058f8a248f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 17:55:29.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eabafa967ec64f558f568917f2ceb4b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e369f18f15c4172bebe3a43cb84eb79 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc1569ca68448a3ba589f99239774fa : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75db2d010f104c038c40ea2f2f2beb8a : SUCCESS in 3m 12s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/9198f1777b8f4631b33d636c37bc706f : SUCCESS in 14m 42s\n- kayobe-seed-images-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/a3125661d04f41e9873d1bd639bed44d : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":23},{"id":"f84eb450259c0ece2df351789efa4835dcc54007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 11:11:29.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"69892d1b0381813fd74336ea2788db27c844c98f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 11:16:35.000000000","message":"Patch Set 24:\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"98483fbf7d27d63ab5adef9bd9547ecd4e58c6cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 11:25:10.000000000","message":"Patch Set 24: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/911422 is needed.","accounts_in_message":[],"_revision_number":24},{"id":"f8d3ff028b89d01df474c76b1f464676b44cedcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 11:25:25.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"6f74a4be2c6d0d95bda7466f4cbc67e6d91aeb0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 14:22:37.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf7fd74c0d6441ba6c6f847cbf524af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0523433100834737ac9aa69934414e6e : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfeb54ca3cfd4ae18c93574772256ce1 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b90a6f75fea4940b034c52c159a851e : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7613dc6f0dd4072a5b9aba7dc8c4c91 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d9b6fd3bf1c4b2da5fc50b24d81f302 : FAILURE in 10m 49s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4dad078f1204a42b5c66ccaa35c50bf : SUCCESS in 11m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d6930854180c4afa9e1ff8d09b2f90c4 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2f31fdcad3bf43f1a9b6690850fd720e : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8b307f97990a4e7d865f69d75f548bda : SUCCESS in 59m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b542cbb8391b4f968c15f9a2bf420579 : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/d837b48b99034b7a93ab9ab07605939c : SUCCESS in 50m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/85820a1d643947709444f958cb2b56b8 : SUCCESS in 1h 15m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8ef2ed3d21344f41b514a1e02da50317 : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/793474f682594c50abea726fa35e8df5 : SUCCESS in 17m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78dcbaf731774b73aa81033c39a0dfc1 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e68e47a5fe94818b9d3618075ca7f14 : SUCCESS in 2h 33m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/356c26d8dbae4448bae222673a940ee2 : SUCCESS in 1h 59m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8af422a925644c6fa12e454746c4f5b6 : SUCCESS in 2h 49m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f71d1bce3b6482fab52efd90c3778f4 : SUCCESS in 2h 05m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8461c077921a466993a0043a24199315 : SUCCESS in 22m 02s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5abdbf194d5141438092ba14e5f06fcc : SUCCESS in 21m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00b6a288300946e2848acf44baaf0f76 : SUCCESS in 13m 50s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/aef1227b079f4b9ab70d97b29e34d061 : SUCCESS in 13m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bb11a23f88b2419db1efdc46bb5b13f9 : SUCCESS in 35m 17s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/86057f3715204bfe996e116dd96d873b : SUCCESS in 32m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/334e40a91048434b9706fa8f818b45d1 : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a821ed6e304d38a0a199eb9845566c : SUCCESS in 23m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/90c1688784224a4ab07e139d94579099 : SUCCESS in 37m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/522ad0ce400b4e7c90959989b2c2c2ee : SUCCESS in 24m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83be45c54785498c812b0774f290d67e : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cce625d49bcc48159c54882ccecd0750 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a88983c4b63f40319b1fd0d2bf903d69 : SUCCESS in 21m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/756580e5219842dd890980860246d65b : SUCCESS in 15m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfefa673306c4152be32bf56c0112db3 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d6108be00854fbc9b1153f8e71be05c : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29ae8e2595624b7aa1446adaf82c9e77 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":25},{"id":"af0f4ce5957add20d476f1a178a1bdc76c59012c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-07 15:55:11.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"9563c6b937646f3c52f7ec0a4c4f010e4d5083f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-07 18:05:45.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"c382d21679c801e9b95255b320da4a2050dac15a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-07 18:07:05.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"4c27f92a1e63cda78666d86e2da352d92a4222bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 21:00:22.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb572f783a294814a5f424527788d332\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f7bcffe1c7430982670d6c884964b3 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5324d9fbb0e48d59c83f334db28521c : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a157373021a04f6194388afbee28d4ce : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bbca2218d17745ac80bf74698f7c7c26 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a89a844eae7045bbbfa97be81accc33a : FAILURE in 11m 36s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c62a210e2f9a46ec817b0e08f9f19e59 : SUCCESS in 12m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/98076cff13cc4345b14850200f8a37cb : FAILURE in 8m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3ea0e557b26d4e3b88cebda9c25547c7 : FAILURE in 9m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/efc066f8b4374968986fb3ebefc25d64 : FAILURE in 10m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b9211da6af4413b22b449d203307ae : FAILURE in 6m 35s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/15827dd9559d47c1b4af3ca773d5a169 : FAILURE in 6m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da2538b5f0a34a1990a2e8d253f290e1 : FAILURE in 10m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3356e96ad4ff4ce2b815ddb029acb6d4 : FAILURE in 8m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2ba933e7142c41bcacd832211ef0ae16 : FAILURE in 10m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3642567ae94492ab4fe53f97cf8f401 : FAILURE in 6m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7cc07bd520b5481c861c3b12f460ddf7 : FAILURE in 2h 06m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ddffee11be4457fbc48325d893c792f : FAILURE in 1h 43m 23s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c2736cf4a32048a8ab2b4753b222170c : FAILURE in 2h 35m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68af8acff912493aa4c6a39c98a3b339 : FAILURE in 1h 56m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b36fe99125524462a5337f17db29a099 : FAILURE in 7m 51s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/23c78d0fb3cb41b3bb4e5445bf39e391 : FAILURE in 9m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df393d8b031b420ab751e2815caf2617 : FAILURE in 7m 33s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/0251a16ee8514c1cb833dff95427405c : FAILURE in 7m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a6ce6989f058419ab5a9e30c1f5a1bf0 : FAILURE in 10m 19s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/d09f2085ef3948b0abda1256a9940238 : FAILURE in 8m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4079c35205864b80be7dda15feec338e : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/961780ab06ff48cba099d0877c98a914 : SUCCESS in 43m 24s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b36846ce01f747da9a93e36f31e3c62e : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9751fbb8082c449ab85e02a5ae99fc84 : SUCCESS in 26m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/67ee7e38dbc841a1ad0f37d0f248602c : FAILURE in 8m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c6bd2625de841278ecce19659ed070a : FAILURE in 7m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/80d75aa576c74cc6ac0dd3529532efde : FAILURE in 8m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/518a42987b5e4d51a6b594364d861a52 : FAILURE in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc84477586e4af7a40a0f4427a8e1e5 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37bb59f5569246d89772db028be62a30 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a86caeeed774f5f8c88e66e97916cd1 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":28},{"id":"c281832e22caa5b54455f6913c1b3ce83ae662f8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-08 09:27:29.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"839d3b6ab80734807e93c5a67d2dd73e1bfd8c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 12:22:50.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a443490809e4bafb392f96d6d790ca2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03bacd9d64d0456e80b8cbe877b43ed1 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485707bb533546cd99aea12aae3fb2dc : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7356356de0d14bdc820f99203a0485fa : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/deb1d589b6174d4eb26a591eacb99d94 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec20423806cc41f99bd2338c63105638 : FAILURE in 13m 17s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/49e4bd7447074e268227e4bc9643fcba : SUCCESS in 10m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/47632568061f46e4bb93628625877ed6 : FAILURE in 9m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/c070a0aef13e44f4adcd87f4aba0888d : FAILURE in 9m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fb1bb86b2ce64c5ca09ace89f62531e0 : FAILURE in 8m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13887247b7a24b2eba08bf9cc4ee2a39 : FAILURE in 7m 18s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/e72776d60b5b43f3a87ce209d8dee506 : FAILURE in 7m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a935f930214742c79e5448a40345e899 : FAILURE in 10m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/385c886b804140dda48d5a600c8c570d : FAILURE in 9m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c78f1834f1ff42268ec69abffc608889 : FAILURE in 11m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/018705f27719476986054f0523de28e9 : FAILURE in 9m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/07732266d9fa49e0bbceb949a848fc9d : FAILURE in 2h 24m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a9e2a4db446432ca81ed59bbc497be6 : FAILURE in 1h 58m 42s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cfd03b6a867b4cc7adede8744b0fb7cd : FAILURE in 2h 30m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc54777808d14fed9698c670b627f838 : FAILURE in 2h 00m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f36dac02af7e4952b97c7b8ff437b624 : FAILURE in 10m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/34a4c5092f42455a951a51fdbb60cdc9 : FAILURE in 8m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fbf75854df945af8099dc22361b5a59 : FAILURE in 7m 32s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/09b37bc27481498db79ea489d8b8fb86 : FAILURE in 7m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/28551eaf1a6042bd88cae336520ab55e : FAILURE in 9m 36s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/63bd9ba5b3c644afbfc7c585ab65e9d9 : FAILURE in 9m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a921c7518654c2987757469fa8fa08a : SUCCESS in 34m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1608fc1fe1c4c6d95614495ca74fd6b : SUCCESS in 24m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6a15460009d34f07b831c86f858c206f : SUCCESS in 32m 51s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ba1a5be53f944588449b2610cc2e83a : SUCCESS in 37m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/17d9fdf7a3b1433e830660fffcf25c13 : FAILURE in 8m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39b44e0a55f34ebcaeb00bc9938c1ad8 : FAILURE in 12m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bb3713edc0a45dca741384eb2d0f4bb : FAILURE in 9m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76144afb4b0b499581b1da11f8630cc9 : FAILURE in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34e1de55e4dd46b29f5a7017db3a7392 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3890c20f34a14885a727e06eec91ca64 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a519e09c2b8f4937b256663e21651e97 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":28},{"id":"34a977b92d98e2af522dd2342a511cfe70ed4397","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 10:29:19.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"1d29fed967a9d4cd21bfe306e430aeeadf4dd3cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 10:56:39.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"c87d83059d369cc3af15b7ffa7b9e9b2f2463159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 13:24:29.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abc6b407628742bda35632564a605e12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ebbf767211e4b25b20f20984b39079a : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/190b8ebeba894627b53953339c60aa26 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da503aa169054e88b8ac22f2905d119f : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14336296c1404f0c87529a7a6e7db7cb : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/336050500f674811b7a0cb307998dbe7 : FAILURE in 12m 50s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ea72f31212746d59239f714dc51dfca : SUCCESS in 10m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cdcdcc9712a446eda7356966a8f6adc3 : FAILURE in 8m 52s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/58c3a4a7e3d14b5fb8ce210a3911534f : FAILURE in 8m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/044b31613d4b4eeca84d2013f0f32ad4 : FAILURE in 9m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f8c06b41b5044aa84c0b0180b04cdbd : FAILURE in 8m 56s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/37b51b6a6cf143d789f59dea6e3369e1 : FAILURE in 7m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8d264bc27e964e47a315e21eac516ab8 : FAILURE in 10m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/de62ea74802641f1a5ac992af7e30aef : FAILURE in 8m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1005581a89ae4546aa51594fc14f81cf : FAILURE in 9m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ace6fdc6f2b1466c9f9b8c7ced5bfd57 : FAILURE in 8m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ec86f075f4eb4f7a87ad0b6c59be5df0 : SUCCESS in 2h 02m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4147fcb58d784c649bf4057428371ae2 : SUCCESS in 2h 05m 58s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1432987c12aa45419b783c70dd59c3d2 : SUCCESS in 2h 13m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ad0369c182c4fd086ff0be20b6e228f : SUCCESS in 2h 16m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/84735e266b2f469592f7e9a8f42c3576 : FAILURE in 7m 54s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a854d7ab56054d218288817edf9b5854 : FAILURE in 7m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c920e6063842b495d723f89b81b531 : FAILURE in 10m 47s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/891c5eca1b8c42b2a16752fa7a4e601b : FAILURE in 7m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8573f14cba1f4df98803c5131bfd6b4e : FAILURE in 8m 37s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/cf883174d96c4658a8574a760ea62e5c : FAILURE in 7m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2695187f56c3453a875a30cf6458815a : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e851493fdf114b099f962149d6792b6b : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/61e3835b62ac423d9460c3167b165c8b : SUCCESS in 30m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4a4b0328bf243838aa7403df35876ff : SUCCESS in 25m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3d14beb34204e2f9612d097306571d7 : FAILURE in 7m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc57ed3091f34fc9a5966667c2877efa : FAILURE in 6m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29829126a9754ba4831c868070e060cc : FAILURE in 9m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b71ebab38314971bf3c6197d2c3bb4e : FAILURE in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6401335bf46a4a8e959cc090c3e7750f : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25263b2882f844debf69fb4cfea18b69 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10f175d4a23b4c07819f8dda5afa2ca6 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":29},{"id":"60915a5c36dccd5dfdaa1fda5da44f2ada3355d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 14:30:59.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"6eb6b2c58cefbe4c1aefc6df3f68bfbd5ead89b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:31:34.000000000","message":"Patch Set 30: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 909685,10","accounts_in_message":[],"_revision_number":30},{"id":"86704897533a6a8f1f79215fc9dd9e8454b95f09","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 18:04:35.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"fb83294d6470c9e828f68584b38b780f0352e046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 18:06:07.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"f41ac8a8a90dfe4280ab182f5d83e1f8e544c2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 20:53:36.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7ab6e6c9964480ebd200237f638899b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/619f4f32e009406e877c869d6285e086 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d49deb749fa42a59cc35eaf61ca090c : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe2d212f62474c92895f4c01559445dc : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc13086a0fc6467b8f9230f3ba5cec5d : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d26d0f543cd945dd979fe2b6ae93dd19 : FAILURE in 16m 07s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5a38705a6fb4a4baaac23da0e7856ff : SUCCESS in 9m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5ea6cf6648d2454095b8991289c71e05 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f1a45424130e4cfbb1daf239bc7c5bef : SUCCESS in 56m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9c157e59d839480faa4cba7879146433 : SUCCESS in 1h 07m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/039891c3d6e348baa39d65faaaf46c3a : SUCCESS in 58m 41s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/00248b0d76614fc98353b6173733f43e : SUCCESS in 59m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/14c2238a03f94447b86a85ec89df22f9 : SUCCESS in 1h 11m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5cbdf3dfb2a34056911896f77678fefe : SUCCESS in 19m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8909ba43f2344c75b879821f674d4438 : SUCCESS in 19m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e3af7434d2845aab64a4de054b99987 : SUCCESS in 16m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ad5ef0051324296bcc61bfcbaf23012 : SUCCESS in 2h 35m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbeb345b8c364fa5bb36aee733af40d7 : SUCCESS in 1h 55m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3c48db4b90a144f68b5108c7c11586b6 : SUCCESS in 2h 27m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/357ee3c1896c4fe986cffd7ae4346bf6 : FAILURE in 52m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3af7510831664f2099c9e1839a2a3a05 : SUCCESS in 20m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/709aba6d68204835a58d340ab5315886 : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fd1f6eb688d425783b5287db2a8c2af : SUCCESS in 15m 32s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/fb32b7894e4541cba2615d05bf934376 : SUCCESS in 12m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9b569633b20c45a2b6702a3b9e15f3b8 : FAILURE in 34m 58s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/63fbdce4ce4449049a39d22fab758664 : FAILURE in 35m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6b34f7f5a6354f38badf0cac0fa6bc1e : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e938b6ef27e34bfc9e85bc5690d49159 : SUCCESS in 20m 18s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c41fb9d681b74518b469a6c143826363 : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8568d41fe846418ebef58421b66ca841 : SUCCESS in 22m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df408c7f7c4841c29c050112ea39d6ea : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/556f1f36df554473afd4706db97d12d1 : SUCCESS in 10m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/33db3c6e7efb4af0aa0d43921bed5fe8 : SUCCESS in 15m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd5622f4f5df45358d60e1bd8d50093e : SUCCESS in 13m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/733f7fbd0583455ea78be4848b77b8ea : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26770d7f30e54aabbbcb49ce2de318bb : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c6b86d9b95b433fa4dcd645a30020bb : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":31},{"id":"a6e5c948a834f2b15de201d08eda8af13231a4ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 13:06:52.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"231a84a3ba04d73f7e8be5844949da6efcacb8d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 14:49:16.000000000","message":"Patch Set 32: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/910017 is needed.","accounts_in_message":[],"_revision_number":32},{"id":"bdeaa33cc5f63684c161c8394fdab1be7b468d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 16:47:10.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"4ea80f14b58b72a3cd9649ed2b4a01f34238defa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 16:49:59.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"d592e8349bcb8553e7a3e332d00ae41e0ebb4476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-18 17:14:41.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"80b9e9b6f10f31e1390b55391ac150938140c114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 20:23:35.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a3a297992ec4aa39ae46145eea0481e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85356eb9623d41f1b21e1f6f19f33009 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b681bf6b9446ce8672c45dd2bf3b9e : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/371e9e8cb9fd4722b6b06dbeb59a583b : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/464be5d1fb4646c2ac0163dd49717feb : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b63bc4a0c2694c5e8b39dbf78b44d306 : FAILURE in 12m 00s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d0e6bcce3cf40ec85726450ae79ebb6 : SUCCESS in 10m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0c71e07c7edb4b59a22c238b5aaa9a0b : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/66fb380fb98b435ea86483b4d9db654e : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/74087b532e8b4784bf9886786150d483 : SUCCESS in 1h 08m 16s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe8b24a18d634add8e17100ff976f7eb : SUCCESS in 55m 34s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/e9c8214d77a243b1a23c7d93dfbbbd7d : SUCCESS in 56m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b0514232ba454a47860678029d9ece6d : SUCCESS in 1h 17m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0949cded5efa4fe3a854d7675f4c1541 : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b5f166f845864533a8bd629bf0310947 : SUCCESS in 18m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91a9bffc18dc4e1191769efc6aa48def : SUCCESS in 13m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/894caa6440464d158ded4686218127ec : SUCCESS in 2h 59m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/431dd31998684cc2a3fb61112ae15a2e : SUCCESS in 1h 57m 32s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b07302b8e1264c7fb68597be94f10a56 : FAILURE in 1h 28m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63377e3ccede4f93a01e6f0704c4d556 : FAILURE in 1h 03m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a68914758bce420ca71454110b21c188 : SUCCESS in 18m 35s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7f2197baed7489280702b2bc156d3c9 : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80d89d014bb849848796ebef4c605e59 : SUCCESS in 15m 22s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/5bb93578d93146fb8e552ab62fd3a658 : FAILURE in 10m 18s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/880655de8b9942e4ba35437155035910 : FAILURE in 23m 44s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7496b71bacd4350b533796f012d5834 : SUCCESS in 42m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3c63673036a441f595744b324729f070 : SUCCESS in 28m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f33850e1fb154de39137fcbce36a4e75 : SUCCESS in 24m 59s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c766b13191264dcd8edbcf7549bc98cc : SUCCESS in 34m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f39c9aa493941f48b1e5f7ae7b82315 : SUCCESS in 25m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61218aa975184fef96fb77354ed6381a : SUCCESS in 13m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc6ada8f3f7e463985bce95a14531416 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a99e2b9f73cf46ce85d4af292c9f9246 : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be041b3dcc024bd2a0fdd8006892e8da : SUCCESS in 14m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1d855f89cae4f29b075372d30e24f5e : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d68ad13fd8b4237938ce798a7b6ec5a : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/896fe7551b0641bc9683a4b8b7e516c4 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":35},{"id":"3b0b2435cf45796f40a66027d8ddb1160de2ac89","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-27 15:22:12.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"37db5638bbbc31e5d62c7323c257cc3263d9eebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 17:56:24.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e39486af5c41a89bb9317d60cfefd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/585425ac244f4ade98ace4c75894b0a4 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6177bdc527eb4c9389e0455fcac8222e : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef6723e9fd924354b80572defe5c30c7 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d87f9edb63c9472f967d58e55ab5c876 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/172e3fe0bc354e1094e6b3b0cfdc7fb3 : FAILURE in 16m 02s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfb06850891e4f268399092ee5587e96 : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/deaf16be04a64afebe517c5f9fbe83e7 : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cf1470732b1b4282bf8e1ea14b0cdf9c : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f31b17e27600482eb5b0ba4b8c1732d2 : SUCCESS in 1h 07m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/183650992a1b4a868a2b810633e5f23c : SUCCESS in 50m 20s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/cd67771d684146d7a55087211da8416e : SUCCESS in 54m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3738854bfacb4797b341c361f5886400 : SUCCESS in 1h 21m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/831f886d476b48409d2bd4e41c73635b : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/df6272308d9949348951cfeea9d23f1b : SUCCESS in 20m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bb2f5a2b454c0e88f595b522cb36ce : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9bb64accdfbc440a9de1c7845cfd80f8 : SUCCESS in 2h 31m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20574543ee7648b19ee30f7af12b8253 : SUCCESS in 2h 08m 38s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ce477dc72dda4c958e5635d13a31b38f : FAILURE in 1h 28m 30s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f54ffd6566a34a8c804041823f9c6b85 : FAILURE in 1h 12m 16s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/95ed240c50654a9795283d8dbf1d45c3 : SUCCESS in 18m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ca7379073c97489f934d899ea29e1ae8 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f242a541e6b4490693224c607d2d14f3 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/8248080dddba4406abe6a830042ceafa : FAILURE in 10m 41s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bcb8ddd7005b459aaf464288e55cfbbf : SUCCESS in 33m 02s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/82455dca727d4eea9c0349c8391e47c5 : SUCCESS in 36m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53d461762757483280f052a610c9f696 : SUCCESS in 28m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50aef934906046e28027628edfff7978 : SUCCESS in 23m 41s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/56c5e0d17cc24d43bf24a224a0a0202b : SUCCESS in 31m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22885be2c1964343bc92d520dd705e75 : SUCCESS in 26m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ece05c8d45114d25bbedf01352bc3e15 : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d580e83a852a464098566d22ea105cfe : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c3732e7b14dc41c592fe33ae00f2591b : SUCCESS in 20m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/420fbff1f4bb455db3e5f2403996c335 : SUCCESS in 17m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7ca103197c74ffeb6b839ae778b8b11 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6bfbf731fdf4e2bbeb53fa59c4021a7 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f22e652a4bbb430896495c146a2bde67 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":35},{"id":"033488c70523992a391a57398dfc561636a203f9","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-04-04 13:30:42.000000000","message":"Patch Set 35:\n\n(8 comments)","accounts_in_message":[],"_revision_number":35},{"id":"fcf35166a4cc86b9cf1d918236848cbdf9922cb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-04-30 11:17:46.000000000","message":"Patch Set 35:\n\n(8 comments)","accounts_in_message":[],"_revision_number":35},{"id":"0ad372737edfd94b642ab0decd9e5ec32934e5eb","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 11:22:26.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":35},{"id":"f1ebfe5f1a8c5ac3d2a23e6b03af993deee211d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 11:22:31.000000000","message":"Patch Set 36: Published edit on patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"705cdd67986885d21c993de13b6df36ab9c43499","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 11:22:40.000000000","message":"Patch Set 36:\n\n(12 comments)","accounts_in_message":[],"_revision_number":36},{"id":"d6c8de4eeef709674cf909027d1a6fb58ba2c35a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 11:35:34.000000000","message":"Patch Set 36:\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"ea32623bb3c586efb960f35d9e80fabd484f7bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 13:56:25.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/192a4f280a9a4ab89106b39a64b97296\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2440a1e0e347839e2762c48a6d06c6 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/549d3015bf1f46bca6454ce6da65d713 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dcbb68501214b87a4a55fb1494d051a : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/429649a0873c4a049f3bcf60887ee8d5 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/348686a781594e43be6108aa0cdd2c7b : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64abe4e32af3477aa2df0f6baadfb5ce : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/95edd64b52794aae90bc097668ead61a : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0fa1f023e30c467ead3568bd6b45a148 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9e45519437ce4e49b394cedb206c1938 : SUCCESS in 1h 03m 34s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa27d0af8420498395b9ee8a9fc3bc54 : FAILURE in 32m 48s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/275b44dfcde8462894cefd9e947ffb0a : SUCCESS in 56m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/27c5208a72e747ac869d3e5dc3a9494e : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/15f89e6589d945c1b7262f1b3a11c6f2 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/28efd9d64ce6461dbcb5efaf2d15b078 : SUCCESS in 16m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c0ca24c3aae425a890342b17806b1ad : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f3e1cc2b0c194110a765fd4608cbd782 : SUCCESS in 2h 28m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de425bbbe98d4d919567e91683b30c65 : SUCCESS in 1h 31m 41s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f3d37e1ec7994e86b5d3c0f30334ba96 : FAILURE in 1h 11m 53s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7204e182f0664ceeb4e7c41474933a21 : FAILURE in 55m 01s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8013ace1689c4ccca882a903cf8d2b90 : SUCCESS in 14m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d2f19e69b34545b2bc50d978f80e5754 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01cd05c8a5994ea0aa09bc0bcffae290 : SUCCESS in 14m 38s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/dedf9acc76834ac2b1b7de16e68e6d4f : SUCCESS in 12m 15s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/651966b314e54fffbdf0d50ce743c4d3 : SUCCESS in 38m 15s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/fcae47fc6d66401b96392301aee269cc : SUCCESS in 40m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/38a50da3d1e64a4f8f4553cf4cb33d0b : SUCCESS in 29m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f4d1ebd2854635843ffa705ad68f8a : SUCCESS in 27m 52s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4eef82f78776494a9f9a844f43bef917 : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6caa365f5258427ba629d7cb5922a88c : SUCCESS in 29m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15bf6a1399a44a4387deb9ab728ca1a8 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b74b7dbd06d84b508d1c68e3b788e2fe : SUCCESS in 9m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8f235e9cbef4899a4912c0e8b4ae78f : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25bf492caf1d48fe9773152f38817f85 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a890afa295b48c9a3db2902aab11a88 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22d9257932b14401a18649d38e7d0043 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8538a3627bdb4063bdac20d8f7f3520a : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":36},{"id":"dedccb82cc78f4ce8613358bccb8051e6c8011df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 17:06:53.000000000","message":"Patch Set 37: Published edit on patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"d21740806f3f3d55a4432c296a330a62e6704ccc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 17:06:59.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"c9d7e77ef6b885008d09173da5956da5a9714c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 19:54:20.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49de0ae184eb47219007a8659796b6aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ff47e77d7c4ff49c5a9acd724a0fb9 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95527608d89848c0a7d2e9b739eed0b7 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3041534270c747c6b8b4ef37d8771400 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3dc0bb7503d34b3397a3d110ab242f7b : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e772a7414ae146e2a09be76df0823399 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ab1df11d8c8403b9643b359d025592d : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/359227071c7d46038676af2d8b0b1134 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/724e74634eb84ba68da2cb9826028d91 : SUCCESS in 58m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8a6ed9b37c8743ec8fbee01bd1dde078 : SUCCESS in 1h 06m 40s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9651e584a4a44f7fb6c7b8e9bf7de8d7 : SUCCESS in 55m 10s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/f6b3f602c69743e2b3a3f25e9d153edf : SUCCESS in 46m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/22bcae8720274415913b18aef76e64f0 : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/56c31becc969492a8c53a0ac5427e5b8 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/be5921aafde44364aa4112b0ee50f87b : SUCCESS in 16m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd0fd0e039244368b476d2a0436a393c : SUCCESS in 16m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c15fd84102348f1aba6909c8a719949 : SUCCESS in 2h 41m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba79abd21cac4ecb9889a26bf94455e0 : SUCCESS in 1h 59m 33s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/834a195cd0d04dc68eb4662dd4e95936 : FAILURE in 1h 28m 57s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d7c678bf9b244bda6c8d76c1d22c0f5 : FAILURE in 55m 09s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/089ad593fd1f44adb4252e971f67f7cc : SUCCESS in 17m 39s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fb85aafc01644e69b5157ee72dbeccf4 : SUCCESS in 18m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/083edc7868a246d7b34bff268e04a37a : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/18696ecb0e6f4d09a64db8a3e9d2a048 : SUCCESS in 14m 11s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2e292f287af645e3922dfdfecbc6d21b : SUCCESS in 35m 32s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/db15d15bc6074666ba15bb4e36dcdd9b : SUCCESS in 38m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6324eb2f511846ebbda407ee8c467495 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b65e96b76c62453b866e2b0168ffb47d : SUCCESS in 22m 14s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/46d7a6a36e204f128693b0b531789c29 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42fc07a47fa44f78b9952ae771e3ac46 : SUCCESS in 24m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ad2734c52514760b36c2aec83d561b5 : SUCCESS in 12m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a3693792d94583a29a75ee8ebf3ea9 : SUCCESS in 11m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/073234f03bc5487bae965ae6283a1a87 : SUCCESS in 19m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9edd3fe9fba54917b315beb7b57a1ba5 : SUCCESS in 14m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bc3e95074d64c1ba8bdc5ce4b69314c : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18e87338c39b4cd9889bc8391414ae68 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e33b9bfdf4b48b8a57f16106c61f212 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":37},{"id":"b77d04ca8f129808b30b3896bf315a9da7d41650","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-02 08:28:53.000000000","message":"Patch Set 38: Published edit on patch set 37.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"c1619d8d19b1ddefabe9b90511e18f5ca6484c00","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-02 08:33:53.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"870257c821214c8ca4aeb0052902299f2a76c364","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-02 08:47:00.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"871a8245aae1a4d18b5b1c844cf2123a36096278","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-02 08:59:22.000000000","message":"Patch Set 39:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":39},{"id":"55f20a6543e4ffb315016b25fff4bfcd63a83776","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-02 09:00:43.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"c76604a0bad09acd5ee8ae8a438da88c1b079c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:36:06.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffbdf45793d4ca18439332ea04da12c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19db556bf9a740b493829a685d22629c : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c4fdcc24be4ece88bff3733e243578 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4321a062df74126ab51ebe255c9603f : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/725ef499360b474c86459899d7eafbc7 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84b70bc5a476495281c368a16c181750 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f57ad4c6176344abbcd3669fd890f85d : SUCCESS in 9m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8a28af3dec6b4c0aa1ddae8ec37edf84 : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0a72441f4e7a4aa08898d0fea919ce65 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3c6303b4a624a798a50d00202e469f2 : SUCCESS in 1h 02m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47608cce0a8940539f3ee3802c27ab21 : SUCCESS in 50m 40s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/b3c3695b97634e1ea9c1efd018b90562 : SUCCESS in 53m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ab075b73cdc14333b4346ecd59c77333 : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f2b9fb51a1c408ba06c045f64fc346f : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/374c1b2bf45446d5978906600d611c58 : SUCCESS in 14m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc1174ac0fe490dacc3fa52f908ae8c : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f236ceca16be40abaed6977db79a62b3 : SUCCESS in 2h 37m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69032de126204253a0157bd190a72ff0 : SUCCESS in 1h 57m 07s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9acb068dc1c84d2cb9595321679e9568 : FAILURE in 1h 22m 01s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24774d028e3d4a4fac4977a394ce8c60 : FAILURE in 54m 21s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/594b2e420c5d444b93e88fd73cd5efd3 : SUCCESS in 18m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/191e6ac601f44b7aac20cf4cf7d3781d : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/026bba3a4e054c2b9cf3c4fad0ce51f1 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/d3c574253ee34696bc446463e69fe01d : SUCCESS in 13m 20s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/65d3988f897a4d99b716badd6277e647 : SUCCESS in 36m 26s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/540d799e6c424dd0a2e96d785296af9b : SUCCESS in 41m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bfbf10b3fbc2438aa2b33928591d5b4f : SUCCESS in 32m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a5731ea7d30416a9614f15c205ee339 : SUCCESS in 25m 45s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6351385c48224e038ce5fc5f7d3c55c4 : SUCCESS in 30m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39b15ad2d8d04efba1bc1ac5a8b27a81 : SUCCESS in 25m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5bfa306d68574824a4adb5932225eccc : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/201dfb37c2544242893b27b4bf8bd783 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51d9386704ca4e9d8d9f55adb4d52567 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0b140eefd5f4a5e9afd50456b21120d : SUCCESS in 15m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04cc34d7c084fceb1ac3b2e4b4c5edc : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acb4990449174b5aa3e1157082444ce0 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf5a56be7e4e4f4c9331e81c9814c0a4 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":39},{"id":"5149ca8eb897aced86e9ba344223811e964b798d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 08:34:26.000000000","message":"Patch Set 39: Review-Priority+1","accounts_in_message":[],"_revision_number":39},{"id":"61d8db8305a909cf46bbe648dbb842a90c901281","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-28 09:38:28.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"a858bce21189e7b0826529739695f01a6617bdc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 10:41:02.000000000","message":"Uploaded patch set 40.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"64545f3ab13826c5c309032c41d5671fcf251be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 10:48:09.000000000","message":"Uploaded patch set 41: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":41},{"id":"839c75eb5dc49496f4e6f1fef026e971a0edb5fc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 13:22:18.000000000","message":"Patch Set 41: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"68cfe64771c40f117c41ee59824ebcdeb29b1134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 13:33:40.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6488bdca47e04f41a9211d926b9692d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/348d169c93754a63b6499a52ce97200f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76cf11b370c94a23af224b68979a2fa7 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fb126cea7944b5dbcbcde838c20025b : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51d47a5e910c4f858b7d6512ebc43e7a : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bbd209baf9040aabaed9b2f5ddd9e44 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61d24f656a1b418da81d5154e34e4258 : SUCCESS in 7m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6cef3b3c37f34bbfb606f94d06b9a686 : SUCCESS in 57m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b75f02ce8e1e41b4b58bb380559333af : FAILURE in 30m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b90be427595041caae32fc9fd1c49610 : SUCCESS in 51m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f48ecf7bc7944a529667a2c4f066b70d : SUCCESS in 47m 43s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/8e9a61227aa744f9b918994c20e579b1 : FAILURE in 22m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fba5593231654676b6ecbbe2177ece2d : SUCCESS in 1h 17m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5d61336371874b85bbdba1d6d397c55d : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/84e9f40afbcb47359f341922d9b37747 : SUCCESS in 15m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd30617fcbf841e9846d4d5edbeceb92 : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cf6f8be8b710481189db223373ab917a : SUCCESS in 2h 34m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4738ca4eed446088ade34822caf2a34 : SUCCESS in 1h 51m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6cf4cf0cba224fc5a242d6a870c93674 : FAILURE in 1h 16m 39s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e430aca74d7642259ddf7e2c66a285f9 : FAILURE in 1h 02m 18s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ccad816a8a8b4128958616167474722f : SUCCESS in 16m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0177be857107403a8f8b59a734959d16 : SUCCESS in 15m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec52f35bb7c048c4841d4963985d360b : SUCCESS in 13m 10s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/facfeaf375874cc197a7175470b19582 : SUCCESS in 15m 37s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/29ea63b68f1d4e4ba912ddc9f6154b8c : FAILURE in 38m 15s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed4653778102425ba81f064114cedb71 : SUCCESS in 31m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fb4fbe2ee3eb445792c8eb2ba111ea84 : SUCCESS in 28m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d813d86e34964c1b98064d6310dc7e72 : SUCCESS in 25m 21s (non-voting)\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/945cc6436b9b4f448a10af5c44ba34e3 : SUCCESS in 32m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0e8727d25294dbe9a0da06a040a0641 : SUCCESS in 24m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c34b7cdb4064281a61f63d62f48f6d5 : SUCCESS in 12m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078686e778834846afa48cbf72870da8 : SUCCESS in 11m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/49e57d9ef6ff45ebb9827e8a183898d6 : SUCCESS in 19m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfa44b9c6d7743b69952d51c31e4a649 : SUCCESS in 11m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b45265d0bd42359147ffe847203d2a : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb3104e38b2d4bbb813b2d3890cf5300 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/287cc74628c4425b90da782f69d80e42 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":41},{"id":"a1f7fa763c16e6ec90285385510cfb0c4e479a33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-20 11:46:58.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"ceb5f467a39a5e189dd96b73bd1370ac995e855b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-20 11:56:03.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"359326f4f8385725acc220e8bc7353c419ddf34c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 13:56:26.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ca6deb1de5047cdbc89db0768fea282\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/399cd6d3469944d7bc9b8452c414d5f7 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c41794f19554989ae0c333f2b197339 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037c3dd33afb450889bbbe89b8a5374d : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/93995e9045fb45d98c7553332fb78b85 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a97e7678fba431bac710256ebf7aaa3 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77d38b1cf0774a43b2579993a0d31110 : SUCCESS in 8m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5ca1d8f03cf47d8abe6079c62de881a : SUCCESS in 1h 26m 23s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5834e49ac433473a9dc4e527941eb5bc : FAILURE in 31m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ef8f1a9be2d641ab8caa624dcd6e76c2 : SUCCESS in 1h 00m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f7c867cd1294b65bc4033bdf2f5d8cb : SUCCESS in 51m 41s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/8b5e1770197c474283465d5ad770d6de : FAILURE in 23m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e619bcd8059d4159a84a88b32d6a2ed0 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/699c397cbb184b02ae3274fb0e6696a8 : SUCCESS in 25m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/936496d266164b66b1d7862f1c5287d8 : SUCCESS in 17m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2158361aa98044aa8f264719961649dd : SUCCESS in 15m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6cea51c329d344f3969f22a7d260d847 : SUCCESS in 1h 48m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e59139e9eed47a0b9ad28d207f773a3 : SUCCESS in 1h 43m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/37803af123674be5ba2ec0ffbd80597a : SUCCESS in 19m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/72b887528f664a3686dd535c802d8825 : SUCCESS in 16m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4033b87a91ea4442820bd20648c7f8b4 : SUCCESS in 14m 40s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/737b75b977b94ec29877793ad947f384 : SUCCESS in 13m 27s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dfaecdee362e4ed99153d2e8d597837c : SUCCESS in 36m 40s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/60f84119573f402f82f258dfb66d25cf : SUCCESS in 1h 03m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b19901ef524a47a195f58e880c580f36 : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18660b450cad45fa89023ed573e840a7 : SUCCESS in 21m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c5f1847c8704c6297a85df91170a49d : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf9326b85d0947f89e76128b530057bc : SUCCESS in 11m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/041089271ccf41148233739808cb8749 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a172c4508ab4b17aeb0dd6b138068ea : SUCCESS in 13m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5360414cfef43138b3a4cf51776a28e : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5b1590cd0204009a772c405c4802b1d : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71342f296f224f5ea3efa9e9fc546835 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":43},{"id":"dc3a2d4b7072c8cb8d713a583f4e5f6ce268de61","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-07-03 13:01:35.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"259c4a9b2e0b00d282671d65dc3f00b16aad4af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 17:23:17.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9463f52bd7c4a70b7cf31518c00992d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d4948b4a4043b5af0e71d7d6b6e864 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b0aada458d42ffaea2f86af0831f06 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afae274c790848629531fa7be64ac200 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/988bd6a57f574434bf629f13a1b82e0b : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab21eb85888d495298e44c56aed569b6 : SUCCESS in 12m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9418e8d4e3cd444e9171b18867e36004 : SUCCESS in 9m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69ea37fdcea74d18b74795eb735cde63 : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/95afa6f56b5040b9ae1c80bd18f17b45 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c9492a0c24f742618128e53bed40a477 : SUCCESS in 1h 05m 04s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdacfaa611364ac5a946bbf2f1a13e7f : SUCCESS in 46m 35s\n- kayobe-overcloud-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/e73dfc89fd964e6ca3e3ad774db7ca42 : SUCCESS in 52m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd5efd470c9540ed941d3f3d24a7f482 : SUCCESS in 1h 16m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/11376a571c3d47c09ce73473fd992a0e : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/79771d12d5fa4672a4bdeb9a59a6cf75 : SUCCESS in 14m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fdcff3bd9334bf583aafe98c4a52060 : SUCCESS in 19m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5db7bf42f82248c19874284bb3633bbc : SUCCESS in 2h 01m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea10e3ad8e5a43c2b7ce0edd4ac65b87 : SUCCESS in 1h 35m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5e54424dbf04022957a1abec2f78e6c : SUCCESS in 18m 13s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/00df86071c2a421abdd4d378c74beae7 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2bdf86879f146a587885d439e04041f : SUCCESS in 13m 51s\n- kayobe-seed-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/37228fe5e6a8480baf059bc12bc69e80 : SUCCESS in 12m 24s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/443b568c065c4bfda0393af5c701e767 : SUCCESS in 42m 47s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/9d905d288df64fe8aaf1083092989b9c : SUCCESS in 1h 04m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6be798e4d8a443b2adae9b23255d7afb : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65dfe38632ef48d3b75baa009d7e9a4b : SUCCESS in 25m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b116bc3f73f4a22af2f9d428a162cd2 : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edf788c229aa4651a28620e87651d807 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b3dbc28a9a444e192d9e4f8e2ab0336 : SUCCESS in 21m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fe8d9834475465ebb1b9509ae985ab9 : SUCCESS in 16m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d19c118918854105a60865d8ab26b7a5 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c12236c6d534b52be3fb5b53f98b621 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16bcb3ea29ea482ba798356c2f453e90 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":43},{"id":"c38fb4a4fa2b970bfa6597fcd1d7101d40a72f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 10:29:27.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"91b074374341b549ee4ad5b488ba262285131392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 12:05:45.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1994f57816a24b9485986fb7f76416ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90cef1ec05c84bedbb2fdf477d00e6d8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d2e3ae44df4731a279df1cdd170c63 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54b3865fa83f4930a35fa41930f62c29 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3891ff6c4cbc4fca9ce43a2acb9995ad : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0a36d4fc3743a49da9b4c7a072549a : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29a88f94a46c400e942630c398589b16 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f828f3600ba427a9264c8f634f0d54c : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9b0e2c9e74f4753a3120bd80d869e5c : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/039dbed6e83a4b07aea2d5ffe322dc9e : SUCCESS in 9m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/112e76aaf5d3403f9e31f4501818a81a : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0eeeda9bc4e74cc5bf838480e9efac74 : SUCCESS in 52m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/43b5f0c614a64769accb1dfc96f33897 : SUCCESS in 37m 57s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ed8d33d5f447edb7e8a6afefae8b94 : SUCCESS in 50m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/53cb5f5e6d2045a4ad460be0532fc380 : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7d0460a4e37645ca9622afe8aa344a08 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c4da9af77244fa8b5c4a403c4d5060f : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/12b6427a45574f92becf3b4a41dc2bb5 : SUCCESS in 17m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4c627c9c8e24fa6bf75e0d430dc0809 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/93016dfef63d4df5bcfb42ae164d77b1 : SUCCESS in 1h 14m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37ae5b8d2c2844a9bf30d944918c5b50 : SUCCESS in 1h 30m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e42fef72aef74b7ba6e1913a4f628104 : SUCCESS in 20m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a52e3ed427f4ca39b86270c8b400d83 : SUCCESS in 18m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45dfe9bb92a345b19d7c7c0ea3683f64 : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b69dcc0e92334e5f8c41823ebc577c62 : SUCCESS in 9m 44s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c583bc25bd4043fe822aa205e0ac41d7 : SUCCESS in 37m 44s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/f8bc3320703e48d7a2cfdc78927ee3a5 : SUCCESS in 1h 08m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0b2b7dc9f4b64ff0827e8b5cd6a0a633 : SUCCESS in 37m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b860204755c42fbab0a3f1f9355fad7 : SUCCESS in 24m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed6f11969e6e465c99ce547ce51745e9 : SUCCESS in 15m 57s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7655610d66bc46ed8033b735049efd0a : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3132d0388fba4eafb37d3d7e05657c27 : SUCCESS in 10m 27s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/952463c8f128449dbbc8bbb5f0d04618 : SUCCESS in 8m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5cadff0a75547d8b73a03a8de80544e : SUCCESS in 21m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4d2e2d28e574f638e0422c2cbaf2792 : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":44},{"id":"bbf2b09e6424dcca355a6f243f17f4c413d84c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-24 17:22:26.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"989429a83558d395e8f9fc695ade63bfcaadbc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 19:13:02.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/097c7a9629c14249ad8630bcd3735aea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bb543bac24c48c59ec0611512e8cc49 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015a780a271c442083484cca0583e18e : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f07090190f7b444dabc29f7b87200dc7 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bfd746795c4786ae06f15aef2791fc : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8e77843c81c41118bfeeceedd5aa1f9 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3d9d4e80dd74dd4842f21b1437e1b07 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea02c9ea4ac542da9539b0746beacd66 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da88e37a13274eaca8407e7918723236 : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8846125d0c3943c28b2b2a10afbb7b3e : SUCCESS in 55m 11s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0503b2aaeb674494b07da78360ddbcf7 : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/77e49685f4a64347a9017db1578d4dae : SUCCESS in 52m 07s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f292f00981aa457b813a1ba0870f9ec5 : SUCCESS in 52m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fbf72c7ee320438684534eaf21f45d3c : SUCCESS in 57m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/61d1c3bcb6d84220b9006777fa4fefde : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cf1ce20b44234b19947785367da99a40 : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6623898bdc8d4068979a8fb1eddf7c56 : SUCCESS in 16m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/572038ee945045688be066e1ebcd1daf : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/18f9abdd5ad54951a73280ee7d4c251e : SUCCESS in 1h 42m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/635edf8b23204012998442403590d1c2 : SUCCESS in 1h 47m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bca8e44fb2424866b2316255c878e5be : SUCCESS in 20m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/42bc0ad959a544cdb47647f3d91230cc : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5e54ae72d6849c28c0640f0115c031c : SUCCESS in 8m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88277a8dc56242a79d96393f09e67668 : SUCCESS in 13m 46s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/12e958264856415c8c1d5bd4d1a4109f : SUCCESS in 44m 24s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/51c290a9464e4f90be9d2c0c6ea52e92 : SUCCESS in 1h 04m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c9018364601b42249ed7a021d54593ff : SUCCESS in 35m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d401bdbf8af4546a2988f9fffd161a0 : SUCCESS in 25m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5eb6447008f74896baca05c0e2fc4015 : SUCCESS in 17m 11s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0a959bf5bcf54a9fa4b6ec49124b5dc5 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc71daaaa00b4f688df303b4205bf31a : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b2275656855f443e9e4253998532ba62 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/013e60990306486da0c0238b3fb18263 : SUCCESS in 21m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/589b7be85356436b95eefd4e16744f0a : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":45},{"id":"4843631d465fb8a6229ca7523efaa12002a789c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 12:44:12.000000000","message":"Patch Set 46: Patch Set 45 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"5fc79c781c45a08d56a2cddab5a781f4a7d3c9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 14:51:03.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22457e7fede6466e95ba27c96c32572e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/925986c121fb41e1b7f076d155436f67 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e0e69f31c04c2e946e295a644ba3c9 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e09d8321038481ca769690d6e153348 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/584a1a5615334b0e9e74fe1b3e6390f8 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fe75f4d36d44ad796d329dc84fae69b : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a06dec1213946f9ac6b706f9e6a7087 : SUCCESS in 3m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60441fdf28664d9d99bd5fd5fa7084b8 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b90d8321a5442ad9261a3b1054d5096 : SUCCESS in 6m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/387f43ac49994897b37a8ca35504227c : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/046e6f5180bd4c7690ecc56384bd02ab : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fea67227d0bd47008bf49e9e436b8a52 : SUCCESS in 32m 20s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2bd2d39aac74cb5bb0866c86059031f : SUCCESS in 58m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/392fb1b8713a4c139d6160c22a32e362 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ddeadaa4d035437297e5796c5baae274 : SUCCESS in 1h 02m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/38a4eda585a14848a5d557147bc846c4 : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/272b03d706244e289430f836a8925ca6 : SUCCESS in 17m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d78c249919374da8b25e67e4d9dd36e9 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8293a4db087c4df58642439992255a54 : SUCCESS in 1h 50m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc55037f70194098ab880161bcb43d1e : SUCCESS in 1h 35m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c01f8f7639ac4000baebc9893f4f2190 : SUCCESS in 19m 00s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/beee3efd77384d8f9072941157ba9ed8 : SUCCESS in 19m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86d082c40fcc483197e4c075007df1ce : SUCCESS in 18m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2bbe3fd780494483a3d4e3c012109df6 : SUCCESS in 13m 12s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a5c410582444849bd1b0479a692c8d5 : SUCCESS in 40m 36s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d0e044297974938960eb9b2db54bb70 : SUCCESS in 1h 04m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cd8d472a93024b77934c216c9847656f : SUCCESS in 37m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8119350bc20049d2b693014e9404c258 : SUCCESS in 25m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71568b0cec5045a88d59809c57d48f58 : SUCCESS in 14m 44s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8d44d1dbc814446294d6509a5f2278d5 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f48d6ef28ee3431c87b52e4a34ba29f3 : SUCCESS in 11m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/bb203065d5194c0c8455233cd5454c35 : SUCCESS in 13m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e364e974a7f14bb989bd1807039c64cb : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ea99024d70446b7931f773985366d22 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":46},{"id":"2bc13046e826fdb8c44d84b63e9c94418b7a954c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-06 09:20:40.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"b9050508d4c83676861a41a781012b1f3203356f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-06 11:35:41.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"7fec0b681be949d47a4c3cb28a47e3868558bb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:45:50.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f9206c010db4eabae4abafd7cd309f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c186cd134643008618757f3848c7fd : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d44f11b2b44139972334b42d1e2ac1 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eedd7e0477f449a0bfe0a181dc294da5 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb19c5fc96845e484742011b74f13ec : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a77a9cfb64014d3b94feebc867b25974 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47d7c10d2c2b4a97af8ad56ee89fd38a : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/121e3c9ca63a409ea661c4934bc918e2 : FAILURE in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e749fccbd00477b887fec7b667cf6fc : FAILURE in 5m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b56ae5c049304d8dad411e82c0dd532a : FAILURE in 1h 56m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9761dab867e147618a9237ca103c86a4 : FAILURE in 2h 00m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0414c1caf2d343f79cbb3c704d1cb75e : FAILURE in 1h 49m 34s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59488a4ffd76437e90045a03d9f9b65b : FAILURE in 1h 50m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/392b201418e54f529c351bca830aeb17 : FAILURE in 1h 53m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/076c11e5d3a84394b63d75658cab0105 : TIMED_OUT in 2h 04m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a6eaeff946844cfaaba8fb83926887cd : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/140893173dcd4535899c1857b99897a5 : SUCCESS in 19m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03357ae4353646b288c577c9215a8dd0 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/851849f974eb4052b58c74d456fbdc4b : FAILURE in 3h 03m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f44d757877d044649cc4d0a85f5c1419 : FAILURE in 2h 38m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/82dbd8459ddf4a768d8dc69cd9a85fdc : SUCCESS in 23m 36s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/70362c0a4d8d444ba7192f78a8871b58 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/254f4e2929754d59b9fd38671f42ecee : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e392d681665d4372a1d984d4b3301443 : SUCCESS in 14m 51s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/217703a491444abcb395641cf219a7f6 : SUCCESS in 38m 06s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/60a6948a45264f79aced3a272d785f30 : SUCCESS in 1h 17m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5dea75f4eea94593b96f85acd2d52e8f : SUCCESS in 40m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48c1854e84db4b8398c7b24f5184334f : SUCCESS in 31m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8326c1ed7f1a47a6a6a14eb62f2d959d : SUCCESS in 18m 08s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b841e69436e14e36b669b24654357a0b : SUCCESS in 18m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8151a190181049f58c050bb615fc08c3 : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/ac15dda85f3e48b08fa038782f8c0026 : SUCCESS in 13m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7f45dda955d4488ab1720d80be001652 : SUCCESS in 18m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95b9ad84a4954de9846fbb078cc2f338 : SUCCESS in 20m 02s","accounts_in_message":[],"_revision_number":47},{"id":"7137b8c2026b025fa946406566e29660eda6a1ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-10 14:53:26.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"a44ae575b7a2bc3438af627dab99f98a38f5f31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:29:49.000000000","message":"Patch Set 47:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b581bc4d6a0e4615aff32fd44efc1670\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d3f9d5b783e434ba57013f53a0e86b3 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830ce20697c04e4cae9f4754dbbd3d0b : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01e7277c53034c1585d7c2acce88a96f : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea030cc2a27d4dd5a2d40dab9098c902 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/303e6763dee54292a9f1de68d2ef06f4 : SUCCESS in 2m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53fdb74ece91404b8a75e181d24a3fb1 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5109fe83f2fa495a970f838f3ec420b3 : FAILURE in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70ffc59314e84ed79db22aa3d710f209 : FAILURE in 8m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f6e098235e24426bb9efb98eaf2b8bee : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/52851e59c9a74be19f14942e42071ed6 : SUCCESS in 1h 14m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7bc0e4e2122643ad868708753078533b : SUCCESS in 1h 09m 26s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c43c7594e564d42a515488405a9430c : SUCCESS in 55m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a10f499c81f4fd7a17c43012fe5ce98 : SUCCESS in 57m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dd3ad3d8cccd406ea1fb43e7faf624c6 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/962ef401167f45aba806b8a0cb9ed7e7 : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e7bc578203b54fd981ef33d5facb7c82 : SUCCESS in 16m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27c2380231ba481e8e8765e958fde38a : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c5159a797994e17a38ab702928713b3 : SUCCESS in 2h 09m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f9e9399624549febb9db70708756d9d : SUCCESS in 1h 37m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7f5b7aa32772479a858cf8c71ecf2419 : SUCCESS in 20m 22s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e65aa60997f54488b20111861bf663cd : SUCCESS in 17m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc45feab6ad74e2ca21304d8601aa1ff : SUCCESS in 15m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b17c83524bb94b41abb9c486d432eacd : SUCCESS in 13m 53s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1738ceda2dd244a1bd0655ac01c1e70f : SUCCESS in 43m 35s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/f12c466c458744dda9343a7523742e9a : SUCCESS in 1h 19m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f569fe5c20ab4f5eb50a513d2119d1c4 : SUCCESS in 30m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a748ab9586af46bb91adb8a8d5039548 : SUCCESS in 23m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79a7f14a545e4c90a2811fcf7cd9d67b : SUCCESS in 11m 30s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/52150e0b1b7b48faa29bd045397da893 : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b533bf30102c422d8cc231d720d9e02a : SUCCESS in 10m 48s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/98e25c1a3fda4a529b86498415d1d2bb : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b57bc2958ab447e589213b26a123fd47 : SUCCESS in 15m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c03c25b5cbcd42a4a0a73ca2683146df : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":47},{"id":"e5630a277fb113f5c65ffbd8adce4030d306978f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-11 13:30:55.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"7c5ae38c1463a28048247da5d430d58f0109119a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 15:31:17.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b001914ff62a44099a517b5ca8e08a13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36d48766d96641ce853c333bde65cbb9 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd05f397d734da0a64e34eaac49966a : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a30c50e055347eda7098d9593dd1d04 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c9f3b545dd49ef8fbb7a5e29d23169 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0148e842d1d04e3baf62c4b59f1c915f : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15cd9018625e4618be1858b923ef5938 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d5e4dae46824ae8be2458ab414c0555 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/38c8583b815046f5b5ce35ac2aaf57f2 : FAILURE in 8m 08s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/64fa37639d3548e7ab413ccb5ef00a6b : SUCCESS in 1h 10m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e489cb7d7bd647649102d5e4f1c83f0b : FAILURE in 56m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e97b0a19a2044fe9d4e9154e0c1815c : SUCCESS in 1h 12m 18s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc3cda8da83d40afa267b53d9cedcb98 : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1e74964f92545329460bb502c0465e7 : FAILURE in 38m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/53324e014f1f4cc3a70d43c5a14ec4b8 : SUCCESS in 1h 17m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4996cc975dff4de9b91c9270d4443958 : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f260943459fe49d8b0ca347278a53686 : SUCCESS in 18m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b280a392e9184003b124a9dc088dd35d : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ed390b6b658a4a3fa1f0243078b78097 : SUCCESS in 1h 56m 46s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/143c1479a0e94ce0b23c826aae46b400 : SUCCESS in 1h 47m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b7fead67a33d45c3902379cdeb9abe3d : SUCCESS in 19m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fb1373134ddf4d42bd218098277a69d9 : SUCCESS in 19m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd17a1f4a8404cfc8683acbb790bb349 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d558b9f80b4e4e0da9da66a929317b7f : SUCCESS in 13m 30s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd734f0dc98c449198d32084c5add628 : SUCCESS in 37m 14s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/41b2faa2b1d54038a1b8a350cd9d9b6e : SUCCESS in 1h 09m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0cf8f45c84994a97bcaf3d85891b00fa : SUCCESS in 32m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8a7b94e67845a3976378562edf4230 : SUCCESS in 26m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c415f5e73ea74eb39da9cf47d63aaf81 : SUCCESS in 18m 19s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3d6b1dc3001d4154ace35ace47bae036 : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af28fbaffddb43f68cce158822010e12 : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2ec0a5da115a4e288e8f4f02beed6827 : SUCCESS in 14m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3da457c66f14d9d8c0d0643e38462e3 : SUCCESS in 23m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5f28f3225604972a68c25c8264057b4 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":48},{"id":"4304c4170d760f3bf2078282210c0f1c17f5d97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 21:11:27.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"abdfaff65d169eaccfd126498a95c60d05877609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 21:21:56.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"e20a5e332d12538674dafba4a1ba7c32227e329a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-12 21:22:09.000000000","message":"Patch Set 49:\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"7be5378b154e9646d2a3ef9c2ff2de3aa6764080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 23:25:26.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4aef4a8d50400ea51699b5f497ac14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2ee9d449bdc42a8a8322a88ddc1fa9b : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c5a03583ec455f84c3bdc850ebf424 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4595d7df1a3243d6be0097c414047c3a : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc784a7e019c465db4201a35e948ffe1 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15609f6fc11f402e9cba4c47bacac7e7 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c7fe707484949b18764bd1f8f028d3e : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b56e258149d54a12ac7058ea5e0af0dc : SUCCESS in 6m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b87ea8b4ce724e72b732adcc5a689faf : FAILURE in 4m 58s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/606b5643d00142578f0cb7042b3469c4 : SUCCESS in 37m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a02a0cb6b20840388012c5aa1da3c268 : SUCCESS in 58m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/321ac749a9b440df8a2ee71c5c93b7a3 : SUCCESS in 1h 00m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8cfe99b9ac946ef95b506a3e5fa510a : SUCCESS in 31m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4eafe1ac86d548239f6ff8b2c15910db : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c7af96faa26548ba8222cdb8b416a1bc : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b25ab484619043549f2837f80700f4c5 : SUCCESS in 14m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f634511af4e0411abf1f481c128439eb : SUCCESS in 15m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d79b79be885e4bb790289c3798e6594d : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9883f4e2bca741e3b98b33dde43a85bc : SUCCESS in 1h 51m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea950a4feed43f0b91e19f047d7cbb3 : SUCCESS in 1h 46m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3705fc4ddc6b4faa8217647e5b194043 : SUCCESS in 19m 18s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3758cfdce67d4f31a876b6d46cad84da : SUCCESS in 22m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d08a612b3bf4bd0b1eab834aa7cde3d : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/97e3b24ace364553b244f2cef05c7e5d : SUCCESS in 14m 29s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1c923f31f9fb4aab8d91521d3155a110 : SUCCESS in 43m 04s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/60a536ac874149fc99def2ad485596d3 : SUCCESS in 1h 19m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c4971dba65d417f947a638bdd3bd7b7 : SUCCESS in 33m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9f9f70ba1b046398e95cd6f4db8bf11 : SUCCESS in 28m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7607250005bd4b3b8540f866bdb2d49c : SUCCESS in 16m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cf033c131e454c25a3cec29d74fe600f : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c82b96970a247b4948ee45639d51a5e : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f68af994b4e54e149e263e539abf2118 : SUCCESS in 11m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2fecc9ef73f4479903cbfc0995f3a06 : SUCCESS in 21m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/709013c19d2e4e8ebd8d4f4dc47591d9 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":50},{"id":"a343ab5afa5ed2ac8b386dc586f794f24dd8324c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 17:34:13.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":51},{"id":"49f6537e6ac0a32428e7f7e875a4eda5e67fc7fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-13 17:40:07.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"2a06ca13007afee79573c13387763378b53b789c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-12-26 18:51:07.000000000","message":"Patch Set 52: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"b84db12c06e48f623f98baa3af86ad24b4988212","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 11:05:59.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"7bf05258896abd793d77a9c4ec37fef4a1f3cfc2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 11:16:14.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"dc1f560647dee4ee57bbdf1c5921358730d96d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 12:03:31.000000000","message":"Patch Set 53: Published edit on patch set 52.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":53},{"id":"6caf3ffa553adab7764041f83c557e8258e48cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-06 12:04:52.000000000","message":"Patch Set 54: Published edit on patch set 53.","accounts_in_message":[],"_revision_number":54},{"id":"b3cab00871b1c9f23817b5af91650ad114a5e081","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-16 12:51:51.000000000","message":"Patch Set 54: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"1a022acf69430df16ccdc697072994a63ff58c96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-17 17:11:39.000000000","message":"Uploaded patch set 55.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":55},{"id":"044138354fb40b67af05b4009d9d6a72e0023713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 19:04:32.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16427cad06614bbfa08e35d319b4e88a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc40aa3b8c4c4433b7dccf900b9aac38 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5df910b39c4aae99fce6279ff73a74 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9056b8f4c2cd4094b1e7d80cc5f92c23 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1f4a5195344eddb86ac9b2e88f8c4c : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf55eb749ece4f37b28520223a9dbbdd : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47b608b4f99449b48c2c2f29751ea0ff : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f43231f88a2a42aca3eac8941bb52bca : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eab687ade6d741bda872cd055c9fdbc5 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a1301e46e12b4f81837c151b123ce492 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/26c552a03b5d45f1bf90608f4b859119 : SUCCESS in 59m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/68e08e66006b44d59eddbb18a48d1ded : SUCCESS in 1h 10m 05s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b713efb0ae34e389dafd9a7d8d821e6 : SUCCESS in 58m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f1775f003f9b40a8b0be0de50657aa66 : SUCCESS in 53m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fe987a0ef208446fa1eb05e696b6b7aa : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7d310609806f4d339e9fb568a3d6e3a0 : SUCCESS in 25m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cd0b3558ec6c44e393f841d63642d477 : SUCCESS in 19m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcb5c191498648b99656668df7c9060e : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a6dc2e4116047098428074ae2c3ed83 : SUCCESS in 1h 51m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7abc79591d564ad29bbaa54d9deb0244 : SUCCESS in 1h 37m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba5929fcede04fd9ac748d6713c78095 : SUCCESS in 19m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/73827a5064854f59a7a5df47a3e6c6b2 : SUCCESS in 20m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/725a5046e9f24b848e6bb90701e17744 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/25fdd6a3cd4e467081954443d344f766 : SUCCESS in 16m 27s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/798c02b4920747af9b8f010fbad6d4d6 : SUCCESS in 38m 55s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/a59a04de9fcb4658b7205f2e55d44709 : SUCCESS in 1h 26m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0648886171634ce985c40e2fc26f9292 : SUCCESS in 39m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19e9610d50554785863dd4c2f2875afd : SUCCESS in 35m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aea92c31b21f414c89364bbdb3d1f009 : SUCCESS in 16m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/48c167ceddde4ca59306db80a74a26fe : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/710aa4dfe88d4c9989463064328b8990 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/02cb99e763de4db3bac036af6d15aaa5 : SUCCESS in 11m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52fd6463a093439b9c208bd068e4e361 : SUCCESS in 20m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a69ebde711d84802bb9bd31a5ee09fb4 : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":55},{"id":"997215c833546a6b9ba53a0a1dadf961600a9e94","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-21 08:50:15.000000000","message":"Patch Set 55:\n\n(2 comments)","accounts_in_message":[],"_revision_number":55},{"id":"7eaf1a3accbbed88c1ad971987913f0a6f07cade","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 10:57:21.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"a25e21dfcf7a822d2cfe797ddfcac73204e05861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:49:59.000000000","message":"Uploaded patch set 56.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":56},{"id":"f69791db1e74736d1ba5391ad6562e939469541e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:50:57.000000000","message":"Patch Set 56:\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"7933f8151d6e6c6009704b35f1d1bd1b61034399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 11:46:25.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a4ab0f23f44e7e97916eff66117a69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b472ecdb0af543b49b3a4ed39ecca590 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef9041bec064ac78915626a1c52c4e3 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f1432b543594242b5d65f91a1dbc109 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30691eea97e042679c8a313db940f121 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a431de6631324155b7ee53501bb74024 : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d6292808f0444099df915ecb5b689d9 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f91309231584fcdb421c3e1f412aebc : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e37a9e10952248418bd6cbda60a74e37 : SUCCESS in 9m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5e82ba7c8a749c78a8a8e860476148a : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6be6461cb93d4963aaa11d05c1c170fd : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/abe4b89654ca4645afcb864538b05926 : SUCCESS in 1h 01m 51s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6892d66186394730be8a9e28425088f6 : SUCCESS in 1h 00m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/987922dd4ae9413293874375da2ca06b : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da35b5a0db7c44fc9a23c7292f6a2fc1 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e1dee448cd48408189c7d7ebd2190f2b : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5d12c601d22d4b3785948db3176c7ffd : SUCCESS in 17m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/417bbb824fe64f3bb807632727d23e7e : SUCCESS in 8m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/23ff324bbed94d279fd2cfc310a4ac41 : SUCCESS in 1h 48m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08e1c1e31dfc4ae4bacc98dd0dfa71ed : SUCCESS in 1h 42m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b1296f07a4f740db96ce09ca7edebf95 : SUCCESS in 37m 47s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e2a4800d1b8f4baf87276af5903afb43 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e640560a2a405fa85d9770bc7c34d6 : SUCCESS in 14m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fbdd07dbc0024f13accf33a37a7b9595 : SUCCESS in 13m 58s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a0d43c4fe5ea45938e528c5f41cedcef : SUCCESS in 22m 20s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/45e333b43444482eb0314a85835e7525 : SUCCESS in 1h 26m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ceb183526f841b6b7a1c4af069af950 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dacd7c17e7cc456487e561fbae47d828 : SUCCESS in 27m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f2cdda71c06470f8282d4a4785da371 : SUCCESS in 14m 52s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4527f9a70c614d3fbb8f76afb987e523 : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2343225ce5144708ac0a799acead1434 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/968833973cce4d7aaec9c17ced62c97c : SUCCESS in 12m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/15554337ec0b4b55be924f0dcdaaf37d : SUCCESS in 13m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3e8c3fded4841689591b9c459cd21ab : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":56},{"id":"7586ebf29e336dd09d6d644d9f3b72aa27477cba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-27 11:40:39.000000000","message":"Patch Set 57: Patch Set 56 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"d5ef232cc1dc8b36f4537ca7b5f18c947f1ad372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 13:32:44.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240e8d3ec22146ea9807240dc0075e23\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c46e3563dea4e89b03819255734b578 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67213683713448868402286375829ead : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d79275ef1b334ae7be5a4e7e3e48fec1 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d9399ee6554777b8951a8a96e04a2b : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c601f37d60a4e5690cb8feadccf214e : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c96c839bb24b40188360575fda8c50d4 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dcd21e978a8b47ecb79eb7c3aae9c17e : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f33fdfd0b9fa4f32ace76229316e2b8f : SUCCESS in 9m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c28a43ad64384fa2a06e82a09b0be974 : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ceee54c4847b45c9b0a81e2a6f7d8a64 : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/205d07e6a64046c494eef4dd872389ed : SUCCESS in 1h 06m 12s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a37d9ac1daa42b2adab08811cdae775 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/772ec320199c4fd5817460802889602a : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/797534108a65436b84c9ad56e147bdd7 : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/36e8db37dc724c2693333dabacad6cbe : SUCCESS in 20m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/376000f21fb2423791de2812f81659bb : SUCCESS in 17m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7cc333f1d65491188c576c0c5d776f9 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0648abe3b09416d8b36fc4a9e7abac6 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f1da3468a63460d8be8869fca1de075 : SUCCESS in 1h 51m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/050e1eb3c75c4538b287a1bf7a1a3f79 : SUCCESS in 13m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0109c7117e02496cb89c2c0ae125a4f4 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a24feeb58c8145dfa5d90d1d7b4d4e77 : SUCCESS in 15m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c9f56e35fd84bb292db1a7a21bd9933 : SUCCESS in 14m 33s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f4a37f24f7c410194691a70c2d4cbab : SUCCESS in 22m 12s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/60b33c87b8ce46fa87ad666cd747c059 : SUCCESS in 1h 12m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab98cd0a01d04c03ae4c2310f579127a : SUCCESS in 34m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95d6bf54c89e41189332e8604252d43e : SUCCESS in 27m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/358a8ff138434c39bb3a0d4333e0024e : SUCCESS in 11m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a84621bd576b49a1849f75408f1d1df2 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2edb13af318b4906b499f5c7cb0c6fd4 : SUCCESS in 11m 27s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/ff794f7a9b2b47c9a10b330e7595fde9 : SUCCESS in 12m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d76cb183b13444e5b126128cda8b7fe2 : SUCCESS in 11m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19fb4bc709ed42a29c2ffa59ca8fb872 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":57},{"id":"cf607dfd9e886c2d04d9a386a76797edab1ad2fa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-27 16:19:25.000000000","message":"Patch Set 57: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"26df7aedf1be46088d22c900169b13b499f56898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-27 16:42:33.000000000","message":"Patch Set 58: Published edit on patch set 57.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"d79058b96e7a37cd345501e0590fa4be094bd41d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-27 16:42:59.000000000","message":"Patch Set 58:\n\n(7 comments)","accounts_in_message":[],"_revision_number":58},{"id":"2bfb57e5efcca1f25163a7ac73d8cc448df2836d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 19:24:47.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e27f48d7e20143ada32f1563a271c02f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2445fd98960c4386ac13b2af6a2b4498 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d886b5341b794964858461abe1beb792 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95f9108308e8421d9d4661e705377332 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7751102714464715b2add3d9346445d3 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a44842ba5b5e4775b93bc3ac2224332c : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03d3df16fd184acc8f0ec2d7c46e2d36 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6195250713074cb2bbc4aefc7823bb01 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c530bd3bfd2d474e9bb8e283c9d245f1 : SUCCESS in 11m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/82765c8c567d4d9bb8c32bc2da441819 : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6f04ee6ae4394a22801ddf605ff6d5ac : SUCCESS in 1h 46m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a1ef43194bc4440ea83ee8e4a724a78a : SUCCESS in 1h 07m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/451973e1d22243069ef140256eb9a008 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9f05fba78f548ba8f93262e1573518f : SUCCESS in 55m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e6a9da2548da42789ae483baca197983 : SUCCESS in 1h 01m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c1597aacd9c547d79c5da2ef4ae5e4ac : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/26a0dd48e9c642c1bab7e494c3620148 : SUCCESS in 16m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1315e8e9de4e47119ababa5aba0b2d68 : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/80092465028c495c972c04b8bbd3228d : SUCCESS in 1h 40m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9d30cebe7854fbc99ce3edaf8d6a869 : SUCCESS in 2h 34m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fbb1e33812cb4ee5b0a7be9623eace4d : SUCCESS in 19m 31s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ea6ce703dfc47a59559d14ba9f7486a : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79df848a2244435e9ebad55fe11909b7 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b441b0d34cbb415d81126fdfea215a82 : SUCCESS in 34m 54s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e206bc7bbd2b4ce18737fae400bc4b33 : SUCCESS in 45m 40s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/407b490621a14211999f9f8c4286ddaf : SUCCESS in 1h 16m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b8ca592c1714f6e97e884417a84424e : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3cab6011654f8180c74a0f7af8ef2d : SUCCESS in 26m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ec6535bec1a420bb8b2cc33cd67acc1 : SUCCESS in 17m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f6b9f8ee8138477d85e756d89a4a7f98 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b90ce5411a341f487ce1764bad900c2 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7c0a217fa89240f9a8e48b24181288d0 : SUCCESS in 11m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3cf9067e752246c9852912e9fc602fe1 : SUCCESS in 16m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e734d71f2e01403b9ae1b2f23f140d10 : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":58},{"id":"97e838d73b58961650d77f47c5686e344cb4bfbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-29 17:04:58.000000000","message":"Patch Set 58: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"9e582ad3509d5ef4044ba74669ec2c205f6be72c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-30 08:56:59.000000000","message":"Patch Set 58: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"c62b792b31edfe1e56ec13f96ce7d87001159107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 08:57:29.000000000","message":"Patch Set 58: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":58},{"id":"7d363f2a18e11724183d4ecbdffd333134e4d053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 11:08:34.000000000","message":"Patch Set 58: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8da60b7afb614096b5894f6fd9226758\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88957db24f134d628a69fa85888275fa : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/239a85143a3346a99f83c32c0d5ec184 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1af1ac537a4239bf9a0c6431cef2a7 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded1bfa5397048989cb3499fadb9d3f4 : SUCCESS in 2m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d83f5048cd604028a8fb2c55dcdfca54 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12afe2c75a1d449097668c8d23427fca : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1487d3c329f345d69823f2e53043c083 : SUCCESS in 12m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9c57ecb1e51245beba47e009f46c3c41 : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/248b91cbf6d94b0ea2eb2bc88544128b : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bd8fd34861245b5a9fac0938ca1309f : FAILURE in 53m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/26c8fd1f27434370bf9a56c6e09f68d7 : SUCCESS in 35m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8f00bfc52a6c4b9aa3edae5e71ed4fe3 : FAILURE in 1h 03m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9a4c5f82662d4f3e99e65b893ce84131 : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33e29ade4aba459d8f6366d1d861a9a3 : SUCCESS in 8m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e88e8287778e4738a5ca12033cce4236 : SUCCESS in 1h 52m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15e6e37868f749b7a5adeee261c1ea48 : SUCCESS in 1h 33m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/04150f4eff9542e185546356d8c17043 : SUCCESS in 16m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/df764e4d76934dc4afb42405c0a31e50 : FAILURE in 13m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d14d23485edb43cbb5488ac71f981a7a : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/58d30416c50e4457a7d5c42ac0aefbb2 : SUCCESS in 28m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4ac05cfba8445c18f28f57718358079 : SUCCESS in 32m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c576cfc39520434a8146ac9534f6fd35 : SUCCESS in 15m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44d18a99a004742b49754bfd19782ab : SUCCESS in 27m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d029723cec2d4879bc794226479e0553 : SUCCESS in 14m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a26aa3ab2bd5449c81338c515c747887 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":58},{"id":"678ee26c1a770b64fac37b32f91d842372864666","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-30 13:29:01.000000000","message":"Patch Set 58: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"a45a4c2033d9743307146ec40379e7c4dce17fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 15:43:41.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb38046b8c8f4374add2f42ad036433f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a188d981d1a346bf9fbafdcfc5762e2d : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c7665c929341b69c811dbaf62e24aa : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aac5f4cbcc04445e99877b92fa1e9663 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17a959571414cdab6100637d89ee01a : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/225b7688e45149c79c7d0747556894b5 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/784dc5d6aab44d508dcb4f584b7692d6 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9604e1153d340bcad27b2e1905d73c7 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/92a8aebe567b4f169172c17a570c6ca3 : SUCCESS in 10m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8af6e92380e84a9ebd607f9dd65a6f80 : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f1841145afcf495396ab7cd591de789f : SUCCESS in 59m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c7ae6e4eceb45ae9f5736bc057fbae6 : SUCCESS in 57m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e067cff45324a3fbd887fa1275c9c2e : SUCCESS in 59m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08675f294ca94483a4d30c94d6455e65 : SUCCESS in 58m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9cd8ddf864b484dbf27cc32660f8947 : FAILURE in 57m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c859e82839f4d29993d5b0e695146cb : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/451ddfb8be5e420eb7cdaf11b79747d3 : SUCCESS in 17m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf474009aea84f848cc17632cb71b75a : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/495673220147496ca09bc21d1c59ab35 : SUCCESS in 2h 07m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f11a9aee9784a0a8116da02bdd0a1d8 : FAILURE in 52m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f60f2d13cfd94931ba8b166e6d0adf4a : SUCCESS in 16m 46s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7aaea21a7037403ab24c0914f88f8c8f : SUCCESS in 19m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0309d6b408ee48079283265d3a1d91d8 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1780fb562d6b441dacf71322e80702cd : SUCCESS in 14m 16s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9fab8588a9364a2c97beb94c5206eee1 : SUCCESS in 38m 10s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/4bd6a35d90e64ab29699f7cfb792b752 : FAILURE in 1h 16m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/69451997c9d5478ab3e034fc4e0d305f : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db93bbfc8e5f40648966a9d84b2f43c1 : SUCCESS in 26m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bde8e76f48ba497684a118a993eb11af : SUCCESS in 12m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/172acf228586419cadb1ccc53b94a9ef : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7647b7ce42284bd2b851cae79e14a61c : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e53b648b1de64e539533abebd2b2f0e1 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51829c5086f04a97b534f5edf7bfbe91 : SUCCESS in 12m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efdb394da3e14f5fa788d734dfae9111 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":58},{"id":"ed6892626e0b27108ebfb6e9eafbdcf47b636bc2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-01 19:00:28.000000000","message":"Patch Set 58:\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"bb01e2eb2ae0ef4c047059b5338b6b755e66a300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:02:48.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87d25d56b86d40f5895202cb8c3bc6e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f74c2e96fd14415b8f8361df5c33efb4 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce48cb9a9714b9583c7b725f227cdd4 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a86dfbce3c94b50b17f65a423fa91eb : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063afaba1cfd4bf88d725935b96a110c : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1002339b6e7c4d8783a722f30dc914ec : SUCCESS in 2m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ad9bfb365e144c7b1e6f53f326519bb : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eff4c245744a4bb3b6d1b14730a8bb1e : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1796acafef754de68ab30ae3c2c23208 : SUCCESS in 7m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8b2f7dd62f04db780b34a9f8f9eef56 : SUCCESS in 49m 18s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2111d20680294058ab0176e0418471ea : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/59ea9bf6841644248d76d29b17a8201b : SUCCESS in 1h 05m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b9ac42adc5746ef961236f5270278ad : SUCCESS in 31m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cfb731e4ec53488392d6997ae20de44d : SUCCESS in 53m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5161959d1a6a473d9ece86c7f48d339b : SUCCESS in 46m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1bc085d46b844102b24b173d44cfb0d5 : SUCCESS in 21m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b08a2d13444b480290e1d28a5a2ac103 : SUCCESS in 16m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05b810f8a4684f15bc03f3a3fa6089fd : SUCCESS in 11m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7edcab384a0c4d0f8c37eb6801311d33 : SUCCESS in 1h 53m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/130a6ac960744121a411896d7465d396 : SUCCESS in 1h 45m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3e26652e4568420287c4d2b4ef632820 : SUCCESS in 16m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f914e5bd84014944a15129a7372da7a4 : SUCCESS in 9m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaf2ad425bb9401f808c03342f474f48 : SUCCESS in 13m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0b626a918fc4fc6991613f7c6622dfc : SUCCESS in 11m 31s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f5ba455c16a04e05b79cf61b5ea92f37 : SUCCESS in 35m 04s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/248dbd0b14444d139c95d457c095c8c3 : FAILURE in 40m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2c506b09b2c147a6b787a22bece65f8d : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1fef93171f34abfbe1040c9e5ac8abd : SUCCESS in 27m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2560fb087eab49438aa7179382bdc9a7 : SUCCESS in 14m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/78c976c49ba74732bba2951367063d10 : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75c5e594420141d894dc56962872b3ea : SUCCESS in 7m 26s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/37301079a61c44ffbea537a3d7ea64c8 : SUCCESS in 14m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e7ca41fa75a64f5995d5b5d64cf39267 : SUCCESS in 10m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eba6544943742d8811003017ce15b5f : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":58},{"id":"54b5c2b2ff0da24ecb971d61d02aca0bd5b5ff81","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-02 13:13:53.000000000","message":"Patch Set 58: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"f474643db8a8802f0918317563de6e38f82213b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:41:01.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/665e1cc0642c462bb1dec12d0e63499f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e30ff9479f614cc5857b0de8e0ef6052 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52cec11938e04be1bd06ee2bb9e22eb1 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc1d310a7d7f49ae9cf2ca50d90b3c41 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc228a6e9cc4b48b9d467f912b04458 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f1e047e9e5d44afbc14edca0373e2c3 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18f4923676fd4f51afd1941bc030146a : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbe2aa8d380c42719d52b55dd0e6c78e : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0ca18daac6742698356899c6f114c82 : SUCCESS in 8m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/602c9890a7514899bee8da918b82b8b2 : SUCCESS in 48m 34s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ccefc8c5dd2c4118bfc91c0187c7ecff : SUCCESS in 51m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/df938a5fbdb84308ab182ce75d623437 : SUCCESS in 33m 56s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/650a95bd1632493e836795a4162d49c5 : SUCCESS in 49m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d17fe9448d8f40478846ab657d26342b : SUCCESS in 50m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/09dfc421a6704777acde77a540ca0d15 : SUCCESS in 37m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f30db22384194dff9a7ff21d07964605 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/971c3c6df5fe418f9e5f0505d535409f : FAILURE in 10m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f128e3cc47b54b5bb4653d4e600392b4 : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/afe42cbdb42c47efb7829def8945f9d3 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a26995c60b0426babf1da4f1125a340 : SUCCESS in 1h 26m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d428771f6014ddebb03e158c5c18d0e : SUCCESS in 15m 51s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a69b7292d074e9695b7a524e3caf9bf : SUCCESS in 10m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e0891ced6fc4831a977ff40b4bebee3 : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e425112ed307454198485ad0646cf6ab : SUCCESS in 9m 51s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/573c3382a3d94bc8a19ad64515e9dac5 : SUCCESS in 33m 22s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ebbb5fafb144ca39c66d3ad1fcbd221 : SUCCESS in 41m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c702159005aa46288895164f2187ac57 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb0cb5ebf86c4076b121c68c8a2e7015 : SUCCESS in 31m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c54ab3ae60e44d759a882c2bc2291390 : SUCCESS in 9m 29s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2ea153ded3e44c5e957387b6c93179b2 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73c9dd651dfe452c8f2f558e92c16d4a : SUCCESS in 12m 27s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/ad27e2817def455298581cebcab6b69e : SUCCESS in 12m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8554f6126a83475e92fd7363b9537366 : SUCCESS in 13m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32bb7bdce4de4eaa9b9e7ce6d5ff1dd7 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":58},{"id":"66f01cc1777bbf7ef9b03e96b44b6f182627588a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 14:41:39.000000000","message":"Patch Set 58: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":58},{"id":"06e856029d3e9eca34f6866975bbb62e70274bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:34:28.000000000","message":"Patch Set 58: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b9b02a6726142549872916be4cda482\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1a1057f0ab4d8193beb99ae2553f32 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c02e531cf5843b4895254b04b406656 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6ef181af614c59b6c32ec3bd25c68f : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b762d2f99e341b5a2825d3c4248631f : SUCCESS in 2m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e16eeac1e3d24d01828047da8f88774e : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2b8984051cd402ba0fae97a689a9911 : SUCCESS in 7m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1d5e796ec1f410c83513b5360f59da6 : SUCCESS in 7m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4c5f36c59be84ca588bfd3980e1bf0ed : SUCCESS in 32m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/28282773ce4d4a49905072257443559f : SUCCESS in 58m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b67f776024a4c329518d9e04941b087 : SUCCESS in 32m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e740703ec5441c1a01a7b33da2293df : SUCCESS in 58m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6ed9b83bb8b84c7c9916949252fa03b6 : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2b59c6a1fe244c199c875befca05e35f : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4980aaa19dd84c22a5efbc8ae55d01ab : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c8507fb5b10b40ffb007aa2418c0b30f : SUCCESS in 1h 47m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fba65f803e3468ab52b72252c8013b6 : SUCCESS in 1h 45m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e88beceec4d4462b9555fbfbe1d42126 : SUCCESS in 16m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/644cef4488e04520a2c24da08e74a465 : SUCCESS in 18m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d2fa2600478458782da6084f060e8ac : SUCCESS in 16m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a6b40ec075c4a1c9bcd0387396cb093 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01b9789d4b93477d91929db3c081a86d : SUCCESS in 22m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02631970ebfd492686b52e3050ae9cb5 : SUCCESS in 10m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9204c020e86d49eba32686f12ad2b951 : SUCCESS in 9m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52845e6cccb34113b9bd6f1ad1edf25f : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50f5c0bc681f4fb39d52c9f6141152a2 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":58},{"id":"6349e058e1e2ff1a13feb33503012b6a20ed58b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:34:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":58},{"id":"829b6b6fbfbc869c00594bd8ee0112ae2eb4834f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 16:36:42.000000000","message":"Patch Set 58:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9086d3fc2234d71b8106d9f0c72591d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4834f0f09dbc463a86019c227fc45dd5 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7913680b17c24abe81d0b1c65cef4598 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":58},{"id":"2c5b0a398ef43104a5f8f52f8aea35a3613c9254","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-03 03:17:04.000000000","message":"Patch Set 58: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"02389adabe88f5269fa0b6e2124501f4f759b892","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-03 03:18:49.000000000","message":"Patch Set 58:\n\n(1 comment)","accounts_in_message":[],"_revision_number":58}],"current_revision_number":58,"current_revision":"988a822259a375bb404fef7b435a74511379c895","revisions":{"bf6e4c845de54eaed2c907c93c3d39e651a87c14":{"kind":"REWORK","_number":1,"created":"2024-02-21 15:36:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/1"}}},"commit":{"parents":[{"commit":"ba99ba4bede218b1e0aa588f0e56ef2238ebc113","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba99ba4bede218b1e0aa588f0e56ef2238ebc113"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:36:03.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf6e4c845de54eaed2c907c93c3d39e651a87c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf6e4c845de54eaed2c907c93c3d39e651a87c14"}]},"branch":"refs/heads/master"},"42a7e4cd8d714d79a5407778c85a10a79df281f9":{"kind":"REWORK","_number":2,"created":"2024-02-21 15:48:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/2"}}},"commit":{"parents":[{"commit":"ba99ba4bede218b1e0aa588f0e56ef2238ebc113","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba99ba4bede218b1e0aa588f0e56ef2238ebc113"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:48:53.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42a7e4cd8d714d79a5407778c85a10a79df281f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42a7e4cd8d714d79a5407778c85a10a79df281f9"}]},"branch":"refs/heads/master"},"337b506dbeffd090aff864776006973b455d044d":{"kind":"REWORK","_number":3,"created":"2024-02-21 15:55:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/3"}}},"commit":{"parents":[{"commit":"ba99ba4bede218b1e0aa588f0e56ef2238ebc113","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba99ba4bede218b1e0aa588f0e56ef2238ebc113"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:55:15.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/337b506dbeffd090aff864776006973b455d044d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/337b506dbeffd090aff864776006973b455d044d"}]},"branch":"refs/heads/master"},"11260b96d4edde24747a7277f17a635497f54282":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-02-21 16:26:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/4"}}},"commit":{"parents":[{"commit":"ba99ba4bede218b1e0aa588f0e56ef2238ebc113","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba99ba4bede218b1e0aa588f0e56ef2238ebc113"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 16:26:21.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11260b96d4edde24747a7277f17a635497f54282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11260b96d4edde24747a7277f17a635497f54282"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1a4c0c2728d0413a23e9d4224dd9e102a2460e1f":{"kind":"REWORK","_number":5,"created":"2024-02-21 18:28:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/5"}}},"commit":{"parents":[{"commit":"e1978fc6d9d775c74635343fe78d5088b6c618da","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1978fc6d9d775c74635343fe78d5088b6c618da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 18:28:41.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a4c0c2728d0413a23e9d4224dd9e102a2460e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a4c0c2728d0413a23e9d4224dd9e102a2460e1f"}]},"branch":"refs/heads/master"},"2ff9df595df570e80e1bb59fa47e95c93d9a7c75":{"kind":"REWORK","_number":6,"created":"2024-02-22 10:49:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/6"}}},"commit":{"parents":[{"commit":"e1978fc6d9d775c74635343fe78d5088b6c618da","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1978fc6d9d775c74635343fe78d5088b6c618da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 10:48:45.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ff9df595df570e80e1bb59fa47e95c93d9a7c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ff9df595df570e80e1bb59fa47e95c93d9a7c75"}]},"branch":"refs/heads/master"},"427daf8d437cc7120d0e84e076992080bf2fbd05":{"kind":"REWORK","_number":7,"created":"2024-02-22 10:51:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/7"}}},"commit":{"parents":[{"commit":"e1978fc6d9d775c74635343fe78d5088b6c618da","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1978fc6d9d775c74635343fe78d5088b6c618da"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 10:51:32.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427daf8d437cc7120d0e84e076992080bf2fbd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427daf8d437cc7120d0e84e076992080bf2fbd05"}]},"branch":"refs/heads/master"},"5fbca30be4125c50700f5710b07e05d7422894af":{"kind":"REWORK","_number":8,"created":"2024-02-23 12:41:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/8"}}},"commit":{"parents":[{"commit":"57cd650136ca423b2eef1f20c1f34834429df68b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57cd650136ca423b2eef1f20c1f34834429df68b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 12:40:21.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fbca30be4125c50700f5710b07e05d7422894af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fbca30be4125c50700f5710b07e05d7422894af"}]},"branch":"refs/heads/master"},"629b979b62ad121840588f86113d6eb0a1260bb9":{"kind":"REWORK","_number":9,"created":"2024-02-23 12:54:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/9"}}},"commit":{"parents":[{"commit":"57cd650136ca423b2eef1f20c1f34834429df68b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57cd650136ca423b2eef1f20c1f34834429df68b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 12:53:17.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/629b979b62ad121840588f86113d6eb0a1260bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/629b979b62ad121840588f86113d6eb0a1260bb9"}]},"branch":"refs/heads/master"},"c82b7f8b6ae8645a84812d1c1acc2649fbae0264":{"kind":"REWORK","_number":10,"created":"2024-02-23 14:50:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/10"}}},"commit":{"parents":[{"commit":"57cd650136ca423b2eef1f20c1f34834429df68b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57cd650136ca423b2eef1f20c1f34834429df68b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 14:50:33.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c82b7f8b6ae8645a84812d1c1acc2649fbae0264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c82b7f8b6ae8645a84812d1c1acc2649fbae0264"}]},"branch":"refs/heads/master"},"475cb5bbae701358dd36f1505f423c38c1e9cfa7":{"kind":"REWORK","_number":11,"created":"2024-02-23 17:25:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/11"}}},"commit":{"parents":[{"commit":"2d9c303e9eece58db9993a00c9e03da0c4a17e91","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d9c303e9eece58db9993a00c9e03da0c4a17e91"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 17:25:29.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/475cb5bbae701358dd36f1505f423c38c1e9cfa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/475cb5bbae701358dd36f1505f423c38c1e9cfa7"}]},"branch":"refs/heads/master"},"66ac026c82054829a9da8749b14afc1df1acb029":{"kind":"REWORK","_number":12,"created":"2024-02-23 18:09:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/12"}}},"commit":{"parents":[{"commit":"2d9c303e9eece58db9993a00c9e03da0c4a17e91","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d9c303e9eece58db9993a00c9e03da0c4a17e91"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 18:05:32.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66ac026c82054829a9da8749b14afc1df1acb029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66ac026c82054829a9da8749b14afc1df1acb029"}]},"branch":"refs/heads/master"},"c4a34e81165c6ac9dc94a177c5bacb26833d9f5c":{"kind":"REWORK","_number":13,"created":"2024-02-23 19:07:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/13"}}},"commit":{"parents":[{"commit":"2d9c303e9eece58db9993a00c9e03da0c4a17e91","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d9c303e9eece58db9993a00c9e03da0c4a17e91"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-23 19:07:31.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a34e81165c6ac9dc94a177c5bacb26833d9f5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4a34e81165c6ac9dc94a177c5bacb26833d9f5c"}]},"branch":"refs/heads/master"},"7620ebc48459c69ec75c96785844cf20572dd0a1":{"kind":"REWORK","_number":14,"created":"2024-02-26 10:47:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/14"}}},"commit":{"parents":[{"commit":"2d9c303e9eece58db9993a00c9e03da0c4a17e91","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d9c303e9eece58db9993a00c9e03da0c4a17e91"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-26 10:47:32.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7620ebc48459c69ec75c96785844cf20572dd0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7620ebc48459c69ec75c96785844cf20572dd0a1"}]},"branch":"refs/heads/master"},"425e37bbe8fb753c716f2ae0db771092509a60e4":{"kind":"REWORK","_number":15,"created":"2024-02-28 13:23:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/15"}}},"commit":{"parents":[{"commit":"f32c348b28cfdf42b801b52f8a8bea81b493a949","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f32c348b28cfdf42b801b52f8a8bea81b493a949"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-28 13:23:00.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n(cherry picked from commit 7620ebc48459c69ec75c96785844cf20572dd0a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425e37bbe8fb753c716f2ae0db771092509a60e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425e37bbe8fb753c716f2ae0db771092509a60e4"}]},"branch":"refs/heads/master"},"8013e053a829dd815eb587af51e3f72157a916f6":{"kind":"REWORK","_number":16,"created":"2024-03-01 19:09:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/16"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-01 19:09:44.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8013e053a829dd815eb587af51e3f72157a916f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8013e053a829dd815eb587af51e3f72157a916f6"}]},"branch":"refs/heads/master"},"119196e9cc23e6c2e540f1c12e864612e8a2008c":{"kind":"REWORK","_number":17,"created":"2024-03-01 19:15:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/17"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-01 19:15:05.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/119196e9cc23e6c2e540f1c12e864612e8a2008c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/119196e9cc23e6c2e540f1c12e864612e8a2008c"}]},"branch":"refs/heads/master"},"4c502b5b04aa90ade72a23a2684fc7db50b0ef9f":{"kind":"REWORK","_number":18,"created":"2024-03-01 19:22:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/18"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-01 19:21:58.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c502b5b04aa90ade72a23a2684fc7db50b0ef9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c502b5b04aa90ade72a23a2684fc7db50b0ef9f"}]},"branch":"refs/heads/master"},"360966697228bb3b149e556c9991a7caa5bba637":{"kind":"REWORK","_number":19,"created":"2024-03-01 19:29:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/19"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-01 19:29:23.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/360966697228bb3b149e556c9991a7caa5bba637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/360966697228bb3b149e556c9991a7caa5bba637"}]},"branch":"refs/heads/master"},"a1742251797ef84dbad344f819dcc5411f4d3a81":{"kind":"REWORK","_number":20,"created":"2024-03-04 09:35:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/20"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-04 09:29:57.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1742251797ef84dbad344f819dcc5411f4d3a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1742251797ef84dbad344f819dcc5411f4d3a81"}]},"branch":"refs/heads/master"},"ab990fd898b5a1493f5df02b20fa136f0c57667b":{"kind":"REWORK","_number":21,"created":"2024-03-04 13:23:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/21"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-04 13:22:11.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab990fd898b5a1493f5df02b20fa136f0c57667b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab990fd898b5a1493f5df02b20fa136f0c57667b"}]},"branch":"refs/heads/master"},"6932046416a9fd4a7e76646803f7424bc995d0b4":{"kind":"REWORK","_number":22,"created":"2024-03-04 16:54:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/22"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-04 16:53:54.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6932046416a9fd4a7e76646803f7424bc995d0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6932046416a9fd4a7e76646803f7424bc995d0b4"}]},"branch":"refs/heads/master"},"e2178a65bc8b879a7a2306296335f6bfbb6c85c2":{"kind":"REWORK","_number":23,"created":"2024-03-04 17:14:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/23"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-04 17:13:01.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2178a65bc8b879a7a2306296335f6bfbb6c85c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2178a65bc8b879a7a2306296335f6bfbb6c85c2"}]},"branch":"refs/heads/master"},"ecae1c5a61be757bdadcdfda0be780c1d0b5bac7":{"kind":"REWORK","_number":24,"created":"2024-03-06 11:11:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/24"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-06 11:10:50.000000000","tz":0},"subject":"WIP: Add podman support","message":"WIP: Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecae1c5a61be757bdadcdfda0be780c1d0b5bac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecae1c5a61be757bdadcdfda0be780c1d0b5bac7"}]},"branch":"refs/heads/master"},"b411479893280531294fede377fb433feccefad9":{"kind":"REWORK","_number":25,"created":"2024-03-06 11:25:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/25"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-06 11:17:14.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b411479893280531294fede377fb433feccefad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b411479893280531294fede377fb433feccefad9"}]},"branch":"refs/heads/master"},"deef32035c29ca12ab0be6e1bf6acd4d9767b1bb":{"kind":"REWORK","_number":26,"created":"2024-03-07 15:55:11.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/26"}}},"commit":{"parents":[{"commit":"3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f59ac9edbd7eb22eb763e6f8b35ee5622b1fa1c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 15:54:59.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deef32035c29ca12ab0be6e1bf6acd4d9767b1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deef32035c29ca12ab0be6e1bf6acd4d9767b1bb"}]},"branch":"refs/heads/master"},"0c13803eb82f817732798b7a07144537767eafba":{"kind":"REWORK","_number":27,"created":"2024-03-07 18:05:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/27"}}},"commit":{"parents":[{"commit":"c7e6dcdc7509ba42a909c8c068082bf829ddbc1d","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7e6dcdc7509ba42a909c8c068082bf829ddbc1d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 18:02:08.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c13803eb82f817732798b7a07144537767eafba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c13803eb82f817732798b7a07144537767eafba"}]},"branch":"refs/heads/master"},"2088322ac219a8b4e85d33ce7fb42d9578837b09":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2024-03-07 18:07:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/28"}}},"commit":{"parents":[{"commit":"e03f6f42a24fe820469069c7547cbc62d5657503","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e03f6f42a24fe820469069c7547cbc62d5657503"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 18:07:05.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2088322ac219a8b4e85d33ce7fb42d9578837b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2088322ac219a8b4e85d33ce7fb42d9578837b09"}]},"branch":"refs/heads/master","description":"Rebase"},"82652a0a12f3f5335ed7c32d6d33b16db85494ca":{"kind":"REWORK","_number":29,"created":"2024-03-13 10:56:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/29"}}},"commit":{"parents":[{"commit":"23b6606f1f460ea0fcec50a05189e371c357ba17","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b6606f1f460ea0fcec50a05189e371c357ba17"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-13 10:48:17.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82652a0a12f3f5335ed7c32d6d33b16db85494ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82652a0a12f3f5335ed7c32d6d33b16db85494ca"}]},"branch":"refs/heads/master"},"e20e909f0609a9bba5e2902a49d1e1d92cbbfe0d":{"kind":"REWORK","_number":30,"created":"2024-03-13 14:30:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/30"}}},"commit":{"parents":[{"commit":"23b6606f1f460ea0fcec50a05189e371c357ba17","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b6606f1f460ea0fcec50a05189e371c357ba17"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-13 14:30:03.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e20e909f0609a9bba5e2902a49d1e1d92cbbfe0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e20e909f0609a9bba5e2902a49d1e1d92cbbfe0d"}]},"branch":"refs/heads/master"},"2633cd96505396fceb8e6621d4cabaaad19aaff7":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-03-13 18:06:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/31"}}},"commit":{"parents":[{"commit":"4a4e54b4430ffcff3286e3b0aa484c58433a42bc","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4e54b4430ffcff3286e3b0aa484c58433a42bc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-13 18:05:24.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2633cd96505396fceb8e6621d4cabaaad19aaff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2633cd96505396fceb8e6621d4cabaaad19aaff7"}]},"branch":"refs/heads/master"},"965929f9e3fce34112eb821e091d131d80ff8551":{"kind":"REWORK","_number":32,"created":"2024-03-18 13:06:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/32"}}},"commit":{"parents":[{"commit":"4a4e54b4430ffcff3286e3b0aa484c58433a42bc","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4e54b4430ffcff3286e3b0aa484c58433a42bc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 13:06:43.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/911422\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/965929f9e3fce34112eb821e091d131d80ff8551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/965929f9e3fce34112eb821e091d131d80ff8551"}]},"branch":"refs/heads/master"},"10c17d7aa8201a35ef0ffe471be8b5cf49432568":{"kind":"REWORK","_number":33,"created":"2024-03-18 16:47:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/33"}}},"commit":{"parents":[{"commit":"4a4e54b4430ffcff3286e3b0aa484c58433a42bc","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4e54b4430ffcff3286e3b0aa484c58433a42bc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 16:46:57.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10c17d7aa8201a35ef0ffe471be8b5cf49432568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10c17d7aa8201a35ef0ffe471be8b5cf49432568"}]},"branch":"refs/heads/master"},"2241c7f9849a606ea44ee8b0703f145777e96343":{"kind":"REWORK","_number":34,"created":"2024-03-18 16:49:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/34"}}},"commit":{"parents":[{"commit":"4a4e54b4430ffcff3286e3b0aa484c58433a42bc","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4e54b4430ffcff3286e3b0aa484c58433a42bc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 16:49:59.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2241c7f9849a606ea44ee8b0703f145777e96343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2241c7f9849a606ea44ee8b0703f145777e96343"}]},"branch":"refs/heads/master"},"04c3d0c10acf74a30cd6e80ee5d7a6cc20aa6062":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2024-03-18 17:14:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/35"}}},"commit":{"parents":[{"commit":"bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-18 17:14:19.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04c3d0c10acf74a30cd6e80ee5d7a6cc20aa6062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04c3d0c10acf74a30cd6e80ee5d7a6cc20aa6062"}]},"branch":"refs/heads/master"},"0cbbec733e79ef077f6e1e5a12719b8f05451810":{"kind":"REWORK","_number":36,"created":"2024-05-01 11:22:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/36"}}},"commit":{"parents":[{"commit":"bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-01 11:22:11.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cbbec733e79ef077f6e1e5a12719b8f05451810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cbbec733e79ef077f6e1e5a12719b8f05451810"}]},"branch":"refs/heads/master"},"d6bb00801f535c59dded669db760321863193644":{"kind":"REWORK","_number":37,"created":"2024-05-01 17:06:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/37"}}},"commit":{"parents":[{"commit":"bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-01 17:06:18.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6bb00801f535c59dded669db760321863193644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6bb00801f535c59dded669db760321863193644"}]},"branch":"refs/heads/master"},"ffa25e181d40d3e8c2ad655e3b78011fe71b2ed6":{"kind":"REWORK","_number":38,"created":"2024-05-02 08:28:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/38"}}},"commit":{"parents":[{"commit":"bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-02 08:26:46.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffa25e181d40d3e8c2ad655e3b78011fe71b2ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffa25e181d40d3e8c2ad655e3b78011fe71b2ed6"}]},"branch":"refs/heads/master"},"be68fbd442955812fe1d46ad3234b785b8e7f738":{"kind":"REWORK","_number":39,"created":"2024-05-02 08:47:00.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/39"}}},"commit":{"parents":[{"commit":"bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd34c1c766b1fd0d46d9d50463ce9b6a89ff5e7b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-02 08:46:07.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be68fbd442955812fe1d46ad3234b785b8e7f738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be68fbd442955812fe1d46ad3234b785b8e7f738"}]},"branch":"refs/heads/master"},"1376f7e1e04d222d47e2325d2ffd4c5ab27ad649":{"kind":"REWORK","_number":40,"created":"2024-05-30 10:41:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/40"}}},"commit":{"parents":[{"commit":"01fe797d51189ff5d57366e74004671a7443d5e5","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01fe797d51189ff5d57366e74004671a7443d5e5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-30 10:40:37.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support add in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1376f7e1e04d222d47e2325d2ffd4c5ab27ad649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1376f7e1e04d222d47e2325d2ffd4c5ab27ad649"}]},"branch":"refs/heads/master"},"1816850cd7c10ec7424be631c2c171b5da433c74":{"kind":"NO_CODE_CHANGE","_number":41,"created":"2024-05-30 10:48:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/41"}}},"commit":{"parents":[{"commit":"01fe797d51189ff5d57366e74004671a7443d5e5","subject":"Support configuring persistent systemd journal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01fe797d51189ff5d57366e74004671a7443d5e5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-21 15:34:28.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-30 10:45:46.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1816850cd7c10ec7424be631c2c171b5da433c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1816850cd7c10ec7424be631c2c171b5da433c74"}]},"branch":"refs/heads/master"},"2c43366fadfab5595e0a30da92f867cefbe100cf":{"kind":"REWORK","_number":42,"created":"2024-06-20 11:46:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/42"}}},"commit":{"parents":[{"commit":"03ceac585491b9266725cdc2867273d40a402f30","subject":"remove migrate_rabbitmq_queues in D cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03ceac585491b9266725cdc2867273d40a402f30"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c43366fadfab5595e0a30da92f867cefbe100cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c43366fadfab5595e0a30da92f867cefbe100cf"}]},"branch":"refs/heads/master"},"9290528730a2b9b9a15548b9a8e467cec95641ed":{"kind":"REWORK","_number":43,"created":"2024-06-20 11:56:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/43"}}},"commit":{"parents":[{"commit":"03ceac585491b9266725cdc2867273d40a402f30","subject":"remove migrate_rabbitmq_queues in D cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03ceac585491b9266725cdc2867273d40a402f30"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:55:54.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/909692\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910016\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910501\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9290528730a2b9b9a15548b9a8e467cec95641ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9290528730a2b9b9a15548b9a8e467cec95641ed"}]},"branch":"refs/heads/master"},"aeba1cc00f00625a1c1e5e458b0b6643a9c4e0e6":{"kind":"REWORK","_number":44,"created":"2024-10-24 10:29:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/44"}}},"commit":{"parents":[{"commit":"9b53c0ee95c02d5d945c7f690f77ce53bab95145","subject":"Merge \"Fix group and mode of /var/log/journal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b53c0ee95c02d5d945c7f690f77ce53bab95145"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 10:29:04.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeba1cc00f00625a1c1e5e458b0b6643a9c4e0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeba1cc00f00625a1c1e5e458b0b6643a9c4e0e6"}]},"branch":"refs/heads/master"},"12edf22715ac6549543ebc4664dc846bcf7af54d":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2024-10-24 17:22:26.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/45"}}},"commit":{"parents":[{"commit":"478347f9dc243aa2565bfbde98fb07d26caa04bf","subject":"Aligns Kayobe Ansible version with Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/478347f9dc243aa2565bfbde98fb07d26caa04bf"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-24 16:19:17.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12edf22715ac6549543ebc4664dc846bcf7af54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12edf22715ac6549543ebc4664dc846bcf7af54d"}]},"branch":"refs/heads/master"},"fd7aa12a03e0a2d4e4854a0ba8922bc1cf33e149":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2024-10-31 12:44:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/46"}}},"commit":{"parents":[{"commit":"95724ec11fdbf5ca32e0269d88e2fc7f1429e37a","subject":"Aligns Kayobe Ansible version with Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95724ec11fdbf5ca32e0269d88e2fc7f1429e37a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-31 12:44:12.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd7aa12a03e0a2d4e4854a0ba8922bc1cf33e149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd7aa12a03e0a2d4e4854a0ba8922bc1cf33e149"}]},"branch":"refs/heads/master","description":"Rebase"},"c1c7656b52485e2ce2e702f1b7286774c8f20cab":{"kind":"REWORK","_number":47,"created":"2024-12-06 11:35:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/47"}}},"commit":{"parents":[{"commit":"3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0","subject":"Merge \"reno: Update master for unmaintained/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c178846a4e40bfc1c5f6f3bed2f5f70bf226cf0"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-06 11:34:58.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c7656b52485e2ce2e702f1b7286774c8f20cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c7656b52485e2ce2e702f1b7286774c8f20cab"}]},"branch":"refs/heads/master"},"f4ce72d397824e15bbfd8aa7f467454e855182b6":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2024-12-11 13:30:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/48"}}},"commit":{"parents":[{"commit":"3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5","subject":"CI: Bump size of Swift device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a926637650b8d820d8a65ec3b5bbf5a2eaba7e5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-11 13:30:35.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4ce72d397824e15bbfd8aa7f467454e855182b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4ce72d397824e15bbfd8aa7f467454e855182b6"}]},"branch":"refs/heads/master"},"01543d31ac09277ea86821359fbe43cd6b1a23a2":{"kind":"REWORK","_number":49,"created":"2024-12-12 21:11:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/49"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 21:07:27.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01543d31ac09277ea86821359fbe43cd6b1a23a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01543d31ac09277ea86821359fbe43cd6b1a23a2"}]},"branch":"refs/heads/master"},"163fa4ab0452e342c74e7fc50c765f284a98b8af":{"kind":"REWORK","_number":50,"created":"2024-12-12 21:21:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/50"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-12 21:21:47.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/163fa4ab0452e342c74e7fc50c765f284a98b8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/163fa4ab0452e342c74e7fc50c765f284a98b8af"}]},"branch":"refs/heads/master"},"d495302d1d5682546206e0f2e147851988cb2588":{"kind":"REWORK","_number":51,"created":"2024-12-13 17:34:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/51"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-13 17:33:40.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d495302d1d5682546206e0f2e147851988cb2588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d495302d1d5682546206e0f2e147851988cb2588"}]},"branch":"refs/heads/master"},"d0e196f7665af597329cfa8e0c950024e2de763e":{"kind":"REWORK","_number":52,"created":"2024-12-13 17:40:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/52"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-12-13 17:40:00.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0e196f7665af597329cfa8e0c950024e2de763e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0e196f7665af597329cfa8e0c950024e2de763e"}]},"branch":"refs/heads/master"},"b036d5a18243f46dac512f59170ef62513f425a4":{"kind":"REWORK","_number":53,"created":"2025-01-06 12:03:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/53"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-06 12:03:30.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b036d5a18243f46dac512f59170ef62513f425a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b036d5a18243f46dac512f59170ef62513f425a4"}]},"branch":"refs/heads/master"},"a582a8d0d86a01ed56cabe5f503e8ad9e8698c39":{"kind":"REWORK","_number":54,"created":"2025-01-06 12:04:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/54"}}},"commit":{"parents":[{"commit":"5198026eedd990cf3ad614736f591c20466dd80e","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5198026eedd990cf3ad614736f591c20466dd80e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-06 12:04:52.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a582a8d0d86a01ed56cabe5f503e8ad9e8698c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a582a8d0d86a01ed56cabe5f503e8ad9e8698c39"}]},"branch":"refs/heads/master"},"312406aff01ed2ee9e69a59973102e58130918a0":{"kind":"REWORK","_number":55,"created":"2025-01-17 17:11:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/909686/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/55"}}},"commit":{"parents":[{"commit":"a6ba61f46282b12ff04d1dd5190595f1216c4693","subject":"Use kayobe_container only when service is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6ba61f46282b12ff04d1dd5190595f1216c4693"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-17 17:11:26.000000000","tz":60},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/312406aff01ed2ee9e69a59973102e58130918a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/312406aff01ed2ee9e69a59973102e58130918a0"}]},"branch":"refs/heads/master"},"d1f81cff8a94c58ac7979eaa462dff58baa3e4b7":{"kind":"REWORK","_number":56,"created":"2025-01-24 09:49:59.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/86/909686/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/56"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-24 09:49:23.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1f81cff8a94c58ac7979eaa462dff58baa3e4b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1f81cff8a94c58ac7979eaa462dff58baa3e4b7"}]},"branch":"refs/heads/master"},"be86554cfd6c3ce28e9fc21f09a30c135732fe74":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2025-01-27 11:40:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/86/909686/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/57"}}},"commit":{"parents":[{"commit":"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a","subject":"Fix Depends-On","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-27 11:40:39.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be86554cfd6c3ce28e9fc21f09a30c135732fe74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be86554cfd6c3ce28e9fc21f09a30c135732fe74"}]},"branch":"refs/heads/master","description":"Rebase"},"988a822259a375bb404fef7b435a74511379c895":{"kind":"REWORK","_number":58,"created":"2025-01-27 16:42:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/909686/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/909686/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/909686/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/909686/58"}}},"commit":{"parents":[{"commit":"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a","subject":"Fix Depends-On","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-20 11:37:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-27 16:42:33.000000000","tz":0},"subject":"Add podman support","message":"Add podman support\n\nAdds support for Podman as an alternative container engine. This builds\non the support added in kolla-ansible in the 2023.2 cycle.\n\nChange-Id: I2c6befbdda7e684228065103feea7250a0ea3826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/988a822259a375bb404fef7b435a74511379c895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/988a822259a375bb404fef7b435a74511379c895"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
