)]}'
{"id":"openstack%2Fkayobe~909911","triplet_id":"openstack%2Fkayobe~master~Ia3fb27d1df2baee78b14f3cd9730c7f3a9daf2ac","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-03-01 09:23:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia3fb27d1df2baee78b14f3cd9730c7f3a9daf2ac","subject":"Fix issue registering inspection rules","status":"ABANDONED","created":"2024-02-22 16:14:59.000000000","updated":"2024-03-01 09:23:04.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"705dce604d4b276157375fc2fad9c66e1b210317","_number":909911,"virtual_id_number":909911,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-27 18:36:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2024-02-22 17:42:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-01 03:22:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-02-28 09:50:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 16:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 17:42:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-28 09:50:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-03-01 03:22:04.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-03-01 08:31:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"5671a8bbd288c596d095dba34d08ccd5a5568969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 16:14:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fea86446c35dad06d0def67cb3c6d7ded759f4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:15:39.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  while parsing a block mapping\n    in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2, column 3\n  did not find expected key\n    in \"openstack/kayobe/zuul.d/project.yaml@master\", line 15, column 4","accounts_in_message":[],"_revision_number":1},{"id":"b5d79d6c91c112589a18696a2ec29ee5e8346245","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 16:17:40.000000000","message":"Abandoned\n\nForgot I can\u0027t depend on kolla-images","accounts_in_message":[],"_revision_number":1},{"id":"32b07d622b0fd9f7cb0c68ee82d30d4b695373fe","tag":"autogenerated:gerrit:restore","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 17:21:45.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"fb8a383119631f2092349e318fcbde9049d99a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-22 17:21:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2de75bb90d82816f27f0a99743eeafc9a9a3a910","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-22 17:42:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8e40a8f234dabe84c4572bf4cfefde8b6fe7bb6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:08:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82a5294f0a9c4118ae20ee0cf7b836b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6cf6ce4daa4634b66df5071e3026bd : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aad8f4bda0441a6b242d290ea02ab8f : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/294fcf0316dc4094a3a3ff13d62939af : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f15856950cff4249b2e9d059a93a9126 : FAILURE in 12m 24s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b431ae5ef9d5428a8cf4546957d46d3c : SUCCESS in 10m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4319b94828fb45379c2d6184627aa6c9 : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/536e6fb96a4846b2996d82193ace3cae : SUCCESS in 1h 10m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dad82dce975c4c50900a71d250f5754a : SUCCESS in 57m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5da35adb017c48c9903f160f305dd1c1 : SUCCESS in 1h 19m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/28725474c7194ab5893f846b8dc8d504 : SUCCESS in 19m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f37d19487a8d4c588f2265292c76033e : SUCCESS in 17m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2ac64ae878f4dcdab3b41699008c5de : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08551b18bb314479a8df130b0432e770 : FAILURE in 2h 32m 23s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a21375b7f24246a6ad99bb09f73eca0d : SUCCESS in 2h 04m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e632b31d8fd34a8986196e4fee5430e5 : FAILURE in 15m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2937de5e04af4e0eb51b7aa2b5e6fbd6 : FAILURE in 15m 19s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f484bf95be1948b3b87f60ba6db7a84f : FAILURE in 18m 01s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/54157e884e2b46268d272ca1cab070a3 : FAILURE in 26m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9885aaa880bc4ad59c2e52cfdcbd68fe : FAILURE in 23m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f2ea3732fb143319fb198cb2cf6997e : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79f316201fed4ef0a8c97a505434f312 : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3bde85d5e2a94aa786759cc63993ee65 : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51627e37449047eeb2df7457c8c4b71e : SUCCESS in 14m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4517f9087234a88bbeefe4c2920ffcd : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8e0f15ecdfb43ec97a9e8567c19ca60 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5b9451fb4004a83afc23bd0f04246b7 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"454f9d3e27b5c9b50436bfac4552768da4f6180c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-27 16:04:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b58a565e89468faa638ce5fb4dfc59b5a8fe073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 18:36:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607c5adf873541e0b50d54bd998b9330\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6a1fd1bffec46ff9e0d98fe5e30bdb3 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e147ab586a4493f95e6817081788e5a : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b603fd2bcc2d416db7947e424a9fd09c : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e148d233f1c84b52a14882df7995de01 : FAILURE in 11m 00s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4387228d4df4698a8e2057e141e5a44 : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2335a3ead25547b5bfd1e51ba90118eb : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/44dc45562d8344af99004c98ecb1481b : SUCCESS in 1h 10m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9261507009444cbd9754f32cdf628f48 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/211c90b1ce574511b6f3c6bdadd2cf40 : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d40ee9cd4fbf4d2c8509933e99ca5509 : SUCCESS in 19m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f5c136db7e69486c9aa999c3cc54ef89 : SUCCESS in 20m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e0bbb4e06ea4fc8ab6319ef8db2d24f : SUCCESS in 18m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/79386dc6cb4642a1b0b1ef1ae711fd79 : SUCCESS in 2h 24m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab8a2dedb0a9411a8f1ec3c910967aa4 : SUCCESS in 2h 21m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/35fd1e45586b494e9295ae31483d01f8 : SUCCESS in 17m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f99288f89e684b298177badb9ad8bae5 : SUCCESS in 14m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0f8c985f8f26434cbd6f37b6293360b5 : FAILURE in 41m 26s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7df66d10dae74f6c9f4ba07676406492 : SUCCESS in 32m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/577e138bc1d043dba98320edaf59043c : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a245ac077b5d44ecbea6de9cf8d7b19c : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/185afa88f9ca47a6b1c2fd62d7d31b24 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ac8ea59ebd294f19a9dac40c62c1af2d : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb8a1409b11e40a08a6625a03a4b4ae3 : SUCCESS in 14m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ee3888868a47ffb05390ad14607396 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e2bc88362d4a778622b468d5c80616 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17207bc63d014b0ab1ac6bf644c4db46 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e54ba32c5c257fc8a469660a869709a7a698dfb1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-28 09:50:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee99e940184b3c2284221c58250b65250172ee8d","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-03-01 03:22:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ab4c59d80e641b7be360978a5a3057e9798f9c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-01 08:31:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"705dce604d4b276157375fc2fad9c66e1b210317","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-01 09:23:04.000000000","message":"Abandoned\n\nDefault changed to true in kolla-ansible.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28520a0e2a286f5cdfb56a6e1cc3830bb95f5e03","revisions":{"981ab470ed95284490ef0ecdc85bd67ae26f46ba":{"kind":"REWORK","_number":1,"created":"2024-02-22 16:14:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/11/909911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/909911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/909911/1"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 16:10:50.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 16:14:40.000000000","tz":0},"subject":"DNM: Test bifrost fixes","message":"DNM: Test bifrost fixes\n\nWaiting for a few fixes to land.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/909910\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/909902\nChange-Id: Ia3fb27d1df2baee78b14f3cd9730c7f3a9daf2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981ab470ed95284490ef0ecdc85bd67ae26f46ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981ab470ed95284490ef0ecdc85bd67ae26f46ba"}]},"branch":"refs/heads/master"},"28520a0e2a286f5cdfb56a6e1cc3830bb95f5e03":{"kind":"REWORK","_number":2,"created":"2024-02-22 17:21:56.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/11/909911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/909911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/909911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/909911/2"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 16:10:50.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-22 17:21:19.000000000","tz":0},"subject":"Fix issue registering inspection rules","message":"Fix issue registering inspection rules\n\nUses support added to kolla-ansible to enable the legacy\nironic-inspector service.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/909910\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/909902\nCloses-Bug: #2054685\nChange-Id: Ia3fb27d1df2baee78b14f3cd9730c7f3a9daf2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28520a0e2a286f5cdfb56a6e1cc3830bb95f5e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28520a0e2a286f5cdfb56a6e1cc3830bb95f5e03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
