)]}'
{"id":"openstack%2Fkayobe~910509","triplet_id":"openstack%2Fkayobe~master~Ie788829432ccf1691f200e5b319da16e0653da46","project":"openstack/kayobe","branch":"master","topic":"continue-on-unreachable","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-10 16:39:20.000000000","reason":"A robot voted negatively on a label"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2026-09-10 16:39:20.000000000","reason":"A robot voted negatively on a label"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2026-06-08 15:20:51.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie788829432ccf1691f200e5b319da16e0653da46","subject":"Allow to continue when some hosts are unreachable - take 2","status":"NEW","created":"2024-02-28 13:58:52.000000000","updated":"2026-09-10 16:39:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dfe391eaf232be53bb1eb094dd389b99f97195c6","_number":910509,"virtual_id_number":910509,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-10 16:39:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-28 16:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 17:31:02.000000000","updated_by":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"reviewer":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-11 15:34:43.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-23 13:06:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-04-28 08:21:06.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"CC"},{"updated":"2026-06-10 11:11:58.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"CC"},{"updated":"2026-08-27 16:19:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"}],"messages":[{"id":"ce5a47e44e8b7988f60a0119d0d19d8107a6733e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-28 13:58:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7027281b760c2341647539136ca8b2fff4e6a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:38:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f82689c0a9a4e7d8e70348439572452\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b16d21d766ed47dcb88f8c491605b58a : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29b803632b341f4bfef9d7b73d84be4 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af21fa18061444fb480e5d5c07cbf27 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8cc9585861554a418644e2a77691cb88 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2a2f56a6c3b4bd794bf789e381ab116 : FAILURE in 11m 10s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4208ffaef5d242698151273841f07b07 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e332c09de499415bb3a1d496e375555f : SUCCESS in 56m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/43de421b2fbc4d30983f7a468a2678c4 : SUCCESS in 57m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b04517153fac4073b1e1be4b820667b8 : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/91274b6440c24830ac9fc93a9941d567 : SUCCESS in 1h 14m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d76d1475fa7e41d39e0399210c985b7b : SUCCESS in 16m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/55a3dc47f4d24d0e8d3b4af0676f9990 : SUCCESS in 17m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c26e035c9f0943f5918ddd4e812238c8 : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/20ac37db10d9453b894f6382bd92e88d : SUCCESS in 2h 33m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a038a41b60148c49271f7ee70b6bdab : SUCCESS in 2h 17m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b1c81dcebaab44eb9c62decf08e6e5d1 : FAILURE in 18m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e720e8459c94d739fdd2e1b38c53e2d : FAILURE in 16m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6ae95cf04737493ca746090566b5ad14 : FAILURE in 19m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b975390c692a45f1bc8c987c4808603a : FAILURE in 33m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7806bf91edf4acebebe8c99cab8a8d2 : FAILURE in 22m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/56126c363bff48ed9c217ab31262fdf7 : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9abec56540f942bc89ff1131ac92987e : SUCCESS in 13m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8cd241fc43fc4c86a773412e78ec02ff : SUCCESS in 19m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17be7d5c4645447181f03f45c92483fd : SUCCESS in 16m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1d41b993da44ad88b09a4e29690a3f : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3de343923a384915838ef8cc183dbe74 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc5ecc967a42446f8c464ee6310b52f1 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"508dcdb0fe8936c617d228ec01e38c4664305553","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-28 17:06:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"958786351d5150382add8eff6901e30053a48d51","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-29 10:59:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"18876ca1d89893d60a4bce3869d3b2dc2995f2f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-02-29 15:51:29.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"70d5e02ccf345fb29e4f2c6a8b5563486c2477ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-06 17:04:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"152b9b483fac9a0bdc6c88aedf060547eae0e477","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-02-24 18:51:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f37215d16824c7f0c78db1e9ec701e3aae645c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 21:21:41.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c941a94d757e4d04b3d813186dbb2468\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a378753dd4804cf9abf910193ba903a1 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46c90c1a4474672b64220920dc4dd5e : FAILURE in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8729bd6e9d74a5085af618d3cf8c943 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94eaab762ca4292964b435c38cf5c40 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e4858e9087445dad764312e337e4dc : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07e27a14a63a44178c20b72a7e55fed6 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7147fb8ff5d4963bc7be44ef9e86acf : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e4b7a0fda6f4f11affbd34ac4a40eaf : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e1d5a5725614bba966d3ff8c086071d : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1524f20e438a45918ad2d195a9acc6ee : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/16a63fbc84d84f63b352f736e1879d96 : FAILURE in 9m 15s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e592398fe59b46d392ee96b6cf0925e6 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bfa5a9ad5ed747fca1448dd05000479b : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9134c964dc564b9483c71c7067acebcc : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/85626b4b2e004659b3131380653f0808 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0cda74e0b6cd4dbcaf669725966dbf6f : FAILURE in 10m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4274f0ddac4a4e4c97d1b7f7b124b03a : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a4db05d0a1b64418bfc347886bc61677 : SUCCESS in 1h 44m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/999c72a8edce442f873a45d3e1a79e56 : SUCCESS in 1h 57m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/76579e7f5ca34197a42ec4c346d7ff0a : FAILURE in 2h 22m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/10cfa64fd6dc4dc38a53a7db3c305762 : SUCCESS in 21m 20s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a081a84a461743c08589e1df2c90d409 : SUCCESS in 21m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/557d537e79b84cfc83cdaca970ac93fe : SUCCESS in 22m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc334c24f1004887b257a93b9893700b : SUCCESS in 20m 54s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f8bd20217e434836a6fbcc8c7f0b316c : SUCCESS in 39m 33s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c45f2909ac24e63bff1a26927b48a8f : SUCCESS in 1h 06m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c4284932703141d48f4d0172779604d7 : SUCCESS in 35m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b21112872274429faf7d273678348d9f : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/536f0de5d9fe48b9bc6c579ca00f2c44 : SUCCESS in 31m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/674e8f8a36ed4232bfff27530e2a8812 : SUCCESS in 12m 50s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cd0d471213254ccd86aab60273263b7d : SUCCESS in 12m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03f1df87083c4901aa6693e9b4faab11 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/67beae596f374212b0c5aa663a4a5ab0 : SUCCESS in 16m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b1b0327bce9494ea55d327da3582ab9 : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/210cc58214f8450c9e94666de610415d : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":6},{"id":"de786e07875d8825bffacbdd3f33dc959520d98e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-02-25 10:23:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d6928d813aa53b0462d8ee41d281d32a17af395d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 13:11:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a388ac7e3e8a4133a040bbe0e4425d8f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f5dcaf18bb64442bc2222496313e7f1 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bfdaae87f1412b8fdbb0e28c7ef6bc : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6eaa2617efa4fd5bfe71aaac4f42652 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db5c19ac3044ef2b3aa154af88afad6 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f61eb4a66dc44eec8ea65162bc37a674 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29ce56c3d8c44e51b8083edf5178c9ec : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8c33f285f434b16a6e29058f3c6c6e6 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a53d2db31dc4cbab6b293f5fb410ef3 : SUCCESS in 9m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1a21fddcbe8743799f90eec9e5cfd996 : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/81b69527fa3e482284114bb40327ddd5 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7607eee12a374c68b2732a654a0c12a1 : FAILURE in 10m 46s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b882f85c23e64b19bdb9183d79a11d5e : SUCCESS in 55m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16b6b795f58b497e890964dfe8ee12a0 : SUCCESS in 57m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0707b6dae2fb4613b1b308423091bcfd : SUCCESS in 1h 15m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/186c74590707478d9a1341ed4477b2b8 : SUCCESS in 20m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9a35ec82a83643778b2ac87fa7b5fb8f : FAILURE in 9m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d339b06b234daebd8bfe63ed91d447 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/db60daae0647471ba51aa52c9ab59866 : SUCCESS in 2h 04m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/195fef9c5f49428daf007cacc6a410d7 : SUCCESS in 1h 59m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5e585ceec702431ea9cf1d8e06910eca : FAILURE in 2h 41m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6667fd2673ef420f951d67b5ce57b9d9 : SUCCESS in 24m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a965bb143e594539ba61f94dc1ce0bce : FAILURE in 19m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a818da600c584401970c5ccd57ffcb57 : SUCCESS in 25m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efd1df8ff8c1464a85ca97dd83be6660 : SUCCESS in 22m 51s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1e3fcb61f4554ba3a78e869211c3627f : SUCCESS in 43m 34s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1834d6744fe47c6b1aca96b378b4065 : FAILURE in 14m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a90dac2394404a0a8f457a67b5a41cd2 : SUCCESS in 43m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bb1969871294d5696ce809d608c289d : SUCCESS in 33m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a204c3dc291f4143b67a6bf04c1be3ec : SUCCESS in 34m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b916ecc8866c4bacabfe5fdbbb7574fe : SUCCESS in 11m 34s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/72f6a749558748419163239c71d19969 : SUCCESS in 17m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f07a75eb9f40febbc389e6ea8216d4 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/066d1f5e95b546e88aa95250dbd2dbb1 : SUCCESS in 18m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2bd5f6ae93b64dbdbe75cc84b9ab7785 : SUCCESS in 18m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8234d5257d84409dacefd6c3fcadee6f : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":7},{"id":"61826a788b41d9b2029bd384423db0bdb68aeb4d","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-02-28 17:33:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e38833a0f791c32086260fa3fd476b16f472ac8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 20:01:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ceb1a4efce74a259cdbf24398f894b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe9c90c818642c687aadb249e91889a : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25de0e557ed742b5ab45f298db8cbbde : SUCCESS in 6m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69d43d480c8444f387faed9c5b5f83cc : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0440bc0099d74c628c93f4798e883dd9 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c00161e69f64580baa67cb268e4741f : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c0eee1b47c44040ac65f93343b0e95f : SUCCESS in 7m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2a70e18c786453a8c3c6ffe91b5df9c : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2655a37dde7c4179ab9094213a9362a1 : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/28151c1772e449b7b7361ab2edb42e41 : SUCCESS in 1h 57m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3493e6d1f5d54cdf9326491347d7e7db : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d28f4ddd44314a188bb7950048c81313 : FAILURE in 19m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5cb4465ad09642349d65adc0e36e3fd3 : SUCCESS in 58m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/16a58599debe45f0b2193dc98ed3f6f0 : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b6ee747f80094f89ad0c484043f52165 : SUCCESS in 25m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff1fd5d88e8c49518afecaf84f86a50f : SUCCESS in 17m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a027abbd1714d89ae303b8586b983e8 : SUCCESS in 2h 11m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d6b0c0bf81e4fcdaff6569cac2fa615 : SUCCESS in 1h 45m 26s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c938357d7afc436bb0543105b0b0d5c1 : SUCCESS in 1h 48m 14s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/862fc5ed56b14c2185850ca629b54b86 : SUCCESS in 21m 04s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/dce1cefaf69b4fd880f14438b0ae9cce : SUCCESS in 22m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13199b9a1a8c4b8da3cc695070341859 : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dffa52109c6747aaba983267aed2d278 : SUCCESS in 21m 35s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0f10772876114f6ca9e71d3597425b40 : SUCCESS in 48m 18s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/77d670127ff6447db0fd886248c40d01 : SUCCESS in 1h 30m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/489bf7c0ff3549c698dbf2b46a1cea34 : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4af4f85fe1547b89048b58a08c65d10 : SUCCESS in 39m 50s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/77a4933d5cf542da85531c5400e0c9de : SUCCESS in 32m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76a4308ab41b47768c21fd6d6481cd3b : SUCCESS in 17m 18s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e25681e28fbf49e69854f2232ba810d9 : SUCCESS in 19m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c953e65e631243e0833dab83420f90b4 : SUCCESS in 18m 10s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/51f30db039344d669e6f117a917eea11 : SUCCESS in 17m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0fa872cf49c84f97ad1a6edc79f91251 : SUCCESS in 19m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7851923a7f2f4a8db43c51d7248873cb : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":7},{"id":"91b69f53d68e24eff742634265d099121c10e6de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-11 15:34:43.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d8a32ffe8b081cd6ffa4452d036ed064615fcf3d","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-11 15:36:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a4ff18b91ae05602fa9f91f71d5f6737ded82b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:13:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00f641fe3d534897a8d921a0c75e599e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b69ea30c984489aa180f5f7bc585c8 : SUCCESS in 5m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea6d2d8cc93e467588d3bb1e72e9fd74 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be0fadd7a65d49119fc1031295d66cad : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f2a3a76248446719e711be24a03df49 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd6da95f24446ac9f35d464e7bc4d17 : FAILURE in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f867795e0a2b4e2e9925bf68d9e2f1de : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47fed55a9d9147ffbb9d99b62796804f : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f8b78598fea471384d9cd4e0a947dd6 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2438aef1bbba4cd880d4331968c2d15c : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1626290a691d4228aff5c967e7cb27b4 : SUCCESS in 14m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fccd98d019a24d90b77c99214d0e4406 : SUCCESS in 13m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8b414d2c6864e3d96057d14cff70848 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f63e665a8445a6937d324f7eabb288 : SUCCESS in 15m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7a4a0f0f856241f3983b3b218593fe2c : SUCCESS in 53m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/96c0e670e77d40418b6b7fb455e0170e : SUCCESS in 57m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4472cd4d54df4e09a49d686a310e88a6 : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00dd2df81421444e8feda0f6d916f80b : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dde60434e56741f59243153441cf9617 : SUCCESS in 58m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/abcd2e3adced44e79e08d7248fad7c78 : SUCCESS in 1h 30m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bcb95a6d7a5f48deae762878d4515308 : SUCCESS in 1h 32m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5598efe4f6604feb847bee6096e04dd5 : SUCCESS in 1h 35m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1608e46a5684fbe895c2d456f7c7056 : SUCCESS in 1h 37m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0adea82fbd244a4f95cb7f05eb98fdc9 : SUCCESS in 15m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/930cff472f8246bb8ffe09969c79dd94 : SUCCESS in 10m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f7e754478ce4d02ab6fe1bf3e99dfa3 : SUCCESS in 17m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c556b6edd2324c2293d01febfb7a9d34 : SUCCESS in 15m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4dce80dc512b43f1a6a8a5ecb00bafdc : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/95e25b3aa824452984bdf98e5dea10e7 : SUCCESS in 38m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35830677888e4549bd32a4237e127513 : SUCCESS in 32m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cff431997c24310b13d4fb7077d52bd : SUCCESS in 28m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dea6c4619fc846f9a196beac7fc693e9 : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9713a9b6f5c4a3ea021d131bca3cb4c : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":8},{"id":"d1b9e339a42f8d3896c7c87980a28c2fd3fd9ffb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-01 09:01:36.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"2a2957cbddcc9af7cd43053b77c572cdf101caa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:39:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f21c1a770fbd43ebb71bbd0a3055540f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7cffcdd2cf41a989fcbd57eef17f69 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b651e0ff85c041d28576b46413910587 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d6752ee18f46d69a498a18ba98de65 : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232aed8a07a14ef18853de9d2acb0a0f : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e19aee15cabf439191688097dc5de202 : FAILURE in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/828a7c2e12434deda4a907753f0b5bdf : FAILURE in 6m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc0cd468a7364ab58913b1d55d456964 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b64a9dc1c6d49138d075527a2636cfb : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e425f563f41444949e3be278c5793933 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2226583b2bd144d5a0ff529cd1cf0d73 : SUCCESS in 14m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5c029e0d540f4b01b78fa3861566bde2 : SUCCESS in 13m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ef39578b41f4664b47596350cae495b : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ef1faaeae4ce47eb9ec23d248c6dbef1 : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60efba55f38940e2a8e67fc467ece443 : SUCCESS in 17m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/670b1f2f87544c4e8bfd84611ece0d25 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7da603a506414d86bea20e6bfa65a8aa : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/896028899c4246d6ba136424a3079ae5 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/987fc987af3b45488f05c19a8b1adf99 : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45cc1477bbbf466fb8cf5230fef0f257 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4238ca26eb8d44099ad582b32ac3b525 : SUCCESS in 1h 37m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6c7ff411f40643bf9f12f8605dcdcf14 : SUCCESS in 1h 31m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adb5f2952fc94f3fabf751c789b40bb3 : SUCCESS in 1h 26m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9e26ac8b8264a038e3d21a46a31679b : SUCCESS in 1h 36m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/732dfeb7b6744921a4535f2b1114acea : SUCCESS in 15m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b0c524d6cb4740d5a3b6f0fe371deeee : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90d3ca57bff14ffb9df9285525821281 : SUCCESS in 21m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a782dd096c234f98a3a7982456a7d97c : SUCCESS in 21m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c7976d3faf74787be34086201aeaeae : SUCCESS in 16m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/707777ed965f4504a93bb8b18651a559 : SUCCESS in 34m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d80f539870554ebd999e4823e787a7cc : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339506c12e1f42698c30dc00d866c5ad : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8d8e9a54441242e4b3b598c4808f91af : SUCCESS in 28m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d6c029d72574bbabad4196acb9f31a7 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":9},{"id":"539482dc80610071c9c7721264f22a8012f265de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-01 10:42:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"5a5870e4745dc35c251cb2014f5274c4459ceb33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 12:22:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5b72d02c4c4cee8c2d2a8492ae0971\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e244dcb47ca4a3996a4392ddf83aa78 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f2c02d9c0cd473a924c918ecb51c1af : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51dd954a3705455cae5fd664faee295b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c3098357024148bd9d94121171cc9e : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68055ff3ace548f395e09550aebfa6d0 : SUCCESS in 2m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fb024f4f9bb466c8baa2f42042e76fe : SUCCESS in 7m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/395bdc0d9a5a4257ae9866023b0217de : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9119da1d5e844ac4a6560ad59e4c3174 : SUCCESS in 3m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f52f6b8461441afa84ce0f046cc9341 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39d59acc22054fc7ab67e1da77b988b8 : SUCCESS in 13m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dfe4051c19cd4d99b091d5e9842b0f46 : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e18fbaaa5e394baf96f5b30188574a42 : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/905a1e52eec54948b5e00e9220f490e7 : SUCCESS in 14m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fdfbffaefdc4a23978857654f34813f : SUCCESS in 20m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9d92a5af175c4e1f851798e887d2165c : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fa82cbbb6f09477e807c66d46e60fbd1 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/96853f3f80b14a41894063b4f3f069cf : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8b6d11b2de44740b056afcd9a741ffc : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/721f919d92a141cf8b960140704ffddb : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2f0ec6ab175c4feca90f296438386eb5 : SUCCESS in 1h 36m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/06b07d2620bc48a5b900f119551b6bfb : SUCCESS in 1h 38m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4bc6176365f414b8b6f074af981be82 : SUCCESS in 1h 29m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc8b10ca6e394e2fac60080066ea359a : SUCCESS in 1h 34m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1ea8f0b396b94d4eac99d350dcd4b2ec : SUCCESS in 13m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/86297070d71b4e799406063141c96a81 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1725fc1d24c44f689bed7d01be4d4cce : SUCCESS in 22m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/996c23de7d69410c9cf6853c558062b9 : SUCCESS in 8m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d052eb510d0c4494857df38052532892 : SUCCESS in 15m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7ec7522e3d2f4f0ba97d95843d5eadc4 : SUCCESS in 17m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed43bed94a374ce6a394dc77c745366c : SUCCESS in 28m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de62a3492c74f7699774a13d95d67f0 : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca5f18898db2418a8ce8a7017f9a0a3f : SUCCESS in 21m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f46720ad28954533ab663982865c1d8b : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":10},{"id":"1ad4dcc4a01b85bd906910ab72a126e0a588c033","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-01 14:14:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"d4d32557919aa116b68614d338fb1b976a931ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 15:48:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6d24a905ba4817b6d23648a1902b60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922038336e7f434681de744a15a83434 : SUCCESS in 8m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a1a902e072604c539065c87374fbe017 : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e6b8b20a624205bf526a4aab4ce708 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbc985c7f07449490f00edcba45b9d7 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0e25f9140524d309c097f8abc131934 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/564c179e794b4299847adb8ba2433417 : SUCCESS in 4m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc739a7bb4a04f0dbaaddfda69c0f32c : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/42b563fae4834e5fba5d7958aa4da9f1 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/241a913850d04f98a76c41b70d9b3139 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1fb392aa9144b67afff9525739dc6a0 : SUCCESS in 13m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/18f52c6c4dcb4b649e1eac3e0bd69e8b : SUCCESS in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce8bffa343c4b988d304a2cc1a1d222 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/414a8a389fac45a486287382d8d3bd16 : SUCCESS in 10m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/313d75e97f1c415c982b9b399bee9ffd : SUCCESS in 11m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ba3bf0315b1a46cfab849d56e392f4c0 : SUCCESS in 59m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7eb648e3c6754168a482196e26f31783 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/21de774a89a047cf8656f0b89b652b04 : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/648cfd52d2624aa0aa1ac1a30be4e519 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5ee73e48bb74ec5a769026514bee5a0 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2de5522b813f486a981387ebbcb498b6 : SUCCESS in 1h 32m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c4bcd731171449efadaaea18d331a48d : SUCCESS in 1h 28m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/983e63d3447b4503a62ad4d6512b5e2d : SUCCESS in 1h 27m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5113cbf58e904c3287b452ab850ad013 : SUCCESS in 1h 32m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/599445a00a7943a78b8f0f0cdd87d629 : SUCCESS in 28m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/407f298eb2464b03b43756394684fb41 : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3998ccaad9d84f6292f30fee30357575 : SUCCESS in 24m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87dbcfac478146e28c4db88423be0366 : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c72639c83c9441e1a1ff11bdd39abe48 : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6e27e46202324904b30ea013fc141cd7 : SUCCESS in 36m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34f59f1dfa8b4060af95f327fc59d2ed : SUCCESS in 31m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c02cd7c8a25a40e79561fa3cd86a13fa : SUCCESS in 29m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0dc86056c7484c79ad71ad59fc4b4761 : SUCCESS in 25m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cedeebe34d1a4447bb00204ecdcb45d6 : SUCCESS in 20m 14s","accounts_in_message":[],"_revision_number":11},{"id":"e4392705ee80c23a4698ff6e64bb958e47dbde72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-11 07:26:23.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"56c3b5ac4b669edbcb4d5b8a8576abad0dcd722e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 08:30:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/03f468020d8f44e88f38d6098bd5acd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0830aa2ade41eea767614d2790cfb1 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9edc7a8d3d743f8abe0741d4f84be06 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4779ce8a05264d579f78dc6281fd38a6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac08352db8a44f5a8768cfdfaa930872 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/855a7ef37e5e4854a383ae415f8c5e7e : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e51813339e04f1294a8991ba7952ec5 : SUCCESS in 8m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b25398b3fe342ca90e39d6b92f0629e : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8391247ce21344afb41bb986d8a87dfe : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f403dd4771a4752bc5618f4b6895a63 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63ced5ec8b5e4155a042ce55fe022d09 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b90fa07b7f7a4f0aa2cb909570c221b4 : SUCCESS in 14m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/697f3005425543d6867ef3ab901edfdb : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/286fc99db21e476fa3ae511d8e24dad6 : SUCCESS in 24m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e59e34b213bb4200b65da5bd6ebbf9be : SUCCESS in 16m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6582a688b0894f19adc00b7bf16422d8 : FAILURE in 21m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb7bfc5fbcd04410b0711648a7f94cbe : FAILURE in 21m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0b120ccd200a48a3a22916e88e455110 : FAILURE in 21m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a9d5bdc34174aac8b3010ea6a0cbe43 : FAILURE in 21m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2433b42dbb3048c18625899d53f125cb : FAILURE in 19m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f351b2b3695f4d75bc4de7792a573947 : FAILURE in 1h 02m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bad3e91f605942a9b90ee40f103f74f7 : FAILURE in 56m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/640c3a7e8f98424aa0906951b9d210fa : FAILURE in 54m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/734cdf7898b84db4bcb08ec086be309c : FAILURE in 57m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/76639aa96379467984740867c08df0c4 : SUCCESS in 17m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/108263013f7a4724b05db4e546254638 : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/411ef81a49c444c689a35a8431691877 : SUCCESS in 13m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b937e03ac9954cfdb2fe012729ac6299 : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a79fdbba8f964a40abb780e462d9e89e : SUCCESS in 15m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/95bae713e8874c5d8d1a708d822a2e1d : SUCCESS in 32m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b8bcce4967c42deab1947b54fa50da7 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05a6dfc476347c1855378a16b0b7476 : SUCCESS in 14m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f526ea270afc4d47b1771426f5d4f149 : SUCCESS in 16m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/992bd0374570430a8a945441356fc797 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":12},{"id":"fb37adc6014ab5b0e36fc924f52f2c63ee120642","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-11 12:25:47.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"cfb5af1d057453a9fef598c5da12f451dfed4c5a","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-11 13:54:17.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b9d47fb637862d2a7ae5583bc2c0f93fd229f8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:12:31.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9621c1b3f53a4bc5a2209430ef6b176a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5173bf1fb1b24782a3ff4688fc2ce4d7 : SUCCESS in 5m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa8be35e2bff42a6817c2dd7a5eb4781 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edca8ccd15534e819bd8d9793ed41f57 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5922e17f90064a94a2c291b9baa3a583 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/198610ec7f2046aabc311d7783766aef : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cb2c4bb0635405ca3f644230e888eae : SUCCESS in 5m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54378c3baca54dac918b24a262304a34 : SUCCESS in 1m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cdfb5bb20aa4123b5dd4037aebced94 : SUCCESS in 3m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ab73ed1feb24181ad51c5d870061e82 : SUCCESS in 6m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/056026095f2c4334942ad0c894bc48b0 : SUCCESS in 6m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf8ed657516d4786999353c08a2e6357 : SUCCESS in 18m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bb5bd77a3d1476683ba0740da373068 : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6e0c1255f7a54d5d99a126d759b7b01b : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d68b0432106c42888ece7940dea8a9ae : SUCCESS in 16m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1a35d3d8fea342e0850ee4dc5a21382c : SUCCESS in 58m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/42b6a85074e0472494fd3d661823e0b8 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/79c287efe0104d8baa829e23754f87a2 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd40df9accb9447e91bec1882f1b7ef3 : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed3e3c0edc574f02a7b9a9f4ab5ea27b : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3f32d30a04fa4c0cb2f3ad259d3e3be8 : SUCCESS in 1h 39m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ca1a9807ea6443f098af6fef2a82f3e6 : SUCCESS in 1h 32m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/764b50cbe2ea40a989cc52de55fcc734 : SUCCESS in 1h 28m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/347f89edb12141c09773000d3e2d32e8 : SUCCESS in 1h 45m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8f474e2410e040ce909bbe5802334a26 : SUCCESS in 16m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2fbd6d597554719bb17f5c79e2f3328 : SUCCESS in 21m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e39348a10554da7a5a0ee02685588b9 : SUCCESS in 23m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/212a803eac66433cb1c8ee74b6500e18 : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d3fde1b1c7774c0cb867d66f7883dfa8 : SUCCESS in 26m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/42f39c7977ba4f49913ca49c0ab8f161 : SUCCESS in 37m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13773f38470d433f8c733dd7b441b7c9 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b81baca856d4550bd5bdc83df6dd885 : SUCCESS in 27m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf9b18a8236d4aa9be05e121bd179cd9 : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77b79fb9ee0f40a9ac4860c2ab7d030a : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":13},{"id":"35d96de49a72be8acfe82ef420553c3650c1f448","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-15 13:14:59.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"82c4d8fbb9363b36c578ca77d4ec145caf3d3694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 15:04:32.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd2d557417bc4f6ab945e59b941a5ada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20419810adb43e6beabe18b0d32e8e6 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/807c1c1852a24521800a4ff8538c0f4a : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/359987e76543470f82c1a94e1d8bc2cf : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b6a4817f8d49e7b804f8d9beaee502 : SUCCESS in 6m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67383ab43ce5469bbf7e37294d49a31d : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3cf2c7ef056d4c69b6d319c361000d28 : SUCCESS in 8m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dfca406d60949ae95a198d2ebcacc47 : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67c95c2e426140468c13cfbb3e62580c : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c932e734bfdc48c6aa3f11535b041f9f : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e39f7b5ea974f56a28d72a94180b518 : SUCCESS in 13m 29s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4c01cd50a7ad4aca8af6e2000d5c380b : SUCCESS in 19m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cc36f31d8e34e04bca827f13a48dbfb : SUCCESS in 15m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c0d66f6d87e3408382a16ea064ffd003 : SUCCESS in 20m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76220369993e4171b9f308ffcf989724 : SUCCESS in 21m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e236cc70914a45509f509a8bbbe78001 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/843b0d8c760045b49bdc7570ada49fad : SUCCESS in 18m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e8ed873943eb455097671a0806f82514 : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9566bcf8c180490f8f7e9cd385f82dc6 : SUCCESS in 58m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/530541e78c054f00846824a1682040a8 : SUCCESS in 1h 06m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48bc4b2d634b4c42b7ad1ac6bb31cb7d : FAILURE in 1h 00m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7df802d6bf854e51baebddbbc9618d78 : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c59ade54e0b43bb9b5e1952ae0eec01 : SUCCESS in 1h 43m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6a872be63a774da1abce5bd6c35181f4 : SUCCESS in 1h 36m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85e54e867ff44bffb032b2ca8c2f36ff : SUCCESS in 1h 36m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8575db6439d943e597ba5d9ae7c783a9 : SUCCESS in 1h 45m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3d7e09d8ca804361b3d93eec2ae7b8a7 : SUCCESS in 22m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8785bae17186423a84901b62fef814b6 : SUCCESS in 15m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eecd8d1533824255bf57495267a77357 : SUCCESS in 26m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f800309411334136974efbb3e7030866 : SUCCESS in 21m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5646714bb6654314a339b8d76b2b909f : SUCCESS in 24m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4e89fc83c02e4fac94bdf1b335dd90f3 : SUCCESS in 37m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17ac2845df34431faa477a43996012ab : SUCCESS in 33m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b4e1f66c2614447a11dfe7bfaa22ffa : SUCCESS in 29m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44f2a0e35fc741efb7a7637c37b112fb : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/002a238f02464143b1d620c9a376d28f : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":14},{"id":"4b04a6b859bc6d99a953f4965cba96a263b6c50a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-21 14:46:03.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"e21f7a8a0bc5b3b01c72288452a93125510c6e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 16:22:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c2c9c4b5dc455d9eba390ac8dc8793\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2393098c6b954894aa423e1a0dcd9923 : SUCCESS in 10m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a6a6acaa8764331bb20749b066f199b : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afcf08f0f36d4090b4c835a8fa08e586 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34b82aee29148d8872fccfaacd8800e : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1f3eb02fa7e4f1d95ad0827eb597734 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/104c337fa7c44a5aa999864f02345b19 : SUCCESS in 8m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8fe77554eaa4253bc9e8f4125f3bc6a : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bb3f0e8f7b04670bb6c31a779b95267 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b118bf633434b74910f93e0357055c5 : SUCCESS in 5m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c8a8f30f5bd4fbbad3df7a69def6598 : SUCCESS in 12m 51s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4e54c9ebb190442f8a30c89f51601e1b : SUCCESS in 12m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cf55a89fdde409b9dec05c02eaf7966 : SUCCESS in 15m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ebc8b232778d44829e372a57394ad919 : SUCCESS in 14m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/918a8e06d5e844dd92776b7f5df6f63e : SUCCESS in 18m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fa9c56be1625426f999316dc23f243a4 : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9a699c4d7f842558e5f3b17db6ab358 : SUCCESS in 16m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3e2942abd77f4f4faff86444f608755b : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3077c052bf3a4cae8e467daa0a8bb41a : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/905ccd4c042543e08d06285bcc5a4c1b : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac66c46fc9954036be9cbd8e2a934881 : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b774f34f582a41cf990a1658a4ba01e4 : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/db689313b4f5449d85c31e0d5c949928 : SUCCESS in 1h 34m 41s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0ca32ff01dea4e2fbfd1fef5821e0e00 : SUCCESS in 1h 32m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d2e0cf3d12944baacf63a893f1c19d4 : SUCCESS in 1h 26m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4abf09b138684d12abff6cda949106a9 : SUCCESS in 1h 34m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0c335fe4569e45c6adaf578487773eaf : SUCCESS in 16m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e7cd0b956c1e4f5e870b88b4f2cb3572 : SUCCESS in 28m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf9ec8b343c348828806bb02690aed63 : SUCCESS in 22m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2474037bb9e546889c496e6ba802e390 : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dabf942b5b794c46a95619b7f3d635cc : SUCCESS in 35m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/875d14d5d6654f23a4de22f5ffbd270c : SUCCESS in 26m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d6334d0b6424b8aa568b86363798886 : SUCCESS in 29m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43c7707db58d4f9dad0e4ea9a413bde0 : SUCCESS in 39m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/848425aa631142a098afcbb685c31104 : SUCCESS in 27m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d30b2497c1c43ddadd3472d03c0eacf : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":15},{"id":"132e6d3c114351ce21897c9d9c04dea0c2062b13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-22 08:23:49.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"171c6567d00cb8b5dc68a33c47109c2302d22646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 10:04:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a3305bedcd4f3e85411754ffa3d823\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67f75073dc24a21b0da602ce1d5570a : SUCCESS in 11m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46714f5aeafc49be8f50037cffa96013 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee21b7cb00aa46dc82dfe054d79360db : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d093d0ff164635b79339ec68334243 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecd9f93499e04f2f8d291950eb9b515e : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd4eacd5ceca495e9828534e65295f59 : SUCCESS in 4m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b21bbda185a941ecb0d3aa4b7ae54b92 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d31d193f3dcc46009f5145d082c34dc8 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c41f3c5a71f04ce9b6d0ab9f415cecd3 : SUCCESS in 14m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8df1c9ae390487eab07a8036f7f1b62 : SUCCESS in 7m 29s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0bbfeae1044b4604918254d721c33ba9 : SUCCESS in 10m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/79ed47b1221d43eea983653257161cbc : SUCCESS in 11m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d13b9659abe04a5fb59c98564d50b887 : SUCCESS in 10m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c4db52559624cf3b8da8a5c25cbbaf0 : SUCCESS in 16m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d261d1b1780544e08810be787d4f53d4 : SUCCESS in 16m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27c6b9405842424092f43926bca1fd35 : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/155aef8a5ce84d1a8024c22dd3c205f9 : SUCCESS in 13m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bdc5d5c2d414852a959a6eb98a4b753 : SUCCESS in 11m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cb73875942f24af59a0e0ff6a1b20b86 : SUCCESS in 1h 00m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4b5dd2edc8a41018059213f934ff39b : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/20ab0790a9934ba2b673f93037f34781 : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7aa8b42632c43539f667702faacc314 : SUCCESS in 55m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b904ccb7f2141568c687ed247f604b0 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/962e93df5dae4ec09331c1268723f28b : SUCCESS in 1h 36m 36s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a7b964968f2844ff9f4cac5984df9c7a : SUCCESS in 1h 31m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/567121cdbbc24bf58b5915814b3df170 : SUCCESS in 1h 26m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc4815edff24e5ba10dbb60b74c589d : SUCCESS in 1h 36m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/971038ab97db4da6a7164f6301dbe0dc : SUCCESS in 22m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fd8447f9572e4af0acefde746913fd83 : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e178c76845444608638a00fc1290760 : SUCCESS in 14m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/830b5f4cb38f47a8acddbb7422d2a6f4 : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/331b2182de174fa592e8d5a9a70b5b8b : SUCCESS in 34m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8013192a3a27463a8611b5793401cce1 : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af980138c8b3411a8e2798a7692e2d2a : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19b8ed8d996248bd8d16672f70596113 : SUCCESS in 27m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ced7609e28954cb2a8a721b1bb29c232 : SUCCESS in 22m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba86e057e40e4b97929aaa7a3fd7dfdd : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":16},{"id":"ff12ed45ab77655ce4331f219d78d2cc40b18c9d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-06-08 15:20:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":16},{"id":"9207429a5339e1182058cd65fad0f6ef184c902b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-27 16:09:38.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"d77a777cbe852c44004dc9c76046b3789dbb7d0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-27 16:17:05.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"d610166ca0c2c942f88d6f7caca2603dcc46bb08","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-27 16:19:20.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a6fb4b01af0ee0842bba7f62ca42c5a4e52d6cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:20:43.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1829f173b544f2091eacdcf8998d5ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c1418ecb9e462cb98a2144a4fa15e2 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b1619270a6847bebf8cf0adf113578b : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bff4d4ab64864204b20764223f9ea135 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6251c873ec72469083065c49464baf3a : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b9b04f40792436b97a39df040db8029 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8c91133903241aba729b34e6d5d269e : SUCCESS in 4m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64322ff8f9af4908a2c696a1fbeecd70 : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/42243b076c9543d799f4296013cb9577 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be62705ccd654884b5fe0b7a4c42abb9 : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ac5fca9759c4486a1d3199fda0c0367 : SUCCESS in 14m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74060d11183848f0acec4a00143b2060 : SUCCESS in 21m 20s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f7783a159fe4ae48836261af9e6fe05 : SUCCESS in 19m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bee11bfce5e049d4a75209ce61fd4737 : SUCCESS in 16m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a48a834633447eca958957f7f57b481 : SUCCESS in 16m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3a33ab92bd40451fa894627fd5c3ea3f : SUCCESS in 14m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c14de0ed5c64fd091fc99c247681a28 : SUCCESS in 14m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0d00b2a0c2c3448a87a06c76b9b58500 : SUCCESS in 11m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f93a12f711e4386874a4a3d75be6098 : SUCCESS in 18m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3d96bff14aa94e18923549a81db0ad59 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/de3ac9db35324d08b89255504e988a64 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5b23bbede6c94ec1b1d0d1e269d78897 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9a0c8a1c0004a74a95d2b1ffed0519f : SUCCESS in 1h 54m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04f9d260097c44d58a537e96ba1c6a24 : SUCCESS in 1h 38m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/74812c3c86b04b0288efc8345fa15802 : SUCCESS in 1h 49m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f79e8fc55bb04a77abab0484131d7050 : SUCCESS in 3h 02m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1d2b55ee14b1421e888a55398b35070c : SUCCESS in 15m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea14ebc6d71e44d19da0bd31f95171a5 : SUCCESS in 10m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c6213c851c44e3aa7d468800c531a74 : SUCCESS in 23m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04b99b04f9b842c9a300f5ab3078793d : SUCCESS in 22m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e776fbaf105e42aea40b39b0d45c969b : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03494d2a7671425b8bcf597744310077 : SUCCESS in 35m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee3423c982aa486598e1b795521e626b : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd3bec122ca140859468c85a1bf037c7 : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":18},{"id":"837a751221225962e04eca88c64956166c715f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 14:34:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"dfe391eaf232be53bb1eb094dd389b99f97195c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 16:39:20.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fefa4b0a2377497f9c6446afd42c5e8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16a1ee9c5b3407482465eaca7d7278a : SUCCESS in 6m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2ce44de779c4483a116e3629897674d : SUCCESS in 10m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d37249816f444ce485440f4f0b581207 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb05f07df904cafaa8534eac4ddea17 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06fc2e1ea6e74e1cb0e983169d3919b9 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/227f614f2d634fa2b964ef7c21501ebb : SUCCESS in 6m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc9c6e48f437425e840759390bfa6089 : FAILURE in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6448814144eb451da017bc2b8a23a073 : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08c99a8a08954e458d50d11b5749f19d : SUCCESS in 20m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ea73f5b50554879b54c2448ab773d20 : SUCCESS in 8m 20s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e42ffbc015484565abc1f2cb64768b09 : SUCCESS in 23m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/fb89479d0ad84c60a5fa7658a7765989 : SUCCESS in 18m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/787ab2e573b64cb6862e4f2a46e84992 : SUCCESS in 26m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6772e7ff8d9f4820a9ce5f485ee9b677 : SUCCESS in 16m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/34fb372a7ee948f4a436d7c08f941e63 : SUCCESS in 22m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fab31b378c884576810fccc160e8acbf : SUCCESS in 29m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fc1fa49715a8454fb8e4e5472c49c547 : SUCCESS in 21m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a90ef2c58bae4880bd53655a0db6fcbd : SUCCESS in 26m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/84d29389e2b641a6a20613daffa1ae09 : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea6aabf1a9d04a07a06c8dd6df8c0f9c : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a3ac0b1b81374464ba6be798b4fe63ec : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b41d43e7bf274d6495774ed1f0b94235 : TIMED_OUT in 2h 02m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d69b433587c5481b9f3babd4e9172aa0 : SUCCESS in 1h 51m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9db44d4080e1477cae26d91f0de32adb : SUCCESS in 2h 03m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bbf38b2394164778b9779bb5c39e902d : SUCCESS in 24m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5f74e572905a4dc7a2d1ca3e5aa8af0f : SUCCESS in 23m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bc520089d974ce88459218e494aa139 : SUCCESS in 31m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6509b860ef449eab6e170146e61f744 : SUCCESS in 21m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1a4d28f4e5a349b2875cf7ff1f851ae7 : SUCCESS in 43m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cccdaa017db841e387187dc7f57e2594 : SUCCESS in 48m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fba0f80b1282459bb6c0bcbabb7ed0fb : SUCCESS in 29m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf10c529ec5c4ee69c521cff63524b74 : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"76e2fbd96a3bca1e27f2460c1d82d72494e93fae","revisions":{"3bc0b67f1bd0601ab34fe789aa06f23bbf286951":{"kind":"REWORK","_number":1,"created":"2024-02-28 13:58:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/910509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/1"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-28 13:56:23.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bc0b67f1bd0601ab34fe789aa06f23bbf286951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bc0b67f1bd0601ab34fe789aa06f23bbf286951"}]},"branch":"refs/heads/master"},"f9660b45f6e0fdf808379d99713ef9a5e01506cd":{"kind":"REWORK","_number":2,"created":"2024-02-28 17:06:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/910509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/2"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-28 17:05:07.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9660b45f6e0fdf808379d99713ef9a5e01506cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9660b45f6e0fdf808379d99713ef9a5e01506cd"}]},"branch":"refs/heads/master"},"01e487d5900501f5bbdd271899b781c2fb17dbd4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-29 10:59:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/910509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/3"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-29 10:58:39.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e487d5900501f5bbdd271899b781c2fb17dbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e487d5900501f5bbdd271899b781c2fb17dbd4"}]},"branch":"refs/heads/master"},"e8af3bd2f438225a012c831217bb95119a7a7135":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-02-29 15:51:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/910509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/4"}}},"commit":{"parents":[{"commit":"5244733d75a05c867f001b1ba637cf8695814359","subject":"Merge \"Make OVS container build regex independent to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5244733d75a05c867f001b1ba637cf8695814359"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-02-29 14:13:03.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910601\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8af3bd2f438225a012c831217bb95119a7a7135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8af3bd2f438225a012c831217bb95119a7a7135"}]},"branch":"refs/heads/master"},"64771cee7d9bcdc906a232ee013481a04a919c24":{"kind":"REWORK","_number":5,"created":"2024-03-06 17:04:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/910509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/5"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-03-06 17:04:13.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910601\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64771cee7d9bcdc906a232ee013481a04a919c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64771cee7d9bcdc906a232ee013481a04a919c24"}]},"branch":"refs/heads/master"},"c9341b1ff394d1cbbaebee8c33e7c235cb88e8e1":{"kind":"REWORK","_number":6,"created":"2025-02-24 18:51:40.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/09/910509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/6"}}},"commit":{"parents":[{"commit":"cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de","subject":"Merge \"CI: create ansible.cfg for pre-upgrade deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-02-24 18:46:45.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910601\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9341b1ff394d1cbbaebee8c33e7c235cb88e8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9341b1ff394d1cbbaebee8c33e7c235cb88e8e1"}]},"branch":"refs/heads/master"},"a17964ba2289c02a531729637580d93b7b6a371a":{"kind":"REWORK","_number":7,"created":"2025-02-25 10:23:11.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/09/910509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/7"}}},"commit":{"parents":[{"commit":"cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de","subject":"Merge \"CI: create ansible.cfg for pre-upgrade deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-02-25 10:23:04.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910601\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/910503\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899615\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/899592\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a17964ba2289c02a531729637580d93b7b6a371a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a17964ba2289c02a531729637580d93b7b6a371a"}]},"branch":"refs/heads/master"},"9824ae2abd3ab7124816a04816bf071fddf39a61":{"kind":"REWORK","_number":8,"created":"2026-02-11 15:34:43.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/8"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-02-11 15:34:07.000000000","tz":0},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9824ae2abd3ab7124816a04816bf071fddf39a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9824ae2abd3ab7124816a04816bf071fddf39a61"}]},"branch":"refs/heads/master"},"ab19535d81886824562d9ee40b153f572c261bbb":{"kind":"REWORK","_number":9,"created":"2026-05-01 09:01:36.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/9"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-01 09:01:21.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab19535d81886824562d9ee40b153f572c261bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab19535d81886824562d9ee40b153f572c261bbb"}]},"branch":"refs/heads/master"},"34ae92b36baa604476da0307cfbceb9f7b3c5370":{"kind":"REWORK","_number":10,"created":"2026-05-01 10:42:47.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/10"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-01 10:42:30.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34ae92b36baa604476da0307cfbceb9f7b3c5370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34ae92b36baa604476da0307cfbceb9f7b3c5370"}]},"branch":"refs/heads/master"},"8f7b2877da97c5d16090678ab522ce5c57a727f4":{"kind":"REWORK","_number":11,"created":"2026-05-01 14:14:19.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/11"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-01 14:13:57.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f7b2877da97c5d16090678ab522ce5c57a727f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f7b2877da97c5d16090678ab522ce5c57a727f4"}]},"branch":"refs/heads/master"},"38435b02e58e72b38ad903c3b69e731bd98e6313":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-05-11 07:26:23.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/12"}}},"commit":{"parents":[{"commit":"1486c00bdb3e66d88a61e7a8cd70957d6c403774","subject":"Merge \"Extend etc-hosts support to all host groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1486c00bdb3e66d88a61e7a8cd70957d6c403774"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-11 07:25:51.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38435b02e58e72b38ad903c3b69e731bd98e6313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38435b02e58e72b38ad903c3b69e731bd98e6313"}]},"branch":"refs/heads/master"},"a330ec76fdff4a06e71422189c92c876751ef183":{"kind":"REWORK","_number":13,"created":"2026-05-11 12:25:47.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/13"}}},"commit":{"parents":[{"commit":"c2a3cde5c3e295a2a0120f32ea97df4363fdb139","subject":"Add support for using Kolla test images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2a3cde5c3e295a2a0120f32ea97df4363fdb139"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-11 12:25:36.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a330ec76fdff4a06e71422189c92c876751ef183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a330ec76fdff4a06e71422189c92c876751ef183"}]},"branch":"refs/heads/master"},"31c194ddbe4b1df238f031fdb52026a472b14198":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-05-15 13:14:59.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/14"}}},"commit":{"parents":[{"commit":"5e864fb929e880696078e669bdc4978e06d5bd1c","subject":"Merge \"[CI] Switch to baseurl for all repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e864fb929e880696078e669bdc4978e06d5bd1c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-15 13:09:25.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31c194ddbe4b1df238f031fdb52026a472b14198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31c194ddbe4b1df238f031fdb52026a472b14198"}]},"branch":"refs/heads/master"},"7c6d9a2b7976af90b211a610d0606cf546b596c5":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-05-21 14:46:03.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/15"}}},"commit":{"parents":[{"commit":"70552faa0b9f1ce82380115b2d7bf83ea298a63b","subject":"Merge \"Fix broken conditional release note formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70552faa0b9f1ce82380115b2d7bf83ea298a63b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-21 14:44:13.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c6d9a2b7976af90b211a610d0606cf546b596c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c6d9a2b7976af90b211a610d0606cf546b596c5"}]},"branch":"refs/heads/master"},"e9101529f87cd62911689b6040d94dbb3b2c3e11":{"kind":"REWORK","_number":16,"created":"2026-05-22 08:23:49.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/09/910509/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/16"}}},"commit":{"parents":[{"commit":"9ea443de09df62ae54cb8c868c149161f68d5f65","subject":"Merge \"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ea443de09df62ae54cb8c868c149161f68d5f65"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-22 08:22:27.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9101529f87cd62911689b6040d94dbb3b2c3e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9101529f87cd62911689b6040d94dbb3b2c3e11"}]},"branch":"refs/heads/master"},"7a6fae08dbc24a9ba00238ceb6d2d6ac74721928":{"kind":"REWORK","_number":17,"created":"2026-08-27 16:09:38.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/09/910509/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/17"}}},"commit":{"parents":[{"commit":"cac4906087360fce1253d1e1fb6c181e51d03bd5","subject":"Merge \"Fix podman support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cac4906087360fce1253d1e1fb6c181e51d03bd5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-08-27 16:09:22.000000000","tz":60},"subject":"DNM: Allow to continue when some hosts are unreachable","message":"DNM: Allow to continue when some hosts are unreachable\n\nExperiment to bundle stats plugin into Kayobe for the below:\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nThis relies on Ansible run statistics collected by an Ansible callback\nplugin, kolla_stats, bundled with Kayobe\n(kayobe.plugins.callback.kolla_stats). It is enabled automatically via\nANSIBLE_CALLBACK_PLUGINS/ANSIBLE_CALLBACKS_ENABLED whenever\n--continue-on-unreachable is used, for both Kayobe\u0027s own playbooks and\nfor Kolla Ansible, with no ansible.cfg changes required.\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\nAssisted-By: Claude Sonnet 5 \u003cnoreply@anthropic.com\u003e\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a6fae08dbc24a9ba00238ceb6d2d6ac74721928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a6fae08dbc24a9ba00238ceb6d2d6ac74721928"}]},"branch":"refs/heads/master"},"78a2f8de76c4ad1c671aeb5adcde64fad254497d":{"kind":"REWORK","_number":18,"created":"2026-08-27 16:17:05.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/09/910509/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/18"}}},"commit":{"parents":[{"commit":"cac4906087360fce1253d1e1fb6c181e51d03bd5","subject":"Merge \"Fix podman support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cac4906087360fce1253d1e1fb6c181e51d03bd5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-08-27 16:16:56.000000000","tz":60},"subject":"Allow to continue when some hosts are unreachable","message":"Allow to continue when some hosts are unreachable\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/910347\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a2f8de76c4ad1c671aeb5adcde64fad254497d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a2f8de76c4ad1c671aeb5adcde64fad254497d"}]},"branch":"refs/heads/master"},"76e2fbd96a3bca1e27f2460c1d82d72494e93fae":{"kind":"REWORK","_number":19,"created":"2026-09-10 14:34:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/910509/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/910509/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/910509/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/910509/19"}}},"commit":{"parents":[{"commit":"cac4906087360fce1253d1e1fb6c181e51d03bd5","subject":"Merge \"Fix podman support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cac4906087360fce1253d1e1fb6c181e51d03bd5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-22 10:28:40.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-10 14:34:40.000000000","tz":120},"subject":"Allow to continue when some hosts are unreachable - take 2","message":"Allow to continue when some hosts are unreachable - take 2\n\nSupport continuing execution when some hosts are unreachable, provided\nthat some hosts reached the end of execution.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/1005021\n\nChange-Id: Ie788829432ccf1691f200e5b319da16e0653da46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76e2fbd96a3bca1e27f2460c1d82d72494e93fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76e2fbd96a3bca1e27f2460c1d82d72494e93fae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
