)]}'
{"id":"openstack%2Fkayobe~911074","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I028e137dade594f51450e47ba4a97c0435ba6796","project":"openstack/kayobe","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-06 12:21:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I028e137dade594f51450e47ba4a97c0435ba6796","subject":"Remove default value from sample configuration","status":"MERGED","created":"2024-03-06 07:57:49.000000000","updated":"2024-03-06 12:26:45.000000000","submitted":"2024-03-06 12:21:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911074","cherry_pick_of_change":911155,"cherry_pick_of_patch_set":1,"meta_rev_id":"ce74bc9442f3ea1e48c2db106387b7be8937b950","_number":911074,"virtual_id_number":911074,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 12:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-06 10:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-03-06 10:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 10:02:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-03-06 10:03:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"413a9f25c17f0614d9b3cbf923ace11424195e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-06 07:57:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"22dff3f5c30a3d35875852f6d942f67d38ee84ce","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 10:02:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ebab019e4915df56b5afbcbf402965dfbe13bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:03:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fee9eccdbe744e5b9d10f811733fbbe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d319b88b3d36410aa9d0056336e0fcb4 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9fb2fea1054960a63ec51534d09f49 : SUCCESS in 9m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bcee710fcdf4733bbd1e9362b622372 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b46aedb5fa594f55814b3e4e0693f8eb : FAILURE in 11m 56s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/000d88b50e59430b9567bc73cd8f37bb : SUCCESS in 11m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f256ce42672646bab532ce27d6b81d3e : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3aa8df5e90b84307b8f25193760a9e7d : SUCCESS in 1h 02m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c59f768c03664374b958468c71c0ed09 : SUCCESS in 57m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b710491c079b430f84adf8eda7005fba : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d795e928b8484b2791835c55e476a095 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ba956ed7c5384c468bc540deab9ebab6 : SUCCESS in 18m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78350ab138084977b9c82ec46cda0340 : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/22f4722e14d7445abeae203d84448e55 : SUCCESS in 1h 57m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/808d83598a4a4f559a5a957794630185 : SUCCESS in 1h 41m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ead88dd9bf7343759914fcb0559a5701 : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b00cbb7171cf4ab5b2e68a3de4362d6c : SUCCESS in 14m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/26ebef3f58fb40549f4623185b8a44f1 : SUCCESS in 35m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f6ea3a34d1074962ba123fdfa279a481 : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd67a1ea88f442a7abe6bd5bc53f9d8d : SUCCESS in 28m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc9e80a04be7491abec14ce59d4ee363 : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e26fde6f2d343d79321de0e6f49a236 : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/476b65facced4d49b3c3d2e65f9d7a73 : SUCCESS in 16m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bf4bca028cd4d3982a78320e2da997e : SUCCESS in 14m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab42a2d7c8b14854afbb5a1f233dafdc : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/276094fffd84478486a2caa05c4a7a9c : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8178e176b5a7432b9cf4ec9e9c2bb509 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a28ec4a911a483e697e387b5c504e61fefe35e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:04:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aa0c3e4f01be71afec9b49e1489a2c4495535e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:21:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77660c00fcb54c759210df6c43654b0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19cb48528f44f5197cc8e36e189936e : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8715ea4f3ab544dfa9999904060384d2 : SUCCESS in 4m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c809db82c084e9ea642b8b47a6c4381 : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b30ee6c2728640a49531cbb6fee2b133 : SUCCESS in 1h 01m 00s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6668102fad724a5a8f7988014a648b70 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/48d723ca47d44895822c0e7ae02ab128 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b7040a107f214e9480fe06b34962c82f : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0ffe22b2ace4bb3a398e31291d8734c : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/957f93cac742471c8d31d97de435811d : SUCCESS in 2h 12m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a992c0d35c364261a3cf7eb45e945a04 : SUCCESS in 1h 52m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d015dc5b6c694148aea491ea0a186f00 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab3acad8f4b74d07862aac334d0bf02a : SUCCESS in 17m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/35adb40b9e5841fdad4649e6170efca7 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e9865dbd16a4f4595dc105886a6a3f7 : SUCCESS in 25m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bf9da161af7b4db7ab60254ebf379236 : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06526bab2a3049299feaff53c6675b54 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fd73e0075faf4d80aba904c2e691f5a2 : SUCCESS in 18m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55b303fd63a54063a184a3da24dbab16 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"44318f8680de6a8bdc1aacdd7e01c81b8c31cc83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:21:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ce74bc9442f3ea1e48c2db106387b7be8937b950","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:26:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b558fdfd947e41d69359ac12f3035ca7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5e8fbfaddf4f269d8ce2bd3739e139 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18644a564e01a32166c183731bdea0408181feef","revisions":{"18644a564e01a32166c183731bdea0408181feef":{"kind":"REWORK","_number":1,"created":"2024-03-06 07:57:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/911074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/911074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/911074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/911074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/911074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/911074/1"}}},"commit":{"parents":[{"commit":"b0700e1fa122b056ff065afd189479bedab96cd4","subject":"Merge \"Add missing space in help output\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0700e1fa122b056ff065afd189479bedab96cd4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-05 17:12:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-06 07:57:49.000000000","tz":0},"subject":"Remove default value from sample configuration","message":"Remove default value from sample configuration\n\nThis was added in change I8129b2a6793ea36bb296ef8337a5461f73d6e557.\n\nChange-Id: I028e137dade594f51450e47ba4a97c0435ba6796\n(cherry picked from commit e460099e1a3e2cc2899942d4fd48b72d8ed5616f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18644a564e01a32166c183731bdea0408181feef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18644a564e01a32166c183731bdea0408181feef"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
