)]}'
{"id":"openstack%2Fkayobe~911075","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I028e137dade594f51450e47ba4a97c0435ba6796","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-06 12:21:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I028e137dade594f51450e47ba4a97c0435ba6796","subject":"Remove default value from sample configuration","status":"MERGED","created":"2024-03-06 07:58:12.000000000","updated":"2024-03-06 12:26:45.000000000","submitted":"2024-03-06 12:21:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911075","cherry_pick_of_change":911155,"cherry_pick_of_patch_set":1,"meta_rev_id":"a13c8719b3b36333dec8609582f6c4b10486fa1c","_number":911075,"virtual_id_number":911075,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 12:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-06 10:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-03-06 10:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 10:02:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-03-06 10:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2870bfaa389d50970df02cff80bffce08303d96e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-06 07:58:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"80d815ddddf290cd842e1d679e05594f1a95f901","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 10:02:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d4210106b4880dfc14bc51f04cfbf2db97cbdf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b605bb12cded4b14ab03dd8e8115817d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2be3e070995b42848ef87c010ed07aa4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d8b6066c764890b98baaf039c47bc2 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76f71d97c21848dcbd309731fc80d3cc : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d437989bd8941e9bae14c7ed7ca7782 : SUCCESS in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8aec813c1a9444d5b36c086c33eff2f2 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae2ae5703ba3433cbdf8b6637c1dff5b : FAILURE in 11m 05s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c105dc45f5ed44d8b5d41dd5e856caa1 : SUCCESS in 11m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/23fcee7bf95345719545549d6daf6efe : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c9910a55f08a45c1bc92f021c5bc8f86 : SUCCESS in 1h 06m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61f3c11e1a544fde95b0658736ea8455 : SUCCESS in 49m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3701ac869ed44c9a87187cb6f6c5ecaf : SUCCESS in 1h 24m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/23495c43ec8844cd93a31e2abff53b4e : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4b32841ec94b419c954435b671cf6e81 : SUCCESS in 20m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00bc4e6d5504420ba0d2ba54c884f682 : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6bd9621ac57c47bdbf442585cf79a53f : SUCCESS in 1h 57m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c1c91ad99aa4b19ac32c3853e2b20a1 : SUCCESS in 2h 04m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e3caa5e19ec74898a5cf9d243c65a354 : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9e9c227e7f44d709aa687cdc9ccf1ea : SUCCESS in 13m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2076048ac1454738aa720b621eb8168d : SUCCESS in 34m 49s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9f9d76ab26de485aa65583fb0b93071d : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d78fe26b46542e793087f7921741ec4 : SUCCESS in 25m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7c9213667da64ea6848ce91f9c5bbaa2 : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9474b8cf1eb4dcf827de0b9815ababb : SUCCESS in 9m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/385f06cadd924da4b4d35a500b980373 : SUCCESS in 18m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/676568fc4c0e488384f7d07e223df78b : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8343d5667e8a957555e950c3bf3949c9354fb234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:10:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55dc33f2cef85c7c9c4abf858971bc934d247627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:21:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cc95c3b9e954609abbe52c6b477113b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9776243dac594912927101002e1925f7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515573d48fee4bf993478a5ff041eb8a : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf90af9c3d64f4597cf78ebeb786ec7 : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02e443f85a3946a5a5ff633d31734232 : SUCCESS in 4m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af26863fab224466bf2170bc9eb75eeb : SUCCESS in 10m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/671f124958134dc3bb1818f13dbb2f55 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e7954531e284862bdbbec6d6470ad70 : SUCCESS in 50m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d0868e8c0a6e4974b6d95ae2c8ffcfb3 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/730c2336daa743a0b62fdd0dcdb646cb : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e786f54c3b7a4e149839930b6d5a7aaa : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8ebc7c5018444a3c9d17f5f597cb0ce6 : SUCCESS in 1h 44m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a205162509724180b421f63a41470860 : SUCCESS in 1h 28m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a639e678493343b08c1224e1cc78dd24 : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8bdbce5553146debdd64ede64ab3105 : SUCCESS in 14m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9ae9efbb7b044bb1a5a18fe1de8ad8ae : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/928ddf886e62492fb28ef52210e9da07 : SUCCESS in 22m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50eed385c1a94774a27266bbe7aec15d : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f53d0cb7c51420582f0fa0fd9204ee4 : SUCCESS in 9m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b871a1916e3c4602b519c7431805a1e7 : SUCCESS in 16m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3efd8552a0544e21a51eaa1390088247 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2b4a0440ed7c558093108e18c34b09f0b083b90c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:21:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a13c8719b3b36333dec8609582f6c4b10486fa1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:26:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd6d417284ae4f6b8bb56acb220fdd46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc84dbe47562461e8ab6e3450a67a991 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f4414f8a4fe5184b6d1e09595993e8bfaeebae7","revisions":{"2f4414f8a4fe5184b6d1e09595993e8bfaeebae7":{"kind":"REWORK","_number":1,"created":"2024-03-06 07:58:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/75/911075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/911075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/911075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/911075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/911075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/911075/1"}}},"commit":{"parents":[{"commit":"1bc5f4cb3be581da0b04c39b22aa20f8878bf691","subject":"Merge \"Add missing space in help output\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bc5f4cb3be581da0b04c39b22aa20f8878bf691"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-05 17:12:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-06 07:58:12.000000000","tz":0},"subject":"Remove default value from sample configuration","message":"Remove default value from sample configuration\n\nThis was added in change I8129b2a6793ea36bb296ef8337a5461f73d6e557.\n\nChange-Id: I028e137dade594f51450e47ba4a97c0435ba6796\n(cherry picked from commit e460099e1a3e2cc2899942d4fd48b72d8ed5616f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4414f8a4fe5184b6d1e09595993e8bfaeebae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4414f8a4fe5184b6d1e09595993e8bfaeebae7"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
