)]}'
{"id":"openstack%2Fkayobe~911076","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I028e137dade594f51450e47ba4a97c0435ba6796","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-06 11:35:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I028e137dade594f51450e47ba4a97c0435ba6796","subject":"Remove default value from sample configuration","status":"MERGED","created":"2024-03-06 07:59:10.000000000","updated":"2024-03-06 11:36:56.000000000","submitted":"2024-03-06 11:35:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911076","cherry_pick_of_change":911155,"cherry_pick_of_patch_set":1,"meta_rev_id":"36b27c8da2e787bb499de10f3eca1f70270fb250","_number":911076,"virtual_id_number":911076,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 11:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-03-06 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-03-06 10:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 09:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 10:02:49.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"cfeb8380ce63855b7a380c57aa14b8536c597afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-06 07:59:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"758b4f61bd868e6c54782d991ee506946ce6c590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 09:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e01d8e607e044e5bdf3bb91738f3c00\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4646be4382a941479e64554f63339a70 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18ae7d5ebbe4d3abf46520ac2e600a1 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0281fe15bf3e4a78a169f19f95095823 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7056c7fc0647778322e11879859523 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/777292d4ebc14f2f956a3d46fa8de7ec : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16deee36ef464966a7e20a1da32a0fb4 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e862abc409f4fd493dce656b8ecab0e : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f07d8734f394528827137d135f7e54b : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c10e174517a411eae9cf76b7ff4ddab : SUCCESS in 12m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e051e4899c99497ab71939b7afd9c399 : SUCCESS in 58m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3c930b52f794588a61fc15fee88298b : SUCCESS in 1h 00m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/827a8a732df64b88ac7f6fe61c840dba : SUCCESS in 49m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9792c873019645c6a28eeff1b58e9c85 : SUCCESS in 55m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c720876e975c49d5ab0cecda21ed0c44 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/43c90d9d23064051832ed02425835e25 : SUCCESS in 22m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4541e10d1121495089cdc8230593c49f : SUCCESS in 16m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d306a56cb7046b3bea11c75166940ef : SUCCESS in 1h 28m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/64fa80433bba42079474712d14b64da7 : SUCCESS in 18m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e03a421a872f46d5bc02eac86bec709a : SUCCESS in 13m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6472ffcb887f48e4af4f3259d1d0f8fc : SUCCESS in 32m 42s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9a3320a6ae6483894c6fa3443225778 : SUCCESS in 31m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d2cd111df554491a686a4e525f30813 : SUCCESS in 12m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5d3916423749119efaef16017266b7 : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dfff9b0991e84453856011a6abdbc33e : SUCCESS in 15m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0d684ba416940358bf4800312fc6ae4 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"44f3f45503c6d07fb5b4d9ab722656256e8eea5c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-06 10:02:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ceb89cbc7bd7693699d3f5783108594d89a88e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"506cfcdc1b087721a8b0c57c8802746444c2e35b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 11:35:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70a8c021c9d494eb6a1fae458160e19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd269985e42486ba04513f5da815b56 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf85a3e8a79d40b1aae30d362d28eee7 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcadcd51775449daa17c8425f72f9b4e : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39706401a0b40458434eb30d6d777af : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c95bccd8b2f14d1ea7cb76381dcdc4cc : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b7c769942564561a69c9ae074b0437a : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4298c1ae7564d1fa9cec4d8baec4e54 : SUCCESS in 12m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4938cd7e39a24626ba12b36fe165b4c6 : SUCCESS in 57m 00s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c5f28eb6920457cab8d8c6ffdce9d6a : SUCCESS in 51m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3f3979c8219a4efca73039167f89f6f9 : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/be9a758822fa48f2bcabddd6ffde7f94 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4431fd4617c4f90ba5af0aebe659ffc : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1f7c888d1254b0892fba0c198aaed5a : SUCCESS in 1h 28m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5cf055eb267048da842abab46941138f : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752d15b46d9c4750956db4de2e9ea22f : SUCCESS in 15m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d447d0f614364592be3e8a88a639e61a : SUCCESS in 25m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4dd8f32f141744069ab227905f7a494f : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa87ef0c5c4f462e978848297ddce1d1 : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/012ae6620b5849a7bccda966ada1e397 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a514e9c28f434af5a36a4757942b8b7c : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"853d3d7079d892d46e8cd42a7b2828decf192a55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 11:35:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"36b27c8da2e787bb499de10f3eca1f70270fb250","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 11:36:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c820025becc84333bc8b639d367b6c0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557727c97e2f414fbe3a213c7e9b1d9c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f113399bdf25b8e7bea475ce061d380837d07a39","revisions":{"f113399bdf25b8e7bea475ce061d380837d07a39":{"kind":"REWORK","_number":1,"created":"2024-03-06 07:59:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/911076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/911076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/911076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/911076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/911076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/911076/1"}}},"commit":{"parents":[{"commit":"a78029236a705c9d90c65f38d44dd51d10de503a","subject":"Fix wipe-disks role to work on util-linux \u003e\u003d 2.37","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a78029236a705c9d90c65f38d44dd51d10de503a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-05 17:12:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-06 07:59:10.000000000","tz":0},"subject":"Remove default value from sample configuration","message":"Remove default value from sample configuration\n\nThis was added in change I8129b2a6793ea36bb296ef8337a5461f73d6e557.\n\nChange-Id: I028e137dade594f51450e47ba4a97c0435ba6796\n(cherry picked from commit e460099e1a3e2cc2899942d4fd48b72d8ed5616f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f113399bdf25b8e7bea475ce061d380837d07a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f113399bdf25b8e7bea475ce061d380837d07a39"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
