)]}'
{"id":"openstack%2Fkayobe~911688","triplet_id":"openstack%2Fkayobe~master~I654f7d6e923c442a8aa08f536cc2b82c5e5b69d1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-03-21 12:14:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-03-19 14:02:44.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-04-02 11:32:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I654f7d6e923c442a8aa08f536cc2b82c5e5b69d1","subject":"Support saving baremetal compute inspection data","status":"MERGED","created":"2024-03-07 10:44:34.000000000","updated":"2024-04-02 11:43:13.000000000","submitted":"2024-04-02 11:32:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911688","meta_rev_id":"fd30e9cfae78d33f27dd1ccd2598f25a009961cb","_number":911688,"virtual_id_number":911688,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-02 11:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2024-04-02 08:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2024-03-19 14:02:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2024-04-02 08:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 13:20:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 10:20:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-03-14 12:41:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"4da8d9edff90c52bb93fac9ef49ad8bd7662edab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-07 10:44:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca692e1b06f4c5885edfda5572f41936013b378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-07 10:45:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b642dade594becfde0f364c35034b56fed1d6445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:20:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d096fe937004c00ae5fb400d1bc7761\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04de033738b54b5d84356bc9896e7a00 : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/966451f6c18846be8cfa005a30426552 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e2ca3df20646fbaac62fe666ea8c9a : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3077ee7c403b4b46912b097f1a051244 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a78a6e825ed4036a8872e7f91970b6e : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f78cedf54e184e71a4d43b2b4f8c750a : FAILURE in 11m 49s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36572b4b8aec40f898f217d707661971 : SUCCESS in 9m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ebd03654afe044619aef908e24bef65e : FAILURE in 49m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/25c4386730e3449a9619fc3b7adfa2b8 : FAILURE in 44m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68bd9e10a87e489c906c342466d5280e : FAILURE in 43m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8428d93de1824214a5b3467c88510e5a : FAILURE in 48m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/df5bac0cb29f4444a3caef5f5cdc8051 : SUCCESS in 17m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3bdab33da8c44070b8dac23f1a326cc8 : SUCCESS in 17m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bf32b5ac8054a839e4bf18ef9f2c934 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e5b1f50e257483097ab62dc71ad6b79 : FAILURE in 2h 28m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d7475c3f1034476bab3cd3bd5b0b8c7 : FAILURE in 2h 05m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a6e0431425954fccba6680dcb762fd54 : FAILURE in 2h 26m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3c099cd16ba4600bf1a9a0812c0489d : FAILURE in 1h 59m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/115b7955d1954006a367a4e79cbbcf06 : SUCCESS in 21m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a20ab8adc04d74b82353b864f65fbc : SUCCESS in 16m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7b2e79784b364a48a0f4df1989e3f40e : SUCCESS in 38m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c9dcb5b1f004115abd9f528c946c758 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/731781789ccb464c8e922ae9510907da : SUCCESS in 23m 22s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/10e865e7dd864a75b5fb5b6c636fe6f0 : SUCCESS in 37m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c5902dd2f64aa683d9c282632db464 : SUCCESS in 25m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66270d8c8cf54c78b86cd966b041cc8c : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bc34e65951042b09a319faaa494a4c4 : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/81b3fbf6c5a74fe0bacca3717c558bfb : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f8af243b7384d5cb06391920ea1bd36 : SUCCESS in 14m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3d00c7a128c4195819ec122c70fc81f : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e82fb0c8f36646a9a078f9734d174b8d : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f3380e1910d47358e5f2b6aa09189c4 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fdc543a22e294e22fa1cff3fbf3938fa727e7b97","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-12 14:41:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2439a58ee3bb07e8fa92a68ba4dad839c3bb6fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:17:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd551fb1f944e908745df39e46610d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bef24b4bc9b414e9b1c166ac993e921 : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041e1253d85e46c39f7b215fdcecc4e2 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366fb636cecc41caa6bb70ef48002f14 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db1f6378f36487e9e2c0c428ae44858 : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf92afc5c8d54efcb0844adeab5e6727 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7038ff034c148d2874e3efb216d83bc : FAILURE in 12m 18s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/755a054a5a8d4555a23c704e366d5a46 : SUCCESS in 10m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ee72423065e54fc7bb1e0b5a3d721fbf : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4db97f2dc18940ed895ab82d92df9ed3 : SUCCESS in 1h 03m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/330f58654a7b4cfbb18eeffc5b0c1f80 : SUCCESS in 54m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f83fd5eda6d44a06b139a0446974e3ec : SUCCESS in 1h 13m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0716ef319d6d455cbb82f6f4693a5ca3 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d66c9e368193452a848bd839c5db7450 : SUCCESS in 17m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0db8a2159e73403b9d22bfe45bf63897 : SUCCESS in 13m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e023640ff014db8b780ec65dd1f777e : SUCCESS in 2h 15m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b538347cca14b94b9c3cca9fa8fab2f : SUCCESS in 2h 11m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/941a833b430847d58152e77d74accab0 : SUCCESS in 2h 32m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67e0b05844ff47dca6086e8c01922b0a : SUCCESS in 2h 00m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f63da3655de240628c12dcca40b505b2 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c3419470ffa409caa16e2ea187c8797 : SUCCESS in 16m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5fff1ce40ec140f09e91e630bec4c4c5 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5ec06440ad9a4055aeddd7eb13e99282 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea3fdba7cc14db298e94377dd9fa57f : SUCCESS in 23m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f456c5f2a2e24973acd08df4424c493a : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29469aab8ddb4a4b9dd80acba3c67ff2 : SUCCESS in 25m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9662d53c052b46728656aad1315332aa : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d3afb6574dc40b3aedd0c287566f86e : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6670bc4e3d83489eab25d81618073e93 : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7ba18b6a0774638bdb0769aef9a3cdc : SUCCESS in 15m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91bf83f2d3f242e2ab6325422c39cf16 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aec6f9834d3f454c98f04374c438f6c9 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1211a45ef5f646619bb5d36bde4cfe59 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"159bd900e95089f79155914c67545823efd717c3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 10:20:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c56ae2c80a017a2b8ba007305db4492c312c6e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 12:41:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5662662d2ca50dc792379ce3115504bbda652e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-14 14:05:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf2faf932daa6c09a39dc7104046a8375d2a73dc","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-14 14:05:49.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f7be5b5ec4911ffe65e60cb4e2273a98746a721e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:46:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3ff519e73a648889570ba1772f7c9ea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8800bf8342594d4ba402b380677cf7a3 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eff1954e99024a379ce83bb6ca890d57 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201edd59c35f4fb08fc163f7bd02190a : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ebcac3c82242e6b6f4f8717ac54a89 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ac19443948304cf88cf29e91b1612522 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/149cac2328624495ba2dfd27f836decc : FAILURE in 11m 24s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b9d8d1aebc94702873eb9617e9703a3 : SUCCESS in 10m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4564e189e4b1411f93eed6fb1aa73010 : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/283da9a107234ecd8fd806610292ce18 : SUCCESS in 1h 07m 28s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1bcd69a6dd488ea9fc82118d97fabf : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/256e30d90e9942589c6a3fb36d0c525b : SUCCESS in 1h 20m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f4752f60696045b493406ed42db543f9 : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7c2dba2db9d74850b4ff8a1317cf54e2 : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd9e8743e46b47abb914158d9ae32c3f : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e028d78d8a184cda8bb18a8f45e0d4e9 : SUCCESS in 2h 35m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42ef7a143a404ba4bdced974f6165f09 : SUCCESS in 1h 56m 26s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/46c2d227d7e6460e938c2902c92334c1 : FAILURE in 1h 28m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ec0536804d7475785c5b7293d90871f : FAILURE in 1h 20m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/407223b1534a45e9812c6306811e1422 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f74caf89538a4b3d956deac966adc542 : SUCCESS in 14m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6ae6af6fd0af4fd295e1f0c1e70836cf : SUCCESS in 41m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c6038eec77c142629cd7a903284b8ab0 : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99baab5e3d92419d80719f3b139b1d26 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/162c61f3fbc4498ea07598f69439aa9a : SUCCESS in 32m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/156315b829304001ada748ca2fa5a7c4 : SUCCESS in 26m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fc97446e0e2e4100b9401848e967190e : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69109fe4030a415e95452f7944fd74d5 : SUCCESS in 14m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/21b544c90b2f45ff864d91c784558f7a : SUCCESS in 14m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3caee3bab37a43e19f74089476ef7fbb : SUCCESS in 12m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25343ff876ad4860a94179bc0f7ab554 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74d17fdae92846c1a10a17577a91d8d8 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05fccb16bf204374b20230493703eafa : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"25e7f39c116acbeb0603c0a50b4d42b67f8b17b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-19 14:02:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2505fc8597fc86b77439a0dcc7e76e0ea703e8ff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-21 12:14:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"923494ce1b3a0a912afe561ca2a6595997522eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 14:39:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8508e2bc3d06474fbf6227def77d62e5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e4e4416685b4c96b83eaf8375fecbbd : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fdf4004e9d5468dbb75043e0a468365 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5909f3a891438caa96f4af139bb581 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/804cd98bf8e445f383c0a637d2b3df58 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35c944c6973740d587bc62c6d829bcb6 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebb387a6b59546578c8b3fc1ddfc8381 : FAILURE in 11m 38s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9809e3c6474b45499eed25675919637e : RETRY_LIMIT in 2m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/11e08d83ebcc4583b0f44962f69d86a5 : SUCCESS in 1h 09m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/433bf8c628b84b968b0c6309af28b5bb : SUCCESS in 1h 03m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c666a6b62c4e4e29a61550a2c7b22350 : SUCCESS in 54m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75266dc18cff4a69b88ebf0c264476ae : SUCCESS in 1h 15m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a5fc0035a94c4e3cb4617d301b0d50dc : SUCCESS in 17m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3856a037478e42888101b24ced0dae64 : SUCCESS in 18m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c917b10f6474d78a71416d5c07efe35 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e986b53de0cd4719bab752280eb8d3ec : SUCCESS in 2h 22m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2e14d5f4c0d4d229a6675d1ee931b1b : SUCCESS in 2h 00m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/92ced2dd551844cc881556b238eb62c4 : FAILURE in 1h 25m 53s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85e8246aa5624988b106ded01d7a8525 : FAILURE in 1h 12m 24s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/474ab6af3a6845e6943848df9319a7ec : SUCCESS in 19m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce93d252224b42638cd08a5f5d88c5e9 : SUCCESS in 14m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f5adb3a5f33d483fac24ce4f9b000486 : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2eea2e2b8d614cf4a87260ff0ce5374c : SUCCESS in 27m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71edf29ef1d74c41a941f2e1a590d26f : SUCCESS in 22m 51s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2750fd520f664cf9ab00ac071f1e38b1 : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2a4b1c9ad7749829000460157818f86 : SUCCESS in 24m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1971597bf8c4a668f83825269770aa7 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49604e0b36974e078b0b72abb2534546 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ce153d48f15457fa38cc59353c6c3ea : SUCCESS in 17m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a87ab0ee19c7466c9b4e726fbd45b6cd : SUCCESS in 14m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2d5e0e311014ab69b60683759abc9f8 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcdad6c0dc1543e7931f1c85e3f16970 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e17be32bc0164c0dbecf5ab7d84a2373 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b333cf428e2b7792720ee6523975ebd51b4278f2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-27 15:11:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f008c958742261846eb7033b34641cf7cf08491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 17:46:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf52509a7685457b8c7d87ded4f8b7c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1960089af4034e7ca8ba4f0a83d7628d : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb219581fcd544b3971b5dedf6bc290a : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f456ca6abb048f391296497c91418d0 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf945887c27d4f0b8940aeb53a322aa9 : SUCCESS in 2m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91f92fb311e04903a83b9e39fe80abcc : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/931463c5fee7484897d4e8d3d4a24c8f : FAILURE in 11m 05s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9124bdce91ca4afd8873cfa49e2dc8b3 : SUCCESS in 10m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d37ea306c24248c18ccc1e4a88a43e0d : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/70db65198445402c8f24401d09bf5fb5 : SUCCESS in 1h 03m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d2649c0f71d4ff982000ee207789b82 : SUCCESS in 54m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6519c66781964b05a197589b31254d31 : SUCCESS in 1h 14m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a0e916165304ed9aaf6b5933aa846b4 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/85492639a5404e98af89934a6b7f497a : SUCCESS in 15m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d53cf3352110440f9eb8524ce117a35e : SUCCESS in 12m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3a9ba3f19f9d46b48a5d04cc424176fa : SUCCESS in 2h 25m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbce5d1b8434431990a14d87d0e2675b : SUCCESS in 1h 55m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3e638175732c43a9a82af2590c51a549 : FAILURE in 1h 26m 42s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6d231e9e16941578e861980e70e064f : FAILURE in 1h 07m 06s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8061f6b822d9412fa920fee26fa0944d : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80fdb2d8d963415285d2c58b99cc7a76 : SUCCESS in 15m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a15b89f75ac47baaffab018a3596171 : SUCCESS in 32m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/58aa6abe1d05403096fe8fd8a6f889e1 : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97f5b017e67b4d9cbc93b3e1708a8dc6 : SUCCESS in 23m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/65e5235b54aa487fb0b19b6098610e93 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/084422fbad1f40f59d84bc58ceca5e9a : SUCCESS in 28m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09ebe007b2f84105b856714987212d20 : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c672cbfc15754b8faef7195a809ed8ad : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a01652f3cfe4403937294d8b809f0c6 : SUCCESS in 18m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ec6e2b558a94ad0bbb2515575bae618 : SUCCESS in 14m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e024b6467ccc40b299fa606cbe3535f8 : POST_FAILURE in 17m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/119fb945ae144f0783d656d1dc69ee89 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/123299087a6c4d549ab8c424ad1ba58f : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5db8192be57446a21597c7c3e15f5a832ec12cbf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-28 12:16:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fbb76666259be5a09aeced484994e372f20e109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:31:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d1f7c375d7e4f0d803098ae0b7c0cf9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4281b1d6a5b34af3bb76fbb758e2822c : SUCCESS in 1m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11519b394de4df39b80fcb33b082eff : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5f5228f1574c2f8222ac011e71045a : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0955bf526754152abdc1d3be35f3500 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6e1599959ec438a999eb35981043b6e : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4947f2663280493bae8ac5aa7f49bdc1 : FAILURE in 11m 26s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe6439691f8640849b256de2c8b6fcee : SUCCESS in 10m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3bb22604dade4425b96bb5276aaa09e4 : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/297d95f6d5c240b594717b1ad1c9550e : SUCCESS in 1h 09m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9f2ed12fec3452b877a7c9405c13608 : SUCCESS in 48m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/252e9862238743c58c9b1e6f7360058b : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52d138184ab1434ebf3fc6b840d2a02b : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/04e0445cabff47eab2e990f0936987e5 : SUCCESS in 16m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cda538d27fb4db684851ee68729074c : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d0be9cf4b7714540950a324f70896f90 : SUCCESS in 1h 53m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f925086a4a2425496ff7343e4c61a38 : SUCCESS in 2h 04m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ef1e39ba2e84415ba4bae00e9e9837ce : FAILURE in 1h 24m 33s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5545c15805c84a09a43d7752579fd35d : FAILURE in 1h 06m 46s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c55ee954bb434f339daa579b752c1acb : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b689b3087314e6f92fda1d0c9bf7feb : SUCCESS in 14m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4558b498c9164a378fef53fe067256f7 : SUCCESS in 34m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5c38a75263a74c9bb8e425cfac9e8df8 : SUCCESS in 27m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4392018898804e5d84f624415d47470d : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/189ba4e86da0426aa3ba19602e4aaf44 : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b214f9d71d2640e3b0d1398f87ce5872 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e037a02f21142938d1b81c7000cb346 : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9aad2a0e79f344608a24731de38faccc : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/030e32b0a09d43a9b5710e4cb6780f73 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad677ced4534da5930833721910c114 : SUCCESS in 14m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d235cc380a414386908b737b85309f16 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d0c621f81634512b18e34836898910d : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27035c9c05bc4a11bbeaf697603981ea : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ebdb34b5ca860c00fb661c7a631aa7dd9d7c33dd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-04-02 08:45:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ed0fde4f28046497d3a781cce692bc678c6ca42f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 08:46:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"92b34ad786b0862488e266ab32c5ebc6887d88af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 11:32:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fcf941be84494b98d554892229b586\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b6890a4c134a4f6780d2a3565a1f4e83 : SUCCESS in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f1197b67c14d97af5bc271cdfa60b0 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b6ff4f64d4417ea388e3125429f870 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9db824684bc641188e86fed229be0527 : SUCCESS in 5m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91d620c14d45453ca82717a2c4ee5d11 : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a6740d35854643c0b98a1514b3daf18e : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3986a3e171cd4e828203fdecf27ca829 : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/946f2adceb904975936741a53f451854 : SUCCESS in 1h 15m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f9a8da4b5d542d5a0a01eb4c376931e : SUCCESS in 16m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18bfea0df52341a4b98e6026b2ca3cb8 : SUCCESS in 16m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4caeae68d8c64e5298469b6c5a83bc13 : SUCCESS in 2h 33m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/915c91189a004ffdb077636132889f4b : SUCCESS in 1h 50m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff9e6a6f88d24fd0a16246db0e58f127 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c2799f685214753a003ecb19f50162a : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ed83cf32a14428d8e24590812ab3467 : SUCCESS in 26m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7613701949a42db84949c62aed75c98 : SUCCESS in 23m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/704325b3ea9a4d3e9ca4d3119c971a46 : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16e611efd8734be6bbeee37d7d2ab024 : SUCCESS in 26m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb7fba848e1b49b6b9f7d93c0b58bffb : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d53d2e0b05274fdbb520cbf6c8155d7b : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bebe97fc4dd84b0eb3c336c3932b5b0e : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/020e0a6603f645f7862f71208c5260e5 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":3},{"id":"9b340f8c3c863dba9ae3a8bdb0d7552f89503e51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 11:32:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fd30e9cfae78d33f27dd1ccd2598f25a009961cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 11:43:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b310f6125594ee88057f2959d58b87c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0836ae4fe13a4adda9e6b66c8c06ebaf : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645f40d701e34090a335011c56cafec6 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c04b6ddf1b6485f5ca56a003645da2fdf43183d8","revisions":{"3d61cb450388dfea6facc544f8ab6d60e3de93a6":{"kind":"REWORK","_number":1,"created":"2024-03-07 10:44:34.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/911688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/911688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/911688/1"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-07 10:41:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-07 10:41:07.000000000","tz":0},"subject":"Support saving baremetal compute inspection data","message":"Support saving baremetal compute inspection data\n\nAdd the command ``kayobe baremetal introspection data save`` to save the\nhardware introspection data gathered by ``kayobe baremetal compute\ninspect``. This mirrors the functionality of ``kayobe overcloud\nintrospection data save``, but for use with the baremetal compute\nIronic, rather than Bifrost.\n\nChange-Id: I654f7d6e923c442a8aa08f536cc2b82c5e5b69d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d61cb450388dfea6facc544f8ab6d60e3de93a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d61cb450388dfea6facc544f8ab6d60e3de93a6"}]},"branch":"refs/heads/master"},"5b0da6153eedd069c1b0100ab08cef8027d22db1":{"kind":"REWORK","_number":2,"created":"2024-03-07 10:45:55.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/911688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/911688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/911688/2"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-07 10:41:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-07 10:45:45.000000000","tz":0},"subject":"Support saving baremetal compute inspection data","message":"Support saving baremetal compute inspection data\n\nAdd the command ``kayobe baremetal introspection data save`` to save the\nhardware introspection data gathered by ``kayobe baremetal compute\ninspect``. This mirrors the functionality of ``kayobe overcloud\nintrospection data save``, but for use with the baremetal compute\nIronic, rather than Bifrost.\n\nChange-Id: I654f7d6e923c442a8aa08f536cc2b82c5e5b69d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b0da6153eedd069c1b0100ab08cef8027d22db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b0da6153eedd069c1b0100ab08cef8027d22db1"}]},"branch":"refs/heads/master"},"c04b6ddf1b6485f5ca56a003645da2fdf43183d8":{"kind":"REWORK","_number":3,"created":"2024-03-14 14:05:45.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/911688/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/911688/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/911688/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/911688/3"}}},"commit":{"parents":[{"commit":"da80a71b40afc2dc23cd9c026a0b1ea1d1b41a6d","subject":"Merge \"Fix: configure etc-hosts for overcloud group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da80a71b40afc2dc23cd9c026a0b1ea1d1b41a6d"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-07 10:41:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-14 14:03:42.000000000","tz":0},"subject":"Support saving baremetal compute inspection data","message":"Support saving baremetal compute inspection data\n\nAdd the command ``kayobe baremetal introspection data save`` to save the\nhardware introspection data gathered by ``kayobe baremetal compute\ninspect``. This mirrors the functionality of ``kayobe overcloud\nintrospection data save``, but for use with the baremetal compute\nIronic, rather than Bifrost.\n\nChange-Id: I654f7d6e923c442a8aa08f536cc2b82c5e5b69d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c04b6ddf1b6485f5ca56a003645da2fdf43183d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c04b6ddf1b6485f5ca56a003645da2fdf43183d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
