)]}'
{"id":"openstack%2Fkayobe~911702","triplet_id":"openstack%2Fkayobe~master~I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-06-05 12:54:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-05 12:25:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-06-05 11:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-06-05 11:05:40.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-05 12:54:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f","subject":"Add seed service destroy","status":"MERGED","created":"2024-03-07 12:10:02.000000000","updated":"2024-06-05 12:55:40.000000000","submitted":"2024-06-05 12:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"911702","meta_rev_id":"3f7686b686a2e90de157fc316bf1c0a7fc7f8eb5","_number":911702,"virtual_id_number":911702,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 12:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-06-05 12:25:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-06-05 11:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-05 11:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-05 11:05:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2024-06-03 16:25:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 14:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 10:53:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-06-01 10:35:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-06-03 08:11:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-06-03 09:53:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-06-05 11:05:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"4a3fce8b91ba1f0fa5df7ea29a8145e97fdec1ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-07 12:10:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d0eca221eea6248c4f33c1bfcdd9edf9b63635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:27:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1e399e9ff9c4335b30eb542a7fe9112\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15adebd2e6994041a695141e45fc433d : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c65b1c4f1cd445c8ced20ba95f59b88 : FAILURE in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f2f9de1ce64f36a60098b844d15c4c : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca588be4823438fb12af8a8c3d5d416 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92e36cb6d9964292a553d4c8a45dc93c : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fea36d7ec8442559daec22b48cddf4c : FAILURE in 11m 35s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd1a41dbdf41448481ad834b03dc728c : SUCCESS in 9m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e83d5f517664464cb330d5501b917157 : FAILURE in 48m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/afba89407b024d23ad9ad6d679a75ed5 : FAILURE in 49m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3bfad6522084afebc391693b06a7248 : FAILURE in 36m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e67401adf27d4262b19920030ef5b152 : FAILURE in 52m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/015208f81af74a049ecbf39a5624e3db : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bcbff0eee3fd4cf5970c79fc4315bba1 : SUCCESS in 16m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2504c2ddcfe41b0b3aedba6682d2dc2 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3968bfafb97c42daa9ad7261c3df3a7e : FAILURE in 2h 15m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e4a78c6e13048b9a5cdd93f8c41af2c : FAILURE in 1h 44m 44s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4b115873e1544bd580f3312d560c0081 : FAILURE in 2h 04m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16345790dcba44a694e09d6856076b4a : FAILURE in 1h 46m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/833e3b8ed3cb4c3c84ed08c20109b66b : FAILURE in 11m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b67a12c7c6d4a31899d43ef2d252563 : FAILURE in 10m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2953665400f4417f8163ce17bd472d7c : FAILURE in 13m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f552cca86a39489cbe859625d5952aac : FAILURE in 25m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9df316b5454e4cd7bd21e2688c2d5254 : FAILURE in 18m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8639fdcea2b6459c9d39707081220468 : FAILURE in 21m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b08a22764369436c8b98e71706dad1f4 : FAILURE in 18m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f7c26ef47dc5403a903f2c48cb13e654 : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55b0a86a34894c11b316205b3a4991fe : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df2310207bcb422b9b5f87372f9eb6c4 : SUCCESS in 16m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4087ae203494b36b0a4d17f109eef91 : SUCCESS in 13m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/968a3b9e14df481e9d694649e670dc17 : FAILURE in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df2b5f8ba819459a928ce2f3f3f1c6da : FAILURE in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c983d38f2e454404a15459e42fcb00ba : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7cb34faa01fe884dfb1cce9b9957efce7934554e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-13 10:49:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"488491e8d470bd4892f126e913aa2f7a3e8606f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 13:19:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9d8def81aba4da094d98f95ca583d95\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04ffa87d309d4faa8768bbc37e8c939b : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbf32067cb624e02ad647b7e2e2f39ed : FAILURE in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4621134d8b465ebea82c0026e82426 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03f16c9c4a6b4d139d538ceb22aa99ad : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b66144ebdf94449adbbce0165be44ea : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/416cea3a9d9c4a42a232d8b1582d8319 : FAILURE in 12m 40s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de3e08b4b92149ebb1870cdfb947321a : SUCCESS in 9m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c4af1c64995e4960822f10b0839ae292 : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1e9d5d1157014bee97026f50378b0ee4 : SUCCESS in 58m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a2916f88a9e4f88bd49bd9317c16c87 : SUCCESS in 53m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ee2c22975e18444a8108c451389ffc49 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bc8c869f4a734832918e6475413885ae : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/abd8214155b6462aa1700cd9fc86bd5f : SUCCESS in 17m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2ad9f533e494be49f721b95ddf1100f : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f07384058ed4dd39e0b76d5e003570d : SUCCESS in 2h 16m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e101e7986a804526a1cecaceb9cb6f5f : SUCCESS in 1h 54m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c93341a853de429788669f57e3443473 : SUCCESS in 2h 24m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ab2b3198b041a5ae8c16163308c785 : SUCCESS in 2h 03m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a537973f5204e78a69b7173051496ca : FAILURE in 11m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff88d035a91042bdb8e3746064ca1eca : FAILURE in 9m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/20dc6c910f9a490bba9f0ef4cb660bd2 : FAILURE in 11m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8ec3d7429444ac7ac4d81e6c6519517 : FAILURE in 20m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5595aebb72d44854a3fbe70339684aad : FAILURE in 18m 05s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/565ae8176a7f4387a963a733bc1abc41 : FAILURE in 21m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c449b05eff549a1889f57ab5c32a2b2 : FAILURE in 19m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03ccd1d70a5041b6a5a6c35c3dbf543f : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c84140ee62e8489d962fbc2d8d09f90e : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/efc31d662f54404cb47f6158461c0473 : SUCCESS in 17m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2014dec615842cc83d5d432c36ab2e8 : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ad118d1490f490e9d986fccb5bbe0e8 : FAILURE in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cea087620dff4df591cf269241ec903e : FAILURE in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9258cba49d9a4f88a2a72859d92e694a : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"cf7fe980fdcf62ef36b9c820d86b92662938e52f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-03-21 12:17:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2dd9af2ffe29c071141b7c6a9e91b7440fde9699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 14:40:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeaeb94fa4554861832b187679ff1d93\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2cf046d5d4ff402d9c5f2a7229822d51 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435fa4e923994372a0fe70fb9b032dc8 : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6ac226440042b9b5c8c423f6217982 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9da02a09ce4e4d7ba2c45e3659d8a083 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c74db0384900401ab1eaec04ea669c31 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b96a36de973643f3962e806d18ef6c93 : FAILURE in 10m 41s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c898d19ae7be472d9872228182f95d77 : RETRY_LIMIT in 3m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b53768e2def457497087d223ed4a79d : FAILURE in 21m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b940592efd5e460ab044bb1557a20b20 : SUCCESS in 47m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9658290e6ca2489885da8d8aeeec9257 : SUCCESS in 52m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eaad26a8584248239698206d8ea86f13 : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/830c19a9703844fa9f30643bf07e5cdc : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dcef920e840241a5b90c18da9c532343 : SUCCESS in 16m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64c4ba1f10684ceeb43c775cf223c12e : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/31dc6ae27a1c41afac6c21cdabe99755 : SUCCESS in 2h 18m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da6d9bc3dfaf484992392263fbcaeada : SUCCESS in 2h 08m 29s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f7a97f2fad7e48f8a6d78f40a99c174c : FAILURE in 1h 15m 06s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9da6897176dc4153a4e35a03e872d8df : FAILURE in 1h 05m 43s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/298ffde8d10847009603691f48c46ea1 : FAILURE in 14m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6d9262ea94147749bf642025d97d0aa : FAILURE in 9m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/283d41d56cc84f1c86e42cc0b673fe2d : FAILURE in 10m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c1263f638595466da9d3bbd165b2a50a : FAILURE in 21m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ace976f1d6e246b893e74d90d680ce98 : FAILURE in 16m 40s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/685e005fdda64bd7bfd614df66102249 : FAILURE in 23m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e6c43059f6c40dd8b437f0a7bb9b5fa : FAILURE in 16m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb290f909a86487aa0f7dcfdb06657dd : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d97b665df4c461ebc5ef00fb120dba9 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f44c597cf5084274896b885a9787bfe9 : SUCCESS in 16m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560ab27799d14b428fecbc3e55fe8d53 : SUCCESS in 13m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65de0c41766a48bcb90181c2525f5043 : FAILURE in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d2fb358a79544cd9bb8645d811119a7 : FAILURE in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d4c7267530c4f45ba3bd0c9103370a8 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e69d057fca2feddb040bfd7e820be3a67d846140","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-16 10:01:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e32c488a6f0d2daca548a6c56eeb6522073755ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 12:36:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b24fdfb6ecb9482da06ff68c752cfd43\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b4d1b5168b5b409590c8d871386dd006 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35ddeec2791456a94d1fcc9365d35f2 : FAILURE in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd8d5be1f6cb47f6bcba3c4503616909 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d7b9e9f18c04795989e44969ce6c64a : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/806d9ba903a44429814f7120f9f08f8d : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/453e9121edd545c799a58aa59e51af4c : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f2e4bc3d26394535ad4358d5ef7ce326 : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bddd1b952d99471bb002f8ba38777388 : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/574b7c3c15a24e8e8bc9ff292615454c : SUCCESS in 53m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7281486e1b4a4977b3f83e519cb508f9 : SUCCESS in 51m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e077b272bd4640389439e0cd4ca5da30 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a36851a5d32f40618c4752148c579bda : SUCCESS in 17m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/46b47b07615b4e83823ecf7416425108 : SUCCESS in 14m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c75eb31f8ad84057b8257f6b52d64dad : SUCCESS in 13m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f97fc53d1de142e6a31e42822a6e2ecb : SUCCESS in 2h 28m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/085d3e2c30a34b019af310f86920da2b : SUCCESS in 2h 06m 59s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/17318ad40a27417cad5609556b1d1d88 : FAILURE in 1h 28m 42s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecaeaf27af0d45e583964c2b32e1d661 : FAILURE in 1h 09m 09s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e1d68944ca8b4444be991e83d00d7285 : FAILURE in 11m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35a9a40ce33349a48694a339ba6a6ade : FAILURE in 9m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/67c3e13ac04f4b0aa0f9dab46e1a8f63 : FAILURE in 11m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5afc69e8fdd84fc18295f809015f609c : FAILURE in 20m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b535e2281fb24eb4a75486e4b22fb9a7 : FAILURE in 20m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2d05994b8c754983a594939bd1f4d1bd : FAILURE in 22m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d92bcd95381544b093f79ff96f801528 : FAILURE in 18m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/47e93dff2d024544a9a81d4c7950b705 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16bf31e6b531403f8c14506f06456023 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aa54eda65a2a49949864e8ebcf020c94 : SUCCESS in 18m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bf15d2f6f104cefaf6a4b539f64f97d : SUCCESS in 13m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90c92ee1fd0643d0bf5c1539ff17bea9 : FAILURE in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97f0abe2967c4ea8a8623d82a45e09f0 : FAILURE in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/734a605da63142eca3cc4d57f593cff2 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5ab37d2527cee38a0e782c9e6e52e2e0ba1f3080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-30 09:49:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"099bfe398e3d931864b31a96553efb197ef7ea8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 12:11:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d58f3da7c8d8426cb5cbf377c4dd6bcc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2565abdcad934b898250d24d046bc755 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7313b938b0ae45e18e90273e8f08d787 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d3ced91a654cf4a6e5b82361c25826 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e7f485782a45b7be1172ec429e7a49 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb2b1851a78246b0a12d3c48f8f43859 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0768b7f6ed5423d97cf9a00e1a5243c : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e71d03e6348d416f83a5dbfa57c0d1a2 : SUCCESS in 8m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5d3c03ad24af4c56a1f4a7fc52a934e5 : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/52f2b02de85242759101d4a7122768c4 : SUCCESS in 46m 28s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffe5010f463943619d738fc144361a5e : SUCCESS in 41m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/86291ce227c24dd680d29da02766b517 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2e81c446f46545d3890e52cee36745ab : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2dc6ca8c90fe4bb8bb61e4e8c1a7d3c8 : SUCCESS in 14m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb131f1922d3470ba67d00682e1177f1 : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5abbd6b4bd7342f4ab83f03738ba17b5 : SUCCESS in 2h 17m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b482b5364474421a4569978f5fc071d : SUCCESS in 2h 06m 05s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/14bbd70bf9614fe0871407be341d8438 : FAILURE in 1h 17m 36s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0e0919332104519a72252b8c4e35c73 : FAILURE in 59m 46s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/630385dace0c4f519e45e9b804d03f32 : FAILURE in 10m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/260dd448a8f34dd881bc1eb95bd519fd : FAILURE in 8m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4e86de19a80a45568ca3d09b3eeb879f : FAILURE in 10m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e30f8264888240bc886b696de6e9959b : FAILURE in 20m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/212d228d4d834e2387f7e4567ab5e9d4 : FAILURE in 14m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/87c2bcea05164465a72e87d787713e4a : FAILURE in 21m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c1a332b3c204874a3610b5e982b79ff : FAILURE in 15m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/484bce71ea9c4a0a994201898cc226f7 : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59751e7ed2714543aade0ba33da10999 : SUCCESS in 9m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/435137adc0df42b8ab0f5e11ef1e80f9 : SUCCESS in 19m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22466e1e1c5544a3b819940aca2a0b02 : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde20df4c861413993e2264a88ce4026 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d3db7b6b5a04c41972d93e85da958ea : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88477dfb423843a1a2a34516cff99522 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"bcb9a7d5fc0018f39ff4e47b8e9c812b10cc83cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-01 08:04:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f45d36bc86978fb92df3f741d53f7712efcd2dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-01 08:35:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"32d847a37617a4d9bbd515cceac04d43fb9f80e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-01 08:45:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71e7a4e4785be759c7ab7b73d95afb7271d600c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-01 10:27:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847da7d297874eb7a3fc8e2c219766e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efdfebd9b27a4163a60c9b04fbe93ee0 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4a5ca206834d7ab072b0bac1eaf0d8 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41292a842854149bfcbe5bda7d41c9d : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bd60cee2b8141678d218b31e678abf0 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/143255a3e02a47e7b0db1d0d6e142728 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12217a7fa82b4b5e8d63e94462eb0e3f : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a9da590f4984e74835d404e6cb6a0ab : SUCCESS in 7m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/74065db1baf14ab7896ea1786111ad90 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/47377d5a90ee4df1b5f99c837e4b1390 : SUCCESS in 45m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f18e5bb71bc4f83a2f41c042300fb0d : SUCCESS in 52m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10283d0a81e74b6f9b1acc3d78607675 : SUCCESS in 59m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4c417bc16ab6405e92d5add36e4efddc : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8baca9bb53a7421fb4b99c85b741b10d : SUCCESS in 15m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39fea19048b74ca9bc6f5162ad0d5baf : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b55d486427e04f19af8fb12544e922b3 : SUCCESS in 1h 34m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a22c1ab3da54db6bdc4fe120fc4caa3 : SUCCESS in 1h 40m 33s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7a9709c751c540469f81eeb4d5069dd0 : FAILURE in 1h 15m 49s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52dda6793574423f9112e7f4a7d8f36a : FAILURE in 1h 14m 53s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a8eb7ba62c2c4320864af5874c924521 : SUCCESS in 14m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe7b95a2ad174ea3a02a1be93d18fc2f : SUCCESS in 18m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/75d1e1e15d2d4f82a2ed4fed6972d75e : SUCCESS in 32m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/47207d4b067a477c8ae1ead7278eaa09 : SUCCESS in 28m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b2d1f7a614d4b50a74ae047a215b1e7 : SUCCESS in 24m 50s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bac73d1de1c74362a1a06cfca79bc45b : SUCCESS in 30m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/048ed2caa7fa4bdc8ba8a808010c6e46 : SUCCESS in 27m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d6b22d08b9040cebd4e4de0a670ef13 : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41f9e3fc684c4d50a8fdf506b54ce20e : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0a85bafa3914c35bb89ea80843c572b : SUCCESS in 18m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0e738b110004b879dc6d31ab3652ce5 : SUCCESS in 12m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eed2244fbf9a4bf39291b726aa51c2e4 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df49552d2fd5484faaaa39a2d994e23f : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d75f89a1895d43c082cd728f53931e51 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":5},{"id":"977d7315546138e0c417a46f9610252e55744ca5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-01 10:35:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"da1fbe7c4312d3364d5d01870769446699d803de","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 07:23:59.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"f491b21c1ded6e99e44c011e8daf6f4d98da1e93","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-03 07:36:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"81315b7a520a0a97654bdfd2ab13115f6f9df72e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-03 08:11:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a0d9b7d38efadc6512ce520dab7f879bfba2fc2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 08:38:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"934f206d5be789600cfcb96785e0abedfb64e76b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 08:45:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff1c296a9279e41c3fabd4a04ad2fd5cede5dcaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 09:12:57.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"54fc3e705ecbde7dc73911b5c5e1c30e8f648434","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-03 09:53:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e56b1a38da7a0db57cc4fa8d38f68484c88ce6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 11:16:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07f1b6c3296b49c497eefac62264460a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0aadbb468bb34a1cadf826b00e49ce37 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e131c45da22c48f582da954a5adc4e12 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/586b22ab82d044ccb2263f2a00211b61 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73af87e5ff974afdb7de8f74bdf4e1e3 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e20d6d92e3854077b081fc50d038bc95 : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7033be16f4c947e19a3be70455fccd82 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e43ac8b27fcb4a8ebf72df1f0454aadf : FAILURE in 5m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/36dcbb2e6ff44fd4bc822af62ebb82de : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/82cedc18726e420585496180674bc532 : SUCCESS in 1h 00m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a0335e59626449c92ea5e073fa29082 : SUCCESS in 59m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b11d5bc6b179441a91d241980b77b6e6 : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6d8dc994848e4092a6a74f4bbafbf341 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/523e3d048d0c481f8192e25fc3506ac2 : SUCCESS in 15m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd3cd711bec94d35aa9be42317508fe7 : SUCCESS in 12m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/da650b90b1044916824214590e36edfe : SUCCESS in 1h 52m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee95dfc756e452e8784dfc9b435f182 : SUCCESS in 1h 37m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/01fa7a1a84174c0f83275e69ff07bca0 : FAILURE in 1h 27m 53s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ecce8c3333a4aa680902af429131543 : FAILURE in 1h 05m 04s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bb8256510fae44ff9d23e449c4e9c7d9 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb5dff75ad2848d7b302eebf9cf2a016 : SUCCESS in 14m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f43947dec6224329987c439e2b01ca08 : SUCCESS in 32m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ea45bd469f9249088bd088073cfc3fee : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6193070b30a14c81b467ba9d3fda9bd2 : SUCCESS in 24m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f06e44b78cd94f579000be90451ebc00 : SUCCESS in 32m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30e183efa5fa4c9496e36fedf57fce09 : SUCCESS in 27m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c795d7bae9b941a6a0c927e74fcc03c4 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/358a2809fd8c4bc4a6401f886685f517 : SUCCESS in 10m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5a771d1559ef4a42b977399dc200bd49 : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/840948edd8554d4685431060e0d0d0ce : SUCCESS in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/716dd7a88e0642008225310fcd5772bf : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/569f5a046f1b4c66882e4fa57ca5ec66 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a76a9098645e421786eebadbf1975ce0 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":6},{"id":"b4552b6e2bca0b7ef1029baa94efd18889d41d47","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:39:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d91f4da1d9d67cca28f4dc45a53d9d341ac16767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:47:25.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"bb1aacd6da6ad23cab26a8a1979150185dc8848c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:48:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"da50e0bf98c3db09e0745655d0f18fc9a78f6a82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:49:24.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"6226ba83cfa65a415915ce69cd4578a20f437891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:50:34.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":9},{"id":"da2132a9ee09d2cb9ada4225a25d8faeffd5f057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 13:51:16.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"35d9767442a7115ea1e0e65893f58120c56141e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 15:58:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a03dbde09c74edb8fccbb358be04449\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b66aaf21b1fe4aa180abc17b2a96d7c4 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4a8d27296164b418d7563ba145e9b28 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaad1bfcfd8d46a99711c3c8ec8e4ce7 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26dfb808100041d0a2f7564cdbb34dd3 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4eab08367f054904abc07ca2a2242fe9 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be384fe0255e4fbe9fbfc0b8af1bf56f : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/96188895a24d444cb950f31864566466 : FAILURE in 4m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/442dbec6b412445986cf3feece40e724 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c409de12604249a39fc334c4191c6c74 : SUCCESS in 1h 05m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a47f692ed8674429b5ae78daa3709d52 : SUCCESS in 56m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b4b2d90599b8464ba2412ed92b2c4813 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c6d8dc7b7f2346a299b5c065283453e9 : SUCCESS in 15m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bc050806f1c8464bb18f499e544f7a87 : SUCCESS in 18m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09b94608869f47e3b2b50c9d2ad709b6 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29995e87ecf74d768697c43eafad7c65 : SUCCESS in 2h 01m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d846f79aa4494c0abc8467b681fac7be : SUCCESS in 1h 38m 39s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1dfdd24681d44efaa2bfde19e9a35317 : FAILURE in 1h 32m 46s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78bca8d58df04c09babbcf6670ac8e7d : FAILURE in 1h 13m 26s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6e7861321fa043d9bf3a588032c7c306 : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5df8492944af485b80bee4ab86a3b78f : SUCCESS in 14m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ec9df9550a4043c9a51c8795a6501d96 : SUCCESS in 43m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/851cb8f4fc38481fb869817890215db0 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90bf9d5c93e843ea9c0eed6beb58a822 : SUCCESS in 22m 38s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2e012f465d77471c99a6cad4b50c4a09 : SUCCESS in 35m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14b22d89711b4e8eacf3d2055995cccf : SUCCESS in 27m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0fd41272aed2482c909d2e8bbf0d8a63 : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd86a3b57a4c4a75afba7c6ee0139238 : SUCCESS in 10m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/880b0bc8cabd4a96950d8a6b519a6c7b : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b1a526648ae478987a779e5be44551a : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef10aea379f24c2ea028d1db038cc873 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38c08c24a1d4468dbea728be48c52084 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10b3fbf34b774d56b0fe82656325b9b5 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":10},{"id":"2809e3c29b85cb93c92300b44a12480fca9ca100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-06-03 16:25:32.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":11},{"id":"e0cfb8e5f0ef21284ea5fb5aa06463fe662b7bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 18:39:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deeaaa264cc74145aac76cf5f43be2bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/917474c6b5b3435196df672e74cd015a : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f60fdacbde434e629262b12299deb863 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d412498b7da34cf283e77436955da197 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa4ba832ee734360baa4726731ce5cb6 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2669a558d2aa4549b4d95d31122a7645 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43bf1f30deb34290b2f197a394ae4716 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0f3c4e4d050494bb2f5ff61182057bf : SUCCESS in 10m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f76ebdcf5cc74ddc8bb376f72affcb8a : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/833377983d464cb0ab4c8967ea92c4c9 : SUCCESS in 54m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddb26cf0d140421fbb5cbee7cfc5346a : SUCCESS in 46m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/890d0dd671554b05841e140f2ef78393 : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f31cd0080f9e4fa49cf7dea57318407d : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a2bb03a6756b4d1cb9c5cef8298de959 : SUCCESS in 16m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e28ea5bf1cc40fb9250cde206be1be3 : SUCCESS in 13m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/698a73d65a6e4015ae4348f5004e13c5 : SUCCESS in 2h 02m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78e2f05d1e224333bcfcfe1c63b6c9ac : SUCCESS in 1h 38m 57s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a8b680372bb04a6581153649ef5f6e85 : FAILURE in 1h 20m 16s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a39ff35250094cd88ee0c8d922c8b639 : FAILURE in 1h 01m 18s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f7eeb208ee5347929a5ca489cd67aade : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d25a9ffc53c4e9fa92fd0f7cb86f56d : SUCCESS in 13m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/eaed12eba34a43c2802d316bbdd92287 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/481a7d89b6cd4dc1908e52f95673a558 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/858260f483264559b9385fa1f18f5aaa : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0ba7332ee4934727bb4d52200f40af9b : SUCCESS in 35m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe86395add174b23a1de69d08cda4fa3 : SUCCESS in 31m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b0bbcd5558134d48a8feac115c7970f2 : SUCCESS in 11m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7684caeb5434159a278029ae1983a8c : SUCCESS in 12m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/839cdf640ed34ad68edceca31cdbb3d6 : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4bcc9d72bb544ec93a94d3e723afa3b : SUCCESS in 13m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ba8421b3364467fae1d949879290b71 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae32f4387a9e482bae26685f798574cb : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f67a55e1d036475ca8caef496093fe6a : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":11},{"id":"32b6bd55b17a908c3e92fc4f70ff92d79bded235","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-06-05 11:04:59.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"30ed21f38160eb6cde14b54b4b01618a2caf2dae","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-05 11:05:40.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"2605f12d0bcaafd6f74b083274c54d5317e0abae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 11:06:04.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"016f4ba1225e12110bbced4b87d55c1c10310a65","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-05 12:25:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"6c6f82b62bdde4390f76fced02296bd936e3a942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:54:07.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/103dd97d67334d52b2ef90dfaa095b12\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f63db252d3824071b22533db86fe1b9f : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e38a4324514e5fbd8935c4dc9e2669 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb7ec9ed85de4e3591854fb0f44ecb81 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd420a5503514c05bbaf50f92c1c4fe4 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21eeddc1e89a4afdaabc0292ef40baf9 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84755a87626149c28352ff52345ddfc6 : SUCCESS in 12m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bdba8c5bc380452e95950b780d1cf377 : SUCCESS in 57m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6aa4aafdf5d4134b4c47d08420fd0df : SUCCESS in 51m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ac982e9b1c9c47e784f4ebc76a02768e : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2ea4ea8a1c44405eabb0a28df9136ac7 : SUCCESS in 18m 43s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b3cdacd5f9c47968396ce97a9a96f31 : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f6e69e16bd2b42b489def9a2c9c2c44e : SUCCESS in 1h 43m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d464fc6c0bf47f6b4beafa4a09a2a00 : SUCCESS in 1h 42m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/271a89a596584286bca4f161e637bc68 : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f194686f8c1470b952d5c4b24f2c85d : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9a11e77bd15e4b3393d3f250f78d0c8b : SUCCESS in 28m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96a81f96f4444302b19d47940f07a05e : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a4de281a35714ca4b47070df3fe99657 : SUCCESS in 37m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0d7545f4519448ba252dd90fa30a74f : SUCCESS in 27m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fef7d3d6c09497abf6384fd5307cac8 : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf0685f623e44186b887ae7e437bcd78 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4a2f79afdd634aef96d345d96dee98a5 : SUCCESS in 20m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/094e507999cd468c92b71f07bbebe357 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":11},{"id":"b6d9858b6e6659442f43dc7a6652484b3f1cc796","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:54:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"3f7686b686a2e90de157fc316bf1c0a7fc7f8eb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:55:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e4c6302aae41c79820d088ed9e6832\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffed6f3ea8342a5adf04c1aeb55c593 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a29c10c43f3f46e291c6c95b5334385d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2c58388ac336456369b5ee277cb6896f58b07c44","revisions":{"cd48ffedb159be72abe2cd6f2434b39624f4a7e3":{"kind":"REWORK","_number":1,"created":"2024-03-07 12:10:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/1"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 12:09:09.000000000","tz":0},"subject":"Adds seed service destroy","message":"Adds seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd48ffedb159be72abe2cd6f2434b39624f4a7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd48ffedb159be72abe2cd6f2434b39624f4a7e3"}]},"branch":"refs/heads/master"},"72da762d100d2632d5d28c3ea9143b56fb304d5e":{"kind":"REWORK","_number":2,"created":"2024-05-30 09:49:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/2"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-30 09:49:22.000000000","tz":60},"subject":"Adds seed service destroy","message":"Adds seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72da762d100d2632d5d28c3ea9143b56fb304d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72da762d100d2632d5d28c3ea9143b56fb304d5e"}]},"branch":"refs/heads/master"},"13bbd29ca647d612f9a2c43751713e32c8c7a362":{"kind":"REWORK","_number":3,"created":"2024-06-01 08:04:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/911702/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/3"}}},"commit":{"parents":[{"commit":"d17f7dc1b8acb447a556d6cdaea69cf46c0daf27","subject":"Merge \"docs: Reword bare metal registration section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d17f7dc1b8acb447a556d6cdaea69cf46c0daf27"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-01 08:04:28.000000000","tz":120},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13bbd29ca647d612f9a2c43751713e32c8c7a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13bbd29ca647d612f9a2c43751713e32c8c7a362"}]},"branch":"refs/heads/master"},"917519307e7fd63662bfca239497657aa9d157cb":{"kind":"REWORK","_number":4,"created":"2024-06-01 08:35:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/911702/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/4"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-01 08:35:30.000000000","tz":120},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/917519307e7fd63662bfca239497657aa9d157cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/917519307e7fd63662bfca239497657aa9d157cb"}]},"branch":"refs/heads/master"},"95818701fe4962fb04f17c38dac0c919b6a81b26":{"kind":"REWORK","_number":5,"created":"2024-06-01 08:45:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/911702/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/5"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-06-01 08:45:17.000000000","tz":120},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95818701fe4962fb04f17c38dac0c919b6a81b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95818701fe4962fb04f17c38dac0c919b6a81b26"}]},"branch":"refs/heads/master"},"3acf156db32b17d61c39f2e92e850805c0243ef5":{"kind":"REWORK","_number":6,"created":"2024-06-03 09:12:57.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/6"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 09:12:57.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3acf156db32b17d61c39f2e92e850805c0243ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3acf156db32b17d61c39f2e92e850805c0243ef5"}]},"branch":"refs/heads/master"},"57e4da59f912bbe7f9317b9cb829c900cc3827ee":{"kind":"REWORK","_number":7,"created":"2024-06-03 13:47:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/7"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 13:47:24.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57e4da59f912bbe7f9317b9cb829c900cc3827ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57e4da59f912bbe7f9317b9cb829c900cc3827ee"}]},"branch":"refs/heads/master"},"8683ef7583c7f5acc56e1a218a327c617bd91d62":{"kind":"REWORK","_number":8,"created":"2024-06-03 13:49:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/8"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 13:49:23.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8683ef7583c7f5acc56e1a218a327c617bd91d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8683ef7583c7f5acc56e1a218a327c617bd91d62"}]},"branch":"refs/heads/master"},"e18e89dbc61937d024b1e2f5670e21d82c2b83e3":{"kind":"REWORK","_number":9,"created":"2024-06-03 13:50:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/9"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 13:50:34.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e18e89dbc61937d024b1e2f5670e21d82c2b83e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e18e89dbc61937d024b1e2f5670e21d82c2b83e3"}]},"branch":"refs/heads/master"},"500b7b8e9141a7e0b6fa2a7f1ae2fcd501812de4":{"kind":"REWORK","_number":10,"created":"2024-06-03 13:51:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/02/911702/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/10"}}},"commit":{"parents":[{"commit":"3ad5a5d9dce12a34b110d39481b68a838b303d41","subject":"CI: Only migrate RMQ queues during SLURP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ad5a5d9dce12a34b110d39481b68a838b303d41"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 13:51:15.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/500b7b8e9141a7e0b6fa2a7f1ae2fcd501812de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/500b7b8e9141a7e0b6fa2a7f1ae2fcd501812de4"}]},"branch":"refs/heads/master"},"2c58388ac336456369b5ee277cb6896f58b07c44":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-06-03 16:25:32.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/02/911702/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/911702/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/911702/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/911702/11"}}},"commit":{"parents":[{"commit":"3b0a74dfc46392c1611b060c3048e3d990476548","subject":"CI: Fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b0a74dfc46392c1611b060c3048e3d990476548"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-03-07 10:56:53.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-03 16:25:32.000000000","tz":0},"subject":"Add seed service destroy","message":"Add seed service destroy\n\nThis can be useful when using a seed VM that is not deployed by kayobe,\nand can therefore not use ``kayobe seed deprovision``, especially when\niterating on kayobe-config during the inital stages of a deployment, or\nfor development, where you want to re-run the playbooks from a clean-ish\nstate.\n\nChange-Id: I43a9c2a57fcfe2c9d84f39903aac7c258f9a897f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c58388ac336456369b5ee277cb6896f58b07c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c58388ac336456369b5ee277cb6896f58b07c44"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
