)]}'
{"id":"openstack%2Fkayobe~912739","triplet_id":"openstack%2Fkayobe~master~Ib8d47408b851493d0843a82c4d7197bfc8965506","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35511":{"account":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"last_update":"2024-03-13 11:59:37.000000000","reason":"\u003cGERRIT_ACCOUNT_35511\u003e replied on the change","reason_account":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-03-14 14:34:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8d47408b851493d0843a82c4d7197bfc8965506","subject":"Fix link in release note","status":"MERGED","created":"2024-03-13 11:27:49.000000000","updated":"2024-03-14 14:35:32.000000000","submitted":"2024-03-14 14:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912739","meta_rev_id":"ab39be9d9122db9f764496f259e9b6ee724c45e7","_number":912739,"virtual_id_number":912739,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 14:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"all":[{"value":1,"date":"2024-03-13 11:59:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":2,"date":"2024-03-14 12:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":1,"date":"2024-03-14 12:05:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 11:28:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"state":"REVIEWER"},{"updated":"2024-03-13 11:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 12:05:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1edfe04838f816f4dd31d1b86e122604757095c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-13 11:27:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81810b5dcf9dce51408ebb00cd6e5932387fb671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 11:36:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b75118bbcb04701ac73449f9fe225d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955df44c1a4f474782c8858eac31fe6d : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaffeac0ffe4415989b77c9f812fcb80 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3f17979831a4eef0e8ce2b6ec8f030803ffaafa0","author":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"date":"2024-03-13 11:59:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3375758514d101fc09a89d40c1ad6b8a9741054f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-14 12:05:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0e06d30ff4fdec3de2dd228bc5ece65990d1d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 12:06:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0505403445fcdd81034e4acba6729c1a18777aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7fa8d08d6f42609689d62801d4864a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb5f08cb4f646a2921890ee927412ae : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11076854da741fa8a51d98ce0fa19be : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"49fb99c703e061365514cf6054f3c6159734e278","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:34:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab39be9d9122db9f764496f259e9b6ee724c45e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 14:35:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caa2c19c139148f0b3bdd2f1de132746\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d79cc699f54cadbd78a826dc28022d : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bfc571ac324440a88d8b07e35c2b769 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"142a22decaab35158fc7d9b6c37f1b37ee31ca89","revisions":{"142a22decaab35158fc7d9b6c37f1b37ee31ca89":{"kind":"REWORK","_number":1,"created":"2024-03-13 11:27:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/912739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/912739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/912739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/912739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/912739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/912739/1"}}},"commit":{"parents":[{"commit":"e460099e1a3e2cc2899942d4fd48b72d8ed5616f","subject":"Remove default value from sample configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460099e1a3e2cc2899942d4fd48b72d8ed5616f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-13 11:27:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-03-13 11:27:38.000000000","tz":60},"subject":"Fix link in release note","message":"Fix link in release note\n\nChange-Id: Ib8d47408b851493d0843a82c4d7197bfc8965506\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/142a22decaab35158fc7d9b6c37f1b37ee31ca89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/142a22decaab35158fc7d9b6c37f1b37ee31ca89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
