)]}'
{"id":"openstack%2Fkayobe~913245","triplet_id":"openstack%2Fkayobe~master~Iccb703eb8c1a4dd7eb5b7882ae0124c5724151cd","project":"openstack/kayobe","branch":"master","topic":"ingress-egress-qos-priority-maps","attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-07-12 16:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_10910\u003e replied on the change","reason_account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iccb703eb8c1a4dd7eb5b7882ae0124c5724151cd","subject":"WIP ingress/egress priority/qos map(s)","status":"NEW","created":"2024-03-14 17:46:13.000000000","updated":"2024-07-12 16:22:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"fc91ae59be7017aef7c8618b90fd59b208ab0e28","_number":913245,"virtual_id_number":913245,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-14 20:02:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}]},"reviewer_updates":[{"updated":"2024-03-14 20:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-12 16:22:09.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"CC"}],"messages":[{"id":"b9eda45652f8a00852a465c51881765da9cf9a8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-14 17:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6eccaecc1550ba81b5daf43e966d25c7fcf0e53","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-03-14 17:46:29.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3fe5fc9126ae859ae440a007d8281c6f0cd43db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 20:02:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b854d5649d3f453e8b8056a10830417c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35744597b00743489dbbcf496d03278b : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b665e7b13c67442da29df75c535667fb : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32d5f2a5469e48818761e66fcd57593b : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f5d3de3177f4eebbeeaee899b543ff2 : FAILURE in 10m 59s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d67870cecfba4c37a7538b0ca3f2a7da : SUCCESS in 9m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d8d5eb2d533b43f9a392c4796e9798c8 : SUCCESS in 56m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0c9f4bc199424770b857022b638708a6 : SUCCESS in 59m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d52251010e464a5d88e6a47bd09e8a55 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3ee33b3a017544e9b3b8336c3879da95 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9efdef84c81945c18049afda63d6c782 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bb1630dd65a74d9c833d84709161a718 : SUCCESS in 17m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0f7d62d9b494e978a46c3d0802da91e : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d36f4a9d812b41a296f7cad5ae46c344 : SUCCESS in 2h 13m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7ec41af83f24d539e94e3b78f71e14d : SUCCESS in 2h 09m 18s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c92afcce14a748e49d9be0f7ae1f0315 : FAILURE in 1h 13m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/990e12f2a28f42a78fcd8cb985c2237d : FAILURE in 1h 02m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cea37aa11bc145ffb6445a3e2eb8dad4 : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/644dc40d6b324899942e36276044f074 : SUCCESS in 15m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2a58285675c240f7a401c2f73d6e8a45 : SUCCESS in 38m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46c6be8762ed4e5b8f62a183dc41103d : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c6a6dee2e8e4107bcdad735e7d2070b : SUCCESS in 22m 45s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d834d3539ae945dba354fed18747c294 : SUCCESS in 31m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9568dd5824ab421fb908337edc3ebeac : SUCCESS in 33m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff28331faf9f4ca2b4e4cde40ebb0b68 : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91f291b4b3b94fc888036622faac0ab4 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7060b9130f7c464a9177121b8f10a453 : SUCCESS in 17m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82fe8ba0ce6c4f6bbd22f266a5c69cee : SUCCESS in 14m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4f983a01fa74f5bb930a420fef7301c : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94f665764c1c409aaf4880dc4da0bb30 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac8bde00fbf84546b728190b6b579d3b : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fc91ae59be7017aef7c8618b90fd59b208ab0e28","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2024-07-12 16:22:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"192fec32fdd89499871aaad43ecac8aca9a1de46","revisions":{"192fec32fdd89499871aaad43ecac8aca9a1de46":{"kind":"REWORK","_number":1,"created":"2024-03-14 17:46:13.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/45/913245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/913245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/913245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/913245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/913245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/913245/1"}}},"commit":{"parents":[{"commit":"2fd1a79db4cf2a5e69273a069775efff6f247e90","subject":"Merge \"Fix link in release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fd1a79db4cf2a5e69273a069775efff6f247e90"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-14 16:32:08.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-03-14 17:45:53.000000000","tz":0},"subject":"WIP ingress/egress priority/qos map(s)","message":"WIP ingress/egress priority/qos map(s)\n\nDepends on:\nhttps://github.com/michaelrigart/ansible-role-interfaces/pull/160\nfor NetworkManager support.\n\nChange-Id: Iccb703eb8c1a4dd7eb5b7882ae0124c5724151cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/192fec32fdd89499871aaad43ecac8aca9a1de46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/192fec32fdd89499871aaad43ecac8aca9a1de46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
