)]}'
{"id":"openstack%2Fkayobe~915330","triplet_id":"openstack%2Fkayobe~master~I32886c0bbbbb659405fdf35af0a49f0d005ac152","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I32886c0bbbbb659405fdf35af0a49f0d005ac152","subject":"CI: Run kayobe-tox-ansible using Ubuntu Focal","status":"ABANDONED","created":"2024-04-09 13:38:59.000000000","updated":"2024-04-11 08:13:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"84d3e1aab22382ca5a4ac602d0137c25e5ec6d9a","_number":915330,"virtual_id_number":915330,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-11 08:09:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-04-09 14:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"336cfc26fa75e4f38a5dd2253f3995b93f388fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-09 13:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce460b3c4758fe63cba38cf0deaa5267a227c237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 14:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96354701ec2b4f7fac0ef1da85e3116c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d058337564d54540a2c4afa214a26c78 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9b1a3d38864e3da987ef095a6a9287 : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4fe13c6c4ebd48a5940f233e40fc2605 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62e78a71efb249e1b46a62fa0ad0f891 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f18d5b7a0072432788b9861ac5871e87 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e29ec6fe22d24f48b9e5d2316835bbd6 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"36a55c69f0d3787aea9661506208458c717253a0","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-09 14:41:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5ce143d7af3788ce1354aa73e05ce28561771373","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-11 07:23:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a33f92254c473f488f667f390b09e86b86a074be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 08:09:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9372a674cbe84e6eb3add31d653013dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9ee595eaf34ed9b01da8b98fa64429 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650625158aef49638c4cc56193139659 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0275314fd7774307800fe5f67060e040 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/107c4067f81b4eaeb5e12716b09cc398 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b289bf4a96094c8583a9da31e61784b1 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b108f010fadb49bb98ace698842b3f4d : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"84d3e1aab22382ca5a4ac602d0137c25e5ec6d9a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-11 08:13:30.000000000","message":"Abandoned\n\nUbuntu Focal is not compatible with ansible\u003e\u003d7.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"875d9f0c7a5d6e8649591adced99313dd1aa63b6","revisions":{"c63253033a87721b0d069225d3cb4839d88b6c07":{"kind":"REWORK","_number":1,"created":"2024-04-09 13:38:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/915330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/915330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/915330/1"}}},"commit":{"parents":[{"commit":"4215a8e68079b5c137a4774edb26129b6e22abfe","subject":"Merge \"Bump KA Ansible versions to match new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4215a8e68079b5c137a4774edb26129b6e22abfe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-04-09 13:28:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-04-09 13:28:17.000000000","tz":120},"subject":"CI: Run kayobe-tox-ansible using Rocky Linux 9","message":"CI: Run kayobe-tox-ansible using Rocky Linux 9\n\nThis is an attempt to work around the partitioning failure bug in\nUbuntu Jammy.\n\nChange-Id: I32886c0bbbbb659405fdf35af0a49f0d005ac152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c63253033a87721b0d069225d3cb4839d88b6c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c63253033a87721b0d069225d3cb4839d88b6c07"}]},"branch":"refs/heads/master"},"875d9f0c7a5d6e8649591adced99313dd1aa63b6":{"kind":"REWORK","_number":2,"created":"2024-04-11 07:23:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/915330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/915330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/915330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/915330/2"}}},"commit":{"parents":[{"commit":"4215a8e68079b5c137a4774edb26129b6e22abfe","subject":"Merge \"Bump KA Ansible versions to match new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4215a8e68079b5c137a4774edb26129b6e22abfe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-04-11 07:23:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-04-11 07:23:00.000000000","tz":120},"subject":"CI: Run kayobe-tox-ansible using Ubuntu Focal","message":"CI: Run kayobe-tox-ansible using Ubuntu Focal\n\nThis is an attempt to work around the partitioning failure bug in\nUbuntu Jammy until it is fixed upstream.\n\nChange-Id: I32886c0bbbbb659405fdf35af0a49f0d005ac152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/875d9f0c7a5d6e8649591adced99313dd1aa63b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/875d9f0c7a5d6e8649591adced99313dd1aa63b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
