)]}'
{"id":"openstack%2Fkayobe~915362","triplet_id":"openstack%2Fkayobe~master~I49522dab189a9b1fb91daed34a85f4cccf654348","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-06-19 14:13:57.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-18 12:01:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-06-19 14:13:57.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-19 14:13:57.000000000","reason":"Change was abandoned"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-06-19 14:13:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I49522dab189a9b1fb91daed34a85f4cccf654348","subject":"CI: fetch Bifrost\u0027s disk-image-create log file","status":"ABANDONED","created":"2024-04-09 15:43:47.000000000","updated":"2024-06-19 14:13:57.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2109de622fa7c520d21b3e799abb38626560b3f9","_number":915362,"virtual_id_number":915362,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-09 23:43:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-04-09 15:44:47.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-09 15:44:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-04-09 15:44:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-04-09 15:44:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-04-09 15:47:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-04-09 18:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"203fcad7516b548edff4c27ca66e354f88b2c9ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-09 15:43:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70b215bce49ff61cd2f5c86c197d17131783c56c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-09 15:44:47.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"1d5df7411e265bcec2a0b009f8bd3356ce075fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 18:24:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7982ee42856a44238b127b604bc1f504\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/235951e3842442b0810611e811c9e3b7 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf196c9a047468b98926dc42d3e9068 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bc565f72efb42a180d6f08ebd3a87a9 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c0ad7646a394b2b9e2d428593af6f3a : SUCCESS in 10m 25s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4253bf5547a749b6ac0f45880c9ea1c7 : SUCCESS in 9m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/407c991af42748a38878afbd6cc6fcbb : FAILURE in 19m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d8da18f2c33144bd8d190bba92e95e55 : SUCCESS in 57m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7faf6eb35a84ec69ad5cb5120e25dc8 : SUCCESS in 1h 33m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/06e28bb2a4a24a369d11dffc55a37403 : SUCCESS in 1h 17m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/702f0b67d4cb4a06940f9277f35ee550 : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/733a56906ce84195896e7e002f894112 : SUCCESS in 17m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6f528699fed4e16ae24e71ed36c0aba : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/698486fdedb045f9a9777bdfa587d8b4 : SUCCESS in 2h 35m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b853d3e08cc444ec858962f6b23996c3 : SUCCESS in 1h 54m 22s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e53f29c3ff334076a4d024b47dfee6fc : FAILURE in 1h 22m 38s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/912cf2a7edb644f6bd7148dc4a346efe : FAILURE in 1h 13m 41s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a62bded2c1444df7ae50aa48552dfc97 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74ce58db19ef468d9c9cee802b9c3dd3 : SUCCESS in 12m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/603fb55f59334632912c82af47eb75e9 : SUCCESS in 31m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c6f1e5d9f46b4d4db9542e9edf2dc27d : SUCCESS in 31m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e0edb59567d44fd9edb62c6edd17335 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2ec331167aa34a35a0a7aff3762f8adb : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3de2e73a6e04d00aa10cd0c1767799e : SUCCESS in 24m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ba0bb7a358d49ffb87c9e70e9f35590 : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eaa8b1f45664c25a8575fa010aec188 : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76767a79b9dc429993209725bd4c174f : SUCCESS in 16m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/480d641cc5b24c80bec002fe6f838f1d : SUCCESS in 13m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/434c2c136e4e4fca94c419c2f8f97a1d : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97c6c887c2a84a3bb0e9c689dd45e521 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/945e237a12b944229e01d757ed32a509 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b43e73de1c3564bbd5119647d14c4db79379ee93","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-09 21:04:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ee6d1aa33e24725955afdb00ed67cdf42981942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 23:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d739bb9d5eed4a508557ce25a30021a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84c697cf72e54b279b5cc8f5f313c836 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bb91d1894548fc87f751c91ff114a3 : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d3c6f2e49d241c4972a7abb7d9c0052 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/042f971600c840f48fa5d63120c3f118 : SUCCESS in 7m 59s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5249234124994a70b562d5f3c55029eb : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dfaea691347a443ca29b7f9aafadfe77 : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7dbd4663833e43f9a2a9e36ee0397bad : SUCCESS in 1h 08m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f871367ccacc4277809020d50b9a89c3 : SUCCESS in 50m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/54ef1acf96c84fec939d7a3b4cb8930f : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1b8c299c47094f7b9ce7c8236c9575e7 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2d5652e4c6174cc98a5b4594209870bb : SUCCESS in 14m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4057f46fc9f9479db8573c9130fff3f3 : SUCCESS in 11m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/018e1dd75ddd422987d621f8ad5abccb : SUCCESS in 2h 30m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4446368395f74b47ae382280634ff0d6 : SUCCESS in 1h 29m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9878462e1c794e48b9a7e0824fa459ce : FAILURE in 1h 27m 00s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4d5bc730d654cddb01e1c367dbfb14a : FAILURE in 48m 24s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2e10c41513ec4a9f8ac90548bef98e4f : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0aadd59cc394806a6ba602e67e99905 : SUCCESS in 11m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1ef8c168f3ed40708c31d7b5894dec36 : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/218c015318994871ba92c1fb18d130ac : SUCCESS in 25m 37s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef67927d522a4635981e34121a9db47a : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6919fde8cb1849a79039cf295aeb3f0e : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0589fba4e56d45fa845f7f33a06f2ac1 : SUCCESS in 24m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77e8e0644f7b467fa1752e8fb24e79f2 : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76d2636782fd455fb180e2ef275db6b9 : SUCCESS in 9m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a78b839a7846455293f5f35123eb49bd : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb502e9d83e44a42bbcbc0ee0fd0e335 : SUCCESS in 14m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf7f73d8a869448e98974b785f666397 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0456355dcef94e1f8deb82c1cc8ff20f : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bba1db97e0b448a582fe5d5fb1cff301 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d32589f390257e2fedf9284a41435a311f8cd137","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-10 11:12:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a2668fc8038f8db56b63e3405e6388bcb03cd5a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-10 13:57:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3eee396f3b78106e37442a6afc6eb7751d0ed836","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-10 14:00:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83ebcd78fae6be991f0009c7bb4bcbbca4f269a4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-10 19:41:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"adce09787df2a6fd735a557cf7ab0cf7b6e7086d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-10 19:45:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7063ab995783cd9423d9a7d748daf6c3d1eeea9b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-04-18 10:40:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"104fffde62f1eeffc2dcc81dd0a51aa6e7244d78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-18 12:01:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2109de622fa7c520d21b3e799abb38626560b3f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-19 14:13:57.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c30f85771c46140392b58789b942df610ca49cc","revisions":{"0c30f85771c46140392b58789b942df610ca49cc":{"kind":"REWORK","_number":1,"created":"2024-04-09 15:43:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/62/915362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/915362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/915362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/915362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/915362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/915362/1"}}},"commit":{"parents":[{"commit":"4215a8e68079b5c137a4774edb26129b6e22abfe","subject":"Merge \"Bump KA Ansible versions to match new defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4215a8e68079b5c137a4774edb26129b6e22abfe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-09 15:39:41.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-09 15:43:30.000000000","tz":180},"subject":"CI: fetch Bifrost\u0027s disk-image-create log file","message":"CI: fetch Bifrost\u0027s disk-image-create log file\n\nFollowup on Iab88b3a28378cc22f767204bef84128bc9627210\n\nChange-Id: I49522dab189a9b1fb91daed34a85f4cccf654348\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c30f85771c46140392b58789b942df610ca49cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c30f85771c46140392b58789b942df610ca49cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
