)]}'
{"id":"openstack%2Fkayobe~916775","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I4901294c8f5912cb35df3866ed13012461b4f644","project":"openstack/kayobe","branch":"stable/2023.2","topic":"fix-apt-keyrings-location-stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-29 11:37:57.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-29 11:37:57.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-04-29 11:37:57.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-04-29 09:05:55.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I4901294c8f5912cb35df3866ed13012461b4f644","subject":"Fix default place for Ubuntu Apt keyrings","status":"MERGED","created":"2024-04-24 20:50:26.000000000","updated":"2024-04-29 11:38:59.000000000","submitted":"2024-04-29 11:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916775-fix-apt-keyrings-location-stable/2023.2","cherry_pick_of_change":916138,"cherry_pick_of_patch_set":2,"meta_rev_id":"43e857f331adb2cad67c0174e3604a9cbe9cf2e8","_number":916775,"virtual_id_number":916775,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-29 11:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"all":[{"value":1,"date":"2024-04-25 07:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":2,"date":"2024-04-29 09:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-25 07:05:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-04-29 09:05:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 23:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 07:05:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-04-25 07:16:18.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-04-26 11:06:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-04-26 11:06:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-26 11:06:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"356fc8f4a525f41de888c9d54947890c388e4fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 20:50:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2dfb7e5ea875f8ee9685124e1ef926db46bd174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 23:35:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d46e7b1ea6a5498886758690fd78d8f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5f7befe59e24cd199df351a5bc1ff19 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6997a08fe0f4cc5ae7f0a06d0c5c676 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13a66bdf95864fe8a36c0ed0fb007c81 : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e63bcf4eeb3a40a8b72b6a45576c1797 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81141a9c1b8f40dcafc4e86c59c95646 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78bd54b754084e248af8b0684d1d9119 : SUCCESS in 8m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/206071b2925b47f692bfa76bf5e79dd9 : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/871c32ce04164a6ab4cfd50dcef9b049 : SUCCESS in 1h 04m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73df667dd98346db834e6a5ef0e15e1b : FAILURE in 20m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/989af8a7fdae412e9b262565136d3d20 : SUCCESS in 1h 17m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0483b478994c4d39900b003e924f1649 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2352db20d833419dae12ca3994267b44 : SUCCESS in 15m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1122331a3bda44a48748afb6dbe703af : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca755f54ed594c108b1543d451cc6ad9 : SUCCESS in 2h 26m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73f12c4d20a1452a82d981ff8c96883a : SUCCESS in 2h 07m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ddd972f04d1f46cc9ef60ba11d70ddf3 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0e008797f334d909d84956ac610394d : SUCCESS in 13m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f893ecd1c4424c17a22efdf6a07335c2 : SUCCESS in 39m 57s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/696c6b0a823a4001a71d345905d924d9 : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/424778fb505e4de695b165dd19b1b9db : SUCCESS in 26m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a4bc6bb498244b8fb2183128a88267bc : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd44347bcc604f13bdade3bd57c21760 : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3710b472e8a44621bb9c631995a3ff1b : SUCCESS in 19m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/624a2a3d38f44219ba390b5b7f7484aa : SUCCESS in 14m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6f8f83ca8c14be9a936a4371bc58d62 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/295438655f2f485888611ba43137697e : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3bf58ce454a485bb041812e37c10a94 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a4ac0f49cba0a19e4633c8f7f5bf70a7c444cda0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-25 00:16:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca8393b369c400391bd6fd456188b47fb60a5118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 02:52:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a82e12b79c14224a8cc3f543c46da6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29a8148b1384026ab96987fafa36cbc : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14e09adb235459e9c361084dbc31211 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1427bc3fa6284985aff5fbef886da461 : SUCCESS in 2m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cafd329c02a469b86003a70f3df4ed9 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ee6095572214899983f4af49c1165ff : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2239dfc2e6148cabccfafad6895fbdd : SUCCESS in 10m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6da24e8c40bd4a0aac470b0f166b8e3b : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cf39cb51a55f4c3a97c763ba0f4a1c7c : SUCCESS in 1h 04m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8437ae90a39f496485a5235a0c436c36 : SUCCESS in 54m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/371940fc867840d8a92f1d9bf42c92b7 : SUCCESS in 1h 07m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/15aab58ee7f64e819682b55593c3e8fb : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7bb4e8f356ce49cc8c5283022f8dcd9b : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf269c5cb5b74bba8650788b47a9d732 : SUCCESS in 14m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b4c600936e8541abacdd5a9a43b899f1 : SUCCESS in 2h 30m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/038d36ec421544148ddad6872e6dd35b : SUCCESS in 2h 10m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/889fed462c884899a322af12a8888654 : SUCCESS in 19m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a38668b7a7bb4977964942babb547ec0 : SUCCESS in 11m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bcd3572cf14d4a66b268b0aedaef7a71 : SUCCESS in 40m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6b5b42ebebaf4b96876e0713898503c0 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b741ac8db8cf439b9d442b7f5682bdf4 : SUCCESS in 20m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05ae98afd1e14f7da4da8206308193d0 : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80d438c6f6c648789b2a97a4bd533ea7 : SUCCESS in 9m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f79aea6313b2476c8a0d827d146aeffd : SUCCESS in 19m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62b30ba55db549919a62e247388ca5c8 : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee5e8a85fdd9412ca83d9fe43d97f1ac : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab1a75f70dec4c3eb5a634f830cb9817 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d4f69ee706745869def0a2f9f8a0606 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"83068b52e83c42b272086ea8602dd36038b9260d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-25 07:05:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7846cf4bd7ce5d7d91f2e970b069a2c3c1b61d19","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-04-25 07:16:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b8d5eea50ac473a0263fa615764b9792a278ae3a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-29 09:05:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"652dbb7d7f4e6dea837a590bb4cc546fa8e06bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 09:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71b24a5f06d6b73fd1aeaa08a7087b48fb002c8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:37:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5537b8de8640fa9b92fec275a4c82a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b95f45b4eab4326a5412ff2623614c4 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b79207d176bc4aa79755d49b012247c4 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df5de3a4d7374bbda411591388ba4ca2 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70e31adf23f14413867073c8b6cd371d : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7d0a8816bdc4f56bb9497ffa378d1aa : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1ed54f16a814451e8d2519a4845f3983 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bf2b95283154542a750f6583507dcfb : SUCCESS in 50m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/849306af4d0a4487a545b8ca04c59cb9 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2661e22e662f4b26822d9ebe9affb0d0 : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7f8fde28df645458291b96eee2fb7bc : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/561dfe3f30664579aee1e17324145a4f : SUCCESS in 2h 27m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/711987f0ddea45b5b7fa3e617b2192a0 : SUCCESS in 2h 14m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/847874876c684bac89e32c147c8c4af4 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2d20d718775461da61408ca9ad42f45 : SUCCESS in 12m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0ae9429f199435a938d210434e15882 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/317e69bfdcd14dafa958fef29a11891b : SUCCESS in 21m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a3f9e9018d0b429894f95a170b44916b : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24a6297a351948d096afa8c7526b961a : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d667ba105db84ceab098050dca59ffbf : SUCCESS in 16m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eeab568a22143d189645bf217c93e43 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a11a6ad3193b20205aa3a5d2c15d7ae6c79a231f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:37:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"43e857f331adb2cad67c0174e3604a9cbe9cf2e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:38:59.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73117add8cb047b4b1c4d97732f266ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a3a76996264bb5b66af926e0fdc5b9 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c22113ee36415fa38d8f757266c3a8 : FAILURE in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92ae295e4c097e6b1b26e685f32a4213d826362a","revisions":{"92ae295e4c097e6b1b26e685f32a4213d826362a":{"kind":"REWORK","_number":1,"created":"2024-04-24 20:50:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/75/916775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/916775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/916775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/916775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/916775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/916775/1"}}},"commit":{"parents":[{"commit":"4ba239497ab7d91d5f77c483c1a82cdc4ca8e9db","subject":"Revert \"Mark kayobe-tox-ansible job as non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ba239497ab7d91d5f77c483c1a82cdc4ca8e9db"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-17 19:29:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-24 20:50:26.000000000","tz":0},"subject":"Fix default place for Ubuntu Apt keyrings","message":"Fix default place for Ubuntu Apt keyrings\n\nThe Ansible openstack.kolla collection already uses [1,2]\nthe recommended location [3] for the Ubuntu Apt keyrings,\nso lets fix this in Kayobe too.\n\n[1] https://opendev.org/openstack/ansible-collection-kolla/src/branch/master/roles/docker/tasks/repo-Debian.yml#L21\n[2] https://opendev.org/openstack/ansible-collection-kolla/src/branch/master/roles/podman_sdk/tasks/main.yml#L13\n[3] https://manpages.ubuntu.com/manpages/jammy/en/man8/apt-key.8.html#deprecation\n\nChange-Id: I4901294c8f5912cb35df3866ed13012461b4f644\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit d539efc94fdd2486e8fe7c08bddbbf6c843301e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92ae295e4c097e6b1b26e685f32a4213d826362a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92ae295e4c097e6b1b26e685f32a4213d826362a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
